)]}'
{"id":"openstack%2Ftrove-dashboard~992015","triplet_id":"openstack%2Ftrove-dashboard~master~Ic05af53eb71508d0ef389d9c19fb44e8680e2fb2","project":"openstack/trove-dashboard","branch":"master","attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-15 19:57:13.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-06-15 11:57:53.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}}},"hashtags":[],"change_id":"Ic05af53eb71508d0ef389d9c19fb44e8680e2fb2","subject":"Launch Instance - add flavor details","status":"NEW","created":"2026-06-07 23:07:45.000000000","updated":"2026-06-15 20:09:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"55bfdf07a22d3dfa05905906d63446f2048637ad","_number":992015,"virtual_id_number":992015,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-15 20:09:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-07 23:19:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 00:42:46.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-06-08 00:42:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"353928d26302824ea3d1ce5d1c39780967510d62","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-07 23:07:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfccd82b2bb6b9662296f88033eaa1256c426f3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-07 23:19:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d64db0338e9641ab8ff2d6135c423559\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ecd52dd4a65a406a83008852ace2f9e2 : SUCCESS in 1m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a12fa77759d1435382d6080a4eeb802b : SUCCESS in 1m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdebc83d07cc487d9a0c11e038ac9c64 : SUCCESS in 3m 57s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/df40171dd28746559be6ec0b84cf36da : SUCCESS in 4m 07s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/6a277711efc24fadb39111eb58861f7e : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f91f99e66bca496fb72bd7c1021048bb : SUCCESS in 1m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1a9e5057cc7452681858679e062d7a7 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bf3b0996fe604e4bb36289eb80720b15 : SUCCESS in 7m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76f75b0b833f4a489ef68f7b30776464 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c85dc812a55a7f5b0f0a256aacb4330683cc6c34","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 00:38:39.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8bd39ef063f1f55801a394eb1013060e217043ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 00:39:24.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"af0195673dce5b0c276427b58f16ce17ef2aa9e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 00:41:57.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4222c0957b1989eaf11d9bc7f30a9f31ceceaed6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 00:42:04.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"ac9f4c87ba27fee5ba61495e1f097ae60f98d15a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 00:42:34.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6382636c6a26f33f9560761305430e05dc826141","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 00:42:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ccfc84da62aac6f077ed1e4f5c60b579c43521a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 00:53:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f795e617e1bd4017bb8dfe98e750cf27\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/398c5bdb888641bfa90b0d1c82900b30 : SUCCESS in 2m 58s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4a93d8dff5cc405f974dc857dc67874f : SUCCESS in 2m 34s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/3a7681eefef942159164ca66cca78a47 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23f394df82e3495a8e174d7bfb2cd848 : SUCCESS in 1m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/424778fe6b794ce3b41a77099ad67a28 : SUCCESS in 2m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3884659c46e84a9898b558e847efa208 : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69ba6f444669472fa8d4783dd357dc52 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":5},{"id":"106cb9e4878de1afc32ed5454f1f3724bf7dbb2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 01:15:04.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"43d017853055a077bf880cd0b94230bafe8caed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 01:27:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80091f54faeb47bc9eebc560f3633281\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c16afa71046547d9b76801167dd02643 : SUCCESS in 2m 13s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/3c69876fd3c54a9d900af070b1296ad6 : SUCCESS in 3m 08s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/ddb82e3b603e4f428f734a75f7947b97 : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71fb6db2935a47398a212d4dfef1fee9 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0900595267724f1fae23311f098a90c7 : SUCCESS in 6m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e138886cb75b48669eb30ed49b605a11 : SUCCESS in 6m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18d4c2fd64a84c0cbd09fd93fc293357 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":6},{"id":"566ae117da89a79885ebd107f05b28afa233b1da","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-15 11:57:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"acca6c7659fd7ce4a4ce44f698d95f155253c561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-15 19:45:18.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"74d2a0a60ba564f7cae70e734e0dee54af25151e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 19:57:13.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6604e8252fd4329857ef4c301087faf\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/519eddbc2905492ea1dcb910e1a430da : FAILURE in 3m 39s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/cbc1c6fc0e0e4877a0f4d1ac7b8b447d : FAILURE in 4m 20s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/7eb9ef34d18145e5a34a7f675a5efcac : FAILURE in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f47171d0b04498b0e89d88eaee79df : FAILURE in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e607f5793bd84bed8e2059ccdd5c930a : FAILURE in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eb3cdb5de1ec44ae830c494494436a63 : FAILURE in 6m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77c48deb13674a5fadb7fa2c8a5b9d97 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":7},{"id":"c62e3d6bf6a8e551f353fed92644271314925892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-15 19:58:40.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"55bfdf07a22d3dfa05905906d63446f2048637ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 20:09:04.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9605f98e23547b9aee10e5e3e102a82\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/480e818f87004398b0a2409905d2fc69 : FAILURE in 2m 32s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/bec086a982da41a6a1a622648e8307b6 : FAILURE in 4m 16s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/f8aee01f23b948ef8c1f640e5700d9e8 : FAILURE in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30485f3e94744dba8d490a8240587c7d : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/adc4c2e5f02945e692f875301acb19c8 : FAILURE in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cf8edf90f85e48499aca6b167a897487 : FAILURE in 6m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/295dfac1e27848d79c7b9c27ecf053cc : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"0cb98cdc7dd783bb9ca7644011574aefd0dcee49","revisions":{"8b332200479c930191d03db6fd713f5cee4b6e5c":{"kind":"REWORK","_number":1,"created":"2026-06-07 23:07:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/992015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/15/992015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/1"}}},"commit":{"parents":[{"commit":"80c13784f75bc6b919261eaecba9094f822c95e4","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/80c13784f75bc6b919261eaecba9094f822c95e4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-07 23:07:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-07 23:07:45.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic05af53eb71508d0ef389d9c19fb44e8680e2fb2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/8b332200479c930191d03db6fd713f5cee4b6e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/8b332200479c930191d03db6fd713f5cee4b6e5c"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"0a49ebd10693194b9ab7798a2c25c7a5198b4fed":{"kind":"REWORK","_number":2,"created":"2026-06-08 00:38:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/992015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/15/992015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/2"}}},"commit":{"parents":[{"commit":"80c13784f75bc6b919261eaecba9094f822c95e4","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/80c13784f75bc6b919261eaecba9094f822c95e4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-07 23:07:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-08 00:38:39.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic05af53eb71508d0ef389d9c19fb44e8680e2fb2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0a49ebd10693194b9ab7798a2c25c7a5198b4fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0a49ebd10693194b9ab7798a2c25c7a5198b4fed"}]},"branch":"refs/heads/master"},"70f1a5eed9ac9b93a78765c1afb5c8fc8d0cace5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-08 00:39:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/992015/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/15/992015/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/3"}}},"commit":{"parents":[{"commit":"80c13784f75bc6b919261eaecba9094f822c95e4","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/80c13784f75bc6b919261eaecba9094f822c95e4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-07 23:07:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-08 00:39:24.000000000","tz":0},"subject":"WIP: Launch Instance - show Flavor Details","message":"WIP: Launch Instance - show Flavor Details\n\nChange-Id: Ic05af53eb71508d0ef389d9c19fb44e8680e2fb2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/70f1a5eed9ac9b93a78765c1afb5c8fc8d0cace5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/70f1a5eed9ac9b93a78765c1afb5c8fc8d0cace5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d689bed6fbe8337ccb435a71f3838d83177a8000":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-06-08 00:41:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/992015/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/15/992015/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/4"}}},"commit":{"parents":[{"commit":"80c13784f75bc6b919261eaecba9094f822c95e4","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/80c13784f75bc6b919261eaecba9094f822c95e4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-07 23:07:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-08 00:41:57.000000000","tz":0},"subject":"WIP: Launch Instance - add flavor details","message":"WIP: Launch Instance - add flavor details\n\nThe modal for creating instance of a database has a flavor dropdown\nbut it is not clear what the properties of the flavor are. This\npatch follows the same convention used in the Launch Instance under\nthe instances panel. It displays the Flavor Details in a table\nunder the help section for the flavor selected.\n\nChange-Id: Ic05af53eb71508d0ef389d9c19fb44e8680e2fb2\nCo-authored-by: Prasoon Telang \u003cprasoontelang@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/d689bed6fbe8337ccb435a71f3838d83177a8000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/d689bed6fbe8337ccb435a71f3838d83177a8000"}]},"branch":"refs/heads/master","description":"Edit commit message"},"718cfacf29104ea4b361912410ec8cbd6e2a89d0":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-06-08 00:42:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/992015/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/15/992015/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/5"}}},"commit":{"parents":[{"commit":"80c13784f75bc6b919261eaecba9094f822c95e4","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/80c13784f75bc6b919261eaecba9094f822c95e4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-07 23:07:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-08 00:42:34.000000000","tz":0},"subject":"Launch Instance - add flavor details","message":"Launch Instance - add flavor details\n\nThe modal for creating instance of a database has a flavor dropdown\nbut it is not clear what the properties of the flavor are. This\npatch follows the same convention used in the Launch Instance under\nthe instances panel. It displays the Flavor Details in a table\nunder the help section for the flavor selected.\n\nChange-Id: Ic05af53eb71508d0ef389d9c19fb44e8680e2fb2\nCo-authored-by: Prasoon Telang \u003cprasoontelang@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/718cfacf29104ea4b361912410ec8cbd6e2a89d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/718cfacf29104ea4b361912410ec8cbd6e2a89d0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"082b0a61ce4746162a8057be9ff64fd5c460aa7e":{"kind":"REWORK","_number":6,"created":"2026-06-08 01:15:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/992015/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/15/992015/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/6"}}},"commit":{"parents":[{"commit":"80c13784f75bc6b919261eaecba9094f822c95e4","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/80c13784f75bc6b919261eaecba9094f822c95e4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-07 23:07:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-08 01:15:03.000000000","tz":0},"subject":"Launch Instance - add flavor details","message":"Launch Instance - add flavor details\n\nThe modal for creating instance of a database has a flavor dropdown\nbut it is not clear what the properties of the flavor are. This\npatch follows the same convention used in the Launch Instance under\nthe instances panel. It displays the Flavor Details in a table\nunder the help section for the flavor selected.\n\nChange-Id: Ic05af53eb71508d0ef389d9c19fb44e8680e2fb2\nCo-authored-by: Prasoon Telang \u003cprasoontelang@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/082b0a61ce4746162a8057be9ff64fd5c460aa7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/082b0a61ce4746162a8057be9ff64fd5c460aa7e"}]},"branch":"refs/heads/master"},"dce5351e1ec40b8b0dd7f7f66a703b9c35b53c1d":{"kind":"REWORK","_number":7,"created":"2026-06-15 19:45:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/992015/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/15/992015/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/7"}}},"commit":{"parents":[{"commit":"80c13784f75bc6b919261eaecba9094f822c95e4","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/80c13784f75bc6b919261eaecba9094f822c95e4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-07 23:07:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-15 19:45:17.000000000","tz":0},"subject":"Launch Instance - add flavor details","message":"Launch Instance - add flavor details\n\nThe modal for creating instance of a database has a flavor dropdown\nbut it is not clear what the properties of the flavor are. This\npatch follows the same convention used in the Launch Instance under\nthe instances panel. It displays the Flavor Details in a table\nunder the help section for the flavor selected.\n\nChange-Id: Ic05af53eb71508d0ef389d9c19fb44e8680e2fb2\nCo-authored-by: Prasoon Telang \u003cprasoontelang@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/dce5351e1ec40b8b0dd7f7f66a703b9c35b53c1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/dce5351e1ec40b8b0dd7f7f66a703b9c35b53c1d"}]},"branch":"refs/heads/master"},"0cb98cdc7dd783bb9ca7644011574aefd0dcee49":{"kind":"REWORK","_number":8,"created":"2026-06-15 19:58:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/992015/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-dashboard","ref":"refs/changes/15/992015/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-dashboard refs/changes/15/992015/8"}}},"commit":{"parents":[{"commit":"80c13784f75bc6b919261eaecba9094f822c95e4","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/80c13784f75bc6b919261eaecba9094f822c95e4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-07 23:07:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-15 19:58:40.000000000","tz":0},"subject":"Launch Instance - add flavor details","message":"Launch Instance - add flavor details\n\nThe modal for creating instance of a database has a flavor dropdown\nbut it is not clear what the properties of the flavor are. This\npatch follows the same convention used in the Launch Instance under\nthe instances panel. It displays the Flavor Details in a table\nunder the help section for the flavor selected.\n\nChange-Id: Ic05af53eb71508d0ef389d9c19fb44e8680e2fb2\nCo-authored-by: Prasoon Telang \u003cprasoontelang@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0cb98cdc7dd783bb9ca7644011574aefd0dcee49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-dashboard/commit/0cb98cdc7dd783bb9ca7644011574aefd0dcee49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
