)]}'
{"id":"openstack%2Ftrove-specs~736487","triplet_id":"openstack%2Ftrove-specs~master~Iad2431eb180426ff51c5ef75bc7cf0db99482f75","project":"openstack/trove-specs","branch":"master","topic":"optimization/pip-no-cache-dir","hashtags":[],"change_id":"Iad2431eb180426ff51c5ef75bc7cf0db99482f75","subject":"optimize size and time using \"--no-cache-dir\"","status":"ABANDONED","created":"2020-06-18 07:58:22.000000000","updated":"2020-06-18 12:23:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7001c575445c1c23f085b546beb7883c1bafba02","_number":736487,"virtual_id_number":736487,"owner":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-18 08:14:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2020-06-18 12:04:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 08:14:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-18 12:04:10.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"08bffc6ea2b2702ccfc6ab9593e4c92d2b90183b","author":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"date":"2020-06-18 07:58:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efe76915c71983d1133a45557e30f6a5506f2736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:14:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c4ac415530e4344a707aa845fe4873d : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":1},{"id":"37a0901de397ae9093667db432f9e3a7509434eb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-18 12:04:10.000000000","message":"Patch Set 1: Code-Review-1\n\nThere\u0027s no need to change a spec, please abandon.","accounts_in_message":[],"_revision_number":1},{"id":"708a276f865d22d4894e9084b2c26c6205e92291","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-18 12:23:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"127dd2cf448c2b2057cd3080a21eaaf0bbeaecc2","revisions":{"127dd2cf448c2b2057cd3080a21eaaf0bbeaecc2":{"kind":"REWORK","_number":1,"created":"2020-06-18 07:58:22.000000000","uploader":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"ref":"refs/changes/87/736487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-specs","ref":"refs/changes/87/736487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-specs refs/changes/87/736487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-specs refs/changes/87/736487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-specs refs/changes/87/736487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-specs refs/changes/87/736487/1"}}},"commit":{"parents":[{"commit":"09b0eaace50c5307bf0a95a249e7515d1ba86ef5","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-specs/commit/09b0eaace50c5307bf0a95a249e7515d1ba86ef5"}]}],"author":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-06-18 04:57:38.000000000","tz":330},"committer":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-06-18 07:57:50.000000000","tz":330},"subject":"optimize size and time using \"--no-cache-dir\"","message":"optimize size and time using \"--no-cache-dir\"\n\nUsing \"--no-cache-dir\" flag in pip install ,make sure dowloaded packages\nby pip don\u0027t cached on system . This is a best practise which make sure\nto fetch ftom repo instead of using local cached one . Further , in case\nof Docker Containers , by restricing caching , we can reduce image size.\nIn term of stats , it depends upon the number of python packages\nmultiplied by their respective size . e.g for heavy packages with a lot\nof dependencies it reduce a lot by don\u0027t caching pip packages.\n\nFurther , more detail information can be found at\n\nhttps://medium.com/sciforce/strategies-of-docker-images-optimization-2ca9cc5719b6\n\nChange-Id: Iad2431eb180426ff51c5ef75bc7cf0db99482f75\nSigned-off-by: Pratik Raj \u003crajpratik71@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-specs/commit/127dd2cf448c2b2057cd3080a21eaaf0bbeaecc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-specs/commit/127dd2cf448c2b2057cd3080a21eaaf0bbeaecc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
