)]}'
{"id":"openstack%2Ftrove-tempest-plugin~1001124","triplet_id":"openstack%2Ftrove-tempest-plugin~master~I74a393503390547e1c1a33aaef644a1149dc0056","project":"openstack/trove-tempest-plugin","branch":"master","topic":"exporters-feature","hashtags":[],"change_id":"I74a393503390547e1c1a33aaef644a1149dc0056","subject":"Add database exporter lifecycle scenarios","status":"NEW","created":"2026-08-17 11:25:41.000000000","updated":"2026-08-17 16:45:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"fad25a10dd4928d4a0713e6fbb0c7dcfaf0115a1","_number":1001124,"virtual_id_number":1001124,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-17 16:45:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-08-17 11:37:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a9f6a9754af83e4f9afa2eaa68712299b6dd126","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-17 11:25:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d333cdb286a93f2e807682a7be28f23a956ae0f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-17 11:26:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b1a240693dd40e468b5786cf159a1ec23ace4500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 11:37:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f3362cd12d84f73b46865946a0a0ecb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1691707fe748df9ad34c05f1b6632f : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc054ddd087648ff86e9058353770296 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"7bdac72fda8d4f1b9d4e24e1d4a4b8156f481f25","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-17 16:10:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c602d913fa9cbacf21379473f81785154cef931e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 16:23:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5c089e46ea64f288ca350e5d384b0dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf9c27d71e3458eb876d836c1ba7d18 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4534254b3e7d4d6bb5cce1c1e7ea6340 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":3},{"id":"70216ffbd33521b6f9088dc0755ff40053ee654e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-17 16:27:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"fad25a10dd4928d4a0713e6fbb0c7dcfaf0115a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 16:45:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12d60ed373ac4db9b6f46da47d40bfbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c9784b3811409dbd3af1b0af1f6053 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623f564f172148dbbb6dbddb871969f1 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f1ec5d94f901133fcbf4280466bcb99ed5d5c878","revisions":{"817770260aaf8ee299fd3adbaa81ed705a6733dc":{"kind":"REWORK","_number":1,"created":"2026-08-17 11:25:41.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/24/1001124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/24/1001124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/1"}}},"commit":{"parents":[{"commit":"e038559f73daf60d56121e624f71b5a294988e8b","subject":"Merge \"Remove jobs for retired branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/e038559f73daf60d56121e624f71b5a294988e8b"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-17 11:10:11.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-17 11:10:11.000000000","tz":300},"subject":"[DNM]  Add database exporter lifecycle scenarios","message":"[DNM]  Add database exporter lifecycle scenarios\n\nAdd opt-in scenarios for MySQL, MariaDB and PostgreSQL that verify\nadmin-only API access, Node Exporter availability on the management\nnetwork, and database exporter show, disable, enable and rebuild\noperations.\n\nBlueprint: exporters-feature\nChange-Id: I74a393503390547e1c1a33aaef644a1149dc0056\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/817770260aaf8ee299fd3adbaa81ed705a6733dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/817770260aaf8ee299fd3adbaa81ed705a6733dc"}]},"branch":"refs/heads/master"},"a0aa9ea8b3240adfa75eb9afd228701a206ff106":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-17 11:26:21.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/24/1001124/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/24/1001124/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/2"}}},"commit":{"parents":[{"commit":"e038559f73daf60d56121e624f71b5a294988e8b","subject":"Merge \"Remove jobs for retired branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/e038559f73daf60d56121e624f71b5a294988e8b"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-17 11:10:11.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-17 11:26:04.000000000","tz":300},"subject":"Add database exporter lifecycle scenarios","message":"Add database exporter lifecycle scenarios\n\nAdd opt-in scenarios for MySQL, MariaDB and PostgreSQL that verify\nadmin-only API access, Node Exporter availability on the management\nnetwork, and database exporter show, disable, enable and rebuild\noperations.\n\nBlueprint: exporters-feature\nChange-Id: I74a393503390547e1c1a33aaef644a1149dc0056\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/a0aa9ea8b3240adfa75eb9afd228701a206ff106"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/a0aa9ea8b3240adfa75eb9afd228701a206ff106"}]},"branch":"refs/heads/master"},"e2d43f8f543c2f8806a9b021871b6e8ac8b4e673":{"kind":"REWORK","_number":3,"created":"2026-08-17 16:10:51.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/24/1001124/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/24/1001124/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/3"}}},"commit":{"parents":[{"commit":"e038559f73daf60d56121e624f71b5a294988e8b","subject":"Merge \"Remove jobs for retired branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/e038559f73daf60d56121e624f71b5a294988e8b"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-17 11:10:11.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-17 16:10:34.000000000","tz":300},"subject":"Add database exporter lifecycle scenarios","message":"Add database exporter lifecycle scenarios\n\nAdd opt-in scenarios for MySQL, MariaDB and PostgreSQL that verify\nadmin-only API access, Node Exporter availability on the management\nnetwork, and database exporter show, disable, enable and rebuild\noperations.\n\nBlueprint: exporters-feature\nChange-Id: I74a393503390547e1c1a33aaef644a1149dc0056\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/e2d43f8f543c2f8806a9b021871b6e8ac8b4e673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/e2d43f8f543c2f8806a9b021871b6e8ac8b4e673"}]},"branch":"refs/heads/master"},"f1ec5d94f901133fcbf4280466bcb99ed5d5c878":{"kind":"REWORK","_number":4,"created":"2026-08-17 16:27:55.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/24/1001124/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/24/1001124/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/24/1001124/4"}}},"commit":{"parents":[{"commit":"e038559f73daf60d56121e624f71b5a294988e8b","subject":"Merge \"Remove jobs for retired branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/e038559f73daf60d56121e624f71b5a294988e8b"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-17 11:10:11.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-17 16:27:44.000000000","tz":300},"subject":"Add database exporter lifecycle scenarios","message":"Add database exporter lifecycle scenarios\n\nAdd opt-in scenarios for MySQL, MariaDB and PostgreSQL that verify\nadmin-only API access, Node Exporter availability on the management\nnetwork, and database exporter show, disable, enable and rebuild\noperations.\n\nBlueprint: exporters-feature\nChange-Id: I74a393503390547e1c1a33aaef644a1149dc0056\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/f1ec5d94f901133fcbf4280466bcb99ed5d5c878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/f1ec5d94f901133fcbf4280466bcb99ed5d5c878"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
