)]}'
{"id":"openstack%2Ftrove-tempest-plugin~1003421","triplet_id":"openstack%2Ftrove-tempest-plugin~master~I2ce5ba38941dc2cec11b8494aa1d81face304e74","project":"openstack/trove-tempest-plugin","branch":"master","topic":"replication-unsync-lag","hashtags":[],"change_id":"I2ce5ba38941dc2cec11b8494aa1d81face304e74","subject":"Test replication lag for UNSYNC replicas","status":"NEW","created":"2026-09-02 09:01:59.000000000","updated":"2026-09-17 09:41:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d2ff76bd442575af5e198b99aa9fe96e5bd36caf","_number":1003421,"virtual_id_number":1003421,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-07 08:50:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 09:06:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11368b021068a5ccd5d9939233750a07f509e1a8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-02 09:01:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e836fff5a86ffab1127e1bf7e9d81a93ce1a0f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 09:06:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac71c6cb4cbb49489a981e604709b320\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642e7544f7974f90aeb08813909c9acf : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05dbd1c5aa4a48e1a81cb72ce94854ab : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f7e70234c0f530f5be48d1770c02ea9ad953b942","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-04 06:53:34.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"98536af7daf5db4280d0bb3d8711916e22fb7315","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-04 09:43:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"72bf5b321ab080a33fbc15ca2fc53177792406ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-07 07:33:29.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f93969a0b2f5be40aed841a669bc6b6283e62692","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-07 07:58:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"949656b8d9a8a81f6ee2b31614447be433c546bc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-07 08:41:23.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"b0682565a548174600681e1ce61e1948f7b24f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 08:50:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eabcf1da7154d43a54fde58e37cbf86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f94451b43d1e4a1c99815c97abad0008 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cbc1453d48f44a0a46cdf72bf59a50d : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":4},{"id":"d2ff76bd442575af5e198b99aa9fe96e5bd36caf","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-17 09:41:16.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"690b5de93eb927e4a8ae35e6e477333a48c64884","revisions":{"5fa3b0cf90283ec0af63dbb89950f19c74a9c121":{"kind":"REWORK","_number":1,"created":"2026-09-02 09:01:59.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/21/1003421/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/21/1003421/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/1"}}},"commit":{"parents":[{"commit":"a265af3d4352c74507e95c25f4878d85740848f7","subject":"Test UNSYNC operating status for replicas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/a265af3d4352c74507e95c25f4878d85740848f7"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-02 08:59:25.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-02 08:59:25.000000000","tz":300},"subject":"Test replication lag for UNSYNC replicas","message":"Test replication lag for UNSYNC replicas\n\nExtend the existing replication scenario to verify that instance details\ninclude replication_lag while a replica is UNSYNC and that its value,\nwhen available, is non-negative.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/1001771\nChange-Id: I2ce5ba38941dc2cec11b8494aa1d81face304e74\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/5fa3b0cf90283ec0af63dbb89950f19c74a9c121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/5fa3b0cf90283ec0af63dbb89950f19c74a9c121"}]},"branch":"refs/heads/master"},"0333493bcbec3f19f0de07c021de0715415b3705":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-04 06:53:34.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/21/1003421/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/21/1003421/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/2"}}},"commit":{"parents":[{"commit":"a265af3d4352c74507e95c25f4878d85740848f7","subject":"Test UNSYNC operating status for replicas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/a265af3d4352c74507e95c25f4878d85740848f7"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-02 08:59:25.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-04 06:53:34.000000000","tz":0},"subject":"Test replication lag for UNSYNC replicas","message":"Test replication lag for UNSYNC replicas\n\nExtend the existing replication scenario to verify that instance details\ninclude replication_lag while a replica is UNSYNC and that its value,\nwhen available, is non-negative.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/1003422\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/1001771\nChange-Id: I2ce5ba38941dc2cec11b8494aa1d81face304e74\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/0333493bcbec3f19f0de07c021de0715415b3705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/0333493bcbec3f19f0de07c021de0715415b3705"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8dcea0c5b88c0a6e224f4f4a234f1c837c0977e7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-07 07:33:29.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/21/1003421/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/21/1003421/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/3"}}},"commit":{"parents":[{"commit":"a265af3d4352c74507e95c25f4878d85740848f7","subject":"Test UNSYNC operating status for replicas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/a265af3d4352c74507e95c25f4878d85740848f7"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-02 08:59:25.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-07 07:33:29.000000000","tz":0},"subject":"Test replication lag for UNSYNC replicas","message":"Test replication lag for UNSYNC replicas\n\nExtend the existing replication scenario to verify that instance details\ninclude replication_lag while a replica is UNSYNC and that its value,\nwhen available, is non-negative.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/1001771\nChange-Id: I2ce5ba38941dc2cec11b8494aa1d81face304e74\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/8dcea0c5b88c0a6e224f4f4a234f1c837c0977e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/8dcea0c5b88c0a6e224f4f4a234f1c837c0977e7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"690b5de93eb927e4a8ae35e6e477333a48c64884":{"kind":"NO_CHANGE","_number":4,"created":"2026-09-07 08:41:23.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/21/1003421/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/21/1003421/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/21/1003421/4"}}},"commit":{"parents":[{"commit":"4c361f771f624383f15b5f97f5f089a71f23af45","subject":"Test UNSYNC operating status for replicas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/4c361f771f624383f15b5f97f5f089a71f23af45"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-02 08:59:25.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-07 08:40:26.000000000","tz":300},"subject":"Test replication lag for UNSYNC replicas","message":"Test replication lag for UNSYNC replicas\n\nExtend the existing replication scenario to verify that instance details\ninclude replication_lag while a replica is UNSYNC and that its value,\nwhen available, is non-negative.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/1001771\nChange-Id: I2ce5ba38941dc2cec11b8494aa1d81face304e74\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/690b5de93eb927e4a8ae35e6e477333a48c64884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/690b5de93eb927e4a8ae35e6e477333a48c64884"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
