)]}'
{"id":"openstack%2Ftrove-tempest-plugin~1004984","triplet_id":"openstack%2Ftrove-tempest-plugin~master~I769bba2a8a01288622a59a121cc20b29ba4be654","project":"openstack/trove-tempest-plugin","branch":"master","topic":"quotas-and-dns-opts","attention_set":{},"removed_from_attention_set":{"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-17 13:35:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I769bba2a8a01288622a59a121cc20b29ba4be654","subject":"Add quota and DNS options for test setup","status":"MERGED","created":"2026-09-10 09:43:48.000000000","updated":"2026-09-17 13:36:18.000000000","submitted":"2026-09-17 13:35:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004984-quotas-and-dns-opts","meta_rev_id":"dae3dedf7a707578af6bbed15c297caba9755561","_number":1004984,"virtual_id_number":1004984,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 13:35:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-17 13:25:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-17 13:25:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-17 13:25:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 09:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 13:25:10.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"4691da6f40846c325c1874d419fbd71609523acd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-10 09:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32227174448b9aab56b679bbf4fc977c50512142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 09:50:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c4de3dae35b4215b54f4ba5a3ea4ee2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4f6171bbd8854baa93f532297ccf5ed4 : SUCCESS in 2m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b9d11cd4bdb48769287491d8489fb3d : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9cbc45e8bdc48c9bbfdfcca46ae3db5 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":1},{"id":"29767d077735b258215ae73dc825b9a4c8dafad4","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-17 09:41:33.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"ae042c26d766ade714971728ded24f58bbce1d70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 13:09:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4de7792591d93a2dc89b7008057a7a620284ced0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:20:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56215cc891a6407d85ad8b230b49622f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7dceb2ab27146d6b7a193fff7349d91 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a1b46c441d4a259d7f3c0213d996cf : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cd73ee1b84d4b1d84ce5a94e59d3ef6 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"151ac047e1b8d48c58ed0116c8acf175ca0d5e0c","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 13:25:10.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a106013f9689a3813e5500ba2b0c218d73860e02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:25:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a24665bc033c4b3f842540a6e292ba41594b4d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:35:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea0538f8c2b04d3887dfaf3b2bc57a26\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bb2b36b26ccb427683a29175ad4709cd : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa37555d710462b8fe1f88ce9c623ed : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624588be99714000a39765d92a01b5d6 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"37fba53875080cf5bd7d015ba74b607e2599d9c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:35:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dae3dedf7a707578af6bbed15c297caba9755561","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:36:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e01e68ec8c943738dc1c4847ab6eb84\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c420b726f8543a7824b75c381baf133 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6aeed8a6bf7b653c467c1a702a52156b78fb77d2","revisions":{"97daf09763513e8bac0266a3f59e313b8575c7cc":{"kind":"REWORK","_number":1,"created":"2026-09-10 09:43:48.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/84/1004984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/84/1004984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/84/1004984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/84/1004984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/84/1004984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/84/1004984/1"}}},"commit":{"parents":[{"commit":"e038559f73daf60d56121e624f71b5a294988e8b","subject":"Merge \"Remove jobs for retired branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/e038559f73daf60d56121e624f71b5a294988e8b"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-10 09:39:24.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-10 09:43:32.000000000","tz":300},"subject":"Add quota and DNS options for test setup","message":"Add quota and DNS options for test setup\n\nAdd options to ensure minimum Trove and Barbican project quotas before\ncreating test resources and configure DNS servers for newly created test\nsubnets. Preserve higher and unlimited quotas and existing Barbican\nproject overrides.\n\nChange-Id: I769bba2a8a01288622a59a121cc20b29ba4be654\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/97daf09763513e8bac0266a3f59e313b8575c7cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/97daf09763513e8bac0266a3f59e313b8575c7cc"}]},"branch":"refs/heads/master"},"6aeed8a6bf7b653c467c1a702a52156b78fb77d2":{"kind":"REWORK","_number":2,"created":"2026-09-17 13:09:10.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/84/1004984/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/84/1004984/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/84/1004984/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/84/1004984/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/84/1004984/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/84/1004984/2"}}},"commit":{"parents":[{"commit":"e038559f73daf60d56121e624f71b5a294988e8b","subject":"Merge \"Remove jobs for retired branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/e038559f73daf60d56121e624f71b5a294988e8b"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-10 09:39:24.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-17 13:08:56.000000000","tz":300},"subject":"Add quota and DNS options for test setup","message":"Add quota and DNS options for test setup\n\nAdd options to ensure minimum Trove and Barbican project quotas before\ncreating test resources and configure DNS servers for newly created test\nsubnets. Preserve higher and unlimited quotas and existing Barbican\nproject overrides.\n\nChange-Id: I769bba2a8a01288622a59a121cc20b29ba4be654\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/6aeed8a6bf7b653c467c1a702a52156b78fb77d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/6aeed8a6bf7b653c467c1a702a52156b78fb77d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
