)]}'
{"id":"openstack%2Ftrove-tempest-plugin~725714","triplet_id":"openstack%2Ftrove-tempest-plugin~master~Ie03a8cfc979cb96d90e2dd47a693043bdaf227ae","project":"openstack/trove-tempest-plugin","branch":"master","topic":"stable-testing-tempest-plugins","hashtags":[],"change_id":"Ie03a8cfc979cb96d90e2dd47a693043bdaf227ae","subject":"Add stable branch jobs on the plugins master gate","status":"ABANDONED","created":"2020-05-05 22:44:09.000000000","updated":"2023-06-11 21:53:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0bb676878fe0be5096afa11d69e0cf6ec8a2ac6","_number":725714,"virtual_id_number":725714,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-05 00:40:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 21:33:08.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-06-05 00:40:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e144c546f3a2ee7edc0669c1fc0f1b3ae1c33b31","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 22:44:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f571b0ea8a1af3426cce6f68100238a9a481cda0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 00:20:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/257b83a192db44788ab7c25753902bc2 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f63c48ac824d4751a7fa2136b960ee0a : SUCCESS in 4m 47s\n- trove-tempest-plugin https://zuul.opendev.org/t/openstack/build/a262046755c349eda60093630246c963 : SUCCESS in 52m 21s\n- trove-tempest-plugin-ussuri https://zuul.opendev.org/t/openstack/build/80a9b022dd0e451aa8de18be377480ca : FAILURE in 1h 13m 23s\n- trove-tempest-plugin-train https://zuul.opendev.org/t/openstack/build/4aa2eb46f45043e39c8872452beb946a : FAILURE in 1h 35m 27s\n- trove-tempest-plugin-stein https://zuul.opendev.org/t/openstack/build/c42eeba9a1e940778312f0f95ce2c971 : FAILURE in 30m 58s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/877132188ab6472da43900126ca79c74 : FAILURE in 1h 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c4285275a9517e658c4b4600cabcf890c58c021","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-07 11:26:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4495600654a2314d6e3e1014b2f55a4333abd58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 12:52:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1167167db69d4203aeaf90ad71c2966e : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23032eb4a48a4455aedca9e06d902910 : SUCCESS in 6m 54s\n- trove-tempest-plugin https://zuul.opendev.org/t/openstack/build/e53c807d734b4238bf706013d7a3808d : SUCCESS in 1h 08m 25s\n- trove-tempest-plugin-ussuri https://zuul.opendev.org/t/openstack/build/3540761da15a4b488cbf40dc7b539c4a : SUCCESS in 1h 09m 13s\n- trove-tempest-plugin-train https://zuul.opendev.org/t/openstack/build/4d6f23fb2b5f4a318db6b00ca7f962fd : FAILURE in 1h 15m 44s\n- trove-tempest-plugin-stein https://zuul.opendev.org/t/openstack/build/842a78c7098c4151bdcd9614ce965680 : FAILURE in 34m 09s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6b2467329c148f5b43fdc92b4e98478 : SUCCESS in 1h 01m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58906b68bf2222fccb7d088b3111e889069260a9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-08 21:32:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ba4bbd4710abd1fa1726a4b3bbb126c5d5bc54c8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-08 21:33:08.000000000","message":"Patch Set 2:\n\ntrove tempest plugin was introduced in stable/train, so stable/stein is not gonna work.","accounts_in_message":[],"_revision_number":2},{"id":"1c05dd79f6d680c6d113638135569820b2c86b90","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-08 21:56:43.000000000","message":"Patch Set 2:\n\n\u003e trove tempest plugin was introduced in stable/train, so\n \u003e stable/stein is not gonna work.\n\n+1, thanks for the update.","accounts_in_message":[],"_revision_number":2},{"id":"2911d8116737bc52780f309e254b873cd698f405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 23:06:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b282cf0c30f74a098838fa1e7b174744 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91bc7b0217894edaa34fdf8b9a6a2128 : SUCCESS in 3m 49s\n- trove-tempest-plugin https://zuul.opendev.org/t/openstack/build/7e534a929515459186e10d62b95be9cc : SUCCESS in 54m 03s\n- trove-tempest-plugin-ussuri https://zuul.opendev.org/t/openstack/build/2a2487fb5a0f4512b3a222e25973fe65 : SUCCESS in 1h 02m 24s\n- trove-tempest-plugin-train https://zuul.opendev.org/t/openstack/build/5e80c427350f446fa93e940aa4fcbf94 : FAILURE in 1h 32m 58s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e307157b79d4a3fb996807d2a60b0ec : SUCCESS in 58m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac10242e077197b03fd59e30b9af7fd2a52f4047","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-04 22:59:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4f6d92d2823be8f76cf04d93d178572fbbc65dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 00:40:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d3f7356b2b94e189412db061f4295df : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0bc5c354c0f45c58354422df66c6a96 : SUCCESS in 3m 13s\n- trove-tempest-plugin https://zuul.opendev.org/t/openstack/build/f6262e1db0a646db8ec144a07bf79561 : SUCCESS in 1h 15m 18s\n- trove-tempest-plugin-ussuri https://zuul.opendev.org/t/openstack/build/e9ccc2f3baf94b50b0cf1470b0d2a9ae : SUCCESS in 35m 28s\n- trove-tempest-plugin-train https://zuul.opendev.org/t/openstack/build/f0183aa6786b40c2944ecddb1b7bc467 : FAILURE in 1h 35m 12s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2fbd785094343ba8a78649680230f1a : FAILURE in 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0bb676878fe0be5096afa11d69e0cf6ec8a2ac6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-06-11 21:53:22.000000000","message":"Abandoned\n\nneed to refresh this for latest stable branches. I will push new one","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ba7e79eb8c50f551a1b661ba11d7c79ce951775d","revisions":{"0899dbdf086b73514f6753e14633f1798cba5f52":{"kind":"REWORK","_number":1,"created":"2020-05-05 22:44:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/14/725714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/14/725714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/1"}}},"commit":{"parents":[{"commit":"702d4a9dd4af3ed863b3d3ca4ef1851135aad500","subject":"Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/702d4a9dd4af3ed863b3d3ca4ef1851135aad500"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:42:06.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:42:46.000000000","tz":-300},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ie03a8cfc979cb96d90e2dd47a693043bdaf227ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/0899dbdf086b73514f6753e14633f1798cba5f52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/0899dbdf086b73514f6753e14633f1798cba5f52"}]},"branch":"refs/heads/master"},"85d6faaffd57d228adef06fd367d4f75bb86fff4":{"kind":"REWORK","_number":2,"created":"2020-05-08 21:32:42.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/14/725714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/14/725714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/2"}}},"commit":{"parents":[{"commit":"702d4a9dd4af3ed863b3d3ca4ef1851135aad500","subject":"Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/702d4a9dd4af3ed863b3d3ca4ef1851135aad500"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:42:06.000000000","tz":-300},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-08 21:32:39.000000000","tz":0},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ie03a8cfc979cb96d90e2dd47a693043bdaf227ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/85d6faaffd57d228adef06fd367d4f75bb86fff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/85d6faaffd57d228adef06fd367d4f75bb86fff4"}]},"branch":"refs/heads/master"},"ba7e79eb8c50f551a1b661ba11d7c79ce951775d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-04 22:59:48.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/14/725714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove-tempest-plugin","ref":"refs/changes/14/725714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove-tempest-plugin refs/changes/14/725714/3"}}},"commit":{"parents":[{"commit":"31894c68e4b6b5b1a68578a7bf62ee305bbb3bc5","subject":"Merge \"Cleanup cruft\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/31894c68e4b6b5b1a68578a7bf62ee305bbb3bc5"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:42:06.000000000","tz":-300},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-04 22:59:48.000000000","tz":0},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ie03a8cfc979cb96d90e2dd47a693043bdaf227ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/ba7e79eb8c50f551a1b661ba11d7c79ce951775d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove-tempest-plugin/commit/ba7e79eb8c50f551a1b661ba11d7c79ce951775d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
