)]}'
{"id":"openstack%2Ftrove~1000734","triplet_id":"openstack%2Ftrove~master~Ida10b61b1fc3a6e0e4ea1823bce7c7dd0714cb3c","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-08-19 12:20:24.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-17 17:51:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ida10b61b1fc3a6e0e4ea1823bce7c7dd0714cb3c","subject":"[DNM] test current master","status":"ABANDONED","created":"2026-08-12 14:13:37.000000000","updated":"2026-09-17 17:51:26.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a02b3c7d8db50d2544be72dfcf04f11725bd04a4","_number":1000734,"virtual_id_number":1000734,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-21 07:02:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-12 15:36:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-19 11:54:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"f0fb8a30c7ed9b6bb6451c9124438aef8710c5b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-12 14:13:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0221559b85310503be782dcb684a10dcf84a2372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 15:36:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d50e018688c44d75bf87b9b33732960c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e71bbc040b141fa9f722ca8e9afccdb : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fe07699899c484b94102d58d895fd62 : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4dc4d0fa701840adb4f944e994a6ada5 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/336dc59a726748fdbfb0bb022cc6e786 : SUCCESS in 6m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a37b7e09d7884a22b158c24543e2d75e : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f8b62cee7cb4d078787db6a54828a86 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a975afa1b654acc86004ba6ac7df0c0 : SUCCESS in 3m 54s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/4e82077da1e440319cffe6defc5ba0d7 : SUCCESS in 43m 27s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/f55f6d693cbb4f07822c54265bf95d34 : SUCCESS in 1h 13m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/3e05278ff18c46e9841f5483629f0eba : SUCCESS in 1h 06m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/e005a3d4aa4b48a8876773667d55d71c : SUCCESS in 1h 15m 15s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/74e066ae096c40a8b2410c930ec8d08d : SUCCESS in 41m 38s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/82d3d5de4aad42d29a8194ef0067a988 : SUCCESS in 1h 19m 29s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/8e8157deaced46fd9123c5d072839ea7 : SUCCESS in 1h 00m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/dcd17249a88d4b85a4922bbcc5730f16 : SUCCESS in 1h 13m 46s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/111b811787de4778b5f64673c06a443a : SUCCESS in 46m 01s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/82c6a16592ca409fad97524579c6a5b3 : SUCCESS in 1h 08m 57s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/3e1785cf39124b53a1bd83c754ab0753 : SUCCESS in 1h 01m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/1e5e07f7a816414badf0e30df9fc7f73 : SUCCESS in 1h 04m 33s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6fa86b5cc13542b39e69fcc43b031622 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"13f330c6e3d1b8aae67a3ecdb2bfe1e80036745e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 16:44:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b99db22ad21456a97bed6005e3a0a3b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/22f68d3b3f074915a0eeb301cb5ed16a : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad141d4e89424e6cafeab1f1bb4c20b5 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d46c410e75db43248398de131985f599 : SUCCESS in 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13f6068ba73dd42756164cc36f95be5c66b5716b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-12 19:17:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a8f0ac2bf33d0a5ee9e0eeabe18d381a330e245a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 19:45:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caafe06327144aa88951933c107ae232\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1bb4b9a5a37e4dd1bc53c20e03273e36 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/53a024b4171c47f888b5f05c27974f17 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6b9371410eb24e63af0257f23e74fc79 : SUCCESS in 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0fc93cb6129b335fab828cc0b9468211f6c3782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 20:38:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45135e3ba9ef46b7a380ecdbe2bad8eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a840c4fe86b34951956741e4d3c96f05 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae3bc4703e6f44c2a6f36dd5f89a25a2 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2483ab676c14bdfb857087ed9786d86 : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afac659f93864df2a82a8178b84dc419 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3564399d890b47b39f668f9545f6c20c : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82461ae260e94673bda3bc28e690f24f : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b32b833f07d40fea09a51e6a5392ef4 : SUCCESS in 7m 41s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c183283ad4a648dc9bdea48e7dd67c1b : SUCCESS in 42m 26s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/322a2d3f70074258a7abfd04068b897f : SUCCESS in 1h 12m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/9621df7a94a845a2aa93df66edac6f05 : SUCCESS in 1h 09m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/0f03e15d51a6484c8a9bdf5c96527fe0 : SUCCESS in 1h 19m 38s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/59799adbce3247eca72e4bf104b2b48b : FAILURE in 44m 05s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/c98f29cd519449fc99df574c02f24055 : FAILURE in 36m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/3290d765517c4c11b5039ce7cc94a15e : SUCCESS in 1h 09m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/39133919754f4634b39cda14361bf748 : SUCCESS in 1h 18m 45s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/a482befb034548bfa7d266e29733c6a9 : SUCCESS in 41m 44s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/f23cfbb1c509406eb1458c7e9b4b9397 : SUCCESS in 1h 04m 48s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/0a4f4407be2f40de8e7f6460f4f1b3a9 : SUCCESS in 58m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/5d84de4010514952abbde4c5fb0d994b : SUCCESS in 1h 00m 21s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e9d387eaf5784e27aebd1ac0462391be : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e7caff372db84204186875f0ed7311507c780d04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-19 11:50:08.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3b4927e60acfae5970ee02e132e8cb8fafccfc0c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-19 11:54:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e279025f7e750f04d8f6d6f0520bedbf938b3355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-19 12:19:54.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ef8cccc6f1a7344193c4f9338be5889e8972b737","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-19 12:20:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"abac3e27df1b51f1c796e8bb6377042b9e6887f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 12:58:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db3d141097694ad8bf53d6a0a7a30f2a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de6a01fb41354fd5b876d51c9b0da3e4 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f8a08e961ad493d9d2ee3089f8ed717 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8f97220746c24526b3e33cb2827a864e : SUCCESS in 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"86bf5d3ecaa74cea5e74cce315c26fff19b69232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 13:38:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d502213fb3f4ed1bf317d73d4f84271\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/401780015c1547c0bde6d6f9f9f7bdaf : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55290d353dba446093d89da7bcec9e0f : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8f823fbb34f415aa3b1b10eb463070a : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8400c4c140cc4145bd75ed70942af1a6 : SUCCESS in 6m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1a5d6a619b7d468198352d9e64284e5b : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8369e2ba9134dfd841d426d4de4282b : SUCCESS in 9m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9c94b7c512145c68dbeb8cb3df9453e : SUCCESS in 7m 38s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/65f66ed3f48e4ce5a99da8d8e119c24b : SUCCESS in 43m 07s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/8d9d69845cb74f4facd23e3217ebf8d5 : SUCCESS in 1h 13m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/f21c07ec81c34160ad979e1378d3a04c : SUCCESS in 1h 05m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/86e556a3af744e268dccd6f82acef54c : SUCCESS in 1h 16m 36s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/147a0cd7974c4ee498b04f08f6ced39a : SUCCESS in 40m 01s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/145de47520bb418f938004925dcd7fde : FAILURE in 30m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/6b1bd16d245349adbec5f81eb38b77d7 : SUCCESS in 1h 04m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/350698aee91a4db18c37b37bc9763fc6 : SUCCESS in 1h 16m 56s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/18b0d8aaf8ad4898a3372ce536bb0d31 : SUCCESS in 42m 49s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/a4c4bffd78fa45a89a4a76cc950b458e : FAILURE in 31m 30s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/2e33fa9ce94d4c7ebd7c74fbcec6189c : SUCCESS in 51m 43s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/141abf09bb3e43f199d327d3af2dbcd0 : SUCCESS in 59m 39s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d9b3bd8e8f704d598ac2476dfc3e38b2 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":4},{"id":"d074ea9eff7855469cd9ba63fcd99c7786bfa98b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-20 18:18:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7eb5d979074affc7f40966f46753f85321600578","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 18:45:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/157bd6d97e25433aa6874fbb5b525363\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/04b84839c9d841e7a327289ad8278b3e : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/48abed62b61d40469d2fa7e633b28651 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a0d156bf47941d1b86e83e49d91fe84 : SUCCESS in 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7b80ec80a2a98a8b963de3b408c066ad0eac4ebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 19:38:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68c7e09ce8da459b820ec4cb3edbfea9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d72441b32994f14bb5a8bca2ed0bdde : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ede07ddd3ab74680a7149a94ecb6b3fc : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3722776529f94afb87ebf0111482d76e : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d50ad0685634b8ea7abff080986c759 : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/51afbbc86a184bffb542b82537c39ef5 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad4b700e07884f19b376387453f4794e : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e28f2bc3d724ed184078293f5e0b5f6 : SUCCESS in 3m 48s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/0af2701b463c46d29be24c1c422be424 : SUCCESS in 39m 37s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/9a6d65b2e53f40d7b3032354e98dde59 : SUCCESS in 1h 15m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/b140215d13fc4f53aeb069b935323e36 : SUCCESS in 1h 05m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/c6718d4e94134d288d1c173bffac1609 : SUCCESS in 1h 18m 26s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/7f12e258728f431a9ce36812278be522 : SUCCESS in 37m 09s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/6bf0141c06e3433b8c1a877162b0403a : SUCCESS in 1h 12m 29s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/17fe117168fd4de6bec697539592f386 : SUCCESS in 1h 03m 25s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/ff93ffcb6afa4b13bdb55462000cc862 : SUCCESS in 1h 12m 09s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/7fe858cd51784865a3094b901f2dcbc2 : SUCCESS in 39m 34s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/ada3b2a6be754576b2a50fa0780e0cc8 : SUCCESS in 1h 03m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/6516738cdfbf4e09be00ab26f29a77f1 : SUCCESS in 55m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/a8fb94d2395f40b08a9518d50511040e : SUCCESS in 1h 01m 31s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a321fd5420954d3db0c39d0306faaede : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":5},{"id":"fe95baa986b22923eff3ac714775623eebcdf612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-21 05:39:31.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"946146da5547d95f587ff70757aa7254cd4d799f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 05:59:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/094fd5d60c4949919065479b57cf444d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/087133643d004464b4a6b987949ef173 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5e8554ce0c774749bd0d06746cdf8788 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/83c12751bae64a85ba96ae3d90f93ce6 : SUCCESS in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f4fb72ef8eebd019ba8f4cec468b898f2ab33ef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 07:02:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70ece1a710e437a8588ff50b971996d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1752179fe6b743a5bbfdce05162fb425 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e5df7b26cc457382683f21947e094a : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c466d34bdea446379b8279aef658b4d8 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d79925dcf7b4a5592f5d33e9d60de0e : SUCCESS in 4m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7ddc8feecb54bafbbad32723ff84506 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2075eccfed5e4a38bff4911acd6c42e9 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36dc4674f17b441093efb57dae6433c4 : SUCCESS in 7m 43s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/7a979686b4ff41fe9ba00058d0a15671 : SUCCESS in 40m 44s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/4bae9953bd494861adb4399743737229 : SUCCESS in 1h 08m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/812f562e57d84ebb926a3a8997baca34 : SUCCESS in 1h 02m 18s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/ece1e7373d9f46eb9022b673c8234c0b : SUCCESS in 56m 10s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/edd3c520981543e9a5dea038be7cc894 : SUCCESS in 1h 02m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/52a3da26508a4739814e7a1fb118a0c5 : SUCCESS in 1h 14m 31s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/412d7c44d4134c7490ae5659fd5a443e : SUCCESS in 40m 06s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/4504b71e4db9445ab2a1e2c0d4607ef5 : SUCCESS in 1h 08m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/271b4daa54b94c3a8236941e8923a316 : SUCCESS in 58m 11s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/ac0ff4970b194c1da479c9591e98f773 : SUCCESS in 51m 30s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/00ba77df8fe94018a50011d5da1cdc3b : SUCCESS in 1h 08m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/15f3f34c27344fd99e1dcc341eede577 : SUCCESS in 1h 17m 13s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/6cc3254762634ad8a6e2415970d12941 : SUCCESS in 45m 08s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/8cd6678fb7964f2bb38bbf8a8b59c8db : SUCCESS in 1h 02m 53s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/ce8085ea404046689439d1b0513e5ce9 : SUCCESS in 42m 07s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/d4ff7b87fa134c1397e771e487dffb26 : SUCCESS in 1h 00m 22s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/74dfa90809d742e484aa353e0c4fdbd2 : SUCCESS in 59m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/7ee9fb409c494abc96ad4e3f950cb0a2 : SUCCESS in 1h 03m 11s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a36ec135f17d4d2e8743b8e219dafe76 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":6},{"id":"a02b3c7d8db50d2544be72dfcf04f11725bd04a4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 17:51:26.000000000","message":"Abandoned\n\nOutdated MR cleanup.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a73ba5ef3bb555ab86698795380b98dee2ed72bc","revisions":{"a7926b7fad590913238429dc08895cdec6442362":{"kind":"REWORK","_number":1,"created":"2026-08-12 14:13:37.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/34/1000734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/1000734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/1000734/1"}}},"commit":{"parents":[{"commit":"af8ebb3604d092b522d9376afe50ac01e58f1e31","subject":"Merge \"Build guest image from local repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af8ebb3604d092b522d9376afe50ac01e58f1e31"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-12 14:12:18.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-12 14:13:27.000000000","tz":300},"subject":"[DNM] test current master","message":"[DNM] test current master\n\nNeed to verify whether cinder tests are working.\nTests in my local DevStack doesn\u0027t work in tests when external\nconnectivity is required.\n\nChange-Id: Ida10b61b1fc3a6e0e4ea1823bce7c7dd0714cb3c\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a7926b7fad590913238429dc08895cdec6442362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a7926b7fad590913238429dc08895cdec6442362"}]},"branch":"refs/heads/master"},"b76d54182a6f2ce022d4b75bbb6df52ed9c44cd4":{"kind":"REWORK","_number":2,"created":"2026-08-12 19:17:06.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/34/1000734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/1000734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/1000734/2"}}},"commit":{"parents":[{"commit":"af8ebb3604d092b522d9376afe50ac01e58f1e31","subject":"Merge \"Build guest image from local repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af8ebb3604d092b522d9376afe50ac01e58f1e31"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-12 14:12:18.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-12 19:16:54.000000000","tz":300},"subject":"[DNM] test current master","message":"[DNM] test current master\n\nNeed to verify whether cinder tests are working.\nTests in my local DevStack doesn\u0027t work in tests when external\nconnectivity is required.\n\nChange-Id: Ida10b61b1fc3a6e0e4ea1823bce7c7dd0714cb3c\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b76d54182a6f2ce022d4b75bbb6df52ed9c44cd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b76d54182a6f2ce022d4b75bbb6df52ed9c44cd4"}]},"branch":"refs/heads/master"},"87202deae13482d5e47275e7f29169b6dbc4f81e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-19 11:50:08.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/34/1000734/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/1000734/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/1000734/3"}}},"commit":{"parents":[{"commit":"4cdc0cdadc76229aaa16195696d3aa2bc295649c","subject":"Merge \"Fix \"no space left on device\" for backup test jobs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4cdc0cdadc76229aaa16195696d3aa2bc295649c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-12 14:12:18.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-19 11:49:47.000000000","tz":300},"subject":"[DNM] test current master","message":"[DNM] test current master\n\nNeed to verify whether cinder tests are working.\nTests in my local DevStack doesn\u0027t work in tests when external\nconnectivity is required.\n\nChange-Id: Ida10b61b1fc3a6e0e4ea1823bce7c7dd0714cb3c\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87202deae13482d5e47275e7f29169b6dbc4f81e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87202deae13482d5e47275e7f29169b6dbc4f81e"}]},"branch":"refs/heads/master"},"41a237cf89742e50dab80a5c7c519247d973653b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-08-19 12:19:54.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/34/1000734/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/1000734/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/1000734/4"}}},"commit":{"parents":[{"commit":"4cdc0cdadc76229aaa16195696d3aa2bc295649c","subject":"Merge \"Fix \"no space left on device\" for backup test jobs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4cdc0cdadc76229aaa16195696d3aa2bc295649c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-12 14:12:18.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-19 12:19:07.000000000","tz":300},"subject":"[DNM] test current master","message":"[DNM] test current master\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/1001367\nChange-Id: Ida10b61b1fc3a6e0e4ea1823bce7c7dd0714cb3c\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/41a237cf89742e50dab80a5c7c519247d973653b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/41a237cf89742e50dab80a5c7c519247d973653b"}]},"branch":"refs/heads/master"},"dbd464109fa62d42e3dd35babd24b6066ced7e53":{"kind":"REWORK","_number":5,"created":"2026-08-20 18:18:17.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/34/1000734/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/1000734/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/1000734/5"}}},"commit":{"parents":[{"commit":"4cdc0cdadc76229aaa16195696d3aa2bc295649c","subject":"Merge \"Fix \"no space left on device\" for backup test jobs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4cdc0cdadc76229aaa16195696d3aa2bc295649c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-12 14:12:18.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-20 18:18:04.000000000","tz":300},"subject":"[DNM] test current master","message":"[DNM] test current master\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/1001367\nChange-Id: Ida10b61b1fc3a6e0e4ea1823bce7c7dd0714cb3c\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dbd464109fa62d42e3dd35babd24b6066ced7e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dbd464109fa62d42e3dd35babd24b6066ced7e53"}]},"branch":"refs/heads/master"},"a73ba5ef3bb555ab86698795380b98dee2ed72bc":{"kind":"REWORK","_number":6,"created":"2026-08-21 05:39:31.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/34/1000734/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/1000734/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/1000734/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/1000734/6"}}},"commit":{"parents":[{"commit":"4cdc0cdadc76229aaa16195696d3aa2bc295649c","subject":"Merge \"Fix \"no space left on device\" for backup test jobs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4cdc0cdadc76229aaa16195696d3aa2bc295649c"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-12 14:12:18.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-21 05:39:20.000000000","tz":300},"subject":"[DNM] test current master","message":"[DNM] test current master\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/1001367\nChange-Id: Ida10b61b1fc3a6e0e4ea1823bce7c7dd0714cb3c\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a73ba5ef3bb555ab86698795380b98dee2ed72bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a73ba5ef3bb555ab86698795380b98dee2ed72bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
