)]}'
{"id":"openstack%2Ftrove~1001926","triplet_id":"openstack%2Ftrove~master~I07b9622c0788edee91d39dd9e5cf1030864d44ca","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-08-27 09:44:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I07b9622c0788edee91d39dd9e5cf1030864d44ca","subject":"Pull embedded docker images with ctr","status":"MERGED","created":"2026-08-22 15:50:59.000000000","updated":"2026-08-27 09:45:57.000000000","submitted":"2026-08-27 09:44:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001926","meta_rev_id":"7ca281cd2cb5e73b4993da6b5cd1ad583dc5aa0f","_number":1001926,"virtual_id_number":1001926,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-27 09:44:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-08-27 09:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-08-27 09:00:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-08-27 09:00:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-22 15:57:07.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-08-22 16:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-22 17:10:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-27 09:00:15.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"d5b3cb9609228fb957ac142e20054e13824ae8c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-08-22 15:50:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85853dc7183287d14751939e00a2bec8914e3be1","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-08-22 15:57:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6248bde24128c8346a1a056f675357d30c7d94c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 16:11:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3e8f1ca92fb42cbafbf4de0b7d0831d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6be4023c94e34b50933420e0e40f546d : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b5e56899b6e4f90863d133ab9ac20cf : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1de9ff7188f54be4bdef4b9f113714fa : SUCCESS in 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0550db4dba4a6b32fb09835d366962bbdd858cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 17:10:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd4f93c99c64f3d8d93a882905ef00b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1a9da32f99b4cffb2a66fda25c56f7a : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf9a6c17909d45f9815c5e387ad4f5a9 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52c33621f26d49d887fa10dec2d5d95a : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ac378b6a243433a8721485757ced2ad : SUCCESS in 3m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a1cbbd2953a4cc397e24102f701c9bf : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad6ea1ed4e849198713aee85a09b5fc : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80f8dfa2c8514ec3b92740e9bdb7d9a0 : SUCCESS in 9m 42s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/465ed42c126d48c786ea9ef3db2f2184 : SUCCESS in 1h 06m 20s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/64fd045f94a24e69b97015db353f0c88 : FAILURE in 35m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/844e45e334234db39ea2fcdfb1415be1 : FAILURE in 35m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/8de143439e224f8dbafcdcf8ca8d195c : SUCCESS in 1h 16m 12s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/e8c456c47ae84be884de9bf1d29eccb7 : SUCCESS in 45m 21s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/e9490ba0def149159d29c4d84ba0d984 : FAILURE in 1h 16m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/8b7f814728664d73aa420b9ac4b6e895 : SUCCESS in 1h 09m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/6956ab0eba454cdab55f1ca93e386031 : SUCCESS in 1h 18m 13s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/1b7a94b5e1ab4e9d8e720917613c504f : SUCCESS in 43m 26s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/a73b449c84ee4324ae85857dc6312956 : SUCCESS in 1h 06m 40s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/2ecb1e8f139545c49cfae9664cb02cfc : SUCCESS in 1h 04m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/16839f6d515e4094bb76ddb97a807920 : SUCCESS in 1h 02m 23s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f4852a12e2f4a2f984fb420a38629c8 : SUCCESS in 5m 00s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/c17e642fb2894f2b91108927f28d3a5f : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a28faf8d20337890f30784a698ef0a5634accb40","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-27 07:12:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b1b0a76a8772018fa5d7adac3c4b486360a2f6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 07:42:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40a5e81c5ae944e4916e50d4738ed9df\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8c0bde7bb816492ab80f99e8429c0fb4 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/465aa79b5af9403599ddabd955c63a9d : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/132f10133f2b4c7ab3a90f0348fcee2a : SUCCESS in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7ad4008f417666bf5100c81e877ac58dafb2288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 08:32:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44c660ca6f7d4f1c8f90f42d17963720\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2f739507ba04510b8d64649020cd80e : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf8bf300e4a7433a892166714e2cb741 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39b39f4243cd42a5b72744b2abbfec8c : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0c1be15295d448a895a95fdc5f81061 : SUCCESS in 4m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/79f033310e5342588419f10ccdd3a5d8 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8d09681d12348a2ac1aa9cf1fec9c32 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/158c70124a124e3aa7c1d1b92fd686c5 : SUCCESS in 4m 30s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/96098fed3ec64d31bd806bd3b1167730 : SUCCESS in 43m 33s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/9e15eccd2d684da48673ee1cfb0e53f4 : SUCCESS in 1h 15m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/48067fe676dc47a0a53bcaf393404c31 : SUCCESS in 1h 05m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/b4d6860e53954d569b99fd10552ea45c : SUCCESS in 1h 18m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/a8cf4be252104cc5ad96dd4509225400 : SUCCESS in 43m 11s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/fd756967fab94b74b33604052b9a0173 : SUCCESS in 1h 08m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/7e58c9cfa34d4d928ec410906201c32d : SUCCESS in 1h 05m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/fd9accfd60814f0fa972961b4f2eba59 : SUCCESS in 1h 13m 33s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/2fa1e9b3f94a48cc975466dc26523b62 : SUCCESS in 44m 42s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/c48a1fa1d37b4771aa30c072377c6f8e : SUCCESS in 1h 06m 38s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/93e0e27b0c604c66b19e707819570440 : SUCCESS in 59m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/68e5e12012d04336ae82ee813930d161 : SUCCESS in 1h 04m 11s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e7c459c0ba64418ba1aa8dcbacac16e8 : SUCCESS in 2m 54s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/19621cd3a1dd4afba44eab572d190abe : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ad2d27d7ab1cef141a1bd4aa6efa62f881e97cba","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-27 09:00:15.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd42ee5d7cb1dc259c1e6491c11760e4d18c5fce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 09:00:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9a767f1c23a81c2d5a810046abaa993013fd935","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 09:44:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cc84358f5084a51b88aaa8cae04e5ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b75e8fdf7c47a9995aaed1e4f44c6e : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11b0507429ea49fcbf166059cb093733 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26fb5f0decb243508f347638761454c9 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f53331c5e7a042a4a8a0304bd8606477 : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/825837630063493bb706ceafbbd9af19 : SUCCESS in 8m 03s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/5156f95554f941069ba1b5483529f2c7 : SUCCESS in 42m 10s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/0e3699b9610c4529b17ea56eabde7e19 : SUCCESS in 40m 03s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8ad81e031c0147cd9087968a49f5f90f : SUCCESS in 43m 20s","accounts_in_message":[],"_revision_number":1},{"id":"003b0e916e67e1ec546fe569edc15a54ff54f20e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 09:44:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ca281cd2cb5e73b4993da6b5cd1ad583dc5aa0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 09:45:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d2d1d99879497f95cab3717590b59b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b05437f2ba04bd6b7b0a4995407b9e0 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5f2728eb6431d76c0d5705cc5fa3bff713444dc","revisions":{"f5f2728eb6431d76c0d5705cc5fa3bff713444dc":{"kind":"REWORK","_number":1,"created":"2026-08-22 15:50:59.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/26/1001926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/1001926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/1001926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/1001926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/1001926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/1001926/1"}}},"commit":{"parents":[{"commit":"4cdc0cdadc76229aaa16195696d3aa2bc295649c","subject":"Merge \"Fix \"no space left on device\" for backup test jobs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4cdc0cdadc76229aaa16195696d3aa2bc295649c"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2026-08-22 15:39:18.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2026-08-22 15:50:49.000000000","tz":480},"subject":"Pull embedded docker images with ctr","message":"Pull embedded docker images with ctr\n\nInstead of pulling images on the build host with docker, saving\nthem to a tarball, and loading them into the guest chroot with\ndockerd, pull images directly into the image\u0027s containerd store\nduring the guest image build. This avoids redundant disk writes\nand removes the build host docker dependency.\n\nRetry with --plain-http when HTTPS pull fails so HTTP local\nregistries work without breaking pulls from HTTPS registries\nsuch as quay.io.\n\nCo-authored-by: Cursor \u003ccursoragent@cursor.com\u003e\nChange-Id: I07b9622c0788edee91d39dd9e5cf1030864d44ca\nSigned-off-by: wu.chunyang \u003cwchy1001@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f5f2728eb6431d76c0d5705cc5fa3bff713444dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f5f2728eb6431d76c0d5705cc5fa3bff713444dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
