)]}'
{"id":"openstack%2Ftrove~1003629","triplet_id":"openstack%2Ftrove~master~I21385586c06a73d61d750a3840381c4cdfd66e0d","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-17 09:23:26.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-17 10:01:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I21385586c06a73d61d750a3840381c4cdfd66e0d","subject":"Use high timeout for database shutdown","status":"MERGED","created":"2026-09-03 06:29:38.000000000","updated":"2026-09-17 10:03:20.000000000","submitted":"2026-09-17 10:01:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003629","meta_rev_id":"ea8d9aa95ef048d73785a9a70b4f64cbbfb37a3d","_number":1003629,"virtual_id_number":1003629,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 10:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-17 09:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-17 09:23:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-17 09:23:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 06:48:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 07:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 11:45:50.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-09-17 09:23:26.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"762e235e6629208a8ab3783576ab7a3edf481bdf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-03 06:29:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f550184c43224663b37c930381b4c8cc19e96951","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 06:48:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c5ccb4f603642039fc076ff7d5d8c22\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6253ba64cb2445fbb9ffe0f138eb2659 : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62617fe3b46744bf874a1a12b65b9eb6 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/14a7d96258c043f99cdd3d1f808b8d6c : SUCCESS in 16m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21b6239a5780eb6f2ee0f285813173c9ce5ba780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 07:46:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05541a30d8af4a8ea83630ffb29dec60\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0468fa4c167e4e558ba3bb4721538f08 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/324e3a5c96b5457cb18f5a38dfa0336e : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34f33b788f4b48f597e77a44324f54d9 : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/596ca10c48f94ba1ae547f477dd89924 : SUCCESS in 2m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26cb22183e014eeab348d48576e694a9 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ab659a32ea4e70b1b44c3523c606f6 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f3712f5dafe4f569f036de078b13053 : SUCCESS in 8m 41s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/defe8bb2cfce424daf6dbb9a1ef7136d : SUCCESS in 45m 31s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/3e855cb105e54ab6bd4d46271c31f59e : SUCCESS in 1h 12m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/7b12dadff5204d9b8957f82ce6002b14 : SUCCESS in 1h 05m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/0a10ba6526304442ac86b92e1efb1052 : FAILURE in 22m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/586a2b6027ba40be84e5b4405a622bee : SUCCESS in 40m 46s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/b8cd8458277e41c48412df19214f4367 : SUCCESS in 1h 16m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/ae2f1e46e37b4a39ad91237f1102efd8 : SUCCESS in 1h 08m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/92db63fb93bd420dbd167602098cf076 : SUCCESS in 1h 12m 42s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/24ee8258461b4db1b495c437ea6558d5 : SUCCESS in 42m 44s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/ef6107d9e00b4ffca41f178b99c82910 : SUCCESS in 1h 07m 34s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/2a58acbb9be142199895a2adc94cd44e : SUCCESS in 1h 01m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/67207ee515124009aa46e485750bfa16 : SUCCESS in 1h 04m 24s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/38c2ce6121834d59b706546f762516aa : SUCCESS in 2m 52s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/fips/enable-fips.yaml for job trove-tempest-ubuntu-replication-mysql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/multinode/enable-multinode.yaml for job trove-tempest-ubuntu-replication-mysql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/pre.yaml for job trove-tempest-ubuntu-replication-mysql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/devstack-tempest.yaml for job trove-tempest-ubuntu-replication-mysql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/post-tempest.yaml for job trove-tempest-ubuntu-replication-mysql.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"c72bc67cc3e7433212d1c3e89d09b25ba1a64128","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-03 11:45:50.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"26bce0bf57f03b30f76d60678bbd6f15a0b6b055","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-04 05:27:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1bfc25945ed4beabfd4589d0669fe88f0ea7fd88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 05:42:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b21893bac04efab986ab9efdd13288\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/859a24a603e9414aad97f670159cdb76 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/50f882e712f64da89e6f657bc6d93b85 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a41ead88058646aaaf4af6079959decb : SUCCESS in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7e4bb4bdb044eefe9d79963c0cdf92a5b1b1c18","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-04 05:43:03.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4c8dfef73e38b83d533ea5aed467423ad9484423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 07:06:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e4ee2dd2b2d4cd0bfb9664970957da8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c06a5ca3b8624bb28e421b852f940a95 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e120ce10b442ad8d90e440a03c0ef2 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53206e7fc4de45bdb6361afe5507a5c3 : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/735fd0145c404f4385323ec44458d786 : SUCCESS in 3m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c998776d196641599b163f7d0ae0498f : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e5cd5f87804ad7879f4b9c4e8781c9 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f0d2382eed3478fa672476f5086a0ae : SUCCESS in 3m 37s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/e209fed35e304eeca0bb9bbf0938201f : SUCCESS in 1h 03m 38s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/0b14b2cfd6904b68902cdf6b345f9070 : FAILURE in 41m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/ade0eb0956254aa19be8af1dbd57ed43 : SUCCESS in 1h 22m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/f26e86d98446401b8f5dce966010d21d : SUCCESS in 1h 38m 45s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/b5c0eff0f16445e4a645a7e0689c0e27 : SUCCESS in 1h 00m 09s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/54242231f3084d01b79ba5bd8af3ca58 : SUCCESS in 1h 33m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/cf23b0f106a24881bbf1af592c4207a9 : SUCCESS in 1h 23m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/74951a0fb8984c68b2d121ef273dd430 : SUCCESS in 1h 32m 07s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/d0a71e8fd2ba469b971509d4ae41a960 : SUCCESS in 1h 01m 34s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/45c46f7c8cb54ace8c9a406038b2423f : SUCCESS in 1h 23m 51s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/96ef485d24164b79bab42d3bcead4844 : SUCCESS in 1h 15m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/23bd2e08e9224839b6e1cd2fc4758449 : SUCCESS in 1h 22m 59s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/49339925bfbb4dbf8b1617c7201f3d41 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":2},{"id":"e154e9482a29da4a7015a6194aedcda1805e0e59","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-04 08:51:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"249434550888191590401fec522390ff0bbb8e95","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-07 07:10:49.000000000","message":"Topic high-timeout-for-shutdown removed","accounts_in_message":[],"_revision_number":2},{"id":"43548c507ce75eec9c8043f95cf680a909755ce1","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-07 07:48:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e022a9753c9d5eb2a77fffe47bf0c1f17c4874d","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 09:23:26.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"bc88c8b90339a1ad471b46efb2ee0fd8244f4698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 09:24:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"37a4f9a5000a1345d9ecd19c25d3d5b52b43a61a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:01:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f089194dd5a46ca980013777aef6847\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6d27a8b118f4762835c08ba259ab394 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c73163bc42d1443a90c5dbd5145fec9e : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3e610bd9af3403688aac0bd529ef161 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69cbadab32324450ae3966cfc1c6c971 : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d4c8fcffb9f47adb131a9e802a22921 : SUCCESS in 8m 01s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/4d60a73905a443809d75d63c92468649 : SUCCESS in 34m 24s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/5cb3be74bf89485bafe777488c6a463f : SUCCESS in 32m 40s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/872913b753d644a283c5ef9d8ea280d6 : SUCCESS in 36m 42s","accounts_in_message":[],"_revision_number":2},{"id":"39ea5029ea4f0ea2a28a8afe7bba0cfe6cfe79ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:01:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ea8d9aa95ef048d73785a9a70b4f64cbbfb37a3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:03:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50853723eed14c58bf33ea0542ce1222\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb6f223725b4ba9a9f0afddf1d6d151 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ab2eed28d3f49959bdd0f198b58dfc4 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de3fb5739fe13341adc7ae0446422172ac82b7ed","revisions":{"bf46308507b8d4e020dcb22f33e259ea9063a6a2":{"kind":"REWORK","_number":1,"created":"2026-09-03 06:29:38.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/29/1003629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/29/1003629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/1003629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/1003629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/1003629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/29/1003629/1"}}},"commit":{"parents":[{"commit":"435f705fbbee993cb740a05c56b519721d90f757","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/435f705fbbee993cb740a05c56b519721d90f757"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-03 06:20:48.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-03 06:20:48.000000000","tz":300},"subject":"Use high timeout for database shutdown","message":"Use high timeout for database shutdown\n\nStopping a database under load can take longer than\nagent_call_low_timeout. This causes flavor resize to fail before the\nguest agent completes the shutdown.\n\nUse agent_call_high_timeout for stop_db requests without changing the\nconfigured timeout value.\n\nChange-Id: I21385586c06a73d61d750a3840381c4cdfd66e0d\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bf46308507b8d4e020dcb22f33e259ea9063a6a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bf46308507b8d4e020dcb22f33e259ea9063a6a2"}]},"branch":"refs/heads/master"},"de3fb5739fe13341adc7ae0446422172ac82b7ed":{"kind":"REWORK","_number":2,"created":"2026-09-04 05:27:07.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/29/1003629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/29/1003629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/1003629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/1003629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/1003629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/29/1003629/2"}}},"commit":{"parents":[{"commit":"435f705fbbee993cb740a05c56b519721d90f757","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/435f705fbbee993cb740a05c56b519721d90f757"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-03 06:20:48.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-04 05:23:16.000000000","tz":300},"subject":"Use high timeout for database shutdown","message":"Use high timeout for database shutdown\n\nStopping a database under load can take longer than\nagent_call_low_timeout. This causes flavor resize to fail before the\nguest agent completes the shutdown.\n\nUse agent_call_high_timeout for stop_db requests without changing the\nconfigured timeout value.\n\nChange-Id: I21385586c06a73d61d750a3840381c4cdfd66e0d\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/de3fb5739fe13341adc7ae0446422172ac82b7ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/de3fb5739fe13341adc7ae0446422172ac82b7ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
