)]}'
{"id":"openstack%2Ftrove~1003917","triplet_id":"openstack%2Ftrove~master~I7d68e4ae76f5ed96663bb8eaa59df224ce343342","project":"openstack/trove","branch":"master","topic":"optimize-zuul-jobs","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-16 16:12:59.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-16 11:46:48.000000000","reason":"\u003cGERRIT_ACCOUNT_39300\u003e replied on the change","reason_account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}}},"hashtags":[],"change_id":"I7d68e4ae76f5ed96663bb8eaa59df224ce343342","subject":"Use lsyncd for guest log collection service","status":"MERGED","created":"2026-09-04 07:20:10.000000000","updated":"2026-09-16 16:50:54.000000000","submitted":"2026-09-16 16:49:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003917-optimize-zuul-jobs","meta_rev_id":"3368312e3d396f1544a10bc53afa3e4dba2a7e29","_number":1003917,"virtual_id_number":1003917,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 16:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-16 16:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":1,"date":"2026-09-16 11:46:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-16 16:12:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-16 16:12:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 07:37:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 08:42:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 09:51:04.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"REVIEWER"}],"messages":[{"id":"3b91f4193232321fdf99a1c3a57d0e65892776ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-04 07:20:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78c29bf6347b9048fd55206bd9f5969387ed3890","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 07:37:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d456a4a15b349febc2542ecd49cdf2c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4c7b735e30054e43bd72837bb59bf183 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/20ad2b6e395042d9b85aa6d33ee08d4a : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/42bd4e2764384444b603c77ba867f5f1 : SUCCESS in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00c1fc44009444311b1663c0de52232cf0135f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 08:42:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f4c5fb954234d77a66142834c93a1bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/083f1db5ad0342df9dc87942f6421689 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7ba4cbe0b1480490eba7ce594869ae : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ce763b088034b14914b1051d524ee8e : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcc515c2c98a4fcdbcbd3ff5076b5f0a : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9be8f676e04245e6bdaae8f34c94a5f4 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5609f0a1f53f474699bbacbf83d84041 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5247c9582ac4c879ccc64ed4de98d55 : SUCCESS in 7m 49s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/2643b5fbf9874a748fd6e235440888f8 : SUCCESS in 43m 31s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/daeda17c35b54523ba9f1f3e05e6803a : SUCCESS in 1h 14m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/f1845403bfec45bfbe9ce0a23cef395e : SUCCESS in 1h 06m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/56fa89b0c0964da88691c4312f235cd0 : SUCCESS in 1h 21m 46s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/d95ea8a8705c4e98a68c2bfe2cbc8f3c : SUCCESS in 41m 44s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/271ecca53ae44b53b9ce5758e7b1f334 : SUCCESS in 1h 17m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/44a3695531344dbdaa8eec2974624450 : SUCCESS in 1h 04m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/3847ac496d0a4d5193ec36b9640a40d3 : SUCCESS in 1h 14m 36s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/f7873ab5b4fe48b7ac290c982e6bfdf2 : SUCCESS in 44m 01s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/9548943011594e33bebee1013769158e : SUCCESS in 1h 12m 31s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/010a2ecf172e46a2b6c293a8f3dd99b9 : SUCCESS in 57m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/437a14fdbca2418587c1c9c8ab20f07d : SUCCESS in 1h 06m 38s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bfad25f9f02b4423bf7fae9efc47f617 : SUCCESS in 5m 05s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/3a2ca89d3074416d882ee49911611010 : SUCCESS in 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"665ae68c177c66c4c92f51f698b6a7fcf9073a72","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-16 09:51:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f47f32f2936c91ce11f8ad05ec5ddc8eb5e45236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-16 10:29:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b4409cc3feb83996f8e4d866d027cdb3d4b7d6ff","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-16 10:30:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81b11b56aa3ace8ba2ac8b874f147f0ce896eb25","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-16 10:31:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"138b341bd295c75631c753346ccd71af9d69bdeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 10:50:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f487f99e7124ae8a4c75bbf81af94a8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d9f32feee666446596c8da3b7ef66d99 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/48306fe9fdd047c19c9085ebe8d22054 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b846a3488ab440959c9f76d9984c7722 : SUCCESS in 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60579b299359e93ea9b39fd45a1d08b73fccdb9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-16 11:21:21.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e5a1303df3624925ca0117613e01a3e2df75dd03","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-16 11:46:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6078a6867d76733f78aed34536671642ebe41cd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 11:50:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/370f6019e87a4a2c9e2bf894b7735f03\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e0c82e64a5e4353a42c372d61863728 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2df52416e27343dbb3fd9ecf310f6ec2 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/82bfc0887a5a42d7ad129fe827246077 : SUCCESS in 16m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d7b341b8fc21cf43d492c28eeaa2277b9e8779b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 13:09:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf58b5c3a0824d84bd1b68e1585196ff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ad7336f78654958aa95227f066b9481 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d7f396dbb9464cb57c0b82319b31d5 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/264d60481ac04d2ab71b89a7ef159a98 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a879615cf51143c9aea2ad4ecadc0ae1 : SUCCESS in 5m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1dc34a33b05a41bdb01f56ea91cb3efb : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea3f86e996dd4a668751187a1bc6a252 : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22e587a6846241fdb4018d4f476591d6 : SUCCESS in 8m 01s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/2100ce1664af41febf2148b92592d046 : SUCCESS in 39m 43s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/d30be804d12144aabfb0d02459af1ed1 : SUCCESS in 1h 09m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/9a25eb27966a4476b1fb4a4a6005ef85 : SUCCESS in 56m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/2e3a54ce363048de80915c2a332b2b77 : FAILURE in 25m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/2042972ee858483aabd51b751d2960dc : SUCCESS in 31m 44s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/451f5f53c0e942ddb84f56744295b75b : SUCCESS in 1h 08m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/b8e9f7668e1d4cb79d75615c44ab705b : SUCCESS in 1h 11m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/0b1ff8db719d479bab66de43b8ab02cb : SUCCESS in 1h 25m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/080e5452c48e4dcb991f86fb3e26fbbd : SUCCESS in 35m 37s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/06a84588cbef43f2a7055bfc95a583bf : SUCCESS in 1h 40m 20s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/a7af6f87169347c4b850a38326b8bd0d : FAILURE in 19m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/f4e615750b0b438d936d80a4e2aef97d : SUCCESS in 56m 31s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ea7211c6a7a480ca71adc2ed22223d8 : SUCCESS in 3m 45s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/bb9400c54c1f425e92d4d42fe7172920 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":3},{"id":"0b8da93f31858eb1773e2adf82791de77c08596a","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-16 16:12:59.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5e7034b52a4d5fa73ec9fe796b3be369123bcd49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:13:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d002d332287bd4c7bcb401ee2b91028947e50646","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:49:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc20afeda31b4ef696b1246b5a729aa6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2fbc98f3df43ffb1b5ef030e25b3ef : SUCCESS in 6m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/022cab822cf44380aebc9426e0aa8fd2 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d66729b9aaa34148b6c96d34980d0823 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43490c1b8c924f058a7dc0385aafa423 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7aff9b1c84f4b51bf333088bd57eabc : SUCCESS in 6m 09s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c34d173debb5450abe4e194b4f2e99f8 : SUCCESS in 33m 11s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/23fc0f9d1e8a4387a946f2715a9a08ea : SUCCESS in 32m 04s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/af7f6c140cdd468e9a854b76a109eb8e : SUCCESS in 34m 49s","accounts_in_message":[],"_revision_number":3},{"id":"01034357f524a375e0e4c07a7b109ab759a1e5ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:49:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3368312e3d396f1544a10bc53afa3e4dba2a7e29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:50:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eba58f1ee784694988e3f5cb19082fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a197096ecc641f684474e500d7ae028 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a1d762fe2108d6dd1aaf4382194e6b2e8f30142f","revisions":{"44b862bc8e451e51ba12666d0c0dcd5052f1dcd2":{"kind":"REWORK","_number":1,"created":"2026-09-04 07:20:10.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/17/1003917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/17/1003917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/1003917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/1003917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/1003917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/17/1003917/1"}}},"commit":{"parents":[{"commit":"435f705fbbee993cb740a05c56b519721d90f757","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/435f705fbbee993cb740a05c56b519721d90f757"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-04 07:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-04 07:17:40.000000000","tz":300},"subject":"Use lsyncd for guest log collection service","message":"Use lsyncd for guest log collection service\n\nReplace the periodic rsync timer with lsyncd to synchronize guest\nlogs to the controller using filesystem events. This provides\nnear-real-time log collection, avoids unnecessary periodic scans\nand rsync invocations when no files have changed.\n\nBlueprint: zuul-jobs-optimiztion-2026\nChange-Id: I7d68e4ae76f5ed96663bb8eaa59df224ce343342\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44b862bc8e451e51ba12666d0c0dcd5052f1dcd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44b862bc8e451e51ba12666d0c0dcd5052f1dcd2"}]},"branch":"refs/heads/master"},"516638ee4849f9bb63f234456c31c8e406b33a2e":{"kind":"REWORK","_number":2,"created":"2026-09-16 10:29:20.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/17/1003917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/17/1003917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/1003917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/1003917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/1003917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/17/1003917/2"}}},"commit":{"parents":[{"commit":"e4a6f3f861e854309838c4857d6286b09fabf4d3","subject":"Merge \"Build guest images asynchronously in DevStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e4a6f3f861e854309838c4857d6286b09fabf4d3"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-04 07:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-16 10:28:55.000000000","tz":300},"subject":"Use lsyncd for guest log collection service","message":"Use lsyncd for guest log collection service\n\nReplace the periodic rsync timer with lsyncd to synchronize guest\nlogs to the controller using filesystem events. This provides\nnear-real-time log collection, avoids unnecessary periodic scans\nand rsync invocations when no files have changed.\n\nBlueprint: zuul-jobs-optimiztion-2026\nChange-Id: I7d68e4ae76f5ed96663bb8eaa59df224ce343342\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/516638ee4849f9bb63f234456c31c8e406b33a2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/516638ee4849f9bb63f234456c31c8e406b33a2e"}]},"branch":"refs/heads/master"},"a1d762fe2108d6dd1aaf4382194e6b2e8f30142f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-16 11:21:21.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/17/1003917/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/17/1003917/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/1003917/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/1003917/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/1003917/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/17/1003917/3"}}},"commit":{"parents":[{"commit":"e4a6f3f861e854309838c4857d6286b09fabf4d3","subject":"Merge \"Build guest images asynchronously in DevStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e4a6f3f861e854309838c4857d6286b09fabf4d3"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-04 07:17:40.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-16 11:20:34.000000000","tz":300},"subject":"Use lsyncd for guest log collection service","message":"Use lsyncd for guest log collection service\n\nReplace the periodic rsync timer with lsyncd to synchronize guest\nlogs to the controller using filesystem events. This provides\nnear-real-time log collection, avoids unnecessary periodic scans\nand rsync invocations when no files have changed.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/1005861\nBlueprint: zuul-jobs-optimiztion-2026\nChange-Id: I7d68e4ae76f5ed96663bb8eaa59df224ce343342\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a1d762fe2108d6dd1aaf4382194e6b2e8f30142f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a1d762fe2108d6dd1aaf4382194e6b2e8f30142f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
