)]}'
{"id":"openstack%2Ftrove~1004378","triplet_id":"openstack%2Ftrove~master~Ice10b7b6aa60d4e687456c75262015ce7caf0452","project":"openstack/trove","branch":"master","topic":"fix-container-status-detection","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-17 12:03:18.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-17 12:44:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice10b7b6aa60d4e687456c75262015ce7caf0452","subject":"Fix database status detection for stopped containers","status":"MERGED","created":"2026-09-07 04:33:32.000000000","updated":"2026-09-17 12:45:26.000000000","submitted":"2026-09-17 12:44:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004378-fix-container-status-detection","meta_rev_id":"a01eb6ebdc6de9c64d35de93e300f9f545115fd1","_number":1004378,"virtual_id_number":1004378,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 12:44:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-17 12:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-17 12:03:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-17 12:03:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 04:48:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 05:52:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 09:19:28.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-09-17 12:03:18.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"f8f9231e7bef0760a8e0a7c8979ad0eab2593b44","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-07 04:33:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcb2adcbe086b47fffff58de10b98817898e4cf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 04:48:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bee0fe7f97247208013de653bf3a418\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b21fef6bdcbe4c178c40778f2997f46f : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c4324c6670804e64904948806137dbf8 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5470735e6b8c49e9b2046b60cd9aa157 : SUCCESS in 13m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6912d095754b5cbea4bf658f31f053d43a947adf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 05:52:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f291d801a734c95a6204c4b3912ef3e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86bf065e7247473b81da2bcd2a44c129 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efab1a0a48c349959d3f10a4ad6620b7 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b87262211ac4424a6c0155e7f7f52ba : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39f2c91fe2b34a959df0652fe9771510 : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a23795c2f4f4f2c81453528ed99f878 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3179933ce464c81a7c1beaecb134c1c : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7e5f9631f0a4498b6d030957b2e9e7f : SUCCESS in 3m 43s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/110c00e069c14c3296a7e207fffab0b7 : SUCCESS in 43m 10s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/56e6dad945644360808563b441546d83 : SUCCESS in 1h 09m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/b657cbba84914bc58db41a11e75e9f93 : SUCCESS in 1h 04m 49s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/ccc0b339541742f68de364ef4d39ccb7 : SUCCESS in 58m 48s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/70ad4cf8e96d456bb9a0e31ac0d7d716 : SUCCESS in 1h 08m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/cc743abf74f841bba49ba9b9d79cf48c : SUCCESS in 1h 17m 25s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/3ad6b5ce6149436d9e415f519a42a2ae : FAILURE in 19m 10s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/9c52a0ca7e024788986137e4310d2901 : SUCCESS in 1h 17m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/1e26bd131fe442b889ceeb4c6100d930 : SUCCESS in 1h 05m 58s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/9052372c71a54a4ba381b8fa58ed3c1a : SUCCESS in 55m 33s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/37c1e51d0fa1481a85014a86724f360d : SUCCESS in 1h 04m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/fc24fbfe94e8404199b223cb384195e1 : SUCCESS in 1h 15m 00s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/d39e63d895cb45e8bad3dbb081be84ed : SUCCESS in 48m 31s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/8421a543baa84293a6da5d218f91c590 : SUCCESS in 1h 08m 17s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/8d46178290364639b93a67e48e91c37d : SUCCESS in 47m 12s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/6537bfde0f0e471d86d942ad43cd742b : SUCCESS in 57m 11s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/ebb0c479c82b423c87f5ab4679428cf2 : SUCCESS in 59m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/72d9d01a0fa14d76ad2cda8de6030029 : SUCCESS in 1h 02m 46s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/24ee75e9e1d54701a373e4921185c380 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9add15015de560731966cd53d5390119600fa373","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-07 07:12:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6a6eae150c97fe73020a3f2da975e472c02af2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 08:31:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d228b243941b4136b8d617e31b602b14\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e27b46c8c5624e1ba767a61735bea3ef : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51834bc904124358a0d07f29b9b1de95 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6376317132a743099fdd975ad24feba9 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54f2ce575b9d4493a374314b85ed20e0 : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc76e0306c5b4f0481f7382739f0b1cb : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b63a3cb18a4243aa61c0184bc10302 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6c8bf784a1e4163b7d6d264557b785f : SUCCESS in 5m 49s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/a8b07c16d43b462d812306350d937deb : SUCCESS in 43m 56s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a5ec826dad9d4b57a7f342b0e462f0f0 : SUCCESS in 1h 11m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/de9d891eefc240db933144a5481ecda8 : FAILURE in 23m 26s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/dda6acfdc30a4b44a9dc76a9cc03594b : SUCCESS in 56m 23s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/d3b79efda3b74a58af081b28f7b04bd5 : SUCCESS in 1h 10m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/d2b341f98e074016b8e5ab1485370554 : SUCCESS in 1h 17m 49s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/3d2d4c67e9d149269dc35393e9c22c6f : SUCCESS in 43m 56s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/1466ffa752ba45548c78b73445a40f4c : SUCCESS in 1h 14m 58s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/27ba0cf8b47c4b5a9e9054868cb4bd02 : SUCCESS in 1h 02m 37s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/d55483a1cb5447ac93f0f9221c7a0168 : SUCCESS in 56m 05s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/cb28286e46e04836a639e181ee601cb5 : SUCCESS in 1h 07m 25s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/00b8f7b4acda4540890a0d791c969364 : SUCCESS in 1h 11m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/e9b83342dfd342c986fbf9c7b1b80db2 : SUCCESS in 47m 51s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/5c7e2c3fdd744018934c94f8ca1623d4 : FAILURE in 35m 17s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/25dab30de5b2474982d398a83131ffe6 : SUCCESS in 44m 34s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/06d2c913629b4adb80c4c2cb3fa794a0 : SUCCESS in 1h 00m 39s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/95dd9bd83b8e4f7e805d732f878151f2 : SUCCESS in 57m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/e3ff83d8089748bebe5bef2d91eee2b0 : SUCCESS in 1h 02m 54s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e8f7d7e6e0a343ab8e2620b1a8b8fbf0 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6e770722bf84b92fa01b5f648473551b22e67c61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 08:57:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb72a96e1ed74dcda4392622d959b4a5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a35e931172574f05bf19e2e539f0690d : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/110636b52ad646f3be469df3ec336519 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/308287cd2ac94aef816555bbb5342819 : SUCCESS in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a08354425357e0d06e01577feee0f65981ad00b5","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 09:19:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7207c73e7b577131e716315f8122bf0870ecd40","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-17 09:38:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"afaf9397a39848f0990a486265772af6b13502f6","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-17 09:39:15.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"6e24590f2f1f797913652499d26d7916020bf246","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 09:57:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b752625670714f3bba2e20df3b5244e8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/361edd50f782419e82a24fd7c1805a81 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5bf1617ce1484a6297c67175bdf973b5 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/30eb61f84c8e4ab3b76f38d890eea300 : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87ec21d42569557fd62d0272da732411a3cbbbe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:56:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/641e9ed7d875468eaf2e1dc8b1127e8f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e2a5eb3a79f419f83daf9c928a0edb3 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a18bf226324f14a30b90f5cc441a6f : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbb2affd46fa4cd1a205df0c70a03c3a : SUCCESS in 6m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8a96b0e9fad4e34ba51b9f9795dbc83 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c299a590589f4b858a5626be1a7192cb : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3826d457f549416ca4caec44af5130be : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd6e1c9bd66740ceac18d3049222082e : SUCCESS in 6m 00s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/3d767d0ca9554a79b4ded3333f458252 : SUCCESS in 52m 00s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/880aba62f13f4663bf66bac457eb7e64 : SUCCESS in 1h 15m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/26f2178dc7574379b3cfc8bb7170012e : SUCCESS in 1h 00m 48s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/b4df25617ad9480ca53397eb9d2e2f47 : SUCCESS in 54m 12s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/75a64b345d9e4725bed9059d5dbca31a : FAILURE in 19m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/bddcfab06cd04f8fa99d15736b19d165 : SUCCESS in 1h 08m 42s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/9dfd19d45bc44f08b5abded459ce976b : SUCCESS in 44m 35s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/21c437e0b5b5478291e168ebc5c1b777 : FAILURE in 58m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/d54ef679e4644d81ba38d7bc16c4b1cc : SUCCESS in 57m 59s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/166f3e078b9b43dc8f90e8be5d5c59ec : SUCCESS in 1h 11m 28s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/15e4939ceae94d31918eee5001b68b49 : SUCCESS in 56m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/2c483cebc3e942cfac9353b43c89673e : SUCCESS in 1h 07m 40s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/a4aef4db19d7473b882f560286a60460 : SUCCESS in 37m 46s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/e16ce2f5e59340bbac81868b52124168 : FAILURE in 20m 49s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/95891a0a6534454daac706b1823d7c2b : SUCCESS in 37m 56s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/94fa92e593774f9ba24f96f09159f1b3 : SUCCESS in 54m 27s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/2c83065b391440ec9edd40137e6f8fcc : FAILURE in 26m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/b6060a85730c4c90b73d491a3848c832 : FAILURE in 17m 39s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6843cf9c708945aa97d7e20213968fb3 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"4fe2816b4378adbfc2746675dda005e2b36ac756","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 12:03:18.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"75f38fba0a6fcdb6e8c96633ae1958a03d2bf288","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 12:04:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9095204be76acb0e50a902736e5a9da2a4c37483","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 12:44:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19690e086f0f4130ab2facc4f5998c45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a42b33a94cf347f7a041ad59bb087383 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e983f907a6042c89202091ae7561a0b : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d070cfa145034552b2051afbfe89b46c : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273f5de8672b45dd9156d9e18174c0b0 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09a97229964a4fba8dd3b794c4e238fa : SUCCESS in 8m 47s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/54e208ef6638485999eb8340b2adc9cf : SUCCESS in 34m 13s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/9b746df7c51e4355bb80fa23234a074a : SUCCESS in 37m 48s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/f76532f425ac415e97c3f86a2c3bc473 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":2},{"id":"c7e1a6ff93014c069c895d34f4945e28bc76032f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 12:44:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a01eb6ebdc6de9c64d35de93e300f9f545115fd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 12:45:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff2d99bdb62d4a92a187d236240c5764\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bf5dccd688e438bbac3c875a2df3cc4 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39c3cd9a26d84c43b8e4bf67f97442a4 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8005a8d551c9078a082e22e06b637dbb0303b4a9","revisions":{"af02e84a80e4e14b739dce7d28bafcd6a3846182":{"kind":"REWORK","_number":1,"created":"2026-09-07 04:33:32.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/78/1004378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/78/1004378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/1004378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/1004378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/1004378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/78/1004378/1"}}},"commit":{"parents":[{"commit":"435f705fbbee993cb740a05c56b519721d90f757","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/435f705fbbee993cb740a05c56b519721d90f757"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-07 04:26:07.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-07 04:26:07.000000000","tz":300},"subject":"Fix database status detection for stopped containers","message":"Fix database status detection for stopped containers\n\nDocker can retain a healthy health status after a database container\nhas exited. Check the container status first so Trove correctly reports\nSHUTDOWN.\n\nChange-Id: Ice10b7b6aa60d4e687456c75262015ce7caf0452\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af02e84a80e4e14b739dce7d28bafcd6a3846182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af02e84a80e4e14b739dce7d28bafcd6a3846182"}]},"branch":"refs/heads/master"},"8005a8d551c9078a082e22e06b637dbb0303b4a9":{"kind":"REWORK","_number":2,"created":"2026-09-17 09:38:46.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/78/1004378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/78/1004378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/1004378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/1004378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/78/1004378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/78/1004378/2"}}},"commit":{"parents":[{"commit":"435f705fbbee993cb740a05c56b519721d90f757","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/435f705fbbee993cb740a05c56b519721d90f757"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-07 04:26:07.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-17 09:37:43.000000000","tz":300},"subject":"Fix database status detection for stopped containers","message":"Fix database status detection for stopped containers\n\nDocker can retain a healthy health status after a database container\nhas exited. Check the container status first so Trove correctly reports\nSHUTDOWN.\n\nChange-Id: Ice10b7b6aa60d4e687456c75262015ce7caf0452\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8005a8d551c9078a082e22e06b637dbb0303b4a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8005a8d551c9078a082e22e06b637dbb0303b4a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
