)]}'
{"id":"openstack%2Ftrove~1005671","triplet_id":"openstack%2Ftrove~master~Ie44a2d49da9c317a97bfff4195e8884a18bde749","project":"openstack/trove","branch":"master","topic":"stuck-service-status","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-16 08:43:10.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-16 09:27:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie44a2d49da9c317a97bfff4195e8884a18bde749","subject":"Use a unique code for RESTARTING status","status":"MERGED","created":"2026-09-15 07:15:19.000000000","updated":"2026-09-16 09:28:40.000000000","submitted":"2026-09-16 09:27:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005671-stuck-service-status","meta_rev_id":"5ec5a9bcff66f92b7ab27c3afd41c91b1d2bb8f5","_number":1005671,"virtual_id_number":1005671,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 09:27:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-16 08:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-16 08:43:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-16 08:43:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 07:17:40.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"},{"updated":"2026-09-15 07:26:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-15 08:33:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0143fefda984b882607d53883550b623d2d305ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-15 07:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22998cff912096f1d1bfdb8a5fc5e3d367fe3527","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 07:26:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a1fc6d31ca4f838290ff1ec2c067e9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1707a7c9f2724ac5a200575d15e38990 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/77e742ea0a9041ae9f18b426dea83c93 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d827ef5a740940adbe2a1e90f7a66b0f : SUCCESS in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b73a5d8a111cc31bd22cf7dc1b91a8f5c64616dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 08:33:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b1bf74bda1743949a62df2d55ced9bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6961e7e0013040bda119949d7e566c11 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4794592f8a64409bae98cabeef296214 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e796afa75d0e436f96bfa298e14f532c : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20d6ac5b45c74a929eedc53661964218 : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/77b47929a772414583f150ed1bd8dbff : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbde1a3597f2408083aa1dbb20d83f70 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d052861177a14c83af3b2c2b86119473 : SUCCESS in 10m 12s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/1ff80c6b842d4ba6b353911afbe2dd2e : SUCCESS in 42m 31s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/1e40ed9f0c1344d7b57389f5d572d7a0 : SUCCESS in 1h 08m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/af819cd19d164feca2c2a98f2a327ec6 : SUCCESS in 1h 06m 37s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/d61e0451883a4bf8ad22561afcc6fd40 : SUCCESS in 1h 01m 42s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/14bd3fee2c6149668f13c537d809291b : SUCCESS in 1h 07m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/71f74c1f6cea4a3b92ab6a666c3184c7 : SUCCESS in 1h 17m 00s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/d16009f8079142a2b66c9901d14f1f44 : SUCCESS in 40m 45s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/b7caeca735dc4e37ba2cdca8ec1c7af8 : SUCCESS in 1h 11m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/497c342ff3154a09b7a7d9b62b43e8a0 : SUCCESS in 1h 07m 25s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/0942b07f72d04dc184c4dcdba5202ce0 : SUCCESS in 52m 08s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/7b5538c0c9124b359cac7a8f22a68233 : SUCCESS in 1h 05m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/bf4d0a495f774ffa84b1678fe4e09e60 : SUCCESS in 1h 13m 45s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/fe945ac5ee95489d9a405b381dfd8625 : SUCCESS in 41m 33s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/a1e657b93d094123b3695f5198a5c47a : SUCCESS in 1h 06m 50s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/083e6023a53840fda307571a2b4c4fab : SUCCESS in 44m 18s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/eb739f92de3843be9fcc417fff87418e : SUCCESS in 58m 09s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/3107aa78e32c489489e4bb4868083c64 : SUCCESS in 55m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/af66531631dc4964bd5e80f4e76eeaa5 : SUCCESS in 1h 02m 38s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b4372377fbf4eed92d7bf3a867f0836 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":1},{"id":"10e37f505c54ee3865533a400bc61aa4e5387b8a","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-16 08:43:10.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9139673ee66b48e9560174670146d146b23e833","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 08:43:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad53d146a13fe361373be662ba6645c92ab52699","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 09:27:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8cedb2c51e047ab981f0e7b734874b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/396764163c4949e18d76496b9c37a0a1 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd16122c3d8942149af28b49d1cb7d7e : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d57a248355974eae8a7132c41e608d05 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57dc87d9a57b4cb69470e8003ecd4862 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f0e43603978488bb3115dca3fcfb7e0 : SUCCESS in 7m 47s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/e390e5b7bd7b45ff9ac0895acf7c5bbb : SUCCESS in 42m 15s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/e9b2b4255aaf4638856d753abe75b686 : SUCCESS in 40m 26s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/6950fd9b7c9e4fb58d84312bf451df3b : SUCCESS in 42m 52s","accounts_in_message":[],"_revision_number":1},{"id":"afe2e36ee26677377618eaf8486bfc78309cf2ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 09:27:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5ec5a9bcff66f92b7ab27c3afd41c91b1d2bb8f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 09:28:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18a3b9e1346d48b7ad22eef3890720c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005428e566f24ccdaaf1044b4f59ed79 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f2b0ed6862d4b1391fe5395b5911fd6 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01418aa838eaf97671b13539a5b73791f4d39e03","revisions":{"01418aa838eaf97671b13539a5b73791f4d39e03":{"kind":"REWORK","_number":1,"created":"2026-09-15 07:15:19.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/71/1005671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/1005671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/1005671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/1005671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/1005671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/1005671/1"}}},"commit":{"parents":[{"commit":"435f705fbbee993cb740a05c56b519721d90f757","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/435f705fbbee993cb740a05c56b519721d90f757"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-15 07:06:32.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-15 07:06:32.000000000","tz":300},"subject":"Use a unique code for RESTARTING status","message":"Use a unique code for RESTARTING status\n\nUPGRADING and RESTARTING currently share the 0x22 service status code.\nSince service status comparisons and lookups are based on the numeric\ncode, the two states cannot be distinguished reliably.\n\nAssign 0x23 to RESTARTING so that both states have unique codes.\n\nPartial-Bug: #2166657\nChange-Id: Ie44a2d49da9c317a97bfff4195e8884a18bde749\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01418aa838eaf97671b13539a5b73791f4d39e03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/01418aa838eaf97671b13539a5b73791f4d39e03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
