)]}'
{"id":"openstack%2Ftrove~1006155","triplet_id":"openstack%2Ftrove~master~I7d1a22a17c3c93e51ce3e74ad118a403b6f30c50","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-18 08:07:56.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}},"hashtags":[],"change_id":"I7d1a22a17c3c93e51ce3e74ad118a403b6f30c50","subject":"Increase guest image size","status":"MERGED","created":"2026-09-18 06:34:00.000000000","updated":"2026-09-18 11:05:44.000000000","submitted":"2026-09-18 11:03:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006155","meta_rev_id":"c708b52d26e73dc88987d12857653b0465fd7715","_number":1006155,"virtual_id_number":1006155,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 11:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-18 10:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-18 10:12:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-18 10:12:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 06:49:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-18 07:49:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eee7177fc5c6bbf25d6365202aee66d2517548c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 06:34:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4df9f7dd1e578a97ee3e5528eab6f20e553dea9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 06:49:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc6e82aa941944d2ad0a308d3bb0abc1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a2b7373955144acca01cfc340093875f : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a9f7fc80e2404d1a92eae4a55be29f7a : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/66ba3aacf8664645aeef4ad88185834f : SUCCESS in 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"961894259bb6affcbc4686c26fec37f6de350506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 07:49:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08d9e2763ec846b78db352a54d5e309d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e4fe8ea66084f38bc4d3b5e1bee5efb : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b558b6fc97c74faa94a3fa9d6ea28be8 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bc4ec28e8194d52b2d538e646e916a9 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89a69b921c624b7a965e4daa8df29247 : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4571396d3de04b6dbdef967fe25ea239 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf61ff4a37b947c5bc7f581023af47f3 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55ce3f5fe022407ea8349c47a4b7e446 : SUCCESS in 7m 02s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/68b87da0f16e4380ae3b424825b64d14 : SUCCESS in 34m 09s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/feab2d37bd9c4e968c27af1a2fc9b073 : SUCCESS in 1h 14m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/5ce629f2cde144fb99c71a03c90360bb : SUCCESS in 59m 15s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/3925cb704b6c4ab18915308ad429dd38 : SUCCESS in 1h 01m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/12caaa31ce7b46819edf00c1f8fc9668 : FAILURE in 17m 36s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/3f609d503d1e4535a56cd0a28af795a9 : POST_FAILURE in 32m 18s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/9719898b06a34134b384b9b6fe002ef9 : SUCCESS in 1h 08m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/4bd7ab450e5b4136ad58af9f450a21b2 : SUCCESS in 56m 09s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/88d520f6005d4f1f955b4a6925440a3b : SUCCESS in 55m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/81bc3fff64604513a334025205bafafe : SUCCESS in 1h 05m 03s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/2f490b9b673c4967be8d1a14def9a838 : SUCCESS in 34m 57s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/d61d3744b891483d899fdb4297fbba2c : SUCCESS in 1h 01m 14s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/6f888f7ab9f9473aa7375f6f40642e2c : SUCCESS in 55m 41s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/f8410180e6e34f00be1540d1e7a3d748 : SUCCESS in 47m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/4928a990464e44e9a36005882964e25a : SUCCESS in 54m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5afc2718cc6d4e9ea6fb0b67f47862fc : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"438c81055f2cf2a0435990b761eefb2230964fb2","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 08:07:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec316873278306121f598ae55854b930fa9f7ed6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 08:23:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39bd844a4d7243aaa48e47446cfe84bf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/173ab0ae0875494487e9b684b0cfa79a : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d84cadc994434d839f0b186a50259f89 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/946b674bfae14bbe8a0b0b034785d543 : SUCCESS in 12m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27c690605dc941230b1651cfb58d460ca7bcc477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 09:32:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/002635e3a7824e3b8b4054eb4622379a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa4c9f59b49b432686c2d0000cb7035f : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/103966124c7d4e05b1fd88cd08868ace : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5ba1e80f12c485f843b120cae91c1bc : SUCCESS in 7m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b5f6f7db6664d98bc2f7007241bfdfb : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/11bbc46b8c5545b2b641403dd36c8a4e : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb956d4062c4e2babb4ce28b7aa0bd2 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9feb0257ad1b4696b4cfd6fe8151c12c : SUCCESS in 7m 43s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/dd6cf80ee275415eb2da0c6051799bd6 : SUCCESS in 35m 52s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/b0d32a835aea41a49917fd0ea4757ee3 : SUCCESS in 1h 20m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/7da12339b28f43138ef39886a91e39cc : SUCCESS in 1h 07m 43s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/fda6fb1e36bb4d26a84308367a382ce3 : SUCCESS in 1h 11m 38s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/c5b94628aff0498cb5ffbd796cb53101 : SUCCESS in 1h 12m 52s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/fa8f718667c04bcf8d247c4963e2d723 : SUCCESS in 35m 32s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/93369466a4254548a977f00e614b8cbc : FAILURE in 33m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/cefcb8ab47c14c939e71abe1ceaa89ce : SUCCESS in 1h 00m 37s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/c4752cc044b14e26ae4e58e184b4fa48 : SUCCESS in 1h 03m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/ce35707884d44716b01a7b94f78a9e0e : SUCCESS in 1h 08m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/1b61ea163f2046f199205dda95ea3a68 : SUCCESS in 37m 22s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/b6858556b51948bd8a8aa5ee1b483292 : SUCCESS in 1h 07m 46s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/1d1ddabf55724efa9e765d153e31e65e : SUCCESS in 56m 56s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/abb45d0a0edb438bab7694e53da08e35 : SUCCESS in 52m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/5e08245ecc754716b756555ab473d769 : SUCCESS in 1h 00m 04s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ade4698931da44559a5c293be3a349cd : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f834c84191e39fd109045e5ba7f881d82ca1ae3c","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 10:12:47.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23fa0e5be74e1ece20e7b36ba16d330060483a61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:13:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffbfb80fa72f951d5890eb62a0aa7cf13256fb03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:03:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abc4937729c549a59e89361f7d07a7c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f37c406766d4eaaa40846ae00dd092d : SUCCESS in 7m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ecd227e2706147deace6f1787951b3c4 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fffc2d7bc7024e8e8cece2f57dfe13d3 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4537fc1811454eae8e1922e5fe923245 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0568417401a24137bdd505461bc4bf7c : SUCCESS in 8m 58s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/f3846ab417514aba8b501d9284072d1c : SUCCESS in 35m 05s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/1598a60958254fa08869e11fc55e8ce6 : SUCCESS in 48m 51s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/ea8a3dbd6c8445c2bb73a0d04aebeb99 : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":1},{"id":"921610caa83a7598ec1f602b2a8d652056802c79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:03:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c708b52d26e73dc88987d12857653b0465fd7715","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:05:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a18f69072bc14fbe80aae9a9ebb39127\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c5651a20794ab998c5e2f6c6dcf66e : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d409657fa1b8c81f740b70f0492556b5e582c7d","revisions":{"6d409657fa1b8c81f740b70f0492556b5e582c7d":{"kind":"REWORK","_number":1,"created":"2026-09-18 06:34:00.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/55/1006155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/1006155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/1006155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/1006155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/1006155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/1006155/1"}}},"commit":{"parents":[{"commit":"56999c6b8b31c6924f0e8b1be1286663bc5bbeed","subject":"Merge \"Fix database status detection for stopped containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56999c6b8b31c6924f0e8b1be1286663bc5bbeed"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 06:31:28.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 06:31:28.000000000","tz":300},"subject":"Increase guest image size","message":"Increase guest image size\n\nThe current 7 GB does not seem to be sufficient because we\nembed Docker images in the guest image now, and the disk space\nconsumed by unpacked images can be significant.\n\nChange-Id: I7d1a22a17c3c93e51ce3e74ad118a403b6f30c50\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d409657fa1b8c81f740b70f0492556b5e582c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d409657fa1b8c81f740b70f0492556b5e582c7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
