)]}'
{"id":"openstack%2Ftrove~1006170","triplet_id":"openstack%2Ftrove~master~I29d293102ebbd2bee631be5d043ac6f849c717f8","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I29d293102ebbd2bee631be5d043ac6f849c717f8","subject":"Use pypi element, set correct pypi mirror address.","status":"MERGED","created":"2026-09-18 08:51:41.000000000","updated":"2026-09-18 11:57:05.000000000","submitted":"2026-09-18 11:55:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006170","meta_rev_id":"9492af60e25aaaaf6cd2d5aee22553b653bf9184","_number":1006170,"virtual_id_number":1006170,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 11:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-18 10:22:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-18 10:22:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-18 10:22:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 09:09:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-18 10:14:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bcedd733a5695b4fb6f3b40d4115f3f1a60f21c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 08:51:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"970827d19df44795516407c0c9ef3911c539006f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 09:09:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a7d8bd401e840d0b86aa7dd2205e655\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a5553234f5c04282b2e52ac04bddcba8 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ee3bda04bab546e28843c0e7e3474ac9 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2387af7051e644f49bd4114f9907fd5d : SUCCESS in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4473bc8fefef8a00821e4db1bbfc9f521d2ef608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:14:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0394131df79e40c5917c186b5a01c273\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf0a1d25b7f7428ea314c1591aae5261 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11cc2cdb5694b71a48d7b23e00af810 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af083ad18ae1400ab116f4c10a0d9e9b : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a40de57057734004afe7aa16f683b78c : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8711de59305547179f2ecd2cca8d8059 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/367c7b656dbd49b9924eb6f8120c966c : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0735f2c8978546feac6aca64e72a8ffd : SUCCESS in 8m 26s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/dac52e6d1a9342ed8fdb78f9a39e20df : SUCCESS in 34m 16s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/7193f0823f3641dba9de125a29b8cf68 : SUCCESS in 1h 19m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/4cafb680bddd4635a2098bae6e662bc6 : SUCCESS in 1h 02m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/1db75296c56542b3adfc2cc84d457ff9 : SUCCESS in 1h 12m 01s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/af05a50fe7d547ec8717270c9524bfdf : SUCCESS in 32m 03s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/85a462dc62f2437695a719e7b27d82b0 : SUCCESS in 1h 13m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/7ff5071d985245f3b43ff8c15519b2cb : SUCCESS in 55m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/c76decc8efac4887aa28f144693271d0 : SUCCESS in 1h 07m 58s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/69a35363b7d64c1dbb8b657c8b66dd1b : SUCCESS in 35m 10s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/eaa8a87ede754c56943765081a5e9629 : FAILURE in 20m 28s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/52af44127ab1464eb05add93a900b3cf : SUCCESS in 52m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/aca33999b9aa4115a6a80e28ba77d07f : SUCCESS in 57m 35s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b495a76d70248bd84a48bd7acae2d20 : SUCCESS in 7m 06s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/104865834895412cad4e1e2775ab26dd : SUCCESS in 15m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6c2b9e3b0281a732462aa17161194193883b189c","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 10:22:19.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5dd7ba03bb9b2de43d0de175e3ed2a477bafb8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:22:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87e44d696622dc661f9691f76c9b54f5a2eee005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:55:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a486f14a1e724839a75f6f83f197e6c2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e10320ae886f4196ba2de6ebbb3f96e9 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41c76f7d490d4d2fac28a16790279908 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c85e31dcb7084e93b16e37def2ef3e9e : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/555b0fdb01104bec81c0726c2afa346c : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d33d6eda5ed3466c9309a31512719faa : SUCCESS in 8m 30s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/ccf8221c73a64eea85a7bef633e4e26d : SUCCESS in 1h 05m 20s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/bf57528ee1bd4ddcae7a9d422f86ae05 : SUCCESS in 34m 39s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/658b5c78faa44b3690e8ac58a74ddfc6 : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f2ed4578c9d46ab13400b82f41c6e0fc258806f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:55:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9492af60e25aaaaf6cd2d5aee22553b653bf9184","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:57:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76478b108544c9c844fe3072b1b91d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b73683579944eed9b19350be1f43bd9 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1cc8058d942246d783692070798609807a1e440","revisions":{"c1cc8058d942246d783692070798609807a1e440":{"kind":"REWORK","_number":1,"created":"2026-09-18 08:51:41.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/70/1006170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/1006170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/1006170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/1006170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/1006170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/1006170/1"}}},"commit":{"parents":[{"commit":"20fb23f5a5a2b76fc997fe28de52b38beff953a5","subject":"Merge \"docs: Fix DevStack install instructions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/20fb23f5a5a2b76fc997fe28de52b38beff953a5"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 08:50:57.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 08:50:57.000000000","tz":300},"subject":"Use pypi element, set correct pypi mirror address.","message":"Use pypi element, set correct pypi mirror address.\n\nChange-Id: I29d293102ebbd2bee631be5d043ac6f849c717f8\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c1cc8058d942246d783692070798609807a1e440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c1cc8058d942246d783692070798609807a1e440"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
