)]}'
{"id":"openstack%2Ftrove~1006248","triplet_id":"openstack%2Ftrove~master~I9b59059967325a81f7486d79dc68b0aabfbc3584","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-21 08:02:52.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}},"hashtags":[],"change_id":"I9b59059967325a81f7486d79dc68b0aabfbc3584","subject":"Fix: \"no space left\" issue","status":"MERGED","created":"2026-09-18 15:37:30.000000000","updated":"2026-09-21 08:42:17.000000000","submitted":"2026-09-21 08:39:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006248","meta_rev_id":"ed5e9735382d533da5034fda89f58a9979ad900b","_number":1006248,"virtual_id_number":1006248,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 08:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-21 08:02:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-21 08:02:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-21 08:02:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 15:58:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-18 17:13:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eca15ca3fcb743d8085d6a1c26b50871acef201d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 15:37:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bda95b33b95ef160a09fe0feac04314fc1004a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 15:58:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/380c2e07f9154dccad20f767a549a41c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5182dfe0aa89443a8e5c2ef641783b5c : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c292e21cf310428f800768ae04c1f792 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7ac6084d857a4cb4832225717dabf95e : SUCCESS in 14m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45bc2373533b4d71f7230d19091c6608af929533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:13:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67f0b0c2db274b8086f4cf1a637fb084\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95e0fa1bf2ce4299a869cdf20d5df86c : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614f48fae96a4a9295a44afb0651521d : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/682a5afd8ef54844b0f194646be46887 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21805461bf4d40268f527d9015a6af87 : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b128167f0db04824a36582d80f865c03 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d77fffe01f94f039bceba3313fce041 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6f9ae79f7a74445b13666180dec3000 : SUCCESS in 8m 27s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/261970b507f14ce089a0402c353db933 : SUCCESS in 1h 13m 46s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/9dced5903ee64e8c8b73ec042921e3ba : SUCCESS in 1h 17m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/ec41acde33e54f819cc76cd095692c67 : SUCCESS in 1h 08m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/a92961421f92428e89560a411ebab747 : SUCCESS in 1h 14m 15s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/280913b3499a46d486040a591e395478 : SUCCESS in 36m 49s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/931c3467306a4cceb36a94ebcc48c55d : SUCCESS in 1h 13m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/2ae78648dad04af197cc8f6b2fc9317f : SUCCESS in 58m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/34d005a3b2ae4c1a8b8820e85911e2f0 : SUCCESS in 1h 05m 58s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8e29b9eba3074c73b2c1ecbcafb2ec42 : SUCCESS in 35m 48s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/83e8d91ffc1a42f58141555c9c6cd9eb : SUCCESS in 1h 10m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/fe8153bdd5a944cc82809e708078629c : SUCCESS in 53m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/12736e9983c948d98585fabd226531c4 : SUCCESS in 1h 33m 40s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aee5c7e110a942a587a17e21dd958b25 : SUCCESS in 6m 17s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/bcda68ac0e35499abf9465e6c1fa7d9c : SUCCESS in 17m 42s","accounts_in_message":[],"_revision_number":1},{"id":"783f7ee1c4d01dcc19e1426856656a523397abac","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 17:41:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d227e3f4f34d1d4100d27f72fb18111ae3efcf89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:57:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ca34a57138343a4bc385094f8334bce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/78987d9f57d34d41acb5fbcbcdbd23a4 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/789729b6b863482097749d36a1af071e : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4ef3739d95d64a5fb2a101b2cd7b1727 : SUCCESS in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9f17f1e4eeccfd822a638f0b2b8884e596076a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 18:58:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90707d1321ca421e8725404c91d9d819\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ce873cd527947c786062f2fa8edd147 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0aa923baae74ca794e900b0e6add234 : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5644b23700104232b2ca0a8e28afc688 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fad7342fb96848f9b9880825f6747169 : SUCCESS in 4m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2effdfa157fd4c8d93e0f7a4aa44fe07 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba1a8efd3694c798b2bd828fc82c1d8 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42c0acb3eb8e4581b1973b703a6dcf80 : SUCCESS in 7m 46s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/af8e60e577764366a250587b042a30b1 : SUCCESS in 33m 13s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/3e99d97d40f14eaca44fb964be9feef9 : SUCCESS in 1h 14m 29s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/e90f1848df1a4bf7b7c3f158fb9129d8 : SUCCESS in 1h 00m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/0fdb3762f06b48df8371c47507be7ec5 : SUCCESS in 1h 10m 15s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/c947ec2cce6f41b5a9d9675233034a2f : SUCCESS in 36m 57s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/c50fc782de0c4bfab1b6985dd75369ec : SUCCESS in 1h 13m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/63795739c2184f72b5977edc33bd143d : SUCCESS in 55m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/53db8d62b7cd4480ac3d466adc9d23cc : SUCCESS in 1h 09m 07s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/7210a3646e06421cbb75cdc835504518 : SUCCESS in 38m 41s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/9901347474fc44738cd6b84625033bc4 : SUCCESS in 1h 05m 29s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/dab0013b7d9e408ca96bd2a82c229329 : SUCCESS in 59m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/6aa40c97a96340eeac27b33225b6e065 : SUCCESS in 1h 02m 46s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2245d88f76e40b3afe8c572b2cb535d : SUCCESS in 6m 38s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/e4406ec1788c4ab89f7715ca77e93843 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b2132d39fb3572aa4f403ab2cb27cc6c1dd97820","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 06:56:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b8016d0ee9555059516de48edfb7abadcbf1bb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 07:00:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/791b17a94bc541188a4477f99e573cac\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/865e78ab01ce41dab8e408dded208f95 : RETRY_LIMIT (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f7b218f1a1214736b1449a44449faff9 : RETRY_LIMIT (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a6d22a7ffbf4467aa586fbb7bee4df6 : RETRY_LIMIT (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da19fe5e117b9a0d787962d14ed9f18efd358306","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 08:33:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4742f8f54f9d28e7839c04c4834c407b07f739ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 08:37:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/852050541ca74544968205d05c5a71f5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c9d61cbe3971400096ae2c9849087883 : RETRY_LIMIT (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/98b6214f7d2d4c0897660c1788cd358d : RETRY_LIMIT (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/22cfcd496a0c460685b766fca887e5b1 : RETRY_LIMIT (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b26a595838c699cdb01ef1edd2ced9f50f3fa5cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 09:40:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e06044a2982425bb88b7f20668375d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f659ec7d41cd497fa992392fbb7f61ca : RETRY_LIMIT\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b469fb96180641dbaf44fa458af1271f : RETRY_LIMIT\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa9bbd15787c4e098f16ca1d0e3bdb2e : RETRY_LIMIT\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60f0b6672f614febb1aa8a5c6cbc2e5f : RETRY_LIMIT\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1211655d48c64323b8d2412e7566caf5 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce5ecf68c8f74674b884c60cee15131d : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/165cd02929c24f59b322b85cf72987bf : SUCCESS in 7m 27s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/1813582eaae54ff2b9f7ae6a61fdd7b7 : RETRY_LIMIT\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/893ebad2cae64d36a4384a6b2bee29fa : RETRY_LIMIT (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/9d8afd57f5f04c71a86ecca5ac882c39 : FAILURE in 20m 01s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/d8e3ab13734646d2a830b0539224acdc : RETRY_LIMIT (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/1da728b1ca4645f1a616edf8f161c311 : RETRY_LIMIT (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/a3b2f1b4e7a14a29af6efe4be50b2f82 : RETRY_LIMIT\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/434ba229022b4a8abf568d56fe7839a3 : RETRY_LIMIT (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/e8dac34c836149ea88c69a9b47d46704 : RETRY_LIMIT (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/b2de12b0e6b7468c886ee54bfac5c7a6 : RETRY_LIMIT (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/7526a6a04dc04e0b9fb714c430e98a30 : RETRY_LIMIT (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8cdcb0db40e4478bbe590d22ec78fb17 : RETRY_LIMIT\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/43f52dddb31c4f8caa687a6e10dc6c2d : SUCCESS in 1h 05m 25s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/44f2ac3662bc49cb911bc4ab4688d6ce : RETRY_LIMIT (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/7a647b85015f4364af38c8765121bbfe : RETRY_LIMIT (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/617c6dc675594f1f9186975cdab18b51 : RETRY_LIMIT (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8f807047add48c18d8760a189879f06 : RETRY_LIMIT\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/9eb9760a92804b46b36d9d4a9d81e42b : RETRY_LIMIT","accounts_in_message":[],"_revision_number":2},{"id":"1d46bde0baf31ff8f194714067a5a96e42ca4aba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 11:01:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e512870129e96c38904efd9edbcc3b950492e8d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 11:12:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"44e55f1e87bc3dba50a91547c059d0ced4e34dfc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 11:26:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc58aaae8b184a70a49175c21f2f3dce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/431b7c15c28e41a49c573ccb7e3ec9db : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cadc6600894f4356aa45aedbd5abc1e3 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e25200b8d3e84bb0983a76bd7098019e : SUCCESS in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1e9b7d38569338e18b37569b0fb8b8481ac20910","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 11:34:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a74b18c4f8894f65125ce2cdd70cf8bc7bc9031c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 11:35:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2ed18f96117c51a288dcbb492830245fae23bb4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 11:50:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dff9249cd3e147089b1bcc120a604875\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4ee63ce189a042fbb0c9cedd90edadc2 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/64d8bea5fd4f48939db930c37582d73c : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7d1864f0908c4b5993cf6efbc296f8db : SUCCESS in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"156804090b80d3e60ed7f0fa6695b368cea90463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 12:02:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f082a21e95fc4a5290df9eed49df62b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dabfa5938a6d4b3f89eb2efcb00ded02 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abff8606356e48ef9c94cd02dcf3f503 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77b4d739c9e14b1da97223feae9c24c8 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efaea8d3c58b4bdfb50f66f8c9b1b928 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d746c79c7bc8473384012cc8d5a2a9c7 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d323c0b94348a89140ddf5daf25bbd : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92948ae9bb2a403991c6edb032e8301d : SUCCESS in 6m 17s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/794a9402072d4fe6830d5ebabe193320 : FAILURE in 19m 38s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/1d9cfe87860f43059c1f04dddd6633a9 : FAILURE in 19m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/1f2bb86350ea42519a9163da679e92d4 : FAILURE in 19m 39s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/fd5a511dc09342be9026d73bd1c566ae : FAILURE in 21m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/77be6bc09a934a0481b8d430747ec8cd : FAILURE in 18m 48s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/f830b670024b47c29a6122b82d9d3fe2 : FAILURE in 18m 54s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/26dc5f125a924600a5337acfe2e8f056 : FAILURE in 19m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/1ff4cb7fba5e42c387169c4090fbab42 : FAILURE in 19m 18s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/097c7d0fc50642e7b9d86c12d09db449 : FAILURE in 20m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/fad3accdf3124bf889c9f439f49edd44 : FAILURE in 18m 22s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/3bc6372f02b3483cbd3cb0ffa3df1221 : FAILURE in 18m 22s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/f01b3b9019394ee09457d40fcf22b883 : FAILURE in 19m 17s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/939a3d48112644c3954e49232bad13a4 : FAILURE in 19m 58s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/daea503c379d471b910360f6fcbbb4a8 : FAILURE in 19m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/f7c63e42aa5b4676b8e072d2da30a9de : FAILURE in 18m 57s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e32c8e51ecf44273b50f6e7de6903f6c : SUCCESS in 7m 58s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/7a64b37224a04da7a82a1f7af451fb5b : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":6},{"id":"26423b9a4dc0717bcfb34c843af516b71447ccff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 12:59:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"9d253f32537a6a3790f49a1401f2eb3989f1b8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 13:01:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b34a33f948272292e52396f2ce2e1544a812c6b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 13:02:19.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0774c7f1f0fe0d5a3f746b4809337cd48e3d1833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 13:03:03.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"dd85a0c28df8dea17d25318cf69a7c6cf50dc20f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 13:19:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dac3ec073dd142e198e612fb8e5d31b2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c7dba858d9b443b0aa77e0c832f8b0ea : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1faaa9533f9a437e87f3b20e784fbb09 : RETRY_LIMIT (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e77ad9f8a78d4d89b45e9680170c6d08 : SUCCESS in 12m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7a68730a7f511bc15172c9652d5b61e433678cb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 13:38:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7c00e3ae449596ba358d9b790965e92e1206f3ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 13:42:01.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d779031f3954e42a50951652221d6c2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8e4a126a30f24487b6eecb213e0feb29 : RETRY_LIMIT (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/baebe9517adc4b749fbd136cb0a36f99 : RETRY_LIMIT (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5829afa5b9264f0db3e7a0b1577115f8 : RETRY_LIMIT (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"72fcc99551e5fa9815875cb7ab52878667b1d63c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 13:44:47.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"fbee747a4b66a4f610500fa346ff78f94e731f01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 13:57:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8d6abd95f084a0c82d8a5e9b1911520\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c45287fb79b84acd8b03879a2f61d026 : RETRY_LIMIT (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1facbce5a45946bdb6203025b8019ad7 : RETRY_LIMIT (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/02596b4dbc37463bb593434af5f7a03d : SUCCESS in 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9e76dd7cea257ae82d254b7cb63f37472aef48d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 13:59:18.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b56cdca14b938363fe951cfe2b3d83f2c4ebb399","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 14:07:09.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99720e05db51456ca3e6bfb82299e3d7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1dc443c9cb294dfebd10cc434ef43954 : RETRY_LIMIT (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ddfcfbaf680f4ead9393901eb6dc4880 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b1e344868a304be69bdad7d2d1649953 : RETRY_LIMIT (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"39a6bc55fe17588e5a84636481a9420ee363dfd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 14:21:14.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13c9966292a3434a9937bda46d797ec6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f6d90023f894107a22d5003ea46495f : RETRY_LIMIT\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a6a996afc224ce88e4efaf708c82269 : RETRY_LIMIT\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07ed2dbdd2aa4c0e8d456df14415de5c : RETRY_LIMIT\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53e0bce813fd413d96f687809503577e : RETRY_LIMIT\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/595adbe596f7410db414064e6d89fd08 : RETRY_LIMIT (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad87859e2c8470793c33b02281afcb1 : RETRY_LIMIT\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a1e8218e2604fa0b63765ae5d3b2da1 : SUCCESS in 6m 20s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/08b1d9e780364fa085ef88a289bfaf05 : RETRY_LIMIT\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/31afe1f097094a4a93aa06e79ac99069 : FAILURE in 18m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/7462a20cb5ba4915af4d58473864adb5 : RETRY_LIMIT (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/0d9ea0605fc545649193cf210be136cf : FAILURE in 19m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/e352986e8fae4feda7df4c5d546f0c0b : FAILURE in 17m 31s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/bd4733a8de8b4da493fedcaab775a31b : RETRY_LIMIT\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/2692f3525ba841e9abf2df3e14aeb400 : FAILURE in 18m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/2622c64d2265406f9d58fcf0e175b7f4 : FAILURE in 18m 49s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/547b503873ca4ea5b8c7ce8b793ba44e : FAILURE in 19m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/9ced7480f0294a339d4b042563f2cff4 : RETRY_LIMIT (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/3b93ced1ad3c4a5d94c75ea5d99025eb : RETRY_LIMIT\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/2609e659c39f48eb9df0d63139f1b8f3 : FAILURE in 18m 56s (non-voting)\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/4494094af0364aecbd58f9196a1128ed : FAILURE in 20m 13s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/b7c84822e6444637b4f7d3497ea0a7d7 : RETRY_LIMIT (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/6ffd41445f34410c80c6a4e95ad26b73 : RETRY_LIMIT (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/397d5a22d0714803b16f20bc908c94a2 : RETRY_LIMIT\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/230b22ff15a14402a23610fd2c656aee : FAILURE in 6m 45s","accounts_in_message":[],"_revision_number":13},{"id":"4b56f860fd53c6a6ddeabd5f07c81501647f176d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 14:56:31.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"f8d9123e65fc2a0a0fade4791e4361b2239635ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 15:11:47.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34bb674629f14c0d9de166127c35df41\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2ba00881da5146f4b27f7a5490b82e37 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9399b4d0dd7545c488dc65942d13aaab : RETRY_LIMIT (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a304907b4f0c43d39f6171bbd45a88d9 : SUCCESS in 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c057d91b57c356dbb830f4bf54a5d9cf82b4a15f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 15:19:39.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"11b37f38f71c63d299e512642199d52499ff4c82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 15:29:00.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"72d946a66b521e8d53c80744f4a0d21db8cda187","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 15:34:49.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"07b0d73bf20ceee0badf0887722c4e7070df5ebb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 15:47:57.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37922a168ee442b78d3b5cf08f003711\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3518c58b22ee45abaca85e86ecc19a6d : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb33861eeec74f649fa449d1a399e818 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b69801711f6746bda0026970744319ef : SUCCESS in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"12842bc3b337d349b347ba138bdf5de222827dd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 16:02:47.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"d48dab155433e3b74be4ff675b22eb0090070fc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 16:17:37.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e1d109c3ad147b2bc002dfac0e199d7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4ea895d0b6bf40a781c8f1659e5bb058 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d657eb7fa93c41fb84edf295a144d316 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1db8399635cd4c27b133a3f7f075ab82 : SUCCESS in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"7b2159757e49cba8ad34ca38573f390140e5032b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 16:56:15.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"bc90a5c598bb2ea964dd7d07e3ce7515aeb40ae7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 17:00:26.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3321e586059041afa6443b53a8ea7305\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f3a5483ee93242a6a281e21b728d0fd9 : RETRY_LIMIT (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0209f335851e47a5b12e75175d7b11a5 : RETRY_LIMIT (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/28d1d9c8a1bc4d91ae2614136e974d19 : RETRY_LIMIT (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"09ff7762f67bd825763dae0ff0f98ecc2091df6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-19 17:35:27.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"17be8e796c97cb15e5d3e39d7d05ded2482d6a32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 17:51:21.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a863bb041bb49a797aa0177ce832bee\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5bf8855fe111476dbbdaf87ac82222c1 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95faede2d30b41d7a0fe4d8250758aee : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f449198464fb4dbaa3022b6e782bbd58 : SUCCESS in 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"068250a60fad617139d8ae7603895a94ded12398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 18:50:32.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/182632dd643545b0a5820d8b87ee0acc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91219bc38f234130a70119f59cbc1684 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62bfed667055435dbc3e7bca28cc7af8 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ee611d605cc4a938724b0dd6dd89d6c : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d604c39065454768ab7bb3bbf86c389a : SUCCESS in 2m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/883d3678d79b414aa968604b4fdb64b8 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19347d47a995440fb0b2b2d32c7dfdda : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80dadbfc4ba64fbbbef89edcfeeec974 : SUCCESS in 5m 27s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/fa3d3ae43ca84061a498952438a8df91 : SUCCESS in 33m 07s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/5ef9057d308241e5aa67a3796f1dedc5 : SUCCESS in 1h 13m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/de8620aa4bca4783ac0693c4198a7f6e : SUCCESS in 1h 01m 07s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/d5986fd9fba349469f2a19df1cf29e2c : SUCCESS in 49m 03s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/cf2e5084a9d04b30b4a3c496e97d98e6 : SUCCESS in 1h 03m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/b9e4185f2293417694cdecf144bef049 : SUCCESS in 1h 09m 30s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/034a97e56f4b4d9e8afeb63adc3b0e96 : SUCCESS in 31m 05s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/12e0d0a1b7d546b1b4ee5fb5697b0c50 : FAILURE in 32m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/7c0e651114f747e194eebdbea5a67312 : SUCCESS in 53m 19s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/c5ff961fbe4b4a988825bf5d86ed9d45 : SUCCESS in 47m 57s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/bf21a756b30340e3b3c304f9792160b1 : SUCCESS in 55m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/cfe3037dee774fb8a341a6a3ff4fd7e1 : SUCCESS in 1h 03m 31s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/d8ed1e401a6b443da6d66236fab9fd7c : SUCCESS in 34m 35s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/3d5dbbc5a21a46cba8d55a2acf7f3361 : SUCCESS in 1h 01m 51s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/0c527d61f81f42cf9748f50629d4878a : SUCCESS in 34m 12s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/aac0decfda974bb7b414ec7cdcce9313 : SUCCESS in 49m 47s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/ace2abb7e3004aa8bb0e6b23cadf5852 : SUCCESS in 49m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/bb5500330abd4834b84ca48bd552743d : SUCCESS in 54m 14s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b91c11531ee046afb48008be4f4339fa : SUCCESS in 3m 07s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/627c84b682c84e15830ee94fe18f4e1d : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":20},{"id":"77f79fad5685094baa8b818c137a3bb3084eabca","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-20 16:51:13.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"66677fd87f94bf850c9eb476995cb3812e7d27a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 17:06:03.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d1ae176e8674e24861a1213ef53246a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3dac1cf363f24b498b3eef070eef43bf : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8a34baba8e264ba897d3aad4d3278808 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a8eb5f0b4ac14fedb52365996f243e7b : SUCCESS in 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"f69afd8f44a2ca6566b184eaa08452b07582f5a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 18:12:36.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40ab7b247fbb438795b49c47161c2da9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cebf0784fbb40d397c8a61c5192a953 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c799209549442c9f75d17438030652 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e204182765144c785f40195290f9bdd : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/871f093655384c93b68cfdd5e11df7f3 : SUCCESS in 6m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5717b73be0bc4fcda47a7f299568530b : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a7ac75b52244a5aa93a76bfeb5105a3 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7252193d73e47019e76bc9209f1f98c : SUCCESS in 7m 47s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c72e78ce077d44fda5a12404a41191ac : SUCCESS in 36m 13s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/e377c88e1d684f068484a94ca52c6166 : SUCCESS in 1h 19m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/6588d7e53cf1448c802a322c32ebbccd : SUCCESS in 1h 02m 26s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/0d653cece67c472194f9b532e82fd143 : SUCCESS in 54m 09s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/8c84aff7e9df48b5aa3cf1dccacf991a : SUCCESS in 1h 05m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/55417ccc6ae94bcf89fe2aea23aae1f9 : FAILURE in 18m 15s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/8c42d17bf6524f4cb6cf2ebb37bb1a30 : SUCCESS in 35m 31s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/7fc5a0fd6c774b358ff8ad7cc4f23f7b : FAILURE in 40m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/a953c9affd8346a6a95ca6713c7bdb38 : SUCCESS in 1h 00m 12s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/ddbaec624bf4491985d5400a8bdd7887 : SUCCESS in 52m 16s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/a2fc1fc8c56d44619f065a7d2ab55ace : SUCCESS in 53m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/1dc2dbb0438749598687cc2dbd1b0842 : SUCCESS in 1h 04m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/e831c01f2dca464d9b1eda2e723289b7 : SUCCESS in 39m 48s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/70015a725ac74fb2b56ed408b6c76ba0 : SUCCESS in 1h 00m 10s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/bc2cbbf347ac42e1b96e3479e698ccbb : SUCCESS in 41m 32s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/b16d855f81734c7dbe5a9b545384cab8 : SUCCESS in 54m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/4a85c41a91d045b28d33358f2c11d393 : SUCCESS in 55m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/d7e317f4525f481bbb8101e2bac67e40 : SUCCESS in 1h 00m 55s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a5bb28ef8f9f44eba42eedd6f245bde1 : SUCCESS in 4m 09s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/f4e403cfa9704640ad5b64035aa65f61 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":20},{"id":"e85edb256c56b6997628d46d10495e1d758462f0","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-20 18:15:34.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"331f54b634405e431662d8b74050b1b6c2fff54e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 18:29:58.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5777e7b80cb44aa998c5390532fa1553\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb58fedc77764398b1a227812217ccc1 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5bce6c8f3c6844ffb9ff02cf7e023bbe : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/33eda566f3e44b5e966436a9e9972a51 : SUCCESS in 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"6251fe3aa76ef68ffbcd56d1803b35a52da0b3fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 19:37:11.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab01d3b467e9484989ae195343a157fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cb2e22559f046989dde9f72e28a8904 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1858cc31e5264333a8506f30b392d6c6 : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1899acbafa994a51a8241fafd6d385d8 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd3c8bd6b3e345b9a2632844962ab9ad : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/95f5a51366f74439bfe4902804315731 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efaac8416bd54f6f9906450ce77e62f4 : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf62b1764e784ad0bb7b7e160a4eef3f : SUCCESS in 6m 53s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/18acea3c055e409aa367b928bb502798 : SUCCESS in 35m 54s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/4b469c822dc7404fbc1e8adda0c3cb49 : SUCCESS in 1h 20m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/a2d6ba05627742e1a182534a7d59660b : SUCCESS in 1h 03m 59s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/e651b6f348c7412f8ab12632718696bf : SUCCESS in 53m 08s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/aa6011d3f4ec4f01b3550df5f20348ba : FAILURE in 43m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/57b2e3924db84881a3e9ffb1396318e5 : SUCCESS in 1h 11m 16s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/624c8212e4be4cf3a150808f62293a86 : SUCCESS in 33m 58s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/e56d50e970c64be88d035a294c235958 : SUCCESS in 1h 17m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/b81607dd84024ed899e2f885848d92bb : SUCCESS in 57m 53s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/2c6df4dc86dd4535a1ee640194c5c8b0 : SUCCESS in 46m 49s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/4817877708a143c3b43434d9f93206c8 : SUCCESS in 1h 12m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/1421ce85301a46cc8bf748dfecdb23e2 : SUCCESS in 1h 01m 12s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/3830e3d57ff94bb695cf5c394853972c : SUCCESS in 36m 32s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/65539f471e0049a6b54049c2f89ff887 : SUCCESS in 1h 05m 53s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/2f04483de8584ccba8faeb264f8c0d2a : SUCCESS in 40m 41s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/2bd1182650a84ce288ba9cd593a306c0 : SUCCESS in 49m 13s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/41705dfc277c4b82b2b934d578e984d3 : SUCCESS in 51m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/3c939b0f46f64e00b0ac941107175b17 : SUCCESS in 1h 06m 54s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fac3a8ab287d4fadb6a1d8bc60888b76 : SUCCESS in 5m 03s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/58936a0adbac4242951200c8be0d1ec0 : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":20},{"id":"28fa71f3295424fb5ba6c2716d55d91f9e23c3d9","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 03:48:06.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"8aafc96bd5976668b2b61bd8cc1155b98360fe7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 03:54:09.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"2eef8af61def01a036b692bc746eb29fd3352bc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 03:54:58.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"adbaf74a00ca1a9744c16a31058361010eb7930e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 03:55:16.000000000","message":"Patch Set 22:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch stable/2025.1.  The\nproblem was:\n\n  Nodeset trove-ubuntu-noble does not match existing definition in branch master\n\nThe problem appears in the \"trove-ubuntu-noble\" nodeset stanza:\n\n  nodeset:\n      name: trove-ubuntu-noble\n      nodes:\n        - name: controller\n          label: nested-virt-ubuntu-noble\n      groups:\n        - name: tempest\n          nodes:\n            - controller\n  ...\n\n  in \"openstack/trove/zuul.d/nodesets.yaml@stable/2025.1\", line 1","accounts_in_message":[],"_revision_number":22},{"id":"1b73d694f70b7c154127d85a47989047af9e0654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 03:55:29.000000000","message":"Patch Set 22: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch stable/2025.1.  The\nproblem was:\n\n  Nodeset trove-ubuntu-noble does not match existing definition in branch master\n\nThe problem appears in the \"trove-ubuntu-noble\" nodeset stanza:\n\n  nodeset:\n      name: trove-ubuntu-noble\n      nodes:\n        - name: controller\n          label: nested-virt-ubuntu-noble\n      groups:\n        - name: tempest\n          nodes:\n            - controller\n  ...\n\n  in \"openstack/trove/zuul.d/nodesets.yaml@stable/2025.1\", line 1","accounts_in_message":[],"_revision_number":22},{"id":"2c7ba66408b79dad5abf5643882e8fc354f925ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 03:58:12.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"d08d9829e2fe57441d94185a1f0762cb41bde571","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 04:06:38.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"ef724a884cd9c5cad66139d58aab4ae36a03931e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 04:12:10.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"9b4093242ca2cfe431fe7a848746143d515600e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 04:30:18.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dc7148d17964f6fa1ad6a7e6259718f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1501c2aac9304076b39d4ef70c8967e6 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0c43c2bd0ac747038fc10aac2081a954 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/188a64db0d6946419fe75a2b533d8ac5 : SUCCESS in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"36a01f90c2ac2d8a86612d8f23963434d914904f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 04:34:16.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"6df8bbff87e6913afe56b99f99f029bb69eec641","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 04:48:49.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba1af5c609ab4284a21e4bd4f99d9a6d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b912d0e5807a47febad2a65a006fcbda : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aec0f2ff0f084307b8c2cc6f0afe81ad : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d218f24272b64e9ea180ad3ab3b9160c : SUCCESS in 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"dc45551866a1cd89f6e87dbadd9270a2914644b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 05:42:52.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"3c01b1f784b3217be5734946f8465676d19e616d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 05:59:28.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74ccdb680f1c4a2cbdc5468be141e55f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/da84cae2877147dba889d3918413d1ca : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7bc9825fa0924447bbe44e1b597f5ab8 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b60895e0f78b42d18daeb7045459829a : SUCCESS in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"bd0e4d53635ecc363d46f316e8e9f701f1c278a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 07:40:38.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ed32f980ac3403ab29dc91876bf46ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76308676fa4f405bb0da88f0da9b5853 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf9739b11b141b0a52b5e326e31b496 : SUCCESS in 6m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cac3e5923cc44aa5adfa6ab131779e63 : SUCCESS in 7m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/173a838bec1d44bdad56a45d545c08f2 : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0bf6aaf0d3c24a8692b3f4c4cdf0e65a : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a92865f65894478a780f9f87c3825a6 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90009110925f41539c3fb7958fe9041d : SUCCESS in 7m 50s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/d22f2989292648e19949fa2b4e4aa4a4 : SUCCESS in 36m 21s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/760b494fbd6e43f98a274aa8fc65800b : SUCCESS in 1h 19m 30s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/4532b1542fa144839766f6d24c933c2b : SUCCESS in 1h 03m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/1b99f96dca79422791d096c6df177936 : FAILURE in 1h 30m 32s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/06b372d082834e799a61f8cfa885864b : SUCCESS in 36m 40s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/2d0923099bb14645aee00cab2901670a : SUCCESS in 1h 10m 52s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/beabce885d254a31aa84990086070dfb : FAILURE in 1h 04m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/cfe43203bf0145a7a80482d7671df484 : FAILURE in 1h 55m 51s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/2796e89f430b4704b7e294bce55d4da2 : SUCCESS in 35m 16s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/789c863031da441d9f51af21ad745647 : SUCCESS in 1h 10m 14s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/89dfbccbbe1e47738b884cd2e07f957e : SUCCESS in 53m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/dccd6bf63c034b1687244818fc762a04 : SUCCESS in 57m 15s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bd4e38197d2c4537bfa96e30a805069c : SUCCESS in 3m 27s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/19b20440ee514a2d9b93a3a941e91479 : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":27},{"id":"b783ce496d13a9b9cbb0bca0e31eff9edde0e2c7","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 08:02:52.000000000","message":"Patch Set 27: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"fecf638509594e1915418ae91af1ed222f4742ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 08:03:31.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"cffc1235b08e1a04d28988a88858b8b85cc15809","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 08:39:13.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a4cc2851149456ea0e5641ab5a5a0a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec61a659dd54c4589fbcf2e0bc24171 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/301e1118fc1045988f8ce37e86dc49d8 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71d0d46a4e2b4202b043bb6762186e08 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea0d2d969a7482880a1d3190025d217 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e39bc369a65141b0b5e2dac62ecd7d4f : SUCCESS in 8m 31s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/48259cf94ff4443889008b1f634d75dd : SUCCESS in 32m 48s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/4fd23b1b999e4d9b8c3e0e2f25a44108 : SUCCESS in 32m 42s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8aa4193b350e4499a0ed026073ec12cc : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":27},{"id":"94bda322111edbc6a4d4d97e81390aef0c40836e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 08:39:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":27},{"id":"ed5e9735382d533da5034fda89f58a9979ad900b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 08:42:17.000000000","message":"Patch Set 27:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aec85a247aa94673a94949112b0a7d41\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89aa263fd0ce43a7bb0666ac72b0ab82 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"8a4090dd8f0315de445ebd7f2c1af99ffb8bd395","revisions":{"6a6ed77f12392a2304998ff4d1e788b8b44406fb":{"kind":"REWORK","_number":1,"created":"2026-09-18 15:37:30.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/1"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a6ed77f12392a2304998ff4d1e788b8b44406fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a6ed77f12392a2304998ff4d1e788b8b44406fb"}]},"branch":"refs/heads/master"},"36a656579b94dc44a54cd69b131fd8617d08e281":{"kind":"REWORK","_number":2,"created":"2026-09-19 08:33:18.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/2"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 08:33:05.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/36a656579b94dc44a54cd69b131fd8617d08e281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/36a656579b94dc44a54cd69b131fd8617d08e281"}]},"branch":"refs/heads/master"},"3a7da1780532f57e77b81e37617caa714eb29d3e":{"kind":"REWORK","_number":3,"created":"2026-09-19 11:01:56.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/3"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 11:01:45.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3a7da1780532f57e77b81e37617caa714eb29d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3a7da1780532f57e77b81e37617caa714eb29d3e"}]},"branch":"refs/heads/master"},"1b951a13d7bf136015f5c66a812bd737077f7f43":{"kind":"REWORK","_number":4,"created":"2026-09-19 11:12:18.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/4"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 11:12:08.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b951a13d7bf136015f5c66a812bd737077f7f43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b951a13d7bf136015f5c66a812bd737077f7f43"}]},"branch":"refs/heads/master"},"cdbc0747a84b7c59415d0528d552c26619067cf7":{"kind":"REWORK","_number":5,"created":"2026-09-19 11:34:00.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/5"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 11:33:49.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cdbc0747a84b7c59415d0528d552c26619067cf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cdbc0747a84b7c59415d0528d552c26619067cf7"}]},"branch":"refs/heads/master"},"6ea885d2cdc4f0030aac947f5c4afa3fdd1609e5":{"kind":"REWORK","_number":6,"created":"2026-09-19 11:35:14.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/6"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 11:35:00.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ea885d2cdc4f0030aac947f5c4afa3fdd1609e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ea885d2cdc4f0030aac947f5c4afa3fdd1609e5"}]},"branch":"refs/heads/master"},"e2b6b49e99e39630fe130cf37e48b99e224b7113":{"kind":"REWORK","_number":7,"created":"2026-09-19 12:59:51.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/7"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 12:59:41.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e2b6b49e99e39630fe130cf37e48b99e224b7113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e2b6b49e99e39630fe130cf37e48b99e224b7113"}]},"branch":"refs/heads/master"},"a71793724c35fcab63e9a5c15fe7bd922775a6c2":{"kind":"REWORK","_number":8,"created":"2026-09-19 13:01:01.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/8"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 13:00:54.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a71793724c35fcab63e9a5c15fe7bd922775a6c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a71793724c35fcab63e9a5c15fe7bd922775a6c2"}]},"branch":"refs/heads/master"},"968b0c203f15ca2f4600261cdb18a2d6328eb2b2":{"kind":"REWORK","_number":9,"created":"2026-09-19 13:02:19.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/9"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 13:02:07.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/968b0c203f15ca2f4600261cdb18a2d6328eb2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/968b0c203f15ca2f4600261cdb18a2d6328eb2b2"}]},"branch":"refs/heads/master"},"8534c86c72fa0a8529221fd099751ed7d1ce6ff4":{"kind":"REWORK","_number":10,"created":"2026-09-19 13:03:03.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/10"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 13:02:54.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8534c86c72fa0a8529221fd099751ed7d1ce6ff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8534c86c72fa0a8529221fd099751ed7d1ce6ff4"}]},"branch":"refs/heads/master"},"df987520c9100412eacd97ab4b024b0f9590b404":{"kind":"REWORK","_number":11,"created":"2026-09-19 13:38:39.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/11"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 13:38:30.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/df987520c9100412eacd97ab4b024b0f9590b404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/df987520c9100412eacd97ab4b024b0f9590b404"}]},"branch":"refs/heads/master"},"c042466ca56863fadc5aba4f03f4b7ff0d629276":{"kind":"REWORK","_number":12,"created":"2026-09-19 13:44:47.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/12"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 13:44:38.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c042466ca56863fadc5aba4f03f4b7ff0d629276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c042466ca56863fadc5aba4f03f4b7ff0d629276"}]},"branch":"refs/heads/master"},"919333673b6ac61f2ad8070f41d993b2d009a288":{"kind":"REWORK","_number":13,"created":"2026-09-19 13:59:18.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/13"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 13:59:08.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/919333673b6ac61f2ad8070f41d993b2d009a288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/919333673b6ac61f2ad8070f41d993b2d009a288"}]},"branch":"refs/heads/master"},"af54abb8193933cf14c6fbe97b3fa6603df604b3":{"kind":"REWORK","_number":14,"created":"2026-09-19 14:56:31.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/14"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 14:56:22.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af54abb8193933cf14c6fbe97b3fa6603df604b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af54abb8193933cf14c6fbe97b3fa6603df604b3"}]},"branch":"refs/heads/master"},"a2dbe8ed8ed1b304b004e831cbc798b6d3651009":{"kind":"REWORK","_number":15,"created":"2026-09-19 15:19:39.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/15"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 15:19:29.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2dbe8ed8ed1b304b004e831cbc798b6d3651009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2dbe8ed8ed1b304b004e831cbc798b6d3651009"}]},"branch":"refs/heads/master"},"73c1cac5ddbd68ed4848658f76b767f98a4c5390":{"kind":"REWORK","_number":16,"created":"2026-09-19 15:29:00.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/16"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 15:28:46.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73c1cac5ddbd68ed4848658f76b767f98a4c5390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73c1cac5ddbd68ed4848658f76b767f98a4c5390"}]},"branch":"refs/heads/master"},"69a24baeda0c49db0f1f5d963d189d0b9a001c66":{"kind":"REWORK","_number":17,"created":"2026-09-19 15:34:49.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/17"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 15:34:40.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69a24baeda0c49db0f1f5d963d189d0b9a001c66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69a24baeda0c49db0f1f5d963d189d0b9a001c66"}]},"branch":"refs/heads/master"},"1044cce393f23181549d883171afc4285fa5f6b6":{"kind":"REWORK","_number":18,"created":"2026-09-19 16:02:47.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/18"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 16:02:38.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1044cce393f23181549d883171afc4285fa5f6b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1044cce393f23181549d883171afc4285fa5f6b6"}]},"branch":"refs/heads/master"},"aa437c245f12194f830d358c7f4080c020d7c0f7":{"kind":"REWORK","_number":19,"created":"2026-09-19 16:56:15.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/19"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 16:56:05.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa437c245f12194f830d358c7f4080c020d7c0f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa437c245f12194f830d358c7f4080c020d7c0f7"}]},"branch":"refs/heads/master"},"04310a2f18475c4fab8e90b73f32b518440b8b5c":{"kind":"REWORK","_number":20,"created":"2026-09-19 17:35:27.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/20"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-19 17:35:17.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04310a2f18475c4fab8e90b73f32b518440b8b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04310a2f18475c4fab8e90b73f32b518440b8b5c"}]},"branch":"refs/heads/master"},"acb879978b29a720894201c5e787da06f69f7290":{"kind":"REWORK","_number":21,"created":"2026-09-21 03:54:09.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/21"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 03:54:00.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/acb879978b29a720894201c5e787da06f69f7290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/acb879978b29a720894201c5e787da06f69f7290"}]},"branch":"refs/heads/master"},"f63063d9a4568d1791b9f29701ed6f5a62d45936":{"kind":"REWORK","_number":22,"created":"2026-09-21 03:54:58.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/22"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 03:54:48.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f63063d9a4568d1791b9f29701ed6f5a62d45936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f63063d9a4568d1791b9f29701ed6f5a62d45936"}]},"branch":"refs/heads/master"},"9350c44ff0bdae05b0a1ea2a54ddcacc55bd313f":{"kind":"REWORK","_number":23,"created":"2026-09-21 03:58:12.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/23"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 03:58:01.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9350c44ff0bdae05b0a1ea2a54ddcacc55bd313f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9350c44ff0bdae05b0a1ea2a54ddcacc55bd313f"}]},"branch":"refs/heads/master"},"43db2d367517b37a04d62b1dad696e78a5639bbb":{"kind":"REWORK","_number":24,"created":"2026-09-21 04:06:38.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/24"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 04:06:27.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/43db2d367517b37a04d62b1dad696e78a5639bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/43db2d367517b37a04d62b1dad696e78a5639bbb"}]},"branch":"refs/heads/master"},"5d92a8e71e73218467fdd1990801d49c2b3d6128":{"kind":"REWORK","_number":25,"created":"2026-09-21 04:12:10.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/25"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 04:12:00.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5d92a8e71e73218467fdd1990801d49c2b3d6128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5d92a8e71e73218467fdd1990801d49c2b3d6128"}]},"branch":"refs/heads/master"},"0f4f8e336247d36fa17bfff7b8e417d82d47fd56":{"kind":"REWORK","_number":26,"created":"2026-09-21 04:34:16.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/26"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 04:34:06.000000000","tz":300},"subject":"WIP debug \"no space left\" error","message":"WIP debug \"no space left\" error\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0f4f8e336247d36fa17bfff7b8e417d82d47fd56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0f4f8e336247d36fa17bfff7b8e417d82d47fd56"}]},"branch":"refs/heads/master"},"8a4090dd8f0315de445ebd7f2c1af99ffb8bd395":{"kind":"REWORK","_number":27,"created":"2026-09-21 05:42:52.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/48/1006248/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/1006248/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/1006248/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/1006248/27"}}},"commit":{"parents":[{"commit":"4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422","subject":"Merge \"Ignore swo files in git\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fbf0280fd4d9cd2625d342b4d6fadf54ebf6422"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 15:36:54.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 05:42:32.000000000","tz":300},"subject":"Fix: \"no space left\" issue","message":"Fix: \"no space left\" issue\n\nDIB uses tmpfs for the target root, with a size of half the total RAM.\nIn some cases, the \"nested-virt-ubuntu-noble\" nodeset provides an\ninstance with 8 GB of RAM, and 4 GB of tmpfs is not enough to build the\ntarget qcow2 image when Docker images are embedded into it.\n\nChange-Id: I9b59059967325a81f7486d79dc68b0aabfbc3584\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8a4090dd8f0315de445ebd7f2c1af99ffb8bd395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8a4090dd8f0315de445ebd7f2c1af99ffb8bd395"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
