)]}'
{"id":"openstack%2Ftrove~1006441","triplet_id":"openstack%2Ftrove~master~I79b7f23b68b7fb3a66eefafcf67a87307ce109e1","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I79b7f23b68b7fb3a66eefafcf67a87307ce109e1","subject":"Use linux-image-virtual kernel version","status":"MERGED","created":"2026-09-21 05:51:02.000000000","updated":"2026-09-21 08:14:04.000000000","submitted":"2026-09-21 08:11:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006441","meta_rev_id":"6e8b268314b983e4d3c15c65d2ad61f78dce1d05","_number":1006441,"virtual_id_number":1006441,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 08:11:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-21 07:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-21 07:32:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-21 07:32:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 06:06:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-21 07:08:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce551ca20977466b2ec5ad80479db31ab53af571","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 05:51:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"feb20caab72d7d316c22b0724c0803787938020e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 06:06:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f96f760a28a14bdc9658761cdcf7e5aa\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/990cb3b2a84143a7a9d95ca3b01ed3e2 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/55cc2f85624b40b9b48f449c94a77669 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bb6347b48e7742dc8fe88b88eff5ea32 : SUCCESS in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c5efccb0471fe85b022fa476b97caa4d83b23d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 07:08:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd739567ebe848309dcacfadd373673d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc2e21e8084a40868d3647dc63e9ce36 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/983edcea13d6466a98f1c5b1b285fa8a : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5f1f79ba7a94071b93441178b8d6e4d : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c077ba6e98145d4899999bb9bf84fb1 : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1e3137445f154577bcf6d1eca45b9754 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7796a5ccc30d47fda0b2a9d23c281f91 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/021daaaaf9d1400fa627df1738547e67 : SUCCESS in 4m 28s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/975cd4f236f3490fbdc9dcc72ddb00fe : SUCCESS in 33m 55s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/d07166f6b3a54b1a948919438762d318 : SUCCESS in 1h 15m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/113240d8dc9b43f2ba0a724825ba51bb : SUCCESS in 1h 01m 33s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/4cf8581f1079486c8e0bf6c708112e7a : SUCCESS in 52m 36s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/d787b84bf2b349feac1c697b323ece6e : SUCCESS in 1h 08m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/37c03f6946ef418586dd750a616a39dc : SUCCESS in 1h 09m 33s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/8e0cd2832e534d1198303b3d571b0c78 : SUCCESS in 31m 38s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/7d6fc18d09a44c38a4763c6530f9fb81 : SUCCESS in 1h 09m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/719d08ce44b04c8caed8a4376418a30c : SUCCESS in 54m 59s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/d494176f718342f49f34bdf69cc7724e : SUCCESS in 45m 51s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/0030740649b048dea60ccb93fbbba4b8 : SUCCESS in 55m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/5585cdcfbeb942918d6b08a192c56352 : SUCCESS in 1h 06m 41s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/d317b72037314e82ba8c5c58c2eca65c : SUCCESS in 36m 53s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/fcaa676ac0154bc298bd720138627069 : FAILURE in 33m 30s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/4d03c24aab664a19ac016c70e795b405 : SUCCESS in 37m 25s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/d0925f5752ca4adbb382c85b5e345069 : SUCCESS in 50m 49s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/c49ac64227b644bc8a9b1ed45a689402 : SUCCESS in 49m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/7f54eefd7505448c98631fc3ddb8db2e : SUCCESS in 56m 19s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/190152a6e1cc400091e90a53ce50850b : SUCCESS in 4m 47s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/c2f3bcd6f16b4495b915bc5ede2bf1fc : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":1},{"id":"5c73f0e14af311bb41e16f60ba4cf92ae236056d","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 07:32:31.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b70f3f052d873c1b8844edf331f223b3ad6813e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 07:33:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dabc47c8fe52badc77f8c3a5638f7119112e1347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 08:11:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3cc07e7eea74e979e75fdf6b8830497\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16d268e2fe0c4a188bb0d6a223b0186c : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1af798603a3242808aaa6c45cc6bcc1a : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf2b455a494041e29c42c6e9f09b2f4a : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21795a9325714adb91af62aadac5511f : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c507a335bec5450588e580b4c2dd993c : SUCCESS in 8m 23s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/874c1965203846cab928d3232b33ba4b : SUCCESS in 34m 55s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/f7de2fe5ac684f7e9b8f68d669c97066 : SUCCESS in 31m 38s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8b0edeed6d894ef88e087b286ff051bd : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e7417276795948ab768356b8b7e6dbeb233dcd62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 08:11:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6e8b268314b983e4d3c15c65d2ad61f78dce1d05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 08:14:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37c3010a0c134b8885382d1729e2cb82\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af766dad89d5434eb37a858bd7d39708 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0737aa312faa05f69824729e42026956f6860e6","revisions":{"e0737aa312faa05f69824729e42026956f6860e6":{"kind":"REWORK","_number":1,"created":"2026-09-21 05:51:02.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/41/1006441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/41/1006441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/1006441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/1006441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/1006441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/41/1006441/1"}}},"commit":{"parents":[{"commit":"ff5d0d51369893ec7e7962a089808369671486e4","subject":"Merge \"Fix: Validate configuration ds version on instance creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ff5d0d51369893ec7e7962a089808369671486e4"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 05:47:05.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 05:47:05.000000000","tz":300},"subject":"Use linux-image-virtual kernel version","message":"Use linux-image-virtual kernel version\n\nSet DIB_UBUNTU_KERNEL\u003dlinux-image-virtual for Zuul jobs, which is\na lightweight kernel package for DIB Ubuntu Minimal element.\n\nThe default kernel package is larger than necessary for our CI images\nand increases the resulting image size and build time. Since these\nimages are used for Zuul CI jobs and do not require a full generic\nkernel, using the virtual kernel is sufficient and helps reduce the\nimage size and resource consumption.\n\nChange-Id: I79b7f23b68b7fb3a66eefafcf67a87307ce109e1\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e0737aa312faa05f69824729e42026956f6860e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e0737aa312faa05f69824729e42026956f6860e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
