)]}'
{"id":"openstack%2Ftrove~1006573","triplet_id":"openstack%2Ftrove~master~Idb12fb7b0cdf580be30df7658d01885c1814e3da","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-21 17:00:46.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}},"hashtags":[],"change_id":"Idb12fb7b0cdf580be30df7658d01885c1814e3da","subject":"CI fix: theck that DIB_PYPI_MIRROR_URL is present","status":"MERGED","created":"2026-09-21 15:21:12.000000000","updated":"2026-09-21 19:00:52.000000000","submitted":"2026-09-21 18:59:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006573","meta_rev_id":"bd304befc1237d95d07f1ccab1893918e7dde7f9","_number":1006573,"virtual_id_number":1006573,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 18:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-21 18:23:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-21 18:23:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-21 18:23:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 15:43:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-21 16:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"231caa14fb11378843286e82c1e7d7f5498b38d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 15:21:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b560c14282a21033080d6c79244a610efe4107c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:43:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b195935cb7f4409a408359be7a37da5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a7f2c8ce441e4df3ae1c127d5d1272d3 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eb470f59093945248bb0531303b3b83b : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d0b3fe68576543bba8b1fc26d23a72a2 : SUCCESS in 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afd3f59246a0d87b9d5f20093d9b7aaaf73f9667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 16:50:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de5cb584481e4aac8c1308892d5f7b63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09c0e9f8258844f99837b1963aa3be21 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2910c7ba42104404a9fae89e3420ef97 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1cf25f2121840be85af706a2d145908 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d92d3f66238b41da99bab24cf44b325e : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4a3c0e7ed604d3b905546988141eb8f : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/438e648ca52c4df5bb3a87c9e758b128 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb8481c47da64c0fbe61dbe495ea0c85 : SUCCESS in 4m 28s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/6e56528a4c1342a6a6bb6b87e1110eb4 : FAILURE in 26m 11s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/4cfe9aad2e054bfb92a608f4aa7480c5 : SUCCESS in 1h 25m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/187c07b167cf4973a1bdcbc04997e01d : FAILURE in 46m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/d97708202fc148bf90e2b39505253e33 : SUCCESS in 1h 16m 30s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/b3d5bd2fccd8458998569aab214ec9d7 : SUCCESS in 30m 08s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/8f8812343e8546d3b3470535156eaddf : SUCCESS in 1h 18m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/1ae7193cf2ef49ce997f3c1e06ea62d9 : SUCCESS in 1h 05m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/dd0c1a23134e47bba0f80ec1043bb8a0 : SUCCESS in 1h 18m 27s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/6c40d4fdf5a447e9b4033dc4f6eabff9 : SUCCESS in 47m 45s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/9aa1e57f8e0e46fca7836a7b47536362 : SUCCESS in 1h 14m 03s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/860c5707e13247ac9e90e76672ece0a2 : SUCCESS in 50m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/31bec19933274b5e975412e656a17490 : SUCCESS in 1h 06m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f17881cea3f94688976a828cf9f9d13c : SUCCESS in 5m 19s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/634ec41c5cc2429db0fa0d8cba3bd581 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"874c9fc24e01607675c863c042d1da254fce8936","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 17:00:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c3539d41a806cdea14972af740063c40ec82fa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:18:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53002456e8af4642b473d57e668e243a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a09ca57006cc4b3091999365220610f8 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6904ac82284342779df9ded598bbcdcb : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/74f52703d24049caa6554741d4ecd5e4 : SUCCESS in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9543b0fd5fb46437edcef985f4838baa35f7cca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 18:22:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2996767205504b03aa719e1e57ad02ff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6edd49336c34f5c8a18f40679f8f02d : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68daba1438324da5838beda278231049 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97c790efb3f5406ab154e3702613502c : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/077c238655d54da4a69e29667fec6ba2 : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e509cf24f8114ddab5f008bdce3103fd : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbbd7721d0746dea1f8fa160927de62 : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c85d79532c9f4504a083d9a6676f1bce : SUCCESS in 8m 16s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/3fe45bc28b4b4b2aadeb753dcad28663 : SUCCESS in 33m 20s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/2bf550c831b348a298b70781778330b5 : SUCCESS in 1h 13m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/db46a9307bf34c438943a6073737a988 : SUCCESS in 1h 00m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/6d3e3b857d314996b06c676f0ac4ebd5 : SUCCESS in 1h 09m 04s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/309c7c7f644047a4a35c623491dd4240 : SUCCESS in 32m 49s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/5e6d0ec1117049a5809447ba0f4d10ab : SUCCESS in 1h 06m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/9f620c04287d49daabdecf91326c7b52 : SUCCESS in 55m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/45d8b10d8e6f47d4bb25a1228d471971 : SUCCESS in 1h 18m 33s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/b4543c2ace4746b48948a74d5a6bc100 : SUCCESS in 34m 47s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/a7115acc292d4c849a4f65122da09e55 : SUCCESS in 1h 00m 38s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/b1b6a1e395a94d3691d76949ca0adeca : SUCCESS in 48m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/c826a1868e1f44f6b2e3182e20cb5720 : SUCCESS in 52m 56s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c7318cdbe3284a019d4adf75456d6535 : SUCCESS in 5m 19s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/bef04b3dfe8e416b91d00ae52e461e3a : SUCCESS in 9m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8e08675158afb5dae795362432c88067bf9d2038","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 18:23:01.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60dd421c629cb34e613a6f01dc45e2d2e40c669a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 18:23:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f492482035a802be09fec1fa44c63b88b2ce2c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 18:59:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e5cd6de80b741fc9d224fd6289e6ed3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a297ec0452dc421399b45fde0ddde57d : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4501eb9a3f294b4490ecd2363f36a978 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/faaa5a3bac1e42c9a385ba51b33df7d9 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd62cfb112ab44af9eddd19515214619 : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e99193a15b6c41429ff95fdf1eadfa4c : SUCCESS in 9m 02s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/66e1099803484eb4a6a95f89ce8b4ec2 : SUCCESS in 32m 31s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/7d260f6c777b489c8b3c6045260dcc86 : SUCCESS in 31m 27s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/b52c08ae7a704e07b29e4a2b78a01f88 : SUCCESS in 35m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6b6c557f3e4bee578a660251a191bd48a98adcc7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 18:59:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bd304befc1237d95d07f1ccab1893918e7dde7f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:00:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fd5833d3e9e4832b9054ee0f122e342\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/823361802c3b412d9d145569a69ddb5a : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5019815f46645272bf4a11552c1dcce83eb7c7e2","revisions":{"5019815f46645272bf4a11552c1dcce83eb7c7e2":{"kind":"REWORK","_number":1,"created":"2026-09-21 15:21:12.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/73/1006573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/73/1006573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/1006573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/1006573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/1006573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/73/1006573/1"}}},"commit":{"parents":[{"commit":"fdd2784272df8c03e9e6c5176d62ad13b3d2b3a6","subject":"Merge \"Doc: update guest image build guide for noble\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fdd2784272df8c03e9e6c5176d62ad13b3d2b3a6"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 15:19:55.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 15:19:55.000000000","tz":300},"subject":"CI fix: theck that DIB_PYPI_MIRROR_URL is present","message":"CI fix: theck that DIB_PYPI_MIRROR_URL is present\n\nChange-Id: Idb12fb7b0cdf580be30df7658d01885c1814e3da\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5019815f46645272bf4a11552c1dcce83eb7c7e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5019815f46645272bf4a11552c1dcce83eb7c7e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
