)]}'
{"id":"openstack%2Ftrove~1006609","triplet_id":"openstack%2Ftrove~master~Ibf993b11d262ad8b7c8145a52812f62d04f95131","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"Ibf993b11d262ad8b7c8145a52812f62d04f95131","subject":"Separate CI-specific DIB elements","status":"MERGED","created":"2026-09-21 17:49:08.000000000","updated":"2026-09-21 19:53:52.000000000","submitted":"2026-09-21 19:52:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006609","meta_rev_id":"62e6a0e468c3085baefc55a9d5ec656b69796d64","_number":1006609,"virtual_id_number":1006609,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 19:52:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-21 19:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-21 19:15:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-21 19:15:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 18:05:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-21 19:10:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b630d21742f5e4e18f19ed7f995c564d74cedde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 17:49:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36a834465560916deb3c3b648d366380b817a39e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 18:05:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/744bbe719a804579aa56fe4dbd1872f3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b23d67554a324ecbb9652989b81b9aff : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5a7d840aa3ea4da9ad260dede8d3a14c : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/74ffa7a646c84181acb7826aca6cb1e6 : SUCCESS in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e19738562dcb609037a0ee0ccd441de8e1286d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:10:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d8110e8ecd840118ee036668c7340bb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f55feb844754946a92960db3eae6c08 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/886ee6701185454a89711ff676ba4753 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ea7c101935e492ea3435ca656cdb627 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0808f7dc96545b383d5d047f99c2bae : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/abc8e2d45e21470a8f80bbc643dab996 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3290fccdbb4fc3b7acf22a052b0ac5 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4fc53fc37d74ae09ba9dfd3ba80e009 : SUCCESS in 4m 56s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/c6e858f4d9454098b8b2bd3a077c2c65 : SUCCESS in 33m 17s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/5f41de77992b4df68596252c658f3a3c : SUCCESS in 1h 19m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/90b13ce46cf943a7a6ddacbf95ff4a0d : SUCCESS in 56m 48s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/3cc8f3d9d0ba44e2ad05c8729da7d616 : SUCCESS in 49m 21s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/bae36e1c5f6e4125aaf77de862cdf9ca : SUCCESS in 1h 06m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/ef42399303e04394a3911c02732b3b36 : SUCCESS in 1h 14m 24s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/34effc8dddfb45858949d4a7bd7a689a : SUCCESS in 32m 06s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/20fafb137e0a4c9083a8e9abb39c3bf0 : SUCCESS in 1h 10m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/4bd3b46b122d482ca44a452e11eee042 : SUCCESS in 52m 52s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/b5ff6776aff94cff979d67658d189a38 : SUCCESS in 51m 00s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/17263e02575f4830a76b94128fbf8075 : SUCCESS in 1h 04m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/436c9db7c98942de952ce9016dae0cb7 : SUCCESS in 1h 13m 04s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/95c0029c0f824cedba71b0e6b77f0363 : SUCCESS in 37m 56s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/19fc4059edf4418f86113dbd772c33df : SUCCESS in 1h 08m 09s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/abc31f93b2b648ee8d90d79486fe9cbf : SUCCESS in 39m 41s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/b06a9953c33a4d20880d7571b27e1317 : SUCCESS in 1h 00m 36s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/1603f536488745cf8fea17ef492ad100 : SUCCESS in 58m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/7f781fdb83e14942a9c4838dcab3ea05 : SUCCESS in 59m 52s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b70f95541f84cda92e82cb0e02c10bd : SUCCESS in 5m 22s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/6f6fadd7ac364b28a50bec1dda84cb96 : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8d55c97698fdb7a5386de8f27b6a726ff5c1beaf","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-21 19:15:56.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9851f1da2bbed70128a22c75c6372dd6f2b1bbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:17:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"666c2d1ae02fc327feedf7def140aaffc623a434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:52:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/935a2bffd5d249fdbfa62ba0894cc037\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13e8ef5ba5c74eedbc56c30acdd4af19 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4668829ccc374ee7b95fdb5ed1f0512c : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89f02f9c02294088a59780bc68192c3a : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fef8a565c214e469fe22dc82297c09d : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5c25ce6b6304144b1adbb69de47c512 : SUCCESS in 8m 01s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/bdbd1388e2bc4539ba626228d13b7133 : SUCCESS in 32m 22s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/82f018f9690944529b33e924d11e805d : SUCCESS in 30m 23s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/9f61ef1b8a1845638d45e3f83748787e : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b534d0b825c8886ec814ff689d14dbe29cfd1afb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:52:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"62e6a0e468c3085baefc55a9d5ec656b69796d64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 19:53:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2189c9e946d64965ae3da4d601efcd6a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5438e7ed06741baa1e82eb150dee927 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e42378953a48a536bba73ba27fea3460fbbc456","revisions":{"3e42378953a48a536bba73ba27fea3460fbbc456":{"kind":"REWORK","_number":1,"created":"2026-09-21 17:49:08.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/09/1006609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/1006609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/1006609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/1006609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/1006609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/1006609/1"}}},"commit":{"parents":[{"commit":"fdd2784272df8c03e9e6c5176d62ad13b3d2b3a6","subject":"Merge \"Doc: update guest image build guide for noble\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fdd2784272df8c03e9e6c5176d62ad13b3d2b3a6"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 17:45:34.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-21 17:45:34.000000000","tz":300},"subject":"Separate CI-specific DIB elements","message":"Separate CI-specific DIB elements\n\nMove the openstack-ci-mirrors and pypi DIB elements out of the\ndefault guest image build.\n\nThese elements were added to improve CI build stability, but they are\nnot required by the guest image itself. Configure them explicitly in\nZuul jobs through TROVE_DIB_EXTRA_ELEMENTS instead.\n\nThis keeps the CI-specific configuration while avoiding CI-specific\ndependencies in other guest image builds.\n\nChange-Id: Ibf993b11d262ad8b7c8145a52812f62d04f95131\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3e42378953a48a536bba73ba27fea3460fbbc456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3e42378953a48a536bba73ba27fea3460fbbc456"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
