)]}'
{"id":"openstack%2Ftrove~1006657","triplet_id":"openstack%2Ftrove~stable%2F2026.1~Ib0b57e9fd8067d52dac6e551130be3f11e75e97c","project":"openstack/trove","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-22 06:29:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0b57e9fd8067d52dac6e551130be3f11e75e97c","subject":"Use MySQL default ROW binlog format instead of MIXED","status":"MERGED","created":"2026-09-21 21:25:37.000000000","updated":"2026-09-22 06:30:40.000000000","submitted":"2026-09-22 06:29:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006657","cherry_pick_of_change":1006655,"cherry_pick_of_patch_set":1,"meta_rev_id":"8affdd449804a9f2c523f388b40caff81fc53d61","_number":1006657,"virtual_id_number":1006657,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 06:29:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-22 05:22:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-22 05:22:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 21:25:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-09-21 23:29:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-22 05:22:32.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"675315180051a35c288fbb08d17acad8da8e9987","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-21 21:25:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.2.","accounts_in_message":[],"_revision_number":1},{"id":"f562bb0997c55c378e32f1125761870c6cb8d02a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 23:29:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc3612ffffb4986ba20d88bf2557d14\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fb10f5cfe5a40eb8ff45092da47105e : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/670caadecb2d4f699fbad7d86ae8fd52 : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/384036c6f1de4f11a86b32ef12610e46 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd284fae512b4ad8948c45039f8b9e4f : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/965c61daa8ef420685eaf6eb09932771 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4781db57c08424a915bb41fda4d909f : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d59a35cc1e5467b9874af7645fc5583 : SUCCESS in 8m 42s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/27523a166cab4c88a44d45178bca65af : SUCCESS in 1h 07m 58s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/e0b0d9f1cd8f4cb18253a51884782565 : SUCCESS in 1h 47m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/e9aa04274f4b43628b15102b7b92cd13 : SUCCESS in 1h 29m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/c0dcfaa993714057b5e3328df1c4def9 : SUCCESS in 2h 02m 28s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/3e31375583e149a8a5bc9e9c6cf50aa8 : SUCCESS in 49m 20s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/fdecaff8edf54a0d8e97c22d177ded8b : SUCCESS in 1h 37m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b0b29762a79b411a9ed6923f9047b4c7 : SUCCESS in 1h 18m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0e50601792ba4f80912105203b075e73 : SUCCESS in 1h 35m 34s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/a4f1258412bf454193664d58b9f5a036 : SUCCESS in 50m 41s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/e41283452fb6434d94f359ca11c91bd3 : FAILURE in 50m 27s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/64164b2797364edab88483d84eb94b25 : SUCCESS in 1h 16m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/91bf315329ad4e15b0fb272d2f3109d7 : SUCCESS in 1h 37m 51s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/27712a0dad5c4f7690c3fd7de8bf65c4 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"37e6411ec6830428402370d90ca5c7fbe1c6c76b","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-22 05:22:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f99ef590e158a5b38d0284f87416a09f2a8356cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 05:23:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a448d6640811a2257b37a03f2ea5cac1c11e57c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 06:29:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a96b53c102f4466a2a0df7e787d0724\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb57fbb8843246f6815cb6da0a8364fb : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/177c860354da4bc7a6b0a00a2343d4f0 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75d2f82260bb4f168261e2247f86414f : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36032b6500e44972acc66849e84d0b8e : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b532f60327f34b7cae27629019d88853 : SUCCESS in 7m 50s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/0908624d772f42009bbc422cc5f6c5fc : SUCCESS in 1h 05m 27s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/63e5e6ff79cc4cb7b3bc95c18664c8d7 : SUCCESS in 50m 19s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/c62303834e7342ac9e75b57e99f95c97 : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":1},{"id":"640d2cceb02aa7cb20d67eb8816e068d8596940a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 06:29:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8affdd449804a9f2c523f388b40caff81fc53d61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 06:30:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aa5e1d2a168401fba599004db42cb06\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f33809a8987426ca09e54e5f83f037d : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdd9fcaf0dd742bebcf1baeb23c37c8f : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2969a35c1cc718249d65ca9e518c11d482ee57bf","revisions":{"2969a35c1cc718249d65ca9e518c11d482ee57bf":{"kind":"REWORK","_number":1,"created":"2026-09-21 21:25:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/57/1006657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/1006657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/1006657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/1006657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/1006657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/1006657/1"}}},"commit":{"parents":[{"commit":"8ca365ba0e32ef3b2be91fd120ed6f5f3f771874","subject":"Fix \u0027swfit\u0027 typo in backup storage driver default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8ca365ba0e32ef3b2be91fd120ed6f5f3f771874"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-19 05:14:20.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-21 21:25:37.000000000","tz":0},"subject":"Use MySQL default ROW binlog format instead of MIXED","message":"Use MySQL default ROW binlog format instead of MIXED\n\nTrove currently explicitly overrides MySQL\u0027s ROW default with\nMIXED when replication is enabled. The MIXED setting was\nintroduced in 2015 for the MySQL 5.6 replication implementation,\nbefore MySQL changed its default to ROW in 5.7.7. Modern MySQL\ndocumentation uses ROW as the default.\n\nChange-Id: Ib0b57e9fd8067d52dac6e551130be3f11e75e97c\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n(cherry picked from commit cfcb0b34f85572d895d687555456be024e88b84d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2969a35c1cc718249d65ca9e518c11d482ee57bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2969a35c1cc718249d65ca9e518c11d482ee57bf"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"b985b018d5ae309eaee8fae91b79c5c169326c52","ours":"8ca365ba0e32ef3b2be91fd120ed6f5f3f771874","theirs":"951c6f41eba113d3987f710bb81e2843c292f329","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
