)]}'
{"id":"openstack%2Ftrove~1006686","triplet_id":"openstack%2Ftrove~master~Id3b1666cbc7152df238c56077854ca7e0024fb97","project":"openstack/trove","branch":"master","topic":"database-removal-error","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-23 12:18:53.000000000","reason":"Reviewer/Cc was removed"},"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-24 04:23:12.000000000","reason":"\u003cGERRIT_ACCOUNT_39300\u003e replied on the change","reason_account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}}},"hashtags":[],"change_id":"Id3b1666cbc7152df238c56077854ca7e0024fb97","subject":"Fix case-sensitive MySQL database removal","status":"NEW","created":"2026-09-22 07:03:28.000000000","updated":"2026-09-24 05:54:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97d9befa834792b167103b9c0cd726ea731ca69f","_number":1006686,"virtual_id_number":1006686,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-24 05:54:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 07:03:28.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-09-22 07:20:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-22 08:05:50.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"},{"updated":"2026-09-22 09:07:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 12:18:53.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"}],"messages":[{"id":"20b092eeaa583b1b5fb19b8cf726452eefa84ee9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-22 07:03:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"185f2edb48da8cdc09321b1a9702bd2ddda1291d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 07:20:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd4267a491554c6d963c964343d68742\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e4caab4ef7104c12a4a14075ad91311d : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4f88498d81940a5ab5ad76334200e0a : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b307da72d9a949a5880f22807a12e00d : SUCCESS in 15m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4f94261a7f65a1d2b4e5bef8d8db43003b9e312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:07:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0381978d6d164019bc5a41e760c56ddc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8180680cbbd845979575bed4d3c3d29a : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4d3122db184ea89ffdcbcfa85a266e : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/226b5e12e466412b8ce5cf6c6253a479 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8c08b3c834243b1a44a054b622224d4 : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aba6ad9370b64c28aba9924016c5fa13 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e001510b444858939602b34c8e2615 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c576a431c25482b8b36873a840658d3 : SUCCESS in 5m 45s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/fc328705ccb44bf3bae6cce247a288a6 : SUCCESS in 34m 06s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/cb8843603b6c44059b810f43c459d7f0 : SUCCESS in 1h 16m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/5b1d2efc44a64967a4f3457efe7c34f4 : SUCCESS in 1h 34m 23s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/734cbfc7f3894153871176055ccb6935 : SUCCESS in 51m 12s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/787f3310c7c64b34bb9ce0d7143f6c65 : FAILURE in 52m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/24c8efbda4d54d3694739ee6aa7fd7d5 : SUCCESS in 2h 01m 48s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/dd9b2e0f0920434ba1ade8015157a7e3 : SUCCESS in 33m 57s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/dc9d6cff543d4955a64c185ba6c288f4 : SUCCESS in 1h 07m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/afddcfc6cb0b4c0f92eabec91c102c31 : SUCCESS in 54m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/4e2d23f702dc4f8481d1ccf03b2db2c3 : SUCCESS in 1h 05m 51s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/def0f995dcfd4fb288209636ffff89ed : SUCCESS in 35m 00s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/d7dd506157394c898774109795d784a6 : FAILURE in 1h 16m 44s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/6774feca61a143bab7038300857cbd79 : FAILURE in 1h 25m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/7159f35c8d6e4a94a7b1cec0374fc2f5 : SUCCESS in 56m 33s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/784e870260714c3a84d2399d6059e698 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f8803a3133314256c65fe5c7855efa86ab355305","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-23 12:18:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b84c46f4d011f81d1496b98a9b8e8eaf6a0c1873","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 12:37:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e50aa50420fa413fa1fff9bb1801bdc3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/31005dbbdd90437d89457e91a630df5f : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/77347cd82e6f41628fc76e2ab24dfc91 : SUCCESS in 14m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b8593d9888d8b5d60a720507e0ae917487aa3fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:41:44.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9f0f0becd434837a336891c182d8a93\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f17104303c94e2db7dd36d79c2bbbcd : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6a4a81c1ab4c2d9bd996ba1efe24f5 : FAILURE in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c57007d8126743df9b329735ccbf31ac : SUCCESS in 5m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3907e2837bb349598b921f061926af55 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977d6234118741e48a274ecc79e2f97f : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0944d0c5b0274464ab3b347010d95917 : SUCCESS in 8m 05s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/5a06cac7794b4ff09200ce3f0d75846d : SUCCESS in 41m 05s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/5e1ee6e4629e4708ac0bc42275f7c7b7 : SUCCESS in 1h 18m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/d2dfc182b1b24e9ba939dc1336c154d4 : SUCCESS in 1h 01m 33s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/462481b9509c475797635cecd168904f : SUCCESS in 57m 02s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/d6afe3d3cb8c4e5d97684541b0aa9296 : SUCCESS in 1h 04m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/d9890b2dafc74ec79e68f1d80dc05415 : SUCCESS in 1h 14m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/818854eacac04808a835ba0e4cdd5b9b : SUCCESS in 32m 34s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/0773a5a89413433f89bf087c8f5b00ae : SUCCESS in 1h 08m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/e630fbbd422f4a5294a264a6f30eb682 : SUCCESS in 55m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/50bf6fa25ee740209fce068e7f20e929 : FAILURE in 22m 33s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8c989d13a2984fc49305add95a6e33a5 : SUCCESS in 41m 41s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/0c546a4765d84725a3bcdd660873df52 : SUCCESS in 1h 02m 41s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/540d07079d5e4098b75bc1614dd183d1 : SUCCESS in 55m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/e7ee3c1b8727431c9fd69821fcf64a7d : SUCCESS in 55m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/769d5727509242c0acac2bb761b38743 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"cb598e4cc347ae5ff6e515300ee352f5cb5a0e5b","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-24 04:23:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5685f0d66c7f02219aa5869a806fde3a9b649fbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-24 04:38:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"67e32a6d103799915422c0ebe53ad77595ea74be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 04:58:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f16420a38954c6daff0350c6437398b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dd00915e560345c2b70d98a19b3125c4 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/92367e410fa448acb5381def13ca7524 : SUCCESS in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"97d9befa834792b167103b9c0cd726ea731ca69f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 05:54:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a37195b6df94c7b8daf02c2dbc17967\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0acdcb40abc94f45afb8c7931e3ed024 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2377f3aad8b449b2bbd3b93466f78232 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4043799699ae4e14b27f7f8cb9d3baba : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0a7b286f56d4550bb73874d62acca2d : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef0c2e6960c4ee48dda0b8e3249886f : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7d2dcc4a16846b19fceec14a826bf27 : SUCCESS in 7m 47s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/b5350e850aef44b788bcd1b8f16f1309 : SUCCESS in 34m 23s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a70a8b39d92b4ce788f670bca1c337a0 : SUCCESS in 1h 15m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/14fd9824e2b341af8e536794d0a0f7e2 : SUCCESS in 1h 03m 14s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/a3f27c370c00462c9c8187c13e409278 : SUCCESS in 50m 41s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/55f4b4951e4d46f09ab1e0ae90b6e034 : SUCCESS in 1h 09m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/40a9977b0f1e4cf9bbd8464cd943daf2 : SUCCESS in 1h 11m 11s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/c465077500da412abe19f542b2fe5f1a : SUCCESS in 33m 10s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/1cb41ec73e1f46d49138ef974da4f88e : SUCCESS in 1h 11m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/f14dcc167c66433d98ccf8f750757682 : SUCCESS in 59m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/89ea38fc90e846f29c27d558440f8aca : SUCCESS in 1h 05m 26s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/59457cee46a64a4f924cbe726c2939f5 : SUCCESS in 36m 43s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/2f606963a43d454684770b1100aeea41 : SUCCESS in 1h 06m 32s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/1130ec824bd44369856ffba0dda0b305 : SUCCESS in 51m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/816e98f580e5410ba052dc97fde1f787 : SUCCESS in 55m 26s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7197aa9073d54216a71cfc1f30aa3faa : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cc6cb297981a2c755828ae875216ff0ecbfc1ab7","revisions":{"851679a2af5522ab5bed75cc2bac8ce7a2077ecb":{"kind":"REWORK","_number":1,"created":"2026-09-22 07:03:28.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/86/1006686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/1006686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/1006686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/1006686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/1006686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/1006686/1"}}},"commit":{"parents":[{"commit":"60b8ff54ed0b564c011305b9b93bb83a65c14a06","subject":"Merge \"Separate CI-specific DIB elements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60b8ff54ed0b564c011305b9b93bb83a65c14a06"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-22 06:59:42.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-22 06:59:42.000000000","tz":300},"subject":"Fix case-sensitive MySQL database removal","message":"Fix case-sensitive MySQL database removal\n\nUse exact matching when checking whether a MySQL or MariaDB database\nexists. Preserve database name case when associating schema privileges\nwith users so that databases whose names differ only by letter case\nremain distinct.\n\nCloses-Bug: #2163657\nChange-Id: Id3b1666cbc7152df238c56077854ca7e0024fb97\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/851679a2af5522ab5bed75cc2bac8ce7a2077ecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/851679a2af5522ab5bed75cc2bac8ce7a2077ecb"}]},"branch":"refs/heads/master"},"e5822ae44882502118f10122a0ee23e321c3dcc2":{"kind":"REWORK","_number":2,"created":"2026-09-23 12:18:53.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/86/1006686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/1006686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/1006686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/1006686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/1006686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/1006686/2"}}},"commit":{"parents":[{"commit":"403555803e33980c714965db09a5fd8579fc8c7e","subject":"Merge \"Fix ignored database validation per datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/403555803e33980c714965db09a5fd8579fc8c7e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-22 06:59:42.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-23 12:15:34.000000000","tz":300},"subject":"Fix case-sensitive MySQL database removal","message":"Fix case-sensitive MySQL database removal\n\nUse exact matching when checking whether a MySQL or MariaDB database\nexists. Preserve database name case when associating schema privileges\nwith users so that databases whose names differ only by letter case\nremain distinct.\n\nCloses-Bug: #2163657\nChange-Id: Id3b1666cbc7152df238c56077854ca7e0024fb97\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e5822ae44882502118f10122a0ee23e321c3dcc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e5822ae44882502118f10122a0ee23e321c3dcc2"}]},"branch":"refs/heads/master"},"cc6cb297981a2c755828ae875216ff0ecbfc1ab7":{"kind":"REWORK","_number":3,"created":"2026-09-24 04:38:31.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/86/1006686/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/1006686/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/1006686/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/1006686/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/1006686/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/1006686/3"}}},"commit":{"parents":[{"commit":"403555803e33980c714965db09a5fd8579fc8c7e","subject":"Merge \"Fix ignored database validation per datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/403555803e33980c714965db09a5fd8579fc8c7e"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-22 06:59:42.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-24 04:38:02.000000000","tz":300},"subject":"Fix case-sensitive MySQL database removal","message":"Fix case-sensitive MySQL database removal\n\nUse exact matching when checking whether a MySQL or MariaDB database\nexists. Preserve database name case when associating schema privileges\nwith users so that databases whose names differ only by letter case\nremain distinct.\n\nCloses-Bug: #2163657\nChange-Id: Id3b1666cbc7152df238c56077854ca7e0024fb97\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc6cb297981a2c755828ae875216ff0ecbfc1ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc6cb297981a2c755828ae875216ff0ecbfc1ab7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
