)]}'
{"id":"openstack%2Ftrove~1006828","triplet_id":"openstack%2Ftrove~stable%2F2026.2~I21385586c06a73d61d750a3840381c4cdfd66e0d","project":"openstack/trove","branch":"stable/2026.2","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-25 06:38:02.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-25 07:21:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I21385586c06a73d61d750a3840381c4cdfd66e0d","subject":"Use high timeout for database shutdown","status":"MERGED","created":"2026-09-22 21:17:17.000000000","updated":"2026-09-25 07:23:15.000000000","submitted":"2026-09-25 07:21:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006828","cherry_pick_of_change":1003629,"cherry_pick_of_patch_set":2,"meta_rev_id":"60df7e8114cefbf211a7f00b0dbd29bf172dfa7f","_number":1006828,"virtual_id_number":1006828,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 07:21:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-25 06:38:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-25 06:38:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"CC":[{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 21:17:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"CC"},{"updated":"2026-09-22 22:36:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-22 23:44:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"4781d3eda25063f825b450c08fed3e39b2a2200f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-22 21:17:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0cbc63215b8def35e59b905195cd1ca1166d70a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 22:36:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f572cdc5007044dd8896172e5dfd5975\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d335afad88b4795bef5898856ae394c : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08069b4cce2f4acfb440887d7d4a4c2c : SUCCESS in 7m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2241b74986e14ee19cc0daa776a2256e : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31ad6a76a4994646b96ccb0bcdfa6e48 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85088f42563b47a6aa9f1b6a418c8bd8 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c89d94bc434446f8fc4cbf2d718aab1 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0548b59c9c404a81a41e29712f9c5a99 : SUCCESS in 3m 57s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/14bcc89a576142859873f16ee06b3a76 : SUCCESS in 45m 00s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/9cfea3a1e75e4485be767690fbe369c2 : SUCCESS in 1h 11m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/911611b198c444dfbaf8d6dc2765e8f9 : SUCCESS in 1h 05m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/0bcfc268d22f44dc89d23bc5cb684e6c : SUCCESS in 1h 17m 03s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/2e0d739db56b420eb590ae1d059fc4c4 : SUCCESS in 43m 44s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/f7b71d7185fa4a10b0c6ed9d1b437cdf : SUCCESS in 1h 12m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/0f579d81bb0e4a78a472374154db42a9 : SUCCESS in 1h 02m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/1dfc578c25ab436790089562590ac75b : SUCCESS in 1h 14m 59s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/9a7a1ce82a994557b0c69a6cd011e5d6 : SUCCESS in 44m 49s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/35fd7065c64842f4abd2d8223388c505 : SUCCESS in 1h 08m 06s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/4d4e514f766f45da84b74d0e56720b1c : SUCCESS in 56m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/68db309084c4451c8e0a94d090a423c8 : SUCCESS in 1h 04m 18s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7e7c4707b09045e4a8cae8a656efbaed : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fd8bb3c1d61a915fee665004c07ad86848fdfef2","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-25 06:38:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18a1f82451a08ad443f108917146eba905ddd2b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 06:38:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95f419d0b14af1cc1e7719aa05a8e7301e1ea190","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:21:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97af560f308a43ba9bdd816b22a797b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d2d4ebc21b24afb9062d10cf78c72c5 : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab7dfd733d954b2b8b57434468eec527 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/771f564e72574683848deefe3c7cfd81 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e698e2ea42f4e73b7c23037732e5457 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b45bc0805aaa455caddd843437b7067d : SUCCESS in 7m 22s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/99a1d6c5d87c4d05b9975e759aae8a73 : SUCCESS in 39m 16s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/fc818f31df0f48519212159efdc5b95c : SUCCESS in 39m 49s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/f120864a28e84928a519918c08c77887 : SUCCESS in 41m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0ad764346d797551cdafd8a6622f99faed85576c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:21:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"60df7e8114cefbf211a7f00b0dbd29bf172dfa7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:23:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daad845b4b5c4a2da1e3a5a121f519b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f83414f2694a588216a491dc41c4c4 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd897a3ffbf643278cb1232f40057db8 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"475beb04512650bf146d8bb3b965128e43cfb175","revisions":{"475beb04512650bf146d8bb3b965128e43cfb175":{"kind":"REWORK","_number":1,"created":"2026-09-22 21:17:17.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/1006828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/28/1006828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/1006828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/1006828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/1006828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/28/1006828/1"}}},"commit":{"parents":[{"commit":"2623e8fb84280794bd211a4f17c4408825bc388b","subject":"Merge \"Use MySQL default ROW binlog format instead of MIXED\" into stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2623e8fb84280794bd211a4f17c4408825bc388b"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-03 06:20:48.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-22 21:17:17.000000000","tz":0},"subject":"Use high timeout for database shutdown","message":"Use high timeout for database shutdown\n\nStopping a database under load can take longer than\nagent_call_low_timeout. This causes flavor resize to fail before the\nguest agent completes the shutdown.\n\nUse agent_call_high_timeout for stop_db requests without changing the\nconfigured timeout value.\n\nChange-Id: I21385586c06a73d61d750a3840381c4cdfd66e0d\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n(cherry picked from commit de3fb5739fe13341adc7ae0446422172ac82b7ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/475beb04512650bf146d8bb3b965128e43cfb175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/475beb04512650bf146d8bb3b965128e43cfb175"}]},"branch":"refs/heads/stable/2026.2","conflicts":{"base":"435f705fbbee993cb740a05c56b519721d90f757","ours":"2623e8fb84280794bd211a4f17c4408825bc388b","theirs":"de3fb5739fe13341adc7ae0446422172ac82b7ed","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
