)]}'
{"id":"openstack%2Ftrove~1006830","triplet_id":"openstack%2Ftrove~master~I8bdb195ba8f6cb967408ffd2ad1f38625c4f7441","project":"openstack/trove","branch":"master","attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-22 22:31:58.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8bdb195ba8f6cb967408ffd2ad1f38625c4f7441","subject":"Fix python shebang","status":"NEW","created":"2026-09-22 21:25:22.000000000","updated":"2026-09-22 23:49:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a15c34125a34d5d179fdeae18df327684a065446","_number":1006830,"virtual_id_number":1006830,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-22 23:49:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 22:31:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"},{"updated":"2026-09-22 22:59:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-22 23:49:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d78ad3fabaea3e4ca5bcfd89c51711ca566fc0b8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-22 21:25:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aec63bb9462da089ebf9bd15f3cc62dd197d9262","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-22 21:32:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"91200b030d730f418a095772c4e4139b3c6127cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-22 21:32:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"1906aaac30e09518858da45e11e7ee60b74570ff","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-22 22:30:21.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"79e0753dbfb23a475349874e4248179dbc334c88","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-22 22:31:53.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"243d9aafe97c7a1e7c94b1f9ecea520d9eb599a9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-22 22:31:58.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"21d62a213b2f36616e612108e15bcf7f33d1751a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 22:59:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed57db2c8e44a4cafe74879c261ae9a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c97bedbb85864f84995ba56ded863181 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5d0f9c53d56d4508896717b711d882ec : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a15c34125a34d5d179fdeae18df327684a065446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 23:49:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38e1a48e23fe487194efb372acd83127\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5fc597c2c934bc89401d42b8e8f23e9 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aff27d87d1fd4c3ab516e5bfbc6a0254 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/974cc6629d8d4ef5b4765f3e1e1173c6 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8805840117cd4f7cb845847becbb8b5b : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f06bc935ea43f39d868bcdee3d7935 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6de804fb73e84c038be97345a890f6c5 : SUCCESS in 7m 45s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/6c28080ac9034770b4215a7f12ae771b : SUCCESS in 38m 09s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/3cd8bfc387104120973de9d0e38977b3 : SUCCESS in 1h 14m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/70c9088d0ae4441d9b47c608024a1c58 : SUCCESS in 1h 05m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/d1fed365590f40aba49a9f65f9ebf28f : SUCCESS in 1h 13m 40s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/cc156b6fd42e4991bd8e15e79a601e62 : SUCCESS in 36m 55s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/7a6631dac2d8469a83e81e870979cd44 : SUCCESS in 1h 11m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/b3b5b1fdf3af4a7cabff3c35e4f8a62d : SUCCESS in 1h 00m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/8a8c19e6b6d641fe9336fb4b9d37d659 : SUCCESS in 1h 09m 58s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/0ebd2d6eb40a442dbc03330116e02f7a : SUCCESS in 38m 30s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/f2822f1439904872ac7fa6b7fc0035be : SUCCESS in 1h 11m 03s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/a3e6fa87790c40298ecea9bf8270a743 : SUCCESS in 58m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/35987720794d4e40857eb9da63ab34eb : SUCCESS in 58m 13s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bb6063d597a54cb4833bd865d3f491ca : SUCCESS in 5m 52s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/794debad337349b68ca3b2d9dd274bc3 : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"40206d64b83ba050b89a72eab35a1d8775990b71","revisions":{"4389bfd72fe4801dc14bedf6c7ea9714c4dd0155":{"kind":"REWORK","_number":1,"created":"2026-09-22 21:25:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/30/1006830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/1006830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/1006830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/1006830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/1006830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/1006830/1"}}},"commit":{"parents":[{"commit":"60b8ff54ed0b564c011305b9b93bb83a65c14a06","subject":"Merge \"Separate CI-specific DIB elements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60b8ff54ed0b564c011305b9b93bb83a65c14a06"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-22 21:25:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-22 21:25:22.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I8bdb195ba8f6cb967408ffd2ad1f38625c4f7441\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4389bfd72fe4801dc14bedf6c7ea9714c4dd0155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4389bfd72fe4801dc14bedf6c7ea9714c4dd0155"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"34ca3f2b729f0e8a27496b339561b677b455a81a":{"kind":"REWORK","_number":2,"created":"2026-09-22 21:32:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/30/1006830/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/1006830/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/1006830/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/1006830/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/1006830/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/1006830/2"}}},"commit":{"parents":[{"commit":"60b8ff54ed0b564c011305b9b93bb83a65c14a06","subject":"Merge \"Separate CI-specific DIB elements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60b8ff54ed0b564c011305b9b93bb83a65c14a06"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-22 21:25:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-22 21:30:54.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I8bdb195ba8f6cb967408ffd2ad1f38625c4f7441\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/34ca3f2b729f0e8a27496b339561b677b455a81a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/34ca3f2b729f0e8a27496b339561b677b455a81a"}]},"branch":"refs/heads/master"},"40206d64b83ba050b89a72eab35a1d8775990b71":{"kind":"REWORK","_number":3,"created":"2026-09-22 22:31:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/30/1006830/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/1006830/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/1006830/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/1006830/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/1006830/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/1006830/3"}}},"commit":{"parents":[{"commit":"60b8ff54ed0b564c011305b9b93bb83a65c14a06","subject":"Merge \"Separate CI-specific DIB elements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60b8ff54ed0b564c011305b9b93bb83a65c14a06"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-22 21:25:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-22 22:31:53.000000000","tz":0},"subject":"Fix python shebang","message":"Fix python shebang\n\nChange-Id: I8bdb195ba8f6cb967408ffd2ad1f38625c4f7441\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40206d64b83ba050b89a72eab35a1d8775990b71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40206d64b83ba050b89a72eab35a1d8775990b71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
