)]}'
{"id":"openstack%2Ftrove~1006866","triplet_id":"openstack%2Ftrove~stable%2F2026.2~Ice10b7b6aa60d4e687456c75262015ce7caf0452","project":"openstack/trove","branch":"stable/2026.2","topic":"fix-container-status-detection-stable/2026.2","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-25 06:40:23.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-25 10:06:36.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ice10b7b6aa60d4e687456c75262015ce7caf0452","subject":"Fix database status detection for stopped containers","status":"MERGED","created":"2026-09-22 23:51:24.000000000","updated":"2026-09-25 12:08:49.000000000","submitted":"2026-09-25 12:07:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006866-fix-container-status-detection-stable/2026.2","cherry_pick_of_change":1004378,"cherry_pick_of_patch_set":2,"meta_rev_id":"117451d98cdb46b371981981bfb667d226cf2397","_number":1006866,"virtual_id_number":1006866,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 12:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-25 06:40:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-25 06:40:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"CC":[{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 23:51:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"CC"},{"updated":"2026-09-22 23:51:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"},{"updated":"2026-09-24 04:42:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85ec8f973580c0a43bd8acb60afcfa5301e78dff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-22 23:51:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"698a9446ed69822f989dc68ed15a2cf7ec540412","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-24 02:57:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"037802b637343457ac4a1efeb6e996ed896fccff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 04:42:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7b44628ff4645ad9623e68b4b5c04ef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50d0f945d44a4abeb9525088bce40828 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d96aa190d64e298c5d49503f951f7f : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24fc806d117a45baadb95fd06f0b4d9f : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6cdab54668d646ff87eca9952b5c2d93 : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/39da489301d7461e9b5a75330d317a89 : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a0d8aaaad7d494cb88856b136786833 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7d9197b0ad0422887d444a1f08e8599 : SUCCESS in 4m 38s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/d0dececd186443fda3f7fb099859f6b2 : SUCCESS in 47m 43s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/f043cceab90547cd8eb854490361189b : SUCCESS in 1h 13m 52s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/5fca5ddc40814e9e8180a66580a5fa6b : SUCCESS in 1h 09m 05s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/a4956e945d974b058be32787fdb90ee9 : SUCCESS in 1h 03m 49s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/af526395018d43ee9ee627c2d1638b6d : SUCCESS in 1h 09m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/3cce8d3250d74659a8828bc2cf080fce : SUCCESS in 1h 18m 50s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/d28056e013e74fd38bad9ec521b3d611 : SUCCESS in 46m 33s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/59405811cdc34858912e61c18391a5d3 : SUCCESS in 1h 16m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/06d759fd4d204736a55c3dede401baeb : SUCCESS in 1h 07m 38s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/16c1ec123d534429ae15a4110a03fbf8 : SUCCESS in 1h 01m 01s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/db3ff89a310b43308729fc5e2ec04c8d : SUCCESS in 1h 08m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/6e6912339d454bb7bbe8dc77034451d6 : SUCCESS in 1h 20m 21s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/060ec591aaea4ae197623462d0779ddd : SUCCESS in 42m 49s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/7389d7475465499dbe09b8998ea947b0 : SUCCESS in 1h 09m 27s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/0605a9f5bf9843a09561eed9fde47412 : SUCCESS in 45m 32s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/4c615df86999413f98f3615c593e6f49 : SUCCESS in 59m 14s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/cdc4740a52874830a6c615611b7a2bf7 : SUCCESS in 1h 01m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/4d55ab1502a9457888c7f05c03c4905e : SUCCESS in 1h 04m 55s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6aa8d817095749b295a64a9e5ee408fe : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ec61b776b9c299f877da488750684d1d7e212eb7","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-25 06:40:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4db08ca555da0d59177b97f072f90cd9332790f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 06:41:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0838153709e36634b894dabf3a6d47dcc3a273db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:24:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62df8fd1c44c433eb1e112e5a3e9faea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1396fe6992804a538663c4b492bf2848 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1973eb7a0a624f408f461ea27d10cf67 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/662b4c1f3e1648d19a47186148428b0b : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17f51d05c8748e39db38ddad266b89f : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10e0f2fca67c4e7abdd30ddf1502b3f6 : SUCCESS in 8m 28s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/caaf3e0ae268460385cdf19e18ec6f19 : FAILURE in 37m 59s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/57032978ecfb4ffc8ba2ef0faaba6589 : SUCCESS in 39m 26s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/212510247fd04132848923772040ed12 : SUCCESS in 42m 52s","accounts_in_message":[],"_revision_number":1},{"id":"35c8aaf31b33a4315fa8b0017439817854f78fb1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-25 07:51:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6fc7a0925f895fb21d6e490d7ee5be951f5f3d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 09:12:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bafb3bc05d814f7183b5d13f370c9f7e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6ca11c592b04619bf886b3e50212a4c : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64ec56adb89241749346a782be3a2053 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/321322b450e149dcb60c45f29c034bef : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7ce1dba75d64825b21069bc65250b94 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b3fa3525332a43c19c17d60262bb76b8 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b764cd4c52684f37a91a30b2a3bc765a : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35a40e61798f498494a55d111e0fe22c : SUCCESS in 3m 56s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/b732cdb94ab746fdbce4eae129468dbb : SUCCESS in 45m 17s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/84fdb3dfc865449a8539e8d968cc1e4e : SUCCESS in 1h 11m 58s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/c9c0d67d4b0a4922a2e0fd80ce23a4d0 : SUCCESS in 1h 06m 39s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/8528f1305ff24b8d86a96c5ef7307d93 : SUCCESS in 1h 01m 13s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/fddcc427f5df40c99ce9c8487220ca84 : SUCCESS in 1h 10m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/954e1d4be68b409c9f6fa7932033cd50 : SUCCESS in 1h 19m 34s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/1064fe4a91f54b9f90185ab49b2b5127 : SUCCESS in 42m 47s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/815a4b0460c348b28f416ffa6c7ad819 : SUCCESS in 1h 15m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/7d87f5d9f7154d10814b39b0b38cdf65 : SUCCESS in 1h 04m 29s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/ce841288cfca4f4ab1cde1a289042b4c : SUCCESS in 56m 33s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/45d9c099d00946d782fe1c0cf77be92a : FAILURE in 38m 38s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/6423a8b645af45a9b1e14fd9e23aedf8 : SUCCESS in 1h 14m 41s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/f3aa6a5aa4e947b99f2eefe68abdcc0b : SUCCESS in 46m 01s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/71b04a2684f949bb984fa066793f6b0a : SUCCESS in 1h 10m 18s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/04dfaa72f4694a7b883624c312e6a130 : SUCCESS in 48m 02s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/d77c8c999fb44e77aff4ff5552c8150e : SUCCESS in 1h 01m 32s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/09332a5034dd48ecb8cc914ad18419f4 : SUCCESS in 1h 00m 20s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/4db0b2f5159747c8907b638facb22fb3 : SUCCESS in 1h 05m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c9c967d1c38146999dee22da5e5fdb20 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9dd85af9d9d3fcb418dfab9b2865609dec4b11b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 09:12:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b912622629903ced5ff0bfeffc4bbbf6c4ae103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 10:00:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48284087dc224c0490a0d0ae38a670fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec598b8020594e9dad5e621cddcde1fa : SUCCESS in 6m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac5d949db35d46d5adb16aa196b8e149 : SUCCESS in 6m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b55ee6a439647759a1ec3d71b95f4ce : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120423bd5d304e74bb8e5d69c34e10a1 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/040acfe11e5444d4bdc3230819d03f3e : SUCCESS in 4m 17s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/da65cee4076b4f6a850aa7bbc5ab35c6 : SUCCESS in 47m 07s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/557f942cd46f435b8647872339c25532 : SUCCESS in 43m 27s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/9ba455c5634d4ba58f48b4b2417c0848 : FAILURE in 38m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b00370880332032466426c63fe34331c3e37cf6d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-25 10:06:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2467365517076365e9e6aa4b49a4e983fc03b877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:23:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b96b95697b54418ba275b05e5700c978\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62afa70520f74e6eb8c5fd29c6928cce : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bcb8f5273a143e49d5a2c74bc59f627 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da9bf41eb24942c7a1e24d6ac0ebcf36 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76bfabd6bcd94b6d8fcabed5973f9331 : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a64311ef64f438b9c67c04f6ec45bcf : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/948ecc7c45794d8eb9d00ce001aebe98 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0161e255309046df9898210843d5133a : SUCCESS in 7m 26s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/cda7020937d441a79cf09766d5210754 : SUCCESS in 45m 32s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/ab88171c7d6e434e8e427d7ad4686a32 : SUCCESS in 1h 13m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/a7bdb3b86c5447c6bd39a6010836d185 : SUCCESS in 1h 08m 49s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/49fc5423535d4e8e95e117d0d0b4f773 : SUCCESS in 58m 09s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/fd4f1628079d4eae92a9539c3ce9dced : SUCCESS in 1h 11m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/e4e85d6e9f9948dda944b285809d9660 : SUCCESS in 1h 14m 29s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/fe6376bbc4cb4774911c43fae57e471a : SUCCESS in 40m 23s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/64a05057ca63420fbfa2b1507c298947 : SUCCESS in 1h 11m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/e382303362a042618922f84480b17195 : SUCCESS in 1h 06m 02s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/c2df9f2fd7a34c88a24f328cc3dd8dcb : SUCCESS in 57m 34s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/51ad318546864127a67a49a39ea3ceb9 : SUCCESS in 1h 05m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/22ac062057814554a68eff2655648d74 : SUCCESS in 1h 14m 46s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/4a4c076b28864d5391dfbdd33981f0bb : SUCCESS in 43m 05s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/9f2cfe275b91469dbccb3085fb490652 : SUCCESS in 1h 08m 57s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/449d958970b94138ae18e01e30344d26 : SUCCESS in 47m 44s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/e1fa32aa56c3442eb12e4372ddb67b1d : SUCCESS in 1h 04m 09s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/d86a263a2289445c9aa3373386399fc5 : SUCCESS in 57m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/39498fe65c844334b68d6a81c65aaefe : SUCCESS in 1h 02m 27s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/935edb6d888b45eb90e0285b585dbc65 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f8947b230159abbaea9f0e13f745cdb81caad945","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:23:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e926e5974b4c660978e1208e5a0f00dee90ec089","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 12:07:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dce8ac92918940e3bba447502b77ef3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9313b69f6b1d45ef85453ef53f4beb32 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68c268d0cc1f4e8196b49b9b2d666d2a : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a8ca6a87407486e99367c5c1947867f : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4b4648e4b44148a485ad9e2637551b : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cb50460d48e4ed99b2f2f1092b76097 : SUCCESS in 7m 55s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/42345e2cd40842f1b97bc7bf48a76612 : SUCCESS in 43m 18s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/b624fc5cda8d4793978cd7cf67c958c1 : SUCCESS in 41m 32s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/4042de0b31664fca8f0a97616af7cd5a : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"27d1e2804fdb85727108d0b3a19f8c455245998e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 12:07:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"117451d98cdb46b371981981bfb667d226cf2397","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 12:08:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/572a6f3f57ce46c9a2298b40a3d978fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844973709bfc4de69bb84b20be72a2de : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe51f134a32b4ce89e8262894fa64f6e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f71022baee39471c81da5b4f8f55d03ca000df42","revisions":{"f71022baee39471c81da5b4f8f55d03ca000df42":{"kind":"REWORK","_number":1,"created":"2026-09-22 23:51:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/66/1006866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/1006866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/1006866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/1006866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/1006866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/1006866/1"}}},"commit":{"parents":[{"commit":"2623e8fb84280794bd211a4f17c4408825bc388b","subject":"Merge \"Use MySQL default ROW binlog format instead of MIXED\" into stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2623e8fb84280794bd211a4f17c4408825bc388b"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-07 04:26:07.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-22 23:51:24.000000000","tz":0},"subject":"Fix database status detection for stopped containers","message":"Fix database status detection for stopped containers\n\nDocker can retain a healthy health status after a database container\nhas exited. Check the container status first so Trove correctly reports\nSHUTDOWN.\n\nChange-Id: Ice10b7b6aa60d4e687456c75262015ce7caf0452\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n(cherry picked from commit 8005a8d551c9078a082e22e06b637dbb0303b4a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f71022baee39471c81da5b4f8f55d03ca000df42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f71022baee39471c81da5b4f8f55d03ca000df42"}]},"branch":"refs/heads/stable/2026.2","conflicts":{"base":"435f705fbbee993cb740a05c56b519721d90f757","ours":"2623e8fb84280794bd211a4f17c4408825bc388b","theirs":"8005a8d551c9078a082e22e06b637dbb0303b4a9","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
