)]}'
{"id":"openstack%2Ftrove~1006869","triplet_id":"openstack%2Ftrove~stable%2F2026.1~Ic8b12fa6b86fdf3987a6e996ed7c8510a478d4a3","project":"openstack/trove","branch":"stable/2026.1","topic":"guest-info-after-upgrade-stable/2026.2-stable/2026.1","attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-24 02:56:44.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic8b12fa6b86fdf3987a6e996ed7c8510a478d4a3","subject":"Update guest info after datastore upgrade","status":"NEW","created":"2026-09-23 00:05:56.000000000","updated":"2026-09-24 02:56:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":1006867,"cherry_pick_of_patch_set":1,"meta_rev_id":"19c403de4d81ca6796a4d7b76b9381d8e919116d","_number":1006869,"virtual_id_number":1006869,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-23 05:17:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"CC":[{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 00:05:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"CC"},{"updated":"2026-09-23 02:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-24 02:56:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"fb8256010e89f8ef1dea42c2610d15d429398920","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-23 00:05:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.2.\n\nThe following files contain Git conflicts:\n* trove/tests/unittests/guestagent/datastore/postgres/test_manager.py","accounts_in_message":[],"_revision_number":1},{"id":"da66f849f02227ea54d3b619519abd11c8fde912","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-23 00:10:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a808eb3da22c7a9f9d22f44045055ed566803547","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 02:02:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df095dd480334115b4ec2c798c44307b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c08b5328867f42379fad15d5972663cc : RETRY_LIMIT in 1m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a40ab28d5c01436aa2f91c3c0a53ac8d : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fb4919513b64f79b48a4a210e3f2e5c : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e682332d5574e67a61eeaad3e54e7e4 : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/56a772d90f3341eaaef576fe696f9700 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c0587e4d8a42edbd4c08cba2bef21a : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e5c17b9701241c89a0dbbfda0fbccd8 : SUCCESS in 7m 54s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/f2ff953ed5f4431f98e505789355b91e : SUCCESS in 1h 04m 27s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/6a27bf1ed8f74b2dab4f5bd72e681866 : SUCCESS in 1h 39m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/e867b6e0406e46e6b0b250ed0349810e : SUCCESS in 1h 22m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/1b482d502ef44d4197a28b1c17829e0b : SUCCESS in 1h 38m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/33ce92678ff44c45b94521c3eae50515 : SUCCESS in 1h 00m 44s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/86aad17fabdb4d4d8792edba47f6e1e4 : FAILURE in 37m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1730c80825f24b0aa060b64689a9e0fd : SUCCESS in 1h 27m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/152c675cfc444f58b179e8673e6296f6 : SUCCESS in 1h 50m 05s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/b54468759b12446fb3dbddaa8607dd19 : SUCCESS in 48m 13s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/168c441511d247a4998f559627c5de05 : FAILURE in 37m 55s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/53f7b83360914853b4c9703b9a449d40 : SUCCESS in 1h 27m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/49ec0ff709074f7598d2e23c1e29cfce : SUCCESS in 1h 19m 06s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fe675bac3dd84c7f9b41777f56975cc6 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"bfea6589e6d527dc332bfbf3285336d505795bbb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-23 02:54:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d29f1a356c460d059b6cacbe2a1aee5b82781d44","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-23 02:55:09.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"0906e2d107b95e19d37c32e95d35e0be953b465c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 05:17:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2543aa58fde4645b6366bcf4720f119\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2623a48225d7450aa94a961cbe16b306 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88c87abf7c0048cab7ab9f37e562c441 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87c56aad6386476cbdf4bd8b02ecf487 : SUCCESS in 6m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dde81a28121d45fbb2adce75ea052875 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0b91fadd88434ddbade3c12c8af91822 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac1368335e149c5840d9d619e671dd4 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8a197d3241844a296b0246029e706bf : SUCCESS in 7m 34s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/80b57bb13de7491aa1cf47d4b06cd247 : SUCCESS in 52m 06s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/1b260d274b6c463ba486cfea25473616 : SUCCESS in 1h 46m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/210bca5c3dfa4057be9ee211ebf6fbef : FAILURE in 2h 00m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/85d8667e7ad74f6f8b7361284ba7cb6f : SUCCESS in 1h 43m 30s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/728bc0906c0d47aa9bd946cd6036af0a : SUCCESS in 46m 59s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2aaf66a0abd344209050a7ad6c101cad : SUCCESS in 1h 34m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/af536974b4524d2f8789512cccf8ddc9 : SUCCESS in 1h 18m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/eaf6d8ce5e1c420ba9c4b8f24b815db5 : FAILURE in 22m 03s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/cf186b908cb84ca9b9b064bcd9dc4f4c : SUCCESS in 1h 06m 56s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/e54c8a606a9541e4ab7532bf4c954c99 : SUCCESS in 1h 33m 59s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/8ba0d817395043af82ae10391b5781e2 : SUCCESS in 1h 17m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/cd6868ff8f7f4c9098220a77d94c0c15 : SUCCESS in 1h 21m 19s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fdd21d87e618472ca49ee1905bc5cb21 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3b68745038834996c23fe8c364dd015540d705e3","revisions":{"136ddb1bf1d1ec177d4c136185e4e100bccfe403":{"kind":"REWORK","_number":1,"created":"2026-09-23 00:05:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/1006869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/1006869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/1006869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/1006869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/1006869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/1006869/1"}}},"commit":{"parents":[{"commit":"fc7ce52e189eed36dd4e1d7116f9d75bb89bcaae","subject":"Merge \"Use MySQL default ROW binlog format instead of MIXED\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fc7ce52e189eed36dd4e1d7116f9d75bb89bcaae"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-06 11:25:50.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-23 00:05:56.000000000","tz":0},"subject":"Update guest info after datastore upgrade","message":"Update guest info after datastore upgrade\n\nUpdate the datastore_version value in guest_info.conf after a\nsuccessful database upgrade.\n\nChange-Id: Ic8b12fa6b86fdf3987a6e996ed7c8510a478d4a3\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n(cherry picked from commit 615d90ba90a21f6c9f2545926db42d6a776a12d4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/136ddb1bf1d1ec177d4c136185e4e100bccfe403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/136ddb1bf1d1ec177d4c136185e4e100bccfe403"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"2623e8fb84280794bd211a4f17c4408825bc388b","ours":"fc7ce52e189eed36dd4e1d7116f9d75bb89bcaae","theirs":"5dbc383fcabdfdb403bffccdabd4aa8682b69530","merge_strategy":"recursive","contains_conflicts":true}},"3b68745038834996c23fe8c364dd015540d705e3":{"kind":"REWORK","_number":2,"created":"2026-09-23 00:10:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/1006869/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/1006869/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/1006869/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/1006869/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/1006869/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/1006869/2"}}},"commit":{"parents":[{"commit":"fc7ce52e189eed36dd4e1d7116f9d75bb89bcaae","subject":"Merge \"Use MySQL default ROW binlog format instead of MIXED\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fc7ce52e189eed36dd4e1d7116f9d75bb89bcaae"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-08-06 11:25:50.000000000","tz":300},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-23 00:10:14.000000000","tz":0},"subject":"Update guest info after datastore upgrade","message":"Update guest info after datastore upgrade\n\nUpdate the datastore_version value in guest_info.conf after a\nsuccessful database upgrade.\n\nChange-Id: Ic8b12fa6b86fdf3987a6e996ed7c8510a478d4a3\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n(cherry picked from commit 615d90ba90a21f6c9f2545926db42d6a776a12d4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b68745038834996c23fe8c364dd015540d705e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b68745038834996c23fe8c364dd015540d705e3"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
