)]}'
{"id":"openstack%2Ftrove~1007827","triplet_id":"openstack%2Ftrove~master~I26d200ca462782765ca4a8f77bfa5a6d46a416ec","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-29 07:15:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I26d200ca462782765ca4a8f77bfa5a6d46a416ec","subject":"Pin pyOpenSSL version to 24.2.1.","status":"ABANDONED","created":"2026-09-29 06:34:11.000000000","updated":"2026-09-29 07:15:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"548405f6bfc89cced93266b348d3557404802850","_number":1007827,"virtual_id_number":1007827,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-29 06:56:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-29 06:41:52.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"CC"},{"updated":"2026-09-29 06:55:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-29 06:56:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76ceac23c65f2636afabbf5d87c885f4edea1df8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-29 06:34:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4409d7e4fc395c025740961dc5d1a32e1e4e17e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-29 06:39:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"40f3f44f0695ec8e2ab03e2e7bb86040fdacf388","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 06:55:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc693a7c4c214b15a9d2a0447980a06d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb1fd02742ec4c7eae93e6ba2f4c6707 : FAILURE in 4m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2938df29485d47a49e2872243605bac0 : FAILURE in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"157ce891637b1b91203f298afae7923d32476099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 06:56:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84702feb3c2f4c63898d3a6eab07f3cc\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c02ca9c218f849128e95fb6d6c9ea76b : FAILURE in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7235b03c484a4d31b00b1a4a54d50118 : FAILURE in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/018b940269a948a5b16d0dd69cf85635 : FAILURE in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7881f7ebfe9b46f1b3a171809beb0d39 : FAILURE in 2m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c1772225bbb464d9f86340d61558599 : FAILURE in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b37a1c8c3be427db7252a027e59b0e2 : FAILURE in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fa55af0fb7a49d38d6495c10d2d7158 : FAILURE in 2m 36s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/3285bca204354eb195b2ea0b8de654ec : FAILURE in 14m 58s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/520411f6f0b842d5851c36bd38b2263a : FAILURE in 15m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/da66523b4c184a0a82b5625a9e29413c : FAILURE in 15m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/85f3ab2b30b04fd4b7a85563d84ddc43 : FAILURE in 14m 52s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/c1673495025841c2987f699acee5808b : FAILURE in 15m 09s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/83743921c91d43b68109db4bb4d2155d : FAILURE in 15m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/0b089b6955b44a3f98ed86d9b947c824 : FAILURE in 15m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/c94b5e276b1147a49f8431c1debc0295 : FAILURE in 14m 47s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/71a16cd25d764470a31228b09036e3e0 : FAILURE in 15m 00s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/8de04ebe076d466facdef44bd309cabc : FAILURE in 15m 05s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/c9dbc2866ab64ac59f7976f9daad714b : FAILURE in 15m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/96a4525027cd4ebd9565c4a6858cc176 : FAILURE in 12m 33s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/fips/enable-fips.yaml for job trove-tempest-ubuntu-base-postgresql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/multinode/enable-multinode.yaml for job trove-tempest-ubuntu-base-postgresql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/pre.yaml for job trove-tempest-ubuntu-base-postgresql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/devstack-tempest.yaml for job trove-tempest-ubuntu-base-postgresql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/post-tempest.yaml for job trove-tempest-ubuntu-base-postgresql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/fips/enable-fips.yaml for job trove-tempest-ubuntu-base-mysql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/multinode/enable-multinode.yaml for job trove-tempest-ubuntu-base-mysql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/pre.yaml for job trove-tempest-ubuntu-base-mysql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/devstack-tempest.yaml for job trove-tempest-ubuntu-base-mysql.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/post-tempest.yaml for job trove-tempest-ubuntu-base-mysql.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"548405f6bfc89cced93266b348d3557404802850","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-29 07:15:12.000000000","message":"Abandoned\n\napproach wasn\u0027t helpful","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"525161c1274f55ac45230774bfb5c3391fb236af","revisions":{"7819ad2792b2967cdac9c84dcca9ee746b2c1da3":{"kind":"REWORK","_number":1,"created":"2026-09-29 06:34:11.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/27/1007827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/1007827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/1007827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/1007827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/1007827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/1007827/1"}}},"commit":{"parents":[{"commit":"9fb744390e479737535e4728d37e7b749861cc79","subject":"Merge \"Fix backup deletion with naive timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9fb744390e479737535e4728d37e7b749861cc79"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-29 06:31:32.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-29 06:31:32.000000000","tz":300},"subject":"Pin pyOpenSSL version to 24.2.1.","message":"Pin pyOpenSSL version to 24.2.1.\n\nTemporarily pin the pyOpenSSL version until it can be safely\nremoved in favor of the cryptography package.\n\nChange-Id: I26d200ca462782765ca4a8f77bfa5a6d46a416ec\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7819ad2792b2967cdac9c84dcca9ee746b2c1da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7819ad2792b2967cdac9c84dcca9ee746b2c1da3"}]},"branch":"refs/heads/master"},"525161c1274f55ac45230774bfb5c3391fb236af":{"kind":"REWORK","_number":2,"created":"2026-09-29 06:39:59.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/27/1007827/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/1007827/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/1007827/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/1007827/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/1007827/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/1007827/2"}}},"commit":{"parents":[{"commit":"9fb744390e479737535e4728d37e7b749861cc79","subject":"Merge \"Fix backup deletion with naive timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9fb744390e479737535e4728d37e7b749861cc79"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-29 06:31:32.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-29 06:39:49.000000000","tz":300},"subject":"Pin pyOpenSSL version to 24.2.1.","message":"Pin pyOpenSSL version to 24.2.1.\n\nTemporarily pin the pyOpenSSL version until it can be safely\nremoved in favor of the cryptography package.\n\nChange-Id: I26d200ca462782765ca4a8f77bfa5a6d46a416ec\nSigned-off-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/525161c1274f55ac45230774bfb5c3391fb236af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/525161c1274f55ac45230774bfb5c3391fb236af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
