)]}'
{"id":"openstack%2Ftrove~1007836","triplet_id":"openstack%2Ftrove~master~Ifbeb90dcd3adcd9e5ad641af95724cb438121846","project":"openstack/trove","branch":"master","topic":"migrate-pyopenssl-to-cryptography","attention_set":{},"removed_from_attention_set":{"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-30 06:26:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifbeb90dcd3adcd9e5ad641af95724cb438121846","subject":"Fix ssl show after pyOpenSSL removed X509.get_extension","status":"MERGED","created":"2026-09-29 09:02:35.000000000","updated":"2026-09-30 06:27:00.000000000","submitted":"2026-09-30 06:26:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007836-migrate-pyopenssl-to-cryptography","meta_rev_id":"c79fb16c8ec1b137f8599d8b9be7df5022d36fbc","_number":1007836,"virtual_id_number":1007836,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-30 06:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-30 05:49:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-30 05:49:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-30 05:49:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-29 09:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-29 12:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-30 05:36:51.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-09-30 05:49:22.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"b0616094cab76610785f97692c6b6cfcbe8a7467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-29 09:02:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23e6e0e9aa370b5e02358badc915b06d6b420638","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 09:46:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be1a2767dc2a41ca87bf97ed723d81a7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/41cce4bc5fd843adbaa3e0dfcdd84fbd : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/919d213e952f4c9c84fd420b04b6d920 : SUCCESS in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f47bc408f41ffb052a0f52cadf830763960ebfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 12:15:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05ebb3337b8b4e2092f23b78520e95f0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b52ddffa34d4efa9e11bab6e84f390c : SUCCESS in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ce6a1856c2343b9ae7bc54ed97727f6 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9357473f46d54de3b8b49b6bb31583f2 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/caff4c45a0e841af9cf0db344c440308 : SUCCESS in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f36a0372e28e419b80031fd4ca8482aa : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b7b43b8a5e4e80a691ef50d34eaa8b : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/367f0b211e8c455289caed5209f2fb6a : SUCCESS in 5m 49s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/9703a85f05784a32ac1b41f36ec72148 : SUCCESS in 47m 20s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/869168966bbd491a97ad346408f9812a : SUCCESS in 2h 02m 30s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/eabb0c3513ba45179fdcefcf2883420e : SUCCESS in 1h 33m 13s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/8eb416dd81f34e3291904a7807649d3e : FAILURE in 54m 49s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/6d79ce78f3514759bc1f8b778e6edfd3 : FAILURE in 1h 03m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/741bd2dad5ce4fafa06e3e73763c5644 : SUCCESS in 2h 11m 44s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/0f83d5b822024772ade0afb38923afc6 : SUCCESS in 50m 10s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/abe08c360e7240e68e7d407c95ecb49e : SUCCESS in 2h 01m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/5fe81afffa0c416e818219f2f3105269 : SUCCESS in 1h 27m 48s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/85e9a2210e0547a9866126d1eb62626c : FAILURE in 1h 01m 34s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/d9a290ceaa7345c2bbb7e4db222fe1c5 : FAILURE in 51m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/c41d4aaa0387464cb863b04f88d25ea6 : SUCCESS in 1h 43m 13s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/338e067eb7da48bf899f0692a5de0cfe : SUCCESS in 50m 39s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/1bb3233c3ed14234b6951dc0ffb17999 : FAILURE in 1h 01m 18s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/9bf5e9af6a8f45d7860fbf3dbb6c2a7f : FAILURE in 50m 22s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/d97dec962e584983a094524272168c95 : FAILURE in 49m 14s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/747eed4c2e6d4243a81d2b3532b239d7 : SUCCESS in 1h 28m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/217eee6cbcc7409594258ff855007e4b : SUCCESS in 1h 29m 12s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1ee7ba733a154d429b5c4bd9daa8569e : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e71c4cb4658b080845775ebb0fb410b779a20ede","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-29 12:25:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8ddab3d310e2b35d7aa91d44340af5feeffce11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 12:47:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e60a0f06278e4c17acaa17ddf04a8751\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/621aa0b16b344c25b4a48890abfac36b : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/182313c7ad4d4a90977abb5a537f8364 : SUCCESS in 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1043a896ea6ebb56d333312a9b48946fe762761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 13:49:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e25ffc2cb825410daae2e090dd209e57\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/215a4fe274a64b39ad6996a361a73fd6 : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32c19f53c4084cc9972a13a63188bd7f : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4317c3e2d9674538889dadc44d0a37fc : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c64c45964390437cb23d79f2a07371bb : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/351e5367d77f456b8e32ab51c46d0817 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f65a294d49450184ed5cba96892f61 : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70ce9114047349ada98a9d76f7dd513e : SUCCESS in 8m 12s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/2f754e96e40b4b2e859a1eb5ffc07b02 : SUCCESS in 40m 40s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/689cfe5d201346e9a9c67717db4c5592 : SUCCESS in 1h 11m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/fa4316a171cd477a9427d02d8d15c92e : SUCCESS in 1h 00m 05s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/ae7b8928b33442bd99eca0d1c857c4c1 : SUCCESS in 52m 42s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/988c276733814b67afe76fce67ba230c : SUCCESS in 1h 13m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/2381382b30d545c48d2bb3d2cd7bdcf9 : SUCCESS in 1h 16m 53s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/1dc0dbf57d514c0aa7d48b0f97e3f79c : SUCCESS in 38m 35s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/f404185fa6f84dc298bc4f82e51f1bff : SUCCESS in 1h 15m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/bb031fa749e446b4be95d7bd8a043fb9 : SUCCESS in 1h 02m 27s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/1af1eabbe76c40019331aae62910f8cb : SUCCESS in 54m 22s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/f643776caa1c4c53ba8528da3725d3ae : SUCCESS in 1h 02m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/41a894550c1e464fbcd152e38629cc22 : SUCCESS in 1h 13m 49s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/f9aeec3a2f754ca6b452633348e41b8b : SUCCESS in 41m 07s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/c77e3b0f624f4430bd0383d7fd037f73 : SUCCESS in 1h 10m 48s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/28cdebe4967b4d85bd8e35089ead0bdd : SUCCESS in 42m 57s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/c470d9ced192412c9ec6fa7cb761ae47 : SUCCESS in 51m 53s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/9e48d5361c494da8943dab702d6396b0 : SUCCESS in 50m 18s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/cb64ca076ed447fd91144b6b619be7ab : SUCCESS in 1h 02m 12s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2b8efaafd2646c1b4130228b0c2392c : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6f200c0c8c5efd77ae992003d9d8221913b9dba5","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-30 05:36:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0590e374608811aca238dff0208241fc19cb0baa","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-30 05:49:22.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"116a8ed9cd49901148b1e60bd73f7f229064ce4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 05:49:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d8ce71d4a1c5b13d018a7d5357c79a022ccd512","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 06:26:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72bba27b59bb4f99986d1046907c4714\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/550d59a8403e4f4c8f122c96a146654d : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8d783fc16864aa99c820f10c9253fb6 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28d2271a1bb34460a5836b1193510868 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b9546d187cab4685804df1705fe266b8 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caeb179e4f9f4dcd9d1cd7e9fa9fc92a : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8b64c45a8794c8aa4454c8197071edf : SUCCESS in 4m 07s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/9e5ed924b5bd418aac3b1e1c53bda65e : SUCCESS in 33m 14s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/91d14e4c182640c2920ac49f44764c4d : SUCCESS in 31m 46s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8f5e0ba0ab9c47c9b160f5c15c5852e7 : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":1},{"id":"548078bd9edae88f4061314034e4d70f30a4db60","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 06:26:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c79fb16c8ec1b137f8599d8b9be7df5022d36fbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 06:27:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a83a2caca0c84ea4b73c06d5a710a957\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b46947154f435d8e4b2db8f390117e : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ab7007ab112406bae98ae01eb642063 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99e694a49d24d2e8310cf621091ba5759f26c1e9","revisions":{"99e694a49d24d2e8310cf621091ba5759f26c1e9":{"kind":"REWORK","_number":1,"created":"2026-09-29 09:02:35.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/36/1007836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/36/1007836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/1007836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/1007836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/1007836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/36/1007836/1"}}},"commit":{"parents":[{"commit":"9fb744390e479737535e4728d37e7b749861cc79","subject":"Merge \"Fix backup deletion with naive timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9fb744390e479737535e4728d37e7b749861cc79"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-29 08:58:13.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-29 08:58:13.000000000","tz":300},"subject":"Fix ssl show after pyOpenSSL removed X509.get_extension","message":"Fix ssl show after pyOpenSSL removed X509.get_extension\n\npyOpenSSL 26.2 removed X509.get_extension(), causing ssl show to\nfail. Read certificate details with cryptography and report DNS and\nIP SAN entries. Keep ssl show available for certificates with\nunsupported general names and document the changed SAN output.\n\nCo-authored-by: Thomas Bechtold \u003cthomasbechtold@jpberlin.de\u003e\nChange-Id: Ifbeb90dcd3adcd9e5ad641af95724cb438121846\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99e694a49d24d2e8310cf621091ba5759f26c1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99e694a49d24d2e8310cf621091ba5759f26c1e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
