)]}'
{"id":"openstack%2Ftrove~1008197","triplet_id":"openstack%2Ftrove~master~Idaac88bf11d96db568e33b6675d927984cddfb4f","project":"openstack/trove","branch":"master","topic":"configuration-validation-before-attach","attention_set":{},"removed_from_attention_set":{"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-10-06 06:29:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idaac88bf11d96db568e33b6675d927984cddfb4f","subject":"Validate KeyDB and Valkey configuration with datastore engines","status":"ABANDONED","created":"2026-10-01 04:56:41.000000000","updated":"2026-10-06 06:29:17.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f796540dd4fde16cd3d593f8c7102c0be4dac689","_number":1008197,"virtual_id_number":1008197,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-02 12:28:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-10-01 04:56:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-01 08:53:41.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"}],"messages":[{"id":"52946d83a588d943b6301d5c3358e691661d1c33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-10-01 04:56:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c33a69aab9ad42fff858c87c2fb1e4f23ac42d43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 04:56:56.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989856,49","accounts_in_message":[],"_revision_number":1},{"id":"b316407ee2c603cc6307a0a6ebe62f519a9c05c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 04:56:56.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989856,49","accounts_in_message":[],"_revision_number":1},{"id":"a4a75002c09da0931e1d38f1d0c4ca6ae39e36cf","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-10-01 08:53:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3fd138440a122b98456fa89094ff3c689363a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 08:53:59.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989856,49","accounts_in_message":[],"_revision_number":1},{"id":"e755d6e286bdc9d64fa053a792d4a8bb748ccfd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 08:54:00.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 989856,49","accounts_in_message":[],"_revision_number":1},{"id":"b6c7ca0bb39ee7c2dcf0886832e683065688263c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-10-01 10:36:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fabb37c4beb0a3e9262866fa7f836135b989dbef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 10:50:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16de0b5f2a3a4267b907074819d22644\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a88c9005d4c34704b1c4b4e6d1b4d2b7 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ed4955a577ae4d33abfa4f5bde2bb12b : SUCCESS in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0029f3d2bba21976837b46a2b293f918025c61ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 11:57:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8de589b540694361b5987fcfa5f52c0a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26d768a716c047c79b4ac5a39a936965 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3fda7479abd417dbb7401c5055904b2 : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ae49794e2cf40d9be0b03ca0345a770 : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b53859e4b74f4099aade99a808693919 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e426fe88534aa195919afcbf87b7d0 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55c9f6020e184bc39451b1e257e4abd6 : SUCCESS in 8m 21s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/e5167ce283ce4ab3891734549336874a : SUCCESS in 43m 17s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/b5b3981b70a64732ab09b197935ff9e9 : SUCCESS in 1h 18m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/2446d7648c61449886ab745652003efe : SUCCESS in 1h 03m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/d589507a89ae4b21982320edab5da99a : SUCCESS in 1h 08m 55s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/5bdad592f397472bbc5dbd1f33f00763 : SUCCESS in 43m 04s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/865fa63a555c43d3866803939cc4e18f : SUCCESS in 1h 08m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/1275384e490940bc9f08249b934ca892 : SUCCESS in 1h 01m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/348d97d5f9644f6f84802eec8d299aac : SUCCESS in 1h 06m 37s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/a9b38871952f486aa4743fca76022da1 : SUCCESS in 46m 01s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/dcf0b963fbc34d77a609afac3c7140ad : SUCCESS in 1h 10m 15s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/36bbe7a9e54745bfbc24aa860a220809 : SUCCESS in 51m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/c96f414fc23e45ae9422e03b900cabf3 : SUCCESS in 57m 09s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/3d1e44556aff4f64a55ca2b41aabeabb : SUCCESS in 36m 54s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/e30490e0794144d3bb90a5c86e7121bf : SUCCESS in 49m 19s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/32fe0d59dcae4803aad9a1da26523535 : SUCCESS in 50m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/eb99b0e00a5e4aa7a6067d45097e7e71 : SUCCESS in 56m 07s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/ef7742c48d114b75a7530329563b1ab7 : SUCCESS in 37m 07s\n- trove-tempest-ubuntu-backup-valkey https://zuul.opendev.org/t/openstack/build/1826145c73a449ddb9f1e2e100a874eb : SUCCESS in 47m 56s (non-voting)\n- trove-tempest-ubuntu-replication-valkey https://zuul.opendev.org/t/openstack/build/5548760a3f28472b9cff755c5e0e22e6 : SUCCESS in 51m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/145bf940a6044e3c92ff554bf008607b : SUCCESS in 53m 27s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98fa5b1759964fb982d76f239d893c29 : FAILURE in 6m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d84bc5c4e77b08a45ae77755b665bb3698bc4bdc","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-10-01 12:13:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63a71b92070ccb992f00a16448c5b86ed749b02e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 12:25:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69b32f25b42748439c33bb540769d44c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f1052e3e9fe14c47a2a093b85b9e5a30 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ac8df7ea770449d5907534283213de60 : SUCCESS in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2715bc36d4c371b593efca58cff728edff487275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 13:51:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3daeedc95da4df5852b882321163acd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83d6a756f9074c849e070bd4f4babeeb : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed137b769ae4205a6fae80f9ebe04f6 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df6514502c7944b8bd8404c8615730c6 : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b589a91f54a74af3aa13a701cc547872 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f1cf9e889c4c9f9818b8c08077c205 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/435b6bf62cc04108b3b937b60795e74f : SUCCESS in 8m 06s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/0e21133130184c1b886e098ad8f24b91 : SUCCESS in 44m 24s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/915469563cda450d9670a9cbec90c6cc : SUCCESS in 1h 12m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/0d65cd5a24db4a44a2bc3c9757549330 : SUCCESS in 1h 00m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/9ea09011f9f7402eb36f8cb7685fd3cc : SUCCESS in 1h 36m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/56e4f7b2ba0e4864b72f9ce060745f93 : SUCCESS in 39m 53s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/cf50ba20e0034986b8eef013d8789c0d : SUCCESS in 1h 07m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/af8c701384a24c3096f0e4b1a77f3a84 : SUCCESS in 55m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/3acb1130a87d4b29aab6e3bf55ae7881 : SUCCESS in 1h 06m 38s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/5461b4d8cd5746caba7df01aeff9d318 : SUCCESS in 45m 05s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/5466c61081604c4dae06566b85bb4560 : SUCCESS in 1h 08m 50s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/e5566db14fd44f7a8dd4dfecdbed9693 : SUCCESS in 1h 06m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/22a9b2c6686e4317b8f879a0d7fb5377 : SUCCESS in 55m 28s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/9aee577b32044fc1b7a6be3a08188e65 : SUCCESS in 49m 11s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/45ea7967028e4d33ae8be98073a4e82f : FAILURE in 20m 12s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/e94483f438cd4e3899cdd6a32b5759dc : SUCCESS in 1h 09m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/d52c0c67f6c343c0800330b91a474fa8 : SUCCESS in 51m 17s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/cc3356e8176f4495b8214aeb87a53751 : FAILURE in 48m 12s\n- trove-tempest-ubuntu-backup-valkey https://zuul.opendev.org/t/openstack/build/4019c29a2d3c4cc991393fb38647cb66 : SUCCESS in 1h 12m 53s (non-voting)\n- trove-tempest-ubuntu-replication-valkey https://zuul.opendev.org/t/openstack/build/5a463827112147a5801a965b8f5ebc7a : SUCCESS in 43m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/7c40a895a8b14483b197d8b8137116d1 : SUCCESS in 52m 32s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dd0389a943cb4a4caca2dfd5864bfac6 : FAILURE in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"3c1fe7727e26ab7f2a2be8f5f732a73087eff49d","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-10-01 14:14:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ae06a59cc44624ab17875a7e9461c1172775e84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 14:46:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d265a6a14fb943d5b3ddb55177b5fba0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/96014a06da8147ff8d83fe0d4864cd56 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a755cb3a47804f21b45ef0f4cdbba0eb : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d41042f7bf7b14eaa37c7b3d9ecdd0fec8477a07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 15:32:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db30b0ca36fc4d888b726b0ff6ba6e7b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11a882872f874d1c8e961e6a518c2d45 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d4ec745df214a889a5d2512176fbbb5 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4cbe2366d254609ab143bc078e5b13e : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eaa12a30ccd245bebc84b70ff8e323c7 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68573a512e974beca0184a8d6bb84dc6 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abb8d2c15b5849fc9298aabfc6632278 : SUCCESS in 4m 11s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/614e60184a4849deac3e805eae74acec : SUCCESS in 46m 49s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/ed27318c149b42b3be1200f481a7f318 : SUCCESS in 1h 15m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/a07112921e894048ae57b7afa26b6c7f : SUCCESS in 1h 02m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/0a10731332854acf8ae657b9d7e45134 : SUCCESS in 1h 13m 30s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/ce712710477a4b199c6c98391fba3a57 : FAILURE in 21m 17s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/99848a83fc794ed7a440665311d334f1 : FAILURE in 22m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/12d3cf48149f4e9b88dfde3810044944 : SUCCESS in 56m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/29f5023d6f7f4376872f711bdbae16eb : SUCCESS in 1h 05m 03s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/9ec7ec04832745abac6335b21a817a55 : SUCCESS in 46m 10s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/7d7d07936a244b138f7e491836e26235 : SUCCESS in 1h 06m 13s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/de5d86a5a40643008e52620066579678 : SUCCESS in 53m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/e5ac43503d3d4f0bbd7d73e09db3baa6 : SUCCESS in 56m 04s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/2f255ed6abf14b2dabb48c480884aaa2 : SUCCESS in 35m 12s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/3a11359ef53f479188e4f30e98060b99 : SUCCESS in 47m 17s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/1189e19d76cc471dbaf617f9d1265bc5 : SUCCESS in 45m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/c9a5e52783504a2d9d09033801597258 : SUCCESS in 50m 33s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/6647296314754f198f604835c0b226fd : SUCCESS in 35m 05s\n- trove-tempest-ubuntu-backup-valkey https://zuul.opendev.org/t/openstack/build/e0f8f15728a146c7b988079d3fbacd7a : FAILURE in 20m 30s (non-voting)\n- trove-tempest-ubuntu-replication-valkey https://zuul.opendev.org/t/openstack/build/e5f551a6d5504f79a9d0274e30c9b132 : SUCCESS in 47m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/2084f2b556fa4c1cb31a75a1359477cf : SUCCESS in 52m 31s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2bb7a66263ae42718b457abe38414e15 : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"be03bd7bd7bb39c8f322a6bf73bdd447bc0956ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-10-02 11:15:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"33edaa40596a264262e98d1242921867b8161556","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 11:30:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84a3bf3aa9314c939362083111284816\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a227c4fa4b0444b09f0bbbfd8ec699cb : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0563dc0c471b4762958767efe625df7a : SUCCESS in 11m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"45823dc57f8aac569e6727141f4ca9aaedd9696c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 12:28:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf52c401f7f24c7aac1371276afca76c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c8b3ba482b945b2a437fdbca745127c : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a6984a42f0d4107b91e64fce6e3f3c1 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b29c0abb1904d60a4293cc306307298 : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f875edf6c248433590a04bf8fcd9b6aa : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/699cf62c224943218a3771bd7cc54bd8 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9f31550fafa4e178cd6d2d16a21f140 : SUCCESS in 6m 29s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/95f307a86d224898801a110e5cb9280e : SUCCESS in 40m 37s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/7c975daaa39e4a5ab7070b7ae9e5f22a : SUCCESS in 1h 08m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/25443edcfcdd4afa9f08064ebf128af7 : SUCCESS in 59m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/829b1a00a65843928d57c88fd9a5e3c7 : SUCCESS in 1h 11m 06s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/758a1859945b4c2ba865fdf51d9f55a8 : SUCCESS in 42m 32s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/a67299e35c3b4179b721939ce14d71fc : SUCCESS in 1h 09m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/fa56c26587754da18fab5806ee129fa8 : SUCCESS in 57m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/8adb804c4e27436691ff75aa9c18fe0f : SUCCESS in 1h 02m 17s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/ed48fa74fe324344bda64a0f05d663a0 : SUCCESS in 46m 29s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/957ef4ccd3f249c7afe83a74fb92c975 : SUCCESS in 1h 07m 25s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/9b5d4c8b94fe455b9a5fe17c02671a33 : SUCCESS in 51m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/75964f85617047d597da820a56baafd6 : SUCCESS in 57m 05s (non-voting)\n- trove-tempest-ubuntu-base-keydb https://zuul.opendev.org/t/openstack/build/02360ec023b246fbb2bdf7c2a97d0d72 : SUCCESS in 38m 00s\n- trove-tempest-ubuntu-backup-keydb https://zuul.opendev.org/t/openstack/build/899e0ee665d9470197f1f0ffc99cdff7 : SUCCESS in 43m 30s (non-voting)\n- trove-tempest-ubuntu-replication-keydb https://zuul.opendev.org/t/openstack/build/ba09591429474a0487bdb2473a02078a : SUCCESS in 44m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-keydb https://zuul.opendev.org/t/openstack/build/96cba210f99c4baeb4ed53a300c3ce2e : SUCCESS in 49m 04s (non-voting)\n- trove-tempest-ubuntu-base-valkey https://zuul.opendev.org/t/openstack/build/0c8f2d762cd84c39b24521c5228b7d09 : SUCCESS in 34m 14s\n- trove-tempest-ubuntu-backup-valkey https://zuul.opendev.org/t/openstack/build/34cdc80749574bc3a663867c202bd3f7 : SUCCESS in 42m 48s (non-voting)\n- trove-tempest-ubuntu-replication-valkey https://zuul.opendev.org/t/openstack/build/4d101b5754bb43d7b141e1ec1d6a6d92 : SUCCESS in 52m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-valkey https://zuul.opendev.org/t/openstack/build/1761648a37614648bec603200ae18733 : SUCCESS in 53m 15s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/014ddf401ad44475a16ad34c6c175289 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f82bd192521294241fe9a8cdd3bae394fc249da6","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-10-05 06:06:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d5febc3a0976ca7915299793ef2a71762455bafa","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-10-05 06:15:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f796540dd4fde16cd3d593f8c7102c0be4dac689","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-10-06 06:29:17.000000000","message":"Abandoned\n\nChanges moved to 999099","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"00e9a843e9b12931a6b1e707a49e68ebc104adf1","revisions":{"84f1f8724c88d12e1e5278b663109932ea6815ee":{"kind":"REWORK","_number":1,"created":"2026-10-01 04:56:41.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/97/1008197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/1008197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/1008197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/1008197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/1008197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/1008197/1"}}},"commit":{"parents":[{"commit":"af90496ae363f40d8cb606836220d9baa67671cc","subject":"Add KeyDB and Valkey datastore support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af90496ae363f40d8cb606836220d9baa67671cc"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-10-01 04:26:26.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-10-01 04:49:27.000000000","tz":300},"subject":"Validate KeyDB and Valkey configuration with datastore engines","message":"Validate KeyDB and Valkey configuration with datastore engines\n\nValidate configuration overrides in an isolated datastore container\nbefore attaching a configuration group or starting a new instance.\nReject invalid values without changing a running instance and report\ncreation failures through the instance fault.\n\nBlueprint: configuration-validation-before-attach\nDepends-On: https://review.opendev.org/c/openstack/trove/+/989856\nDepends-On: https://review.opendev.org/c/openstack/trove/+/999099\nAssisted-by: GPT-6\nChange-Id: Idaac88bf11d96db568e33b6675d927984cddfb4f\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/84f1f8724c88d12e1e5278b663109932ea6815ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/84f1f8724c88d12e1e5278b663109932ea6815ee"}]},"branch":"refs/heads/master"},"a6815a6615a76e09d2004ab6601ca4c57bbf250b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2026-10-01 10:36:08.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/97/1008197/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/1008197/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/1008197/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/1008197/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/1008197/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/1008197/2"}}},"commit":{"parents":[{"commit":"217db27fd133aa5a49ce519d59507b6c53120043","subject":"Add KeyDB and Valkey datastore support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/217db27fd133aa5a49ce519d59507b6c53120043"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-10-01 04:26:26.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-10-01 10:35:47.000000000","tz":300},"subject":"Validate KeyDB and Valkey configuration with datastore engines","message":"Validate KeyDB and Valkey configuration with datastore engines\n\nValidate configuration overrides in an isolated datastore container\nbefore attaching a configuration group or starting a new instance.\nReject invalid values without changing a running instance and report\ncreation failures through the instance fault.\n\nBlueprint: configuration-validation-before-attach\nDepends-On: https://review.opendev.org/c/openstack/trove/+/999099\nAssisted-by: GPT-6\nChange-Id: Idaac88bf11d96db568e33b6675d927984cddfb4f\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a6815a6615a76e09d2004ab6601ca4c57bbf250b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a6815a6615a76e09d2004ab6601ca4c57bbf250b"}]},"branch":"refs/heads/master"},"00e9a843e9b12931a6b1e707a49e68ebc104adf1":{"kind":"REWORK","_number":3,"created":"2026-10-02 11:15:35.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/97/1008197/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/1008197/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/1008197/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/1008197/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/1008197/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/1008197/3"}}},"commit":{"parents":[{"commit":"217db27fd133aa5a49ce519d59507b6c53120043","subject":"Add KeyDB and Valkey datastore support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/217db27fd133aa5a49ce519d59507b6c53120043"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-10-01 04:26:26.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-10-02 11:15:30.000000000","tz":300},"subject":"Validate KeyDB and Valkey configuration with datastore engines","message":"Validate KeyDB and Valkey configuration with datastore engines\n\nValidate configuration overrides in an isolated datastore container\nbefore attaching a configuration group or starting a new instance.\nReject invalid values without changing a running instance and report\ncreation failures through the instance fault.\n\nBlueprint: configuration-validation-before-attach\nDepends-On: https://review.opendev.org/c/openstack/trove/+/999099\nAssisted-by: GPT-6\nChange-Id: Idaac88bf11d96db568e33b6675d927984cddfb4f\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00e9a843e9b12931a6b1e707a49e68ebc104adf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00e9a843e9b12931a6b1e707a49e68ebc104adf1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
