)]}'
{"id":"openstack%2Ftrove~35166","triplet_id":"openstack%2Ftrove~master~I34268f768a5cdb8085607269e2c2cb95974a539d","project":"openstack/trove","branch":"master","topic":"swift","hashtags":[],"change_id":"I34268f768a5cdb8085607269e2c2cb95974a539d","subject":"Change the swift file deletion to use the manifest.","status":"MERGED","created":"2013-07-01 17:43:47.000000000","updated":"2013-07-16 21:10:06.000000000","submitted":"2013-07-16 21:10:06.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fd5af588daf2ab530738ea9a8771cd6bf1fde40a","_number":35166,"virtual_id_number":35166,"owner":{"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-07-16 21:10:06.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-07-16 20:20:10.000000000","_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},{"value":0,"_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},{"value":0,"_account_id":971,"name":"Daniel Salinas","email":"imsplitbit@gmail.com","username":"imsplitbit"},{"value":0,"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},{"value":0,"_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},{"value":0,"_account_id":1375,"name":"Ed Cranford","email":"ed.cranford@gmail.com","username":"ed--cranford"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-07-16 19:19:08.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},{"value":2,"date":"2013-07-16 19:21:12.000000000","_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},{"value":0,"_account_id":971,"name":"Daniel Salinas","email":"imsplitbit@gmail.com","username":"imsplitbit"},{"value":0,"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},{"value":2,"date":"2013-07-16 21:04:54.000000000","_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},{"value":0,"_account_id":1375,"name":"Ed Cranford","email":"ed.cranford@gmail.com","username":"ed--cranford"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},{"value":0,"date":"2013-07-16 19:21:12.000000000","_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},{"value":0,"_account_id":971,"name":"Daniel Salinas","email":"imsplitbit@gmail.com","username":"imsplitbit"},{"value":0,"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},{"value":1,"date":"2013-07-16 21:04:54.000000000","_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},{"value":0,"_account_id":1375,"name":"Ed Cranford","email":"ed.cranford@gmail.com","username":"ed--cranford"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},{"value":0,"_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},{"value":0,"_account_id":971,"name":"Daniel Salinas","email":"imsplitbit@gmail.com","username":"imsplitbit"},{"value":0,"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},{"value":0,"_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},{"value":0,"_account_id":1375,"name":"Ed Cranford","email":"ed.cranford@gmail.com","username":"ed--cranford"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},{"_account_id":971,"name":"Daniel Salinas","email":"imsplitbit@gmail.com","username":"imsplitbit"},{"_account_id":1375,"name":"Ed Cranford","email":"ed.cranford@gmail.com","username":"ed--cranford"},{"_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},{"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-07-01 18:53:55.000000000","updated_by":{"_account_id":971,"name":"Daniel Salinas","email":"imsplitbit@gmail.com","username":"imsplitbit"},"reviewer":{"_account_id":971,"name":"Daniel Salinas","email":"imsplitbit@gmail.com","username":"imsplitbit"},"state":"REVIEWER"},{"updated":"2013-07-03 20:55:27.000000000","updated_by":{"_account_id":1375,"name":"Ed Cranford","email":"ed.cranford@gmail.com","username":"ed--cranford"},"reviewer":{"_account_id":1375,"name":"Ed Cranford","email":"ed.cranford@gmail.com","username":"ed--cranford"},"state":"REVIEWER"},{"updated":"2013-07-16 19:21:12.000000000","updated_by":{"_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},"reviewer":{"_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},"state":"REVIEWER"},{"updated":"2013-07-16 20:20:10.000000000","updated_by":{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},"reviewer":{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-07-16 21:04:54.000000000","updated_by":{"_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},"reviewer":{"_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},"state":"REVIEWER"},{"updated":"2013-07-16 21:10:06.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"a9dba24956c4e67ae6e9a412b4d338beaccbb43b","author":{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},"date":"2013-07-01 17:43:51.000000000","message":"Patch Set 1:\n\nBuild Started https://rdjenkins.dyndns.org/job/Trove-Gate/431/ ","accounts_in_message":[],"_revision_number":1},{"id":"1a692fa67ba3807f3d83589a3f000d764d6fcd10","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-01 17:46:41.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/35166/1/check/gate-trove-pep8/106 : FAILURE in 43s\n- http://logs.openstack.org/35166/1/check/gate-trove-python26/106 : SUCCESS in 2m 35s\n- http://logs.openstack.org/35166/1/check/gate-trove-python27/106 : FAILURE in 47s\n","accounts_in_message":[],"_revision_number":1},{"id":"fd44d49e4ecb41782cb7f5daea10d59d45587661","author":{"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},"date":"2013-07-01 18:20:21.000000000","message":"Patch Set 1:\n\nrecheck: no bug","accounts_in_message":[],"_revision_number":1},{"id":"3929ebcb2446f6f2ce5037947ed71b965381232f","author":{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},"date":"2013-07-01 18:44:43.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild Failed \n\nhttps://rdjenkins.dyndns.org/job/Trove-Gate/431/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"310406cb8ff0a8c4a6175496e9da446c3c175f61","author":{"_account_id":971,"name":"Daniel Salinas","email":"imsplitbit@gmail.com","username":"imsplitbit"},"date":"2013-07-01 18:53:55.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nSee one inline comment.","accounts_in_message":[],"_revision_number":1},{"id":"51aaec38c6bbbea283e26f6e88d2132ce878632d","author":{"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},"date":"2013-07-01 19:05:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aee3b4e64027fbe0877be6298e63c463f63b1342","author":{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},"date":"2013-07-01 19:05:15.000000000","message":"Patch Set 2:\n\nBuild Started https://rdjenkins.dyndns.org/job/Trove-Gate/434/ ","accounts_in_message":[],"_revision_number":2},{"id":"dd3395075de0283401b84e799fb191b570d41c42","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-01 19:08:07.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/35166/2/check/gate-trove-pep8/108 : SUCCESS in 2m 32s\n- http://logs.openstack.org/35166/2/check/gate-trove-python26/108 : SUCCESS in 2m 36s\n- http://logs.openstack.org/35166/2/check/gate-trove-python27/108 : SUCCESS in 1m 57s\n","accounts_in_message":[],"_revision_number":2},{"id":"194facc51537db4a0030976c292eb96873e83e91","author":{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},"date":"2013-07-01 20:06:55.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild Failed \n\nhttps://rdjenkins.dyndns.org/job/Trove-Gate/434/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c4b0a1ee4dce91255927620e66524dcf0acdb290","author":{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},"date":"2013-07-02 07:15:27.000000000","message":"Patch Set 2: No score\n\nBuild Started https://rdjenkins.dyndns.org/job/Trove-Gate/440/ ","accounts_in_message":[],"_revision_number":2},{"id":"3d3df193eab0c4887c9f2e639ed719e76f5e23f8","author":{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},"date":"2013-07-02 08:17:00.000000000","message":"Patch Set 2: Works for me\n\nBuild Successful \n\nhttps://rdjenkins.dyndns.org/job/Trove-Gate/440/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ac152f4b631fdff0d10f975a6b055bda98108d72","author":{"_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},"date":"2013-07-03 20:53:11.000000000","message":"Patch Set 2: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":2},{"id":"566b6185c78a563b4cfc046e5daaf6600b54c4a7","author":{"_account_id":1375,"name":"Ed Cranford","email":"ed.cranford@gmail.com","username":"ed--cranford"},"date":"2013-07-03 20:55:27.000000000","message":"Patch Set 2: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":2},{"id":"9458fd9183c62b118b08dcaa8b5f6749d5077cee","author":{"_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},"date":"2013-07-10 07:56:39.000000000","message":"Patch Set 2: (1 inline comment)\n\nJust had one clarification.\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"bc543553a6cf35d142d9db49854cbb324eaabb53","author":{"_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},"date":"2013-07-11 19:06:02.000000000","message":"Patch Set 2: (1 inline comment)\n\n@SlickNik: Responded to your comments (Rob is out for awhile).","accounts_in_message":[],"_revision_number":2},{"id":"c24c0be5e524eec27d4984314acf386593234c1a","author":{"_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},"date":"2013-07-16 17:47:56.000000000","message":"Patch Set 2: Looks good to me (core reviewer); Approved\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"43f91496a7d96594e05cf3d304319c23e5b90b0d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-16 17:47:57.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"c393e06bb1fd649c9fe2e54904cf9fcccc7a76d8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-16 17:47:58.000000000","message":"Patch Set 2: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"30af2989c97f34c6ad39c697207e55579d99db34","author":{"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},"date":"2013-07-16 19:19:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd975ddb29929a06989893c7cfa2953f84e353d3","author":{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},"date":"2013-07-16 19:19:08.000000000","message":"Patch Set 3:\n\nBuild Started https://rdjenkins.dyndns.org/job/Trove-Gate/505/ ","accounts_in_message":[],"_revision_number":3},{"id":"37483ea3758e5eceb84540e3391c6e7ec0610aa7","author":{"_account_id":694,"name":"Tim Simpson","email":"tim.simpson@rackspace.com","username":"tim-simpson"},"date":"2013-07-16 19:21:12.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"79becfb3c144bfc353335fa6028afb4575aa0138","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-16 19:22:36.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/35166/3/check/gate-trove-pep8/178 : SUCCESS in 2m 55s\n- http://logs.openstack.org/35166/3/check/gate-trove-python26/177 : SUCCESS in 3m 00s\n- http://logs.openstack.org/35166/3/check/gate-trove-python27/178 : SUCCESS in 3m 30s\n","accounts_in_message":[],"_revision_number":3},{"id":"76b9da436e971c7d35a40ef054477a81119cba70","author":{"_account_id":7092,"name":"Deprecated Trove CI","email":"reddwarf-jenkins@googlegroups.com","username":"reddwarf","tags":["SERVICE_USER"]},"date":"2013-07-16 20:20:10.000000000","message":"Patch Set 3: Works for me\n\nBuild Successful \n\nhttps://rdjenkins.dyndns.org/job/Trove-Gate/505/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"dfc4c627d2e45f2c71f92cec6d9c4f493e8b4fa1","author":{"_account_id":5293,"name":"Nikhil","email":"SlickNik@gmail.com","username":"slicknik"},"date":"2013-07-16 21:04:54.000000000","message":"Patch Set 3: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":3},{"id":"af2a58697f85cfdcd71de3e393058f446ac927f8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-16 21:04:55.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"5de0f298fc9400f93d0b5ab800aafb629f4ff903","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-16 21:10:06.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3},{"id":"e7e8f3d64a21bcc0d6ecb76dc3ced671b5de569d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-16 21:10:06.000000000","message":"Patch Set 3: Verified\n\nBuild succeeded.\n\n- http://logs.openstack.org/35166/3/gate/gate-trove-pep8/179 : SUCCESS in 2m 54s\n- http://logs.openstack.org/35166/3/gate/gate-trove-python26/178 : SUCCESS in 2m 30s\n- http://logs.openstack.org/35166/3/gate/gate-trove-python27/179 : SUCCESS in 5m 09s\n","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1815d5d56635d98dc95b302b5bf869543d05b8cb","revisions":{"c3e4f2b72359964552742c4bd461e0a56ef8c257":{"kind":"REWORK","_number":1,"created":"2013-07-01 17:43:47.000000000","uploader":{"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},"ref":"refs/changes/66/35166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/35166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/35166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/35166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/35166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/35166/1"}}},"commit":{"parents":[{"commit":"eafc62f61b1a3b1294f557f676e33f831af0ce77","subject":"Merge \"Fixed backups GET and DELETE to restrict access to the owner of backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eafc62f61b1a3b1294f557f676e33f831af0ce77"}]}],"author":{"name":"Robert Myers","email":"robert.myers@rackspace.com","date":"2013-07-01 14:43:24.000000000","tz":-300},"committer":{"name":"Robert Myers","email":"robert.myers@rackspace.com","date":"2013-07-01 17:41:35.000000000","tz":-300},"subject":"Change the swift file deletion to use the manifest.","message":"Change the swift file deletion to use the manifest.\n\n* Adding a _parse_manifest helper to parse the x-object-manifest header\n* Delete file using the prefix in the manifest.\n* Fix error handling in the delete_backup method.\n\nFixes Bug: #1194653\n\nChange-Id: I34268f768a5cdb8085607269e2c2cb95974a539d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c3e4f2b72359964552742c4bd461e0a56ef8c257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c3e4f2b72359964552742c4bd461e0a56ef8c257"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eafc62f61b1a3b1294f557f676e33f831af0ce77","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"392463ad4312882708a01271de72206b87add194":{"kind":"REWORK","_number":2,"created":"2013-07-01 19:05:11.000000000","uploader":{"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},"ref":"refs/changes/66/35166/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/35166/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/35166/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/35166/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/35166/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/35166/2"}}},"commit":{"parents":[{"commit":"eafc62f61b1a3b1294f557f676e33f831af0ce77","subject":"Merge \"Fixed backups GET and DELETE to restrict access to the owner of backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eafc62f61b1a3b1294f557f676e33f831af0ce77"}]}],"author":{"name":"Robert Myers","email":"robert.myers@rackspace.com","date":"2013-07-01 14:43:24.000000000","tz":-300},"committer":{"name":"Robert Myers","email":"robert.myers@rackspace.com","date":"2013-07-01 19:04:57.000000000","tz":-300},"subject":"Change the swift file deletion to use the manifest.","message":"Change the swift file deletion to use the manifest.\n\n* Adding a _parse_manifest helper to parse the x-object-manifest header\n* Delete file using the prefix in the manifest.\n* Fix error handling in the delete_backup method.\n\nFixes Bug: #1194653\n\nChange-Id: I34268f768a5cdb8085607269e2c2cb95974a539d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/392463ad4312882708a01271de72206b87add194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/392463ad4312882708a01271de72206b87add194"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eafc62f61b1a3b1294f557f676e33f831af0ce77","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1815d5d56635d98dc95b302b5bf869543d05b8cb":{"kind":"REWORK","_number":3,"created":"2013-07-16 19:19:04.000000000","uploader":{"_account_id":6268,"name":"Robert Myers","email":"myer0052@gmail.com","username":"rmyers"},"ref":"refs/changes/66/35166/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/35166/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/35166/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/35166/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/35166/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/35166/3"}}},"commit":{"parents":[{"commit":"8c339954f04981d8fdc6bee0b3077349fd8c79f3","subject":"Merge \"Bump pbr to the version 0.5.16\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c339954f04981d8fdc6bee0b3077349fd8c79f3"}]}],"author":{"name":"Robert Myers","email":"robert.myers@rackspace.com","date":"2013-07-01 14:43:24.000000000","tz":-300},"committer":{"name":"Robert Myers","email":"robert.myers@rackspace.com","date":"2013-07-16 19:17:37.000000000","tz":-300},"subject":"Change the swift file deletion to use the manifest.","message":"Change the swift file deletion to use the manifest.\n\n* Adding a _parse_manifest helper to parse the x-object-manifest header\n* Delete file using the prefix in the manifest.\n* Fix error handling in the delete_backup method.\n\nFixes Bug: #1194653\n\nChange-Id: I34268f768a5cdb8085607269e2c2cb95974a539d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1815d5d56635d98dc95b302b5bf869543d05b8cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1815d5d56635d98dc95b302b5bf869543d05b8cb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8c339954f04981d8fdc6bee0b3077349fd8c79f3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
