)]}'
{"id":"openstack%2Ftrove~447839","triplet_id":"openstack%2Ftrove~master~I98d67b34b59ad47dc774afb16395861870dd09b7","project":"openstack/trove","branch":"master","topic":"bug/1674566","hashtags":[],"change_id":"I98d67b34b59ad47dc774afb16395861870dd09b7","subject":"Remove log translations","status":"ABANDONED","created":"2017-03-21 05:59:41.000000000","updated":"2017-08-04 01:08:00.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f0c178f833976b0fd9c45b9371b2593e5261f9a1","_number":447839,"virtual_id_number":447839,"owner":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},{"value":-1,"date":"2017-03-21 13:21:50.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":25395,"name":"liyi","email":"liyi7155@fiberhome.com","username":"liyi8611"},{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},{"date":"2017-03-23 02:10:44.000000000","_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"all":[{"value":-1,"date":"2017-03-21 12:23:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25395,"name":"liyi","email":"liyi7155@fiberhome.com","username":"liyi8611"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":25395,"name":"liyi","email":"liyi7155@fiberhome.com","username":"liyi8611"},{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":25395,"name":"liyi","email":"liyi7155@fiberhome.com","username":"liyi8611"},{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},{"_account_id":25395,"name":"liyi","email":"liyi7155@fiberhome.com","username":"liyi8611"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-21 08:01:05.000000000","updated_by":{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},"reviewer":{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},"state":"REVIEWER"},{"updated":"2017-03-21 11:37:44.000000000","updated_by":{"_account_id":25395,"name":"liyi","email":"liyi7155@fiberhome.com","username":"liyi8611"},"reviewer":{"_account_id":25395,"name":"liyi","email":"liyi7155@fiberhome.com","username":"liyi8611"},"state":"REVIEWER"},{"updated":"2017-03-21 12:23:30.000000000","updated_by":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"reviewer":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"state":"REVIEWER"},{"updated":"2017-03-21 13:21:50.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"7ba1168c65cc08b705e40e1451ea25bbad3e3285","author":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"date":"2017-03-21 05:59:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c745f24e7a32be625299b63362e9c76fb8e6ea9f","author":{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},"date":"2017-03-21 08:01:05.000000000","message":"Patch Set 1:\n\nIt seems that this is not a complete removal, the proposal can be divided according to the file into a different patch submission. The best advice is to follow the way we have added to delete, which is conducive to review the speed","accounts_in_message":[],"_revision_number":1},{"id":"e9d05254ada360d8764df78f3d39cd7ca76fcf0c","author":{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},"date":"2017-03-21 08:05:41.000000000","message":"Patch Set 1:\n\nThis is a reference.https://review.openstack.org/#/q/status:merged+project:openstack/trove+i18n","accounts_in_message":[],"_revision_number":1},{"id":"8f8f5f83b1dbe00ad0a99859e2a2d1db8426d155","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-21 08:39:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-trove-docs-ubuntu-xenial http://docs-draft.openstack.org/39/447839/1/check/gate-trove-docs-ubuntu-xenial/e87f154//doc/build/html/ : SUCCESS in 2m 13s\n- gate-trove-pep8-ubuntu-xenial http://logs.openstack.org/39/447839/1/check/gate-trove-pep8-ubuntu-xenial/5872292/ : FAILURE in 10m 58s\n- gate-trove-python27-ubuntu-xenial http://logs.openstack.org/39/447839/1/check/gate-trove-python27-ubuntu-xenial/ef50c93/ : FAILURE in 10m 33s\n- gate-trove-python35 http://logs.openstack.org/39/447839/1/check/gate-trove-python35/6f8d993/ : FAILURE in 2m 15s\n- gate-trove-functional-dsvm-mysql-ubuntu-xenial http://logs.openstack.org/39/447839/1/check/gate-trove-functional-dsvm-mysql-ubuntu-xenial/0b9a8f0/ : SUCCESS in 59m 07s\n- gate-trove-scenario-dsvm-mariadb-single-ubuntu-xenial-nv http://logs.openstack.org/39/447839/1/check/gate-trove-scenario-dsvm-mariadb-single-ubuntu-xenial-nv/6765e90/ : SUCCESS in 1h 04m 36s (non-voting)\n- gate-trove-scenario-dsvm-mariadb-multi-ubuntu-xenial-nv http://logs.openstack.org/39/447839/1/check/gate-trove-scenario-dsvm-mariadb-multi-ubuntu-xenial-nv/7946472/ : SUCCESS in 1h 04m 24s (non-voting)\n- gate-trove-scenario-dsvm-mysql-single-ubuntu-xenial http://logs.openstack.org/39/447839/1/check/gate-trove-scenario-dsvm-mysql-single-ubuntu-xenial/b2db760/ : SUCCESS in 2h 08m 36s\n- gate-trove-scenario-dsvm-mysql-multi-ubuntu-xenial http://logs.openstack.org/39/447839/1/check/gate-trove-scenario-dsvm-mysql-multi-ubuntu-xenial/6258bc9/ : SUCCESS in 2h 17m 26s\n- gate-trove-scenario-dsvm-postgresql-single-ubuntu-xenial-nv http://logs.openstack.org/39/447839/1/check/gate-trove-scenario-dsvm-postgresql-single-ubuntu-xenial-nv/2816871/ : SUCCESS in 2h 39m 07s (non-voting)\n- gate-trove-scenario-dsvm-postgresql-multi-ubuntu-xenial-nv http://logs.openstack.org/39/447839/1/check/gate-trove-scenario-dsvm-postgresql-multi-ubuntu-xenial-nv/843b006/ : SUCCESS in 2h 03m 55s (non-voting)\n- gate-trove-scenario-dsvm-pxc-single-ubuntu-xenial-nv http://logs.openstack.org/39/447839/1/check/gate-trove-scenario-dsvm-pxc-single-ubuntu-xenial-nv/7965afa/ : SUCCESS in 51m 25s (non-voting)\n- gate-trove-scenario-dsvm-pxc-multi-ubuntu-xenial-nv http://logs.openstack.org/39/447839/1/check/gate-trove-scenario-dsvm-pxc-multi-ubuntu-xenial-nv/ac7b46c/ : SUCCESS in 48m 39s (non-voting)\n- gate-trove-scenario-dsvm-redis-single-ubuntu-xenial-nv http://logs.openstack.org/39/447839/1/check/gate-trove-scenario-dsvm-redis-single-ubuntu-xenial-nv/706c371/ : SUCCESS in 1h 17m 40s (non-voting)\n- gate-trove-scenario-dsvm-redis-multi-ubuntu-xenial-nv http://logs.openstack.org/39/447839/1/check/gate-trove-scenario-dsvm-redis-multi-ubuntu-xenial-nv/92254b0/ : SUCCESS in 1h 17m 15s (non-voting)\n- gate-trove-tox-fakemodetests-ubuntu-xenial http://logs.openstack.org/39/447839/1/check/gate-trove-tox-fakemodetests-ubuntu-xenial/2bd4381/ : SUCCESS in 2m 38s\n- gate-trove-tox-apiexamples-ubuntu-xenial http://logs.openstack.org/39/447839/1/check/gate-trove-tox-apiexamples-ubuntu-xenial/3b874a0/ : SUCCESS in 3m 52s\n- gate-trove-pylint-ubuntu-xenial http://logs.openstack.org/39/447839/1/check/gate-trove-pylint-ubuntu-xenial/d4e7519/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"16285a87870c0cdddca86d7eaa20acfaf71a06b0","author":{"_account_id":25395,"name":"liyi","email":"liyi7155@fiberhome.com","username":"liyi8611"},"date":"2017-03-21 11:37:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nfix pep8","accounts_in_message":[],"_revision_number":1},{"id":"a9ca13755bfdf1d8cd3ea8e3aa6aee06ba7ec577","author":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"date":"2017-03-21 11:50:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"adc33bd3cd575599fdc22552ff62699cf1902955","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-03-21 12:23:30.000000000","message":"Patch Set 2: Code-Review-1\n\nwait for https://review.openstack.org/#/c/446762/ to merge","accounts_in_message":[],"_revision_number":2},{"id":"b3f747541b4f3e78ce2f4460fe7feac1f8f82296","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-21 13:21:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-trove-docs-ubuntu-xenial http://docs-draft.openstack.org/39/447839/2/check/gate-trove-docs-ubuntu-xenial/779e082//doc/build/html/ : SUCCESS in 4m 26s\n- gate-trove-pep8-ubuntu-xenial http://logs.openstack.org/39/447839/2/check/gate-trove-pep8-ubuntu-xenial/7c90f6c/ : FAILURE in 7m 19s\n- gate-trove-python27-ubuntu-xenial http://logs.openstack.org/39/447839/2/check/gate-trove-python27-ubuntu-xenial/f18cebd/ : FAILURE in 8m 30s\n- gate-trove-python35 http://logs.openstack.org/39/447839/2/check/gate-trove-python35/ff039a8/ : FAILURE in 2m 53s\n- gate-trove-functional-dsvm-mysql-ubuntu-xenial http://logs.openstack.org/39/447839/2/check/gate-trove-functional-dsvm-mysql-ubuntu-xenial/3e376f9/ : SUCCESS in 1h 22m 38s\n- gate-trove-scenario-dsvm-mariadb-single-ubuntu-xenial-nv http://logs.openstack.org/39/447839/2/check/gate-trove-scenario-dsvm-mariadb-single-ubuntu-xenial-nv/cacdbf6/ : SUCCESS in 1h 25m 44s (non-voting)\n- gate-trove-scenario-dsvm-mariadb-multi-ubuntu-xenial-nv http://logs.openstack.org/39/447839/2/check/gate-trove-scenario-dsvm-mariadb-multi-ubuntu-xenial-nv/8a44d3f/ : FAILURE in 1h 09m 29s (non-voting)\n- gate-trove-scenario-dsvm-mysql-single-ubuntu-xenial http://logs.openstack.org/39/447839/2/check/gate-trove-scenario-dsvm-mysql-single-ubuntu-xenial/7feccac/ : FAILURE in 1h 09m 35s\n- gate-trove-scenario-dsvm-mysql-multi-ubuntu-xenial http://logs.openstack.org/39/447839/2/check/gate-trove-scenario-dsvm-mysql-multi-ubuntu-xenial/eb5a783/ : SUCCESS in 54m 06s\n- gate-trove-scenario-dsvm-postgresql-single-ubuntu-xenial-nv http://logs.openstack.org/39/447839/2/check/gate-trove-scenario-dsvm-postgresql-single-ubuntu-xenial-nv/a4adc9c/ : SUCCESS in 1h 25m 00s (non-voting)\n- gate-trove-scenario-dsvm-postgresql-multi-ubuntu-xenial-nv http://logs.openstack.org/39/447839/2/check/gate-trove-scenario-dsvm-postgresql-multi-ubuntu-xenial-nv/35490ff/ : SUCCESS in 1h 04m 53s (non-voting)\n- gate-trove-scenario-dsvm-pxc-single-ubuntu-xenial-nv http://logs.openstack.org/39/447839/2/check/gate-trove-scenario-dsvm-pxc-single-ubuntu-xenial-nv/99d998c/ : SUCCESS in 50m 51s (non-voting)\n- gate-trove-scenario-dsvm-pxc-multi-ubuntu-xenial-nv http://logs.openstack.org/39/447839/2/check/gate-trove-scenario-dsvm-pxc-multi-ubuntu-xenial-nv/c3296be/ : SUCCESS in 39m 39s (non-voting)\n- gate-trove-scenario-dsvm-redis-single-ubuntu-xenial-nv http://logs.openstack.org/39/447839/2/check/gate-trove-scenario-dsvm-redis-single-ubuntu-xenial-nv/75d0948/ : SUCCESS in 1h 16m 32s (non-voting)\n- gate-trove-scenario-dsvm-redis-multi-ubuntu-xenial-nv http://logs.openstack.org/39/447839/2/check/gate-trove-scenario-dsvm-redis-multi-ubuntu-xenial-nv/e1f7b4c/ : SUCCESS in 1h 21m 52s (non-voting)\n- gate-trove-tox-fakemodetests-ubuntu-xenial http://logs.openstack.org/39/447839/2/check/gate-trove-tox-fakemodetests-ubuntu-xenial/aed8fd5/ : SUCCESS in 3m 44s\n- gate-trove-tox-apiexamples-ubuntu-xenial http://logs.openstack.org/39/447839/2/check/gate-trove-tox-apiexamples-ubuntu-xenial/c3fe8d1/ : SUCCESS in 2m 39s\n- gate-trove-pylint-ubuntu-xenial http://logs.openstack.org/39/447839/2/check/gate-trove-pylint-ubuntu-xenial/1624847/ : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"db2cfe60b69215f9126072f4a3f4781510da14cc","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-03-22 02:17:00.000000000","message":"Patch Set 2:\n\nThere is a lot more that must be done for this fix; this fix is woefully inadequate.","accounts_in_message":[],"_revision_number":2},{"id":"d9e06a8714a2f6a91fcaaea1bc655ee00cea4881","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-03-22 13:19:22.000000000","message":"Patch Set 2:\n\nsee https://review.openstack.org/448567","accounts_in_message":[],"_revision_number":2},{"id":"51da097106e705625e5220edba01ae400f772f27","author":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"date":"2017-03-23 02:10:44.000000000","message":"Patch Set 2:\n\n\u003e see https://review.openstack.org/448567\n\nI have searched all the codes in this trove, \nthere is nothing else to change but what I changed.\nAnd in your change, you delete \"_\", and haven\u0027t delete \"(\" and \")\".","accounts_in_message":[],"_revision_number":2},{"id":"2d82acc86b7ef45453a16b794bfeed2386d7b106","author":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"date":"2017-08-04 01:08:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fca9e8bf5bdeef226fb611b4127329c21dc1d6d0","revisions":{"5b744da0539f9e127d6d983dd12398197685ed71":{"kind":"REWORK","_number":1,"created":"2017-03-21 05:59:41.000000000","uploader":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"ref":"refs/changes/39/447839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/39/447839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/39/447839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/39/447839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/39/447839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/39/447839/1"}}},"commit":{"parents":[{"commit":"ee70b1d0e56537b2a78ab0ce50191754b6c0920c","subject":"Merge \"Remove redstack reference\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee70b1d0e56537b2a78ab0ce50191754b6c0920c"}]}],"author":{"name":"yfzhao","email":"dsware@126.com","date":"2017-03-21 05:57:24.000000000","tz":480},"committer":{"name":"yfzhao","email":"dsware@126.com","date":"2017-03-21 05:59:43.000000000","tz":480},"subject":"Remove log translations","message":"Remove log translations\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nSee:\nhttp://lists.openstack.org/pipermail/openstack-i18n/2016-November/002574.html\nhttp://lists.openstack.org/pipermail/openstack-dev/2017-March/113365.html\n\nChange-Id: I98d67b34b59ad47dc774afb16395861870dd09b7\nCloses-bug: #1674566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b744da0539f9e127d6d983dd12398197685ed71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b744da0539f9e127d6d983dd12398197685ed71"}]},"branch":"refs/heads/master"},"fca9e8bf5bdeef226fb611b4127329c21dc1d6d0":{"kind":"REWORK","_number":2,"created":"2017-03-21 11:50:42.000000000","uploader":{"_account_id":25067,"name":"yfzhao","email":"dsware@126.com","username":"yfzhao"},"ref":"refs/changes/39/447839/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/39/447839/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/39/447839/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/39/447839/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/39/447839/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/39/447839/2"}}},"commit":{"parents":[{"commit":"ee70b1d0e56537b2a78ab0ce50191754b6c0920c","subject":"Merge \"Remove redstack reference\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee70b1d0e56537b2a78ab0ce50191754b6c0920c"}]}],"author":{"name":"yfzhao","email":"dsware@126.com","date":"2017-03-21 05:57:24.000000000","tz":480},"committer":{"name":"yfzhao","email":"dsware@126.com","date":"2017-03-21 11:50:15.000000000","tz":480},"subject":"Remove log translations","message":"Remove log translations\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, _LC, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nSee:\nhttp://lists.openstack.org/pipermail/openstack-i18n/2016-November/002574.html\nhttp://lists.openstack.org/pipermail/openstack-dev/2017-March/113365.html\n\nChange-Id: I98d67b34b59ad47dc774afb16395861870dd09b7\nCloses-bug: #1674566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fca9e8bf5bdeef226fb611b4127329c21dc1d6d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fca9e8bf5bdeef226fb611b4127329c21dc1d6d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
