)]}'
{"id":"openstack%2Ftrove~502398","triplet_id":"openstack%2Ftrove~master~I75c0bfc350c8c21d4491d85a3d65a858645620d4","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-02-26 11:11:25.000000000","reason":"removed on reply"},"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2025-03-04 08:42:58.000000000","reason":"Change was submitted"},"26311":{"account":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"last_update":"2025-03-04 08:11:43.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"13562":{"account":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"last_update":"2025-02-22 18:29:16.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-02-26 16:45:45.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-03-04 07:35:16.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-03-04 08:11:13.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I75c0bfc350c8c21d4491d85a3d65a858645620d4","subject":"Change the default configuration value","status":"MERGED","created":"2017-09-11 06:48:10.000000000","updated":"2025-03-04 08:44:07.000000000","submitted":"2025-03-04 08:42:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"502398","meta_rev_id":"63835f509b01e4875c2de89de8d99d279bbe031a","_number":502398,"virtual_id_number":502398,"owner":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},{"value":0,"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},{"value":0,"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"value":0,"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-04 08:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-02-26 11:11:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},{"value":0,"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},{"value":0,"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"value":0,"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},{"value":1,"date":"2025-02-26 16:45:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-04 07:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-03-04 08:11:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},{"value":0,"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},{"value":0,"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"value":0,"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-04 07:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},{"value":0,"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},{"value":0,"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"value":0,"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},{"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-09-11 09:35:52.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-09-28 10:02:36.000000000","updated_by":{"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},"reviewer":{"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},"state":"REVIEWER"},{"updated":"2017-11-02 03:19:23.000000000","updated_by":{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},"reviewer":{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},"state":"REVIEWER"},{"updated":"2017-11-02 05:23:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-02 12:03:18.000000000","updated_by":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"reviewer":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"state":"REVIEWER"},{"updated":"2025-02-20 14:47:58.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2025-02-22 21:04:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-02-22 21:04:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-02-22 21:04:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-02-24 18:13:46.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REMOVED"},{"updated":"2025-02-24 18:13:46.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-02-24 18:14:03.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-02-26 11:13:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2025-03-04 08:11:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"f1ed37beb4e9144256197beabf738318c811de36","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-09-11 06:48:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75e37d3f9f38bd4d4d915ba72211ed1e487d6560","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-11 09:35:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-trove-docs-ubuntu-xenial http://docs-draft.openstack.org/98/502398/1/check/gate-trove-docs-ubuntu-xenial/a4a0d64//doc/build/html/ : SUCCESS in 3m 35s\n- gate-trove-pep8-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-pep8-ubuntu-xenial/6d2906f/ : SUCCESS in 10m 02s\n- gate-trove-python27-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-python27-ubuntu-xenial/968e011/ : SUCCESS in 9m 51s\n- gate-trove-python35 http://logs.openstack.org/98/502398/1/check/gate-trove-python35/4a0d290/ : SUCCESS in 11m 40s\n- gate-trove-functional-dsvm-mysql-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-functional-dsvm-mysql-ubuntu-xenial/187a841/ : FAILURE in 2h 36m 10s\n- gate-trove-scenario-dsvm-mariadb-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mariadb-single-ubuntu-xenial-nv/7b3db89/ : SUCCESS in 1h 42m 57s (non-voting)\n- gate-trove-scenario-dsvm-mariadb-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mariadb-multi-ubuntu-xenial-nv/b3f3f62/ : FAILURE in 1h 37m 48s (non-voting)\n- gate-trove-scenario-dsvm-mysql-single-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mysql-single-ubuntu-xenial/83bcd7b/ : FAILURE in 1h 42m 37s\n- gate-trove-scenario-dsvm-mysql-multi-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mysql-multi-ubuntu-xenial/f1c40e1/ : SUCCESS in 2h 02m 26s\n- gate-trove-scenario-dsvm-postgresql-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-postgresql-single-ubuntu-xenial-nv/a9566ed/ : SUCCESS in 2h 27m 01s (non-voting)\n- gate-trove-scenario-dsvm-postgresql-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-postgresql-multi-ubuntu-xenial-nv/6a004a8/ : SUCCESS in 1h 55m 08s (non-voting)\n- gate-trove-scenario-dsvm-pxc-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-pxc-single-ubuntu-xenial-nv/c45b010/ : SUCCESS in 1h 55m 11s (non-voting)\n- gate-trove-scenario-dsvm-pxc-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-pxc-multi-ubuntu-xenial-nv/238caf2/ : SUCCESS in 42m 36s (non-voting)\n- gate-trove-scenario-dsvm-redis-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-redis-single-ubuntu-xenial-nv/3f8cca4/ : SUCCESS in 1h 22m 18s (non-voting)\n- gate-trove-scenario-dsvm-redis-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-redis-multi-ubuntu-xenial-nv/0382c5b/ : SUCCESS in 1h 26m 43s (non-voting)\n- gate-trove-tox-fakemodetests-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-tox-fakemodetests-ubuntu-xenial/bf76b3c/ : SUCCESS in 4m 21s\n- gate-trove-tox-apiexamples-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-tox-apiexamples-ubuntu-xenial/7c074c8/ : SUCCESS in 3m 43s\n- gate-trove-pylint-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-pylint-ubuntu-xenial/a9e6018/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5e9a37af524393dc41897b2cb6d2afe17c207aa6","author":{"_account_id":25257,"name":"limeng","email":"limeng@awcloud.com","username":"Linmeng"},"date":"2017-09-28 10:02:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7b775650d5a795b5becfdeef1fd1ba9e4c893827","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-10-13 10:55:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3cde9a035d5640e2239319bd5f833dc907afc526","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-13 12:57:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-trove-docs-ubuntu-xenial http://docs-draft.openstack.org/98/502398/1/check/gate-trove-docs-ubuntu-xenial/e2f8919//doc/build/html/ : SUCCESS in 6m 00s\n- gate-trove-pep8-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-pep8-ubuntu-xenial/fb3d75b/ : SUCCESS in 11m 25s\n- gate-trove-python27-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-python27-ubuntu-xenial/9312686/ : FAILURE in 11m 20s\n- gate-trove-python35 http://logs.openstack.org/98/502398/1/check/gate-trove-python35/2f987bb/ : FAILURE in 11m 29s\n- gate-trove-functional-dsvm-mysql-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-functional-dsvm-mysql-ubuntu-xenial/7419d0f/ : FAILURE in 38m 40s\n- gate-trove-scenario-dsvm-mariadb-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mariadb-single-ubuntu-xenial-nv/22b3eb0/ : FAILURE in 57m 52s (non-voting)\n- gate-trove-scenario-dsvm-mariadb-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mariadb-multi-ubuntu-xenial-nv/f555fe9/ : FAILURE in 39m 53s (non-voting)\n- gate-trove-scenario-dsvm-mysql-single-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mysql-single-ubuntu-xenial/2dcb5a4/ : FAILURE in 37m 39s\n- gate-trove-scenario-dsvm-mysql-multi-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mysql-multi-ubuntu-xenial/664b75d/ : FAILURE in 34m 35s\n- gate-trove-scenario-dsvm-postgresql-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-postgresql-single-ubuntu-xenial-nv/569d7f5/ : FAILURE in 37m 53s (non-voting)\n- gate-trove-scenario-dsvm-postgresql-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-postgresql-multi-ubuntu-xenial-nv/4523f19/ : FAILURE in 48m 52s (non-voting)\n- gate-trove-scenario-dsvm-pxc-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-pxc-single-ubuntu-xenial-nv/68b71f1/ : FAILURE in 34m 39s (non-voting)\n- gate-trove-scenario-dsvm-pxc-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-pxc-multi-ubuntu-xenial-nv/e0f2482/ : SUCCESS in 37m 41s (non-voting)\n- gate-trove-scenario-dsvm-redis-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-redis-single-ubuntu-xenial-nv/5c1dab0/ : FAILURE in 35m 35s (non-voting)\n- gate-trove-scenario-dsvm-redis-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-redis-multi-ubuntu-xenial-nv/5aa2428/ : FAILURE in 37m 08s (non-voting)\n- gate-trove-tox-fakemodetests-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-tox-fakemodetests-ubuntu-xenial/fa4aa1b/ : SUCCESS in 5m 19s\n- gate-trove-tox-apiexamples-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-tox-apiexamples-ubuntu-xenial/2fcdc2d/ : SUCCESS in 4m 27s\n- gate-trove-pylint-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-pylint-ubuntu-xenial/601d789/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"83d307f755ba8ad25cfbd341a9dd257efa239a48","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-10-13 13:00:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"123e001929e3e5281ca6b052bd3b70c1e9535859","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-13 13:48:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-trove-docs-ubuntu-xenial http://docs-draft.openstack.org/98/502398/1/check/gate-trove-docs-ubuntu-xenial/978082f//doc/build/html/ : SUCCESS in 5m 39s\n- gate-trove-pep8-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-pep8-ubuntu-xenial/e18625f/ : SUCCESS in 13m 39s\n- gate-trove-python27-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-python27-ubuntu-xenial/16d115e/ : FAILURE in 9m 41s\n- gate-trove-python35 http://logs.openstack.org/98/502398/1/check/gate-trove-python35/bf839e7/ : FAILURE in 10m 34s\n- gate-trove-functional-dsvm-mysql-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-functional-dsvm-mysql-ubuntu-xenial/e0955fe/ : FAILURE in 36m 40s\n- gate-trove-scenario-dsvm-mariadb-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mariadb-single-ubuntu-xenial-nv/4a4c96a/ : FAILURE in 47m 33s (non-voting)\n- gate-trove-scenario-dsvm-mariadb-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mariadb-multi-ubuntu-xenial-nv/d7c7405/ : FAILURE in 44m 13s (non-voting)\n- gate-trove-scenario-dsvm-mysql-single-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mysql-single-ubuntu-xenial/4ba5416/ : FAILURE in 44m 41s\n- gate-trove-scenario-dsvm-mysql-multi-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-mysql-multi-ubuntu-xenial/3465609/ : FAILURE in 41m 40s\n- gate-trove-scenario-dsvm-postgresql-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-postgresql-single-ubuntu-xenial-nv/1dd7854/ : FAILURE in 32m 30s (non-voting)\n- gate-trove-scenario-dsvm-postgresql-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-postgresql-multi-ubuntu-xenial-nv/027b566/ : FAILURE in 44m 33s (non-voting)\n- gate-trove-scenario-dsvm-pxc-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-pxc-single-ubuntu-xenial-nv/c4dcd99/ : FAILURE in 35m 16s (non-voting)\n- gate-trove-scenario-dsvm-pxc-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-pxc-multi-ubuntu-xenial-nv/a91052e/ : SUCCESS in 37m 20s (non-voting)\n- gate-trove-scenario-dsvm-redis-single-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-redis-single-ubuntu-xenial-nv/a7f6aae/ : FAILURE in 44m 29s (non-voting)\n- gate-trove-scenario-dsvm-redis-multi-ubuntu-xenial-nv http://logs.openstack.org/98/502398/1/check/gate-trove-scenario-dsvm-redis-multi-ubuntu-xenial-nv/61717c6/ : FAILURE in 35m 08s (non-voting)\n- gate-trove-tox-fakemodetests-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-tox-fakemodetests-ubuntu-xenial/a5612ac/ : SUCCESS in 5m 09s\n- gate-trove-tox-apiexamples-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-tox-apiexamples-ubuntu-xenial/e83828f/ : SUCCESS in 7m 10s\n- gate-trove-pylint-ubuntu-xenial http://logs.openstack.org/98/502398/1/check/gate-trove-pylint-ubuntu-xenial/f04fce2/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5a60533b61b74505bf091eb88b3b089479cd38ac","author":{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},"date":"2017-10-16 03:36:02.000000000","message":"Patch Set 1:\n\nI see your bug description and commit-msg is not the same, please modify, so that we can understand the reasons for such changes, and reproduce the wrong way.","accounts_in_message":[],"_revision_number":1},{"id":"4cfa38b3a1645ceb956cde23a12fedbdf24c9276","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-10-17 01:33:52.000000000","message":"Patch Set 1:\n\n\u003e I see your bug description and commit-msg is not the same, please\n \u003e modify, so that we can understand the reasons for such changes, and\n \u003e reproduce the wrong way.\nThis bug description is not raised by me, and I had add messages in the comments","accounts_in_message":[],"_revision_number":1},{"id":"8f989e67d58ecc0e5c3b842aa052b68ba676d99f","author":{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},"date":"2017-10-17 02:14:00.000000000","message":"Patch Set 1:\n\n\u003e \u003e I see your bug description and commit-msg is not the same, please\n \u003e \u003e modify, so that we can understand the reasons for such changes,\n \u003e and\n \u003e \u003e reproduce the wrong way.\n \u003e This bug description is not raised by me, and I had add messages in\n \u003e the comments\n\nI understand this problem. It seems that in the environment where neutron is used to provide the network, network_label_regex is set to. *, And ^ private $ is used in other network environments. I was thinking about whether this would need to be changed, and in the puppet deployment environment, the value was modified during the deployment process. Which in itself depends only on whether the user has used nova-network or neutron.Keep watch and discuss","accounts_in_message":[],"_revision_number":1},{"id":"9aaf040d1e81468b780f7087970f834397f58a6b","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-10-26 08:00:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e6008101303f19a9f90db65fec22785ee07617a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-26 10:40:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/98/502398/1/check/openstack-tox-pep8/4921881/ : SUCCESS in 8m 06s\n- openstack-tox-py27 http://logs.openstack.org/98/502398/1/check/openstack-tox-py27/f6c5a2b/ : SUCCESS in 11m 18s\n- openstack-tox-py35 http://logs.openstack.org/98/502398/1/check/openstack-tox-py35/3ae0c64/ : SUCCESS in 13m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/502398/1/check/build-openstack-sphinx-docs/d408c44/html/ : SUCCESS in 5m 31s\n- legacy-trove-functional-dsvm-mysql http://logs.openstack.org/98/502398/1/check/legacy-trove-functional-dsvm-mysql/641cef6/ : SUCCESS in 2h 09m 11s\n- legacy-trove-scenario-dsvm-mariadb-single http://logs.openstack.org/98/502398/1/check/legacy-trove-scenario-dsvm-mariadb-single/261cc20/ : SUCCESS in 1h 07m 52s (non-voting)\n- legacy-trove-scenario-dsvm-mariadb-multi http://logs.openstack.org/98/502398/1/check/legacy-trove-scenario-dsvm-mariadb-multi/5ae5836/ : FAILURE in 1h 16m 07s (non-voting)\n- legacy-trove-scenario-dsvm-mysql-single http://logs.openstack.org/98/502398/1/check/legacy-trove-scenario-dsvm-mysql-single/17684ea/ : SUCCESS in 1h 56m 11s\n- legacy-trove-scenario-dsvm-mysql-multi http://logs.openstack.org/98/502398/1/check/legacy-trove-scenario-dsvm-mysql-multi/ac6696d/ : SUCCESS in 1h 56m 17s\n- legacy-trove-scenario-dsvm-postgresql-single http://logs.openstack.org/98/502398/1/check/legacy-trove-scenario-dsvm-postgresql-single/dfb9b85/ : FAILURE in 2h 36m 48s (non-voting)\n- legacy-trove-scenario-dsvm-postgresql-multi http://logs.openstack.org/98/502398/1/check/legacy-trove-scenario-dsvm-postgresql-multi/c0ef970/ : FAILURE in 36m 53s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-single http://logs.openstack.org/98/502398/1/check/legacy-trove-scenario-dsvm-pxc-single/246bdf6/ : SUCCESS in 1h 50m 26s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-multi http://logs.openstack.org/98/502398/1/check/legacy-trove-scenario-dsvm-pxc-multi/495d3bb/ : SUCCESS in 37m 51s (non-voting)\n- legacy-trove-scenario-dsvm-redis-single http://logs.openstack.org/98/502398/1/check/legacy-trove-scenario-dsvm-redis-single/993576d/ : SUCCESS in 1h 25m 34s (non-voting)\n- legacy-trove-scenario-dsvm-redis-multi http://logs.openstack.org/98/502398/1/check/legacy-trove-scenario-dsvm-redis-multi/1b169bb/ : SUCCESS in 1h 08m 14s (non-voting)\n- legacy-trove-tox-fakemodetests http://logs.openstack.org/98/502398/1/check/legacy-trove-tox-fakemodetests/e61c14a/ : SUCCESS in 6m 08s\n- legacy-trove-tox-apiexamples http://logs.openstack.org/98/502398/1/check/legacy-trove-tox-apiexamples/ca91e45/ : SUCCESS in 8m 53s\n- openstack-tox-pylint http://logs.openstack.org/98/502398/1/check/openstack-tox-pylint/1b315a3/ : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a6b33ec9c6d5abfe5736b4ea11bfd0ad46153c32","author":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"date":"2017-11-01 16:59:20.000000000","message":"Patch Set 1: Code-Review-2\n\nThis code should not merge.\n\nThe bug was marked invalid by Amrith before it was reopened for this proposed fix.  The issue is occurring because of a misconfiguration of Trove.  The fix is to properly configure the network_label_regex config property, not to completely open it up with .*.\n\nThe one fix I could see for this bug is what was suggested in the bug description, check for None and raise a better error message than index out of range.","accounts_in_message":[],"_revision_number":1},{"id":"5ca90a20f6846c8278bb77de33fd74d596390a21","author":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"date":"2017-11-01 18:13:02.000000000","message":"Patch Set 1: Code-Review-1\n\nI made a mistake.  I still hold this is a configuration issue. However, the sample conf does document that the value should be set to .* when you are using Neutron.\n\nNova-network is deprecated.  So these changes could probably go in and resolve common configuration errors.\n\nThe bug should still be invalidated, the commit message should not list it as closes, and the commit message should say this change is to change the default configuration value because Nova network (the current value), is deprecated.","accounts_in_message":[],"_revision_number":1},{"id":"63a760d6e895d8030765f519584b976a3e827cf4","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-11-02 02:15:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"19dd9450aeaf802524cfe049aaa673f21d661849","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-11-02 02:17:23.000000000","message":"Topic bug/1627565 removed","accounts_in_message":[],"_revision_number":2},{"id":"324700eca1bb1f1b0b551273f8dd2b0e41761715","author":{"_account_id":22694,"name":"jian.song","email":"jiansong19941206@gmail.com","username":"jiansong"},"date":"2017-11-02 03:19:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"85f7062907f81390fbf73f18265a9634d8e513c4","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-11-02 03:25:16.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ecd071258b2f339691fb2dc4757c31e6aeb0cb44","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-11-02 03:32:49.000000000","message":"Patch Set 3:\n\n(1 comment)\n\n\u003e (1 comment)\n\nI think it can be deleted","accounts_in_message":[],"_revision_number":3},{"id":"ebcf195fad6603b4903710b8930c593244db1736","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-02 05:23:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/502398/3/check/openstack-tox-pep8/646cc7e/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/98/502398/3/check/openstack-tox-py27/c11cfab/ : FAILURE in 10m 43s\n- openstack-tox-py35 http://logs.openstack.org/98/502398/3/check/openstack-tox-py35/25e1bea/ : FAILURE in 11m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/502398/3/check/build-openstack-sphinx-docs/2da5450/html/ : SUCCESS in 6m 03s\n- legacy-trove-functional-dsvm-mysql http://logs.openstack.org/98/502398/3/check/legacy-trove-functional-dsvm-mysql/f1a5689/ : SUCCESS in 1h 30m 45s\n- legacy-trove-scenario-dsvm-mariadb-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mariadb-single/f6a96eb/ : SUCCESS in 1h 25m 58s (non-voting)\n- legacy-trove-scenario-dsvm-mariadb-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mariadb-multi/12bc007/ : FAILURE in 1h 31m 28s (non-voting)\n- legacy-trove-scenario-dsvm-mysql-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mysql-single/a186428/ : SUCCESS in 1h 36m 44s\n- legacy-trove-scenario-dsvm-mysql-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mysql-multi/e6b7ce4/ : SUCCESS in 1h 55m 46s\n- legacy-trove-scenario-dsvm-postgresql-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-postgresql-single/075e604/ : FAILURE in 33m 49s (non-voting)\n- legacy-trove-scenario-dsvm-postgresql-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-postgresql-multi/020e004/ : FAILURE in 32m 10s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-pxc-single/f0fe94f/ : SUCCESS in 1h 45m 57s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-pxc-multi/9564c34/ : SUCCESS in 39m 53s (non-voting)\n- legacy-trove-scenario-dsvm-redis-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-redis-single/42e39cd/ : FAILURE in 1h 02m 18s (non-voting)\n- legacy-trove-scenario-dsvm-redis-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-redis-multi/de47b0e/ : SUCCESS in 1h 22m 16s (non-voting)\n- legacy-trove-tox-fakemodetests http://logs.openstack.org/98/502398/3/check/legacy-trove-tox-fakemodetests/9193188/ : FAILURE in 7m 46s\n- legacy-trove-tox-apiexamples http://logs.openstack.org/98/502398/3/check/legacy-trove-tox-apiexamples/5b8112b/ : FAILURE in 5m 26s\n- openstack-tox-pylint http://logs.openstack.org/98/502398/3/check/openstack-tox-pylint/2f70b88/ : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":3},{"id":"c3c5833339038efbb256c39938709d63a5e84588","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-11-02 05:32:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7312dd7f2348884d68713a7004493d6f2731e133","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-02 08:08:57.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/502398/3/check/openstack-tox-pep8/a481499/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/98/502398/3/check/openstack-tox-py27/157bac8/ : FAILURE in 12m 34s\n- openstack-tox-py35 http://logs.openstack.org/98/502398/3/check/openstack-tox-py35/948913e/ : FAILURE in 12m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/502398/3/check/build-openstack-sphinx-docs/d31b115/html/ : SUCCESS in 4m 59s\n- legacy-trove-functional-dsvm-mysql http://logs.openstack.org/98/502398/3/check/legacy-trove-functional-dsvm-mysql/6756422/ : SUCCESS in 2h 29m 56s\n- legacy-trove-scenario-dsvm-mariadb-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mariadb-single/c6ac4dd/ : SUCCESS in 1h 17m 38s (non-voting)\n- legacy-trove-scenario-dsvm-mariadb-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mariadb-multi/e39681e/ : FAILURE in 1h 27m 20s (non-voting)\n- legacy-trove-scenario-dsvm-mysql-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mysql-single/3fbc511/ : SUCCESS in 1h 23m 43s\n- legacy-trove-scenario-dsvm-mysql-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mysql-multi/95ae4d4/ : SUCCESS in 1h 58m 33s\n- legacy-trove-scenario-dsvm-postgresql-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-postgresql-single/b8197e0/ : SUCCESS in 2h 19m 52s (non-voting)\n- legacy-trove-scenario-dsvm-postgresql-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-postgresql-multi/75d8162/ : SUCCESS in 1h 02m 40s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-pxc-single/d9f88f5/ : SUCCESS in 1h 47m 31s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-pxc-multi/28742ee/ : SUCCESS in 1h 18m 28s (non-voting)\n- legacy-trove-scenario-dsvm-redis-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-redis-single/1487af5/ : FAILURE in 1h 06m 27s (non-voting)\n- legacy-trove-scenario-dsvm-redis-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-redis-multi/78fa0b6/ : SUCCESS in 1h 32m 36s (non-voting)\n- legacy-trove-tox-fakemodetests http://logs.openstack.org/98/502398/3/check/legacy-trove-tox-fakemodetests/fe96d66/ : FAILURE in 5m 15s\n- legacy-trove-tox-apiexamples http://logs.openstack.org/98/502398/3/check/legacy-trove-tox-apiexamples/48872a3/ : FAILURE in 4m 35s\n- openstack-tox-pylint http://logs.openstack.org/98/502398/3/check/openstack-tox-pylint/61c8ab3/ : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":3},{"id":"74b3aa13011c75a8983fa35b3b27dcd415d6c6e9","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2017-11-02 11:22:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7993ade9ad2da8454c9b81892fb089ee140e08d6","author":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"date":"2017-11-02 12:03:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"57ec63533537b35eccf3986dda55ab11e2acd271","author":{"_account_id":13562,"name":"Samuel Matzek","email":"smatzek@us.ibm.com","username":"smatzek"},"date":"2017-11-02 12:34:10.000000000","message":"Patch Set 3:\n\nAfter more thought I think there is more work to do here.\n\nIf a deployment is using nova network, does not have the value set in the conf file because it\u0027s the default, and then upgrades to Queens they will be broken.\n\nAt a minimum we will need a release note, and we also need to look into what is normally done in situations like this, i.e. do we update the config to put the old default value in?","accounts_in_message":[],"_revision_number":3},{"id":"3394927d4285e5234be3d461110f5df094555449","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-02 13:41:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/502398/3/check/openstack-tox-pep8/46d1c02/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/98/502398/3/check/openstack-tox-py27/566aa9c/ : FAILURE in 10m 41s\n- openstack-tox-py35 http://logs.openstack.org/98/502398/3/check/openstack-tox-py35/7ec05ad/ : FAILURE in 12m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/502398/3/check/build-openstack-sphinx-docs/34996ee/html/ : SUCCESS in 5m 57s\n- legacy-trove-functional-dsvm-mysql http://logs.openstack.org/98/502398/3/check/legacy-trove-functional-dsvm-mysql/1039a0b/ : SUCCESS in 2h 16m 05s\n- legacy-trove-scenario-dsvm-mariadb-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mariadb-single/845fecf/ : SUCCESS in 1h 36m 15s (non-voting)\n- legacy-trove-scenario-dsvm-mariadb-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mariadb-multi/c189a9a/ : FAILURE in 1h 37m 16s (non-voting)\n- legacy-trove-scenario-dsvm-mysql-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mysql-single/871b95e/ : SUCCESS in 1h 19m 53s\n- legacy-trove-scenario-dsvm-mysql-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-mysql-multi/b281ec3/ : SUCCESS in 2h 06m 58s\n- legacy-trove-scenario-dsvm-postgresql-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-postgresql-single/9822a1f/ : FAILURE in 47m 30s (non-voting)\n- legacy-trove-scenario-dsvm-postgresql-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-postgresql-multi/59a8744/ : FAILURE in 55m 19s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-pxc-single/99d99d4/ : SUCCESS in 1h 46m 24s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-pxc-multi/64b8470/ : SUCCESS in 36m 57s (non-voting)\n- legacy-trove-scenario-dsvm-redis-single http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-redis-single/97a9da4/ : FAILURE in 52m 03s (non-voting)\n- legacy-trove-scenario-dsvm-redis-multi http://logs.openstack.org/98/502398/3/check/legacy-trove-scenario-dsvm-redis-multi/a7d94e6/ : SUCCESS in 57m 21s (non-voting)\n- legacy-trove-tox-fakemodetests http://logs.openstack.org/98/502398/3/check/legacy-trove-tox-fakemodetests/70ced60/ : FAILURE in 4m 26s\n- legacy-trove-tox-apiexamples http://logs.openstack.org/98/502398/3/check/legacy-trove-tox-apiexamples/3368fc9/ : FAILURE in 4m 44s\n- openstack-tox-pylint http://logs.openstack.org/98/502398/3/check/openstack-tox-pylint/fb8f9d6/ : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":3},{"id":"695d4e878717b40d96a9e84cf02fbce7d0e36021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-22 18:20:22.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9dd4ad0fb739ebc25dd4089e80834ee9ac759977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 18:21:02.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 502398,4","accounts_in_message":[],"_revision_number":4},{"id":"1c4011ac88087112bd6ebf6267af78b2c288a354","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 18:21:02.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 502398,4","accounts_in_message":[],"_revision_number":4},{"id":"c15adf6482cda6c20143f4ecbba17499129e5523","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-22 18:26:10.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a0756545ec06c360743b3010b11d5981857732a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 18:26:24.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 502398,5","accounts_in_message":[],"_revision_number":5},{"id":"7a3ef68f846397b8581cb65ee91cd7abdc5e25c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 18:26:24.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 502398,5","accounts_in_message":[],"_revision_number":5},{"id":"1df61d28bad62d5a50e5c7d9278bc9400254f762","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-22 18:27:07.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cfa1690df745211b8f273af75f4db0302e22ef5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-22 18:27:10.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"2eef1534f32da9485d0fa4dfd176de76fef18825","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-22 18:29:00.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"dd6d25676c7a52efe1b9a1b36df7608376463467","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 18:42:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a56a1f9dc87c4bff90b34e1d2406ec55\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c40da8d347454c5fbe392d03ef294554 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67e83e30a6274077b467c73c3bdd7976 : SUCCESS in 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a0699cd9617f273b868ce5663369d71fc0ac3e0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-22 18:42:39.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"e92e10ca9af23bb602d098d4c0dfa4ecc2115cdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-22 18:43:13.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"33109114ec74725299b488d94e5e4b4eec10ee07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 18:57:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dddaf83f2354cb6b0197e68efeccf5c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1440d2852ca445b3a89015324abfbdea : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f284ce0c222744999eb9117744f62e88 : SUCCESS in 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fc4efe8a5bab4e07c3927c401bf49e37b4cbfe66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 20:47:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7909e2d4104f470887f4c89d36608bd9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66a1c767c8534796a76fdfa147071de4 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe1625d3bdd4738a9d9dc7217b22c75 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1876801a7d14ad6b6bb86d6056c8e19 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e421edf3374b4ef3bae9b9966b50cd31 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f866cd68cde4be8865e4216813c72a1 : SUCCESS in 4m 01s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/13d54f72b9344685a870bc4a3974a0c4 : SUCCESS in 32m 29s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/251b627dc8b44fb0be4e233379e24e38 : SUCCESS in 40m 24s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/5b534af4a5384a22aecc0cfcf944e828 : SUCCESS in 1h 25m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/66e8166a2a344005bbe2d960ac865d54 : SUCCESS in 1h 23m 53s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/68b093c5a1f24fd5bf060a719be5d93f : SUCCESS in 37m 10s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f7e93c04090a408da2524998e19d9516 : SUCCESS in 1h 33m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8fdd09e6938c4ea19bb63e4681962f73 : SUCCESS in 1h 48m 36s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/44d66b27d06b4fabbdbb2a62a805f2d6 : SUCCESS in 1h 04m 05s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4ff6403e6eb14f6a93a40cb15b9b3236 : SUCCESS in 2h 01m 32s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/44b10aeaddad4211b0ffada636e2db37 : SUCCESS in 1h 42m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/052b5733cdd049b7bd5555dd481d10d1 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":10},{"id":"990148935b0dea2be2de1c75f587607cc0b95322","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-24 18:13:46.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_37598\u003e.","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":10},{"id":"68a80fb968c9f2cc026151092d01cb562e7afbfe","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-02-26 11:11:25.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"a012865811eabf80580930255665072a4f0510f5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-26 16:45:45.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"ff165caa3493013ebcf302f7aedba856d0f5f3bc","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-03-04 07:35:16.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5377f052f2e542cbe1a0611c610e790368d8ebf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 07:35:45.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"8f26fbd3acecae8f23bf8def0420e8609f8510ca","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-04 08:11:32.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"8f8f8230a911644e14d4142a1e501c815bb80480","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 08:42:58.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2476f49ba8154c7e963bff3690a9d47d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e85987aecea848d48478f131a5c69388 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d601a71dd2b475fb62f682274e15efc : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3c83238ff0d402f863d2cb4343069ec : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b18bd74f6e1c4600a9c61b4f9bd666d2 : SUCCESS in 4m 15s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/559c25902fc0444ca9163b99e3b9ea0b : SUCCESS in 57m 29s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/30ec190c7df74d2e8e40f1cb1054dcb2 : SUCCESS in 58m 39s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/94136f3dd2f64fae90c4af8dc111680b : SUCCESS in 53m 38s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b54a0914d60b42d18a7e3a9cc4613995 : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":10},{"id":"59f82fc3bc019dc349df3a61d42c54b4a92d287e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 08:42:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"63835f509b01e4875c2de89de8d99d279bbe031a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 08:44:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64abd94c3d2a47fea9cf86c74a1bf1e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372280e931eb4ac193619ed96f34df4a : SUCCESS in 49s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a94e7f390df324f93bb4ab948be4eee78b44deae","revisions":{"2ecb45c3ca95785969a7d4ed3b6ca54dbca0fd7f":{"kind":"REWORK","_number":1,"created":"2017-09-11 06:48:10.000000000","uploader":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"ref":"refs/changes/98/502398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/502398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/502398/1"}}},"commit":{"parents":[{"commit":"b0175ff03acd157d740b4eb4854f959c06bd2677","subject":"Creating cluster in multiple networks environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0175ff03acd157d740b4eb4854f959c06bd2677"}]}],"author":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"committer":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"subject":"Mongodb cluster cause error:list index out of range","message":"Mongodb cluster cause error:list index out of range\n\nWhen I create a mongodb cluster, if I assign a network to the member\ninstance, and the network name is diffrent from that defined in the conf\nand cfg.py file, it will cause the error: list index out of range\n\nChange-Id: I75c0bfc350c8c21d4491d85a3d65a858645620d4\nCloses-bug: #1627565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2ecb45c3ca95785969a7d4ed3b6ca54dbca0fd7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2ecb45c3ca95785969a7d4ed3b6ca54dbca0fd7f"}]},"branch":"refs/heads/master"},"22c80f5ec0ff2f0379cbafe5a26eca7bc04ff289":{"kind":"REWORK","_number":2,"created":"2017-11-02 02:15:47.000000000","uploader":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"ref":"refs/changes/98/502398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/502398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/502398/2"}}},"commit":{"parents":[{"commit":"b0175ff03acd157d740b4eb4854f959c06bd2677","subject":"Creating cluster in multiple networks environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0175ff03acd157d740b4eb4854f959c06bd2677"}]}],"author":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"committer":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-11-02 02:15:21.000000000","tz":0},"subject":"Change the default configuration value","message":"Change the default configuration value\n\nNova-network will be deprecated, this change is to change the default \nconfiguration value because Nova network(the current value), is \ndeprecated\n\nChange-Id: I75c0bfc350c8c21d4491d85a3d65a858645620d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/22c80f5ec0ff2f0379cbafe5a26eca7bc04ff289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/22c80f5ec0ff2f0379cbafe5a26eca7bc04ff289"}]},"branch":"refs/heads/master"},"e329dcce45f13543e4843444a95b308794efbccf":{"kind":"REWORK","_number":3,"created":"2017-11-02 03:25:16.000000000","uploader":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"ref":"refs/changes/98/502398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/502398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/502398/3"}}},"commit":{"parents":[{"commit":"b0175ff03acd157d740b4eb4854f959c06bd2677","subject":"Creating cluster in multiple networks environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0175ff03acd157d740b4eb4854f959c06bd2677"}]}],"author":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"committer":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-11-02 03:23:54.000000000","tz":0},"subject":"Change the default configuration value","message":"Change the default configuration value\n\nNova-network will be deprecated, this change is to change the default \nconfiguration value because Nova network(the current value), is \ndeprecated\n\nChange-Id: I75c0bfc350c8c21d4491d85a3d65a858645620d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e329dcce45f13543e4843444a95b308794efbccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e329dcce45f13543e4843444a95b308794efbccf"}]},"branch":"refs/heads/master"},"cbe16a368e16a63b86c2da5b81b0a0bbdff47dd6":{"kind":"REWORK","_number":4,"created":"2025-02-22 18:20:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/502398/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/502398/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/502398/4"}}},"commit":{"parents":[{"commit":"b0175ff03acd157d740b4eb4854f959c06bd2677","subject":"Creating cluster in multiple networks environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0175ff03acd157d740b4eb4854f959c06bd2677"}]}],"author":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-22 18:20:12.000000000","tz":0},"subject":"Change the default configuration value","message":"Change the default configuration value\n\nNova-network will be deprecated, this change is to change the default \nconfiguration value because Nova network(the current value), is \ndeprecated\n\nChange-Id: I75c0bfc350c8c21d4491d85a3d65a858645620d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cbe16a368e16a63b86c2da5b81b0a0bbdff47dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cbe16a368e16a63b86c2da5b81b0a0bbdff47dd6"}]},"branch":"refs/heads/master"},"e5a3921e3156818c8bc61bf0b0b3f2e349eb5f63":{"kind":"NO_CHANGE","_number":5,"created":"2025-02-22 18:26:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/502398/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/502398/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/502398/5"}}},"commit":{"parents":[{"commit":"b0175ff03acd157d740b4eb4854f959c06bd2677","subject":"Creating cluster in multiple networks environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0175ff03acd157d740b4eb4854f959c06bd2677"}]}],"author":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-22 18:26:09.000000000","tz":0},"subject":"Change the default configuration value","message":"Change the default configuration value\n\nNova-network will be deprecated, this change is to change the default \nconfiguration value because Nova network(the current value), is \ndeprecated\n\nChange-Id: I75c0bfc350c8c21d4491d85a3d65a858645620d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e5a3921e3156818c8bc61bf0b0b3f2e349eb5f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e5a3921e3156818c8bc61bf0b0b3f2e349eb5f63"}]},"branch":"refs/heads/master"},"cd40c7d768f9024d24e26c924fd2f28418b394dd":{"kind":"REWORK","_number":6,"created":"2025-02-22 18:27:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/502398/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/502398/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/502398/6"}}},"commit":{"parents":[{"commit":"b0175ff03acd157d740b4eb4854f959c06bd2677","subject":"Creating cluster in multiple networks environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0175ff03acd157d740b4eb4854f959c06bd2677"}]}],"author":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-22 18:27:02.000000000","tz":0},"subject":"Change the default configuration value","message":"Change the default configuration value\n\nNova-network will be deprecated, this change is to change the default \nconfiguration value because Nova network(the current value), is \ndeprecated\n\nChange-Id: I75c0bfc350c8c21d4491d85a3d65a858645620d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd40c7d768f9024d24e26c924fd2f28418b394dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd40c7d768f9024d24e26c924fd2f28418b394dd"}]},"branch":"refs/heads/master"},"a2d3372cac4f652726dc458e46604f4e05df2260":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-02-22 18:27:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/502398/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/502398/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/502398/7"}}},"commit":{"parents":[{"commit":"2c4f4c33ff63d2fb3b5acd2b885410d6108fb806","subject":"Merge \"Install and configure the Trove dashboard - several improvements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2c4f4c33ff63d2fb3b5acd2b885410d6108fb806"}]}],"author":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-22 18:27:10.000000000","tz":0},"subject":"Change the default configuration value","message":"Change the default configuration value\n\nNova-network will be deprecated, this change is to change the default \nconfiguration value because Nova network(the current value), is \ndeprecated\n\nChange-Id: I75c0bfc350c8c21d4491d85a3d65a858645620d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2d3372cac4f652726dc458e46604f4e05df2260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2d3372cac4f652726dc458e46604f4e05df2260"}]},"branch":"refs/heads/master","description":"Rebase"},"27fa8787ca8a4fdb0e6c1871f981cd3e6e63dd6f":{"kind":"REWORK","_number":8,"created":"2025-02-22 18:29:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/502398/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/502398/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/502398/8"}}},"commit":{"parents":[{"commit":"2c4f4c33ff63d2fb3b5acd2b885410d6108fb806","subject":"Merge \"Install and configure the Trove dashboard - several improvements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2c4f4c33ff63d2fb3b5acd2b885410d6108fb806"}]}],"author":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-22 18:29:00.000000000","tz":0},"subject":"Change the default configuration value","message":"Change the default configuration value\n\nNova-network will be deprecated, this change is to change the default \nconfiguration value because Nova network(the current value), is \ndeprecated\n\nChange-Id: I75c0bfc350c8c21d4491d85a3d65a858645620d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/27fa8787ca8a4fdb0e6c1871f981cd3e6e63dd6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/27fa8787ca8a4fdb0e6c1871f981cd3e6e63dd6f"}]},"branch":"refs/heads/master"},"951e76359a19a5cab23a4e5fa72be19cdc0fafed":{"kind":"REWORK","_number":9,"created":"2025-02-22 18:42:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/502398/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/502398/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/502398/9"}}},"commit":{"parents":[{"commit":"2c4f4c33ff63d2fb3b5acd2b885410d6108fb806","subject":"Merge \"Install and configure the Trove dashboard - several improvements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2c4f4c33ff63d2fb3b5acd2b885410d6108fb806"}]}],"author":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-22 18:42:39.000000000","tz":0},"subject":"Change the default configuration value","message":"Change the default configuration value\n\nNova-network will be deprecated, this change is to change the default \nconfiguration value because Nova network(the current value), is \ndeprecated\n\nChange-Id: I75c0bfc350c8c21d4491d85a3d65a858645620d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/951e76359a19a5cab23a4e5fa72be19cdc0fafed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/951e76359a19a5cab23a4e5fa72be19cdc0fafed"}]},"branch":"refs/heads/master"},"a94e7f390df324f93bb4ab948be4eee78b44deae":{"kind":"REWORK","_number":10,"created":"2025-02-22 18:43:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/98/502398/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/502398/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/502398/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/502398/10"}}},"commit":{"parents":[{"commit":"2c4f4c33ff63d2fb3b5acd2b885410d6108fb806","subject":"Merge \"Install and configure the Trove dashboard - several improvements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2c4f4c33ff63d2fb3b5acd2b885410d6108fb806"}]}],"author":{"name":"tianhui","email":"tianhui@awcloud.com","date":"2017-09-11 06:31:35.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-22 18:43:13.000000000","tz":0},"subject":"Change the default configuration value","message":"Change the default configuration value\n\nNova-network will be deprecated, this change is to change the default \nconfiguration value because Nova network(the current value), is \ndeprecated\n\nChange-Id: I75c0bfc350c8c21d4491d85a3d65a858645620d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a94e7f390df324f93bb4ab948be4eee78b44deae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a94e7f390df324f93bb4ab948be4eee78b44deae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
