)]}'
{"id":"openstack%2Ftrove~523295","triplet_id":"openstack%2Ftrove~master~I6702bdd5483f9b84efaefaf2585f6e7305a0c64d","project":"openstack/trove","branch":"master","topic":"datastore-flavors","hashtags":[],"change_id":"I6702bdd5483f9b84efaefaf2585f6e7305a0c64d","subject":"List flavors based on datastore/version associations","status":"ABANDONED","created":"2017-11-28 04:34:50.000000000","updated":"2026-09-18 04:43:45.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e3bfb0f1e8fa9709a4f3361287d43e4967af3912","_number":523295,"virtual_id_number":523295,"owner":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-11 01:30:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":-1,"date":"2020-11-11 20:23:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},{"date":"2020-11-10 23:08:58.000000000","_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-11-28 14:22:44.000000000","updated_by":{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},"reviewer":{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},"state":"REVIEWER"},{"updated":"2017-11-29 06:58:26.000000000","updated_by":{"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},"reviewer":{"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},"state":"REVIEWER"},{"updated":"2017-11-29 12:38:51.000000000","updated_by":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"reviewer":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"state":"REVIEWER"},{"updated":"2020-11-10 23:08:58.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-11-11 01:30:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-11 20:23:38.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2026-08-31 03:53:43.000000000","updated_by":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"reviewer":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"state":"REMOVED"}],"messages":[{"id":"03bf03bd82d1a6bfbc7e065db3ee244a93514c10","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2017-11-28 04:34:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c5c92102b1dabe8df5f1adfb733882526afb06b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-28 07:05:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/95/523295/1/check/openstack-tox-pep8/e0ddec5/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/95/523295/1/check/openstack-tox-py27/75193b3/ : SUCCESS in 10m 16s\n- openstack-tox-py35 http://logs.openstack.org/95/523295/1/check/openstack-tox-py35/6012966/ : SUCCESS in 10m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/95/523295/1/check/build-openstack-sphinx-docs/3946999/html/ : SUCCESS in 5m 31s\n- legacy-trove-functional-dsvm-mysql http://logs.openstack.org/95/523295/1/check/legacy-trove-functional-dsvm-mysql/90a6b63/ : SUCCESS in 1h 57m 18s\n- legacy-trove-scenario-dsvm-mariadb-single http://logs.openstack.org/95/523295/1/check/legacy-trove-scenario-dsvm-mariadb-single/391968b/ : SUCCESS in 1h 22m 23s (non-voting)\n- legacy-trove-scenario-dsvm-mariadb-multi http://logs.openstack.org/95/523295/1/check/legacy-trove-scenario-dsvm-mariadb-multi/2d4cf51/ : FAILURE in 1h 57m 43s (non-voting)\n- legacy-trove-scenario-dsvm-mysql-single http://logs.openstack.org/95/523295/1/check/legacy-trove-scenario-dsvm-mysql-single/5c8dda1/ : SUCCESS in 1h 57m 43s\n- legacy-trove-scenario-dsvm-mysql-multi http://logs.openstack.org/95/523295/1/check/legacy-trove-scenario-dsvm-mysql-multi/21ec54d/ : SUCCESS in 1h 58m 25s\n- legacy-trove-scenario-dsvm-postgresql-single http://logs.openstack.org/95/523295/1/check/legacy-trove-scenario-dsvm-postgresql-single/8aa79e5/ : SUCCESS in 2h 28m 18s (non-voting)\n- legacy-trove-scenario-dsvm-postgresql-multi http://logs.openstack.org/95/523295/1/check/legacy-trove-scenario-dsvm-postgresql-multi/4cb1e08/ : SUCCESS in 1h 49m 09s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-single http://logs.openstack.org/95/523295/1/check/legacy-trove-scenario-dsvm-pxc-single/7803062/ : SUCCESS in 1h 39m 06s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-multi http://logs.openstack.org/95/523295/1/check/legacy-trove-scenario-dsvm-pxc-multi/2cbd63e/ : SUCCESS in 34m 08s (non-voting)\n- legacy-trove-scenario-dsvm-redis-single http://logs.openstack.org/95/523295/1/check/legacy-trove-scenario-dsvm-redis-single/f4be727/ : SUCCESS in 1h 15m 07s (non-voting)\n- legacy-trove-scenario-dsvm-redis-multi http://logs.openstack.org/95/523295/1/check/legacy-trove-scenario-dsvm-redis-multi/5313667/ : SUCCESS in 1h 12m 44s (non-voting)\n- legacy-trove-tox-fakemodetests http://logs.openstack.org/95/523295/1/check/legacy-trove-tox-fakemodetests/51c7c11/ : SUCCESS in 5m 46s\n- legacy-trove-tox-apiexamples http://logs.openstack.org/95/523295/1/check/legacy-trove-tox-apiexamples/536390e/ : SUCCESS in 4m 57s\n- openstack-tox-pylint http://logs.openstack.org/95/523295/1/check/openstack-tox-pylint/03a9ce4/ : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"52a2daca68ee7360efd4d9cdae093de19a70130d","author":{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},"date":"2017-11-28 14:22:44.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good.","accounts_in_message":[],"_revision_number":1},{"id":"1cc068e13333a939a248cb22508fe3a6f9f122d4","author":{"_account_id":24739,"name":"Yao Wang","email":"wangyao@cmss.chinamobile.com","username":"wangyao"},"date":"2017-11-28 14:28:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35b52f754c7909f03b6fe9be781264bd27738ce4","author":{"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},"date":"2017-11-28 15:08:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9ae2d71319ec352078d21a9174d921be3087cd9","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2017-11-28 22:19:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97f20acb6ed2abda4638cd92fb5135e10c4c45e0","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2017-11-28 22:24:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36765d7d3ab3cfce0168b88fc12c5bf7288e294c","author":{"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},"date":"2017-11-29 01:21:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebe9a8b67e65a5840a126bb9d6fb21df14d79748","author":{"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},"date":"2017-11-29 01:24:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cdfd6f9831660d6e780c092b25010b46bd6da0d4","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2017-11-29 04:16:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"67214d963b9e8c525991bd26be462e0c8f2b10fb","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2017-11-29 04:18:31.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n@zhanggang: Does patch set 2 look better?","accounts_in_message":[],"_revision_number":2},{"id":"2b8ed6a717c103b9412d030f3ac4b62b335fa883","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-29 05:04:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/95/523295/2/check/openstack-tox-pep8/c1e4678/ : FAILURE in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/95/523295/2/check/openstack-tox-py27/285649a/ : SUCCESS in 9m 19s\n- openstack-tox-py35 http://logs.openstack.org/95/523295/2/check/openstack-tox-py35/9b8d75d/ : SUCCESS in 11m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/95/523295/2/check/build-openstack-sphinx-docs/5b8d5fa/html/ : SUCCESS in 4m 40s\n- legacy-trove-functional-dsvm-mysql http://logs.openstack.org/95/523295/2/check/legacy-trove-functional-dsvm-mysql/580eae9/ : FAILURE in 38m 25s\n- legacy-trove-scenario-dsvm-mariadb-single http://logs.openstack.org/95/523295/2/check/legacy-trove-scenario-dsvm-mariadb-single/7605779/ : FAILURE in 32m 52s (non-voting)\n- legacy-trove-scenario-dsvm-mariadb-multi http://logs.openstack.org/95/523295/2/check/legacy-trove-scenario-dsvm-mariadb-multi/de0f65e/ : FAILURE in 34m 38s (non-voting)\n- legacy-trove-scenario-dsvm-mysql-single http://logs.openstack.org/95/523295/2/check/legacy-trove-scenario-dsvm-mysql-single/740a19f/ : FAILURE in 31m 56s\n- legacy-trove-scenario-dsvm-mysql-multi http://logs.openstack.org/95/523295/2/check/legacy-trove-scenario-dsvm-mysql-multi/40c9e54/ : FAILURE in 37m 23s\n- legacy-trove-scenario-dsvm-postgresql-single http://logs.openstack.org/95/523295/2/check/legacy-trove-scenario-dsvm-postgresql-single/f0d60cc/ : FAILURE in 37m 06s (non-voting)\n- legacy-trove-scenario-dsvm-postgresql-multi http://logs.openstack.org/95/523295/2/check/legacy-trove-scenario-dsvm-postgresql-multi/207ad0b/ : FAILURE in 38m 17s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-single http://logs.openstack.org/95/523295/2/check/legacy-trove-scenario-dsvm-pxc-single/b0d4d8f/ : FAILURE in 38m 37s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-multi http://logs.openstack.org/95/523295/2/check/legacy-trove-scenario-dsvm-pxc-multi/982e10e/ : SUCCESS in 33m 53s (non-voting)\n- legacy-trove-scenario-dsvm-redis-single http://logs.openstack.org/95/523295/2/check/legacy-trove-scenario-dsvm-redis-single/5252cda/ : FAILURE in 39m 00s (non-voting)\n- legacy-trove-scenario-dsvm-redis-multi http://logs.openstack.org/95/523295/2/check/legacy-trove-scenario-dsvm-redis-multi/ad1107d/ : FAILURE in 35m 57s (non-voting)\n- legacy-trove-tox-fakemodetests http://logs.openstack.org/95/523295/2/check/legacy-trove-tox-fakemodetests/68aeee7/ : FAILURE in 6m 35s\n- legacy-trove-tox-apiexamples http://logs.openstack.org/95/523295/2/check/legacy-trove-tox-apiexamples/6f6f1ab/ : FAILURE in 3m 48s\n- openstack-tox-pylint http://logs.openstack.org/95/523295/2/check/openstack-tox-pylint/0537951/ : FAILURE in 6m 48s","accounts_in_message":[],"_revision_number":2},{"id":"5bf80bdd63e9151b0ca90bbb1ca8665d2a2a889a","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2017-11-29 06:08:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0ea0370362878307e2367d7ee276b72b9ee859b0","author":{"_account_id":22608,"name":"zhanggang","email":"zhanggang@cmss.chinamobile.com","username":"zhanggang"},"date":"2017-11-29 06:58:26.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nHi Andy, add two comments. Others looks good.","accounts_in_message":[],"_revision_number":3},{"id":"071e439a26d8d411e309f56f265564f5c94b2e8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-29 08:32:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/95/523295/3/check/openstack-tox-pep8/7041012/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/95/523295/3/check/openstack-tox-py27/cda971a/ : SUCCESS in 10m 21s\n- openstack-tox-py35 http://logs.openstack.org/95/523295/3/check/openstack-tox-py35/b58f355/ : SUCCESS in 12m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/95/523295/3/check/build-openstack-sphinx-docs/a4dfc47/html/ : SUCCESS in 5m 48s\n- legacy-trove-functional-dsvm-mysql http://logs.openstack.org/95/523295/3/check/legacy-trove-functional-dsvm-mysql/6c5e7bf/ : FAILURE in 40m 15s\n- legacy-trove-scenario-dsvm-mariadb-single http://logs.openstack.org/95/523295/3/check/legacy-trove-scenario-dsvm-mariadb-single/399ac34/ : FAILURE in 33m 15s (non-voting)\n- legacy-trove-scenario-dsvm-mariadb-multi http://logs.openstack.org/95/523295/3/check/legacy-trove-scenario-dsvm-mariadb-multi/ad1d700/ : FAILURE in 39m 47s (non-voting)\n- legacy-trove-scenario-dsvm-mysql-single http://logs.openstack.org/95/523295/3/check/legacy-trove-scenario-dsvm-mysql-single/fe4b1df/ : FAILURE in 2h 18m 42s\n- legacy-trove-scenario-dsvm-mysql-multi http://logs.openstack.org/95/523295/3/check/legacy-trove-scenario-dsvm-mysql-multi/5e184db/ : FAILURE in 40m 54s\n- legacy-trove-scenario-dsvm-postgresql-single http://logs.openstack.org/95/523295/3/check/legacy-trove-scenario-dsvm-postgresql-single/9ea2e88/ : FAILURE in 48m 53s (non-voting)\n- legacy-trove-scenario-dsvm-postgresql-multi http://logs.openstack.org/95/523295/3/check/legacy-trove-scenario-dsvm-postgresql-multi/18a648c/ : FAILURE in 2h 12m 43s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-single http://logs.openstack.org/95/523295/3/check/legacy-trove-scenario-dsvm-pxc-single/664ebf0/ : FAILURE in 37m 11s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-multi http://logs.openstack.org/95/523295/3/check/legacy-trove-scenario-dsvm-pxc-multi/cbacde6/ : SUCCESS in 36m 57s (non-voting)\n- legacy-trove-scenario-dsvm-redis-single http://logs.openstack.org/95/523295/3/check/legacy-trove-scenario-dsvm-redis-single/7e0b4c7/ : FAILURE in 49m 07s (non-voting)\n- legacy-trove-scenario-dsvm-redis-multi http://logs.openstack.org/95/523295/3/check/legacy-trove-scenario-dsvm-redis-multi/f96634f/ : FAILURE in 43m 50s (non-voting)\n- legacy-trove-tox-fakemodetests http://logs.openstack.org/95/523295/3/check/legacy-trove-tox-fakemodetests/5a3f6aa/ : FAILURE in 4m 22s\n- legacy-trove-tox-apiexamples http://logs.openstack.org/95/523295/3/check/legacy-trove-tox-apiexamples/671c77a/ : FAILURE in 6m 34s\n- openstack-tox-pylint http://logs.openstack.org/95/523295/3/check/openstack-tox-pylint/ee91932/ : FAILURE in 6m 09s","accounts_in_message":[],"_revision_number":3},{"id":"018287a6a1fb23f87d97553d0b3b146d8a4c6b5d","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2017-11-29 09:38:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d3ab48cd024eb3d1ee0060caf6ff218c7938eaa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-29 12:25:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/95/523295/4/check/openstack-tox-pep8/be8e8c7/ : SUCCESS in 5m 54s\n- openstack-tox-py27 http://logs.openstack.org/95/523295/4/check/openstack-tox-py27/0072f60/ : SUCCESS in 12m 18s\n- openstack-tox-py35 http://logs.openstack.org/95/523295/4/check/openstack-tox-py35/8f3e54a/ : SUCCESS in 12m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/95/523295/4/check/build-openstack-sphinx-docs/3cb1094/html/ : SUCCESS in 4m 52s\n- legacy-trove-functional-dsvm-mysql http://logs.openstack.org/95/523295/4/check/legacy-trove-functional-dsvm-mysql/dd8a56b/ : SUCCESS in 1h 42m 15s\n- legacy-trove-scenario-dsvm-mariadb-single http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-mariadb-single/6793060/ : SUCCESS in 1h 33m 52s (non-voting)\n- legacy-trove-scenario-dsvm-mariadb-multi http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-mariadb-multi/2225ed2/ : FAILURE in 1h 34m 35s (non-voting)\n- legacy-trove-scenario-dsvm-mysql-single http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-mysql-single/f00777e/ : FAILURE in 2h 34m 46s\n- legacy-trove-scenario-dsvm-mysql-multi http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-mysql-multi/6b059b7/ : SUCCESS in 1h 56m 40s\n- legacy-trove-scenario-dsvm-postgresql-single http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-postgresql-single/8b7226b/ : SUCCESS in 2h 18m 44s (non-voting)\n- legacy-trove-scenario-dsvm-postgresql-multi http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-postgresql-multi/4c05d84/ : SUCCESS in 1h 45m 19s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-single http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-pxc-single/d52ceca/ : SUCCESS in 2h 14m 04s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-multi http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-pxc-multi/99d5d54/ : SUCCESS in 33m 48s (non-voting)\n- legacy-trove-scenario-dsvm-redis-single http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-redis-single/150ce52/ : SUCCESS in 1h 26m 36s (non-voting)\n- legacy-trove-scenario-dsvm-redis-multi http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-redis-multi/1858be0/ : SUCCESS in 1h 13m 54s (non-voting)\n- legacy-trove-tox-fakemodetests http://logs.openstack.org/95/523295/4/check/legacy-trove-tox-fakemodetests/32e7b6c/ : SUCCESS in 5m 41s\n- legacy-trove-tox-apiexamples http://logs.openstack.org/95/523295/4/check/legacy-trove-tox-apiexamples/8d0545d/ : SUCCESS in 5m 06s\n- openstack-tox-pylint http://logs.openstack.org/95/523295/4/check/openstack-tox-pylint/2dd0887/ : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":4},{"id":"af074ecf3994305d79fc1a22b502cd7d10de646a","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-11-29 12:38:51.000000000","message":"Patch Set 4:\n\nthis change breaks the current API change rules; the \u0027proper\u0027 way to do this is to rev the API version.","accounts_in_message":[],"_revision_number":4},{"id":"28d81532dec6a4032f079bf01670ebca61d41f7a","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2017-12-05 22:45:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"736e94e6b57a4ce3b0c13d88296406d0a848e9a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-06 01:02:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/95/523295/4/check/openstack-tox-pep8/36f95c1/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/95/523295/4/check/openstack-tox-py27/1123cdf/ : SUCCESS in 9m 34s\n- openstack-tox-py35 http://logs.openstack.org/95/523295/4/check/openstack-tox-py35/caedadc/ : SUCCESS in 12m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/95/523295/4/check/build-openstack-sphinx-docs/8bedca3/html/ : SUCCESS in 4m 13s\n- legacy-trove-functional-dsvm-mysql http://logs.openstack.org/95/523295/4/check/legacy-trove-functional-dsvm-mysql/7f17579/ : SUCCESS in 2h 13m 40s\n- legacy-trove-scenario-dsvm-mariadb-single http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-mariadb-single/2908e64/ : SUCCESS in 1h 29m 31s (non-voting)\n- legacy-trove-scenario-dsvm-mariadb-multi http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-mariadb-multi/e10ffef/ : FAILURE in 2h 00m 27s (non-voting)\n- legacy-trove-scenario-dsvm-mysql-single http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-mysql-single/c971ecd/ : SUCCESS in 1h 47m 26s\n- legacy-trove-scenario-dsvm-mysql-multi http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-mysql-multi/35aa911/ : SUCCESS in 1h 56m 09s\n- legacy-trove-scenario-dsvm-postgresql-single http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-postgresql-single/ebe2f31/ : FAILURE in 27m 34s (non-voting)\n- legacy-trove-scenario-dsvm-postgresql-multi http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-postgresql-multi/ab7788a/ : FAILURE in 57m 23s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-single http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-pxc-single/1491a9c/ : SUCCESS in 1h 43m 33s (non-voting)\n- legacy-trove-scenario-dsvm-pxc-multi http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-pxc-multi/2bcb6da/ : SUCCESS in 33m 41s (non-voting)\n- legacy-trove-scenario-dsvm-redis-single http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-redis-single/d3175bc/ : SUCCESS in 1h 19m 12s (non-voting)\n- legacy-trove-scenario-dsvm-redis-multi http://logs.openstack.org/95/523295/4/check/legacy-trove-scenario-dsvm-redis-multi/6c7a924/ : SUCCESS in 1h 15m 56s (non-voting)\n- legacy-trove-tox-fakemodetests http://logs.openstack.org/95/523295/4/check/legacy-trove-tox-fakemodetests/8ac2a5f/ : SUCCESS in 4m 41s\n- legacy-trove-tox-apiexamples http://logs.openstack.org/95/523295/4/check/legacy-trove-tox-apiexamples/1bf7e72/ : SUCCESS in 4m 44s\n- openstack-tox-pylint http://logs.openstack.org/95/523295/4/check/openstack-tox-pylint/b1c13b1/ : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":4},{"id":"d6386562d82f5b4cb54ac82db03308d302fcb45b","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-10 23:08:58.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d174366350504d8d2e88ca1bc3ad05b5c7fccd9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 01:30:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23c50bcc036e40419d2754b77b02266c : FAILURE in 6m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42870a629c504ba1872a7f01a1bf6102 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd9acbd976f94f089dc32c030066ea1c : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae247c6976ec4e1481af95c35aa4246d : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c285ebb8c8734f3094303ca32f8a9f8d : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d30b1e68646d46e2bc6b1df9ee68eda9 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ff03fa23ec4789814272cace558d19 : SUCCESS in 14m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1833ad4f4969432991ae8dc6b9fec08e : SUCCESS in 6m 22s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9d967b62b0c64758932cc62c56f5d503 : TIMED_OUT in 2h 13m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/baba7084f77746fe954decf91c993043 : SUCCESS in 58m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"467e124d9b592d639a9fc73d80fcdb95adde3d28","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-11 20:23:38.000000000","message":"Patch Set 5: Code-Review-1\n\nI don\u0027t think this patch is ready.\n\n1. No tests.\n2. If a datastore version has flavors associated but others do not, the flavors in the flavor API response only contains those flavors. This is not what we expect.","accounts_in_message":[],"_revision_number":5},{"id":"750e23ea04897b065f0b33e70364ca7ebef70390","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-11 20:24:06.000000000","message":"Patch Set 5:\n\nAlso needs release note and API doc change.","accounts_in_message":[],"_revision_number":5},{"id":"3c2b4b3fc365fa02587202c2853e1a4b6f312c2f","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2026-08-31 03:53:43.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_9664\u003e.","accounts_in_message":[{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"}],"_revision_number":5},{"id":"e3bfb0f1e8fa9709a4f3361287d43e4967af3912","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 04:43:45.000000000","message":"Abandoned\n\nFixed in bf1ee5e8f29f286d2dee853db19912ea01ea4c78","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"56c0d6359160efbc1bc919063e931fd6a2ac5037","revisions":{"e6531bf9aca76c32ed4877d8c0947557da8f26e4":{"kind":"REWORK","_number":1,"created":"2017-11-28 04:34:50.000000000","uploader":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"ref":"refs/changes/95/523295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/95/523295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/95/523295/1"}}},"commit":{"parents":[{"commit":"a6a9259d42ed7e10fc05bd2206371a31b90820dc","subject":"Merge \"fix typos in cluster/test_models.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a6a9259d42ed7e10fc05bd2206371a31b90820dc"}]}],"author":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-11-27 00:25:48.000000000","tz":660},"committer":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-11-28 04:34:35.000000000","tz":660},"subject":"List flavors based on datastore/version associations","message":"List flavors based on datastore/version associations\n\nThis commit provides a better filtered list of flavors.\n\nIf there are any datastore/version flavor mappings, then provide\na list of all flavors defined (across all datastore/versions).\nThis is most useful when using the dashboard, and a list of flavors\nis presented to the user before they\u0027ve had a chance to choose a\ndatastore.\n\nIn the case where no specific datastore/flavor mappings exist,\nit lists all available flavors as before.\n\nChange-Id: I6702bdd5483f9b84efaefaf2585f6e7305a0c64d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e6531bf9aca76c32ed4877d8c0947557da8f26e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e6531bf9aca76c32ed4877d8c0947557da8f26e4"}]},"branch":"refs/heads/master"},"1a7ea0485fb287ab446cd4ff989a6dca0ef9e72d":{"kind":"REWORK","_number":2,"created":"2017-11-29 04:16:58.000000000","uploader":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"ref":"refs/changes/95/523295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/95/523295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/95/523295/2"}}},"commit":{"parents":[{"commit":"a6a9259d42ed7e10fc05bd2206371a31b90820dc","subject":"Merge \"fix typos in cluster/test_models.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a6a9259d42ed7e10fc05bd2206371a31b90820dc"}]}],"author":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-11-27 00:25:48.000000000","tz":660},"committer":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-11-29 04:16:04.000000000","tz":660},"subject":"List flavors based on datastore/version associations","message":"List flavors based on datastore/version associations\n\nThis commit provides a better filtered list of flavors.\n\nIf there are any datastore/version flavor mappings, then provide\na list of all flavors defined (across all datastore/versions).\nThis is most useful when using the dashboard, and a list of flavors\nis presented to the user before they\u0027ve had a chance to choose a\ndatastore.\n\nIn the case where no specific datastore/flavor mappings exist,\nit lists all available flavors as before.\n\nChange-Id: I6702bdd5483f9b84efaefaf2585f6e7305a0c64d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a7ea0485fb287ab446cd4ff989a6dca0ef9e72d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a7ea0485fb287ab446cd4ff989a6dca0ef9e72d"}]},"branch":"refs/heads/master"},"87d0f5de469b1d17a1023c0df23b242fca89168b":{"kind":"REWORK","_number":3,"created":"2017-11-29 06:08:39.000000000","uploader":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"ref":"refs/changes/95/523295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/95/523295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/95/523295/3"}}},"commit":{"parents":[{"commit":"a6a9259d42ed7e10fc05bd2206371a31b90820dc","subject":"Merge \"fix typos in cluster/test_models.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a6a9259d42ed7e10fc05bd2206371a31b90820dc"}]}],"author":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-11-27 00:25:48.000000000","tz":660},"committer":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-11-29 06:08:20.000000000","tz":660},"subject":"List flavors based on datastore/version associations","message":"List flavors based on datastore/version associations\n\nThis commit provides a better filtered list of flavors.\n\nIf there are any datastore/version flavor mappings, then provide\na list of all flavors defined (across all datastore/versions).\nThis is most useful when using the dashboard, and a list of flavors\nis presented to the user before they\u0027ve had a chance to choose a\ndatastore.\n\nIn the case where no specific datastore/flavor mappings exist,\nit lists all available flavors as before.\n\nChange-Id: I6702bdd5483f9b84efaefaf2585f6e7305a0c64d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87d0f5de469b1d17a1023c0df23b242fca89168b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87d0f5de469b1d17a1023c0df23b242fca89168b"}]},"branch":"refs/heads/master"},"00bc34620697c9c1f53c193fe325bac8fa107740":{"kind":"REWORK","_number":4,"created":"2017-11-29 09:38:33.000000000","uploader":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"ref":"refs/changes/95/523295/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/95/523295/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/95/523295/4"}}},"commit":{"parents":[{"commit":"a6a9259d42ed7e10fc05bd2206371a31b90820dc","subject":"Merge \"fix typos in cluster/test_models.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a6a9259d42ed7e10fc05bd2206371a31b90820dc"}]}],"author":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-11-27 00:25:48.000000000","tz":660},"committer":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-11-29 09:31:30.000000000","tz":660},"subject":"List flavors based on datastore/version associations","message":"List flavors based on datastore/version associations\n\nThis commit provides a better filtered list of flavors.\n\nIf there are any datastore/version flavor mappings, then provide\na list of all flavors defined (across all datastore/versions).\nThis is most useful when using the dashboard, and a list of flavors\nis presented to the user before they\u0027ve had a chance to choose a\ndatastore.\n\nIn the case where no specific datastore/flavor mappings exist,\nit lists all available flavors as before.\n\nChange-Id: I6702bdd5483f9b84efaefaf2585f6e7305a0c64d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00bc34620697c9c1f53c193fe325bac8fa107740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00bc34620697c9c1f53c193fe325bac8fa107740"}]},"branch":"refs/heads/master"},"56c0d6359160efbc1bc919063e931fd6a2ac5037":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-11-10 23:08:58.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/95/523295/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/95/523295/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/95/523295/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/95/523295/5"}}},"commit":{"parents":[{"commit":"78772cef68887a59adbde42057ef43d586e28b85","subject":"Merge \"Make guest agent api changes backwards compatible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78772cef68887a59adbde42057ef43d586e28b85"}]}],"author":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-11-27 00:25:48.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-10 23:08:18.000000000","tz":660},"subject":"List flavors based on datastore/version associations","message":"List flavors based on datastore/version associations\n\nThis commit provides a better filtered list of flavors.\n\nIf there are any datastore/version flavor mappings, then provide\na list of all flavors defined (across all datastore/versions).\nThis is most useful when using the dashboard, and a list of flavors\nis presented to the user before they\u0027ve had a chance to choose a\ndatastore.\n\nIn the case where no specific datastore/flavor mappings exist,\nit lists all available flavors as before.\n\nChange-Id: I6702bdd5483f9b84efaefaf2585f6e7305a0c64d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56c0d6359160efbc1bc919063e931fd6a2ac5037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56c0d6359160efbc1bc919063e931fd6a2ac5037"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
