)]}'
{"id":"openstack%2Ftrove~652068","triplet_id":"openstack%2Ftrove~master~I2efb69d3d50344914a875b773f62a227dba2ccaf","project":"openstack/trove","branch":"master","topic":"service-tenant-model","hashtags":[],"change_id":"I2efb69d3d50344914a875b773f62a227dba2ccaf","subject":"Enable service tenant deployment model by default in DevStack","status":"MERGED","created":"2019-04-12 13:12:12.000000000","updated":"2019-06-08 07:57:46.000000000","submitted":"2019-06-08 07:57:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652068-1559980666645-48a02ede","meta_rev_id":"f371ca42bd3f03d8141f0cb1d87040f0ee2ec07a","_number":652068,"virtual_id_number":652068,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"value":2,"date":"2019-06-08 07:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-08 05:46:35.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-08 05:46:35.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 08:09:04.000000000","updated_by":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"reviewer":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"state":"REVIEWER"},{"updated":"2019-06-08 07:57:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd1ba6b37bc464a41e16443202355b32ce97bc1a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-12 13:12:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d03cf42b873beccf74f40f5b2fc1cb6f8bf1112b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 14:00:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/652068/1/check/openstack-tox-cover/f3302e0/cover/ : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/1/check/openstack-tox-lower-constraints/35929a2/ : SUCCESS in 12m 34s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/1/check/openstack-tox-pep8/c69b8db/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/1/check/openstack-tox-py27/efeaf20/ : SUCCESS in 11m 17s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/1/check/openstack-tox-py35/3e0e8f1/ : SUCCESS in 13m 32s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/1/check/openstack-tox-py36/c368ed1/ : SUCCESS in 13m 06s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/1/check/openstack-tox-py37/78a89f6/ : SUCCESS in 17m 42s\n- openstack-tox-docs http://logs.openstack.org/68/652068/1/check/openstack-tox-docs/b55dba7/html/ : SUCCESS in 5m 29s\n- trove-functional-mysql http://logs.openstack.org/68/652068/1/check/trove-functional-mysql/ce49f53/ : FAILURE in 40m 25s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/1/check/trove-scenario-mariadb-single/f55bb1c/ : FAILURE in 30m 47s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/1/check/trove-scenario-mariadb-multi/9fdbad2/ : FAILURE in 42m 39s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/1/check/trove-scenario-mysql-single/9b53fc1/ : FAILURE in 41m 04s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/1/check/trove-scenario-mysql-multi/8fa3719/ : FAILURE in 41m 01s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/1/check/trove-scenario-postgresql-single/781bf76/ : FAILURE in 39m 43s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/1/check/trove-scenario-postgresql-multi/66ad49a/ : FAILURE in 39m 58s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/1/check/trove-scenario-pxc-single/f9a67f5/ : FAILURE in 38m 40s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/1/check/trove-scenario-pxc-multi/bc8512c/ : FAILURE in 42m 02s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/1/check/trove-scenario-redis-single/bca0462/ : FAILURE in 40m 13s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/1/check/trove-scenario-redis-multi/7aacb3c/ : FAILURE in 38m 28s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/1/check/trove-tox-fakemodetests/5be7915/ : SUCCESS in 6m 01s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/1/check/trove-tox-apiexamples/12f06ab/ : SUCCESS in 5m 33s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/1/check/trove-tox-bandit-baseline/da15009/ : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/1/check/openstack-tox-pylint/7a8d3b4/ : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c2e16716666f4de7a5c7b636badad9138ed53211","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-13 11:38:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7bb08c80aea0241e3e567d1b41a753ca688736e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-13 12:35:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/652068/2/check/openstack-tox-cover/57ac4cd/cover/ : SUCCESS in 18m 06s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/2/check/openstack-tox-lower-constraints/755cb28/ : SUCCESS in 13m 32s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/2/check/openstack-tox-pep8/e065b93/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/2/check/openstack-tox-py27/0b5a5bf/ : SUCCESS in 10m 57s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/2/check/openstack-tox-py35/c9bb078/ : SUCCESS in 13m 59s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/2/check/openstack-tox-py36/f382f21/ : SUCCESS in 12m 03s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/2/check/openstack-tox-py37/74348ea/ : SUCCESS in 17m 49s\n- openstack-tox-docs http://logs.openstack.org/68/652068/2/check/openstack-tox-docs/2affbb7/html/ : SUCCESS in 4m 40s\n- trove-functional-mysql http://logs.openstack.org/68/652068/2/check/trove-functional-mysql/e0db2bd/ : FAILURE in 41m 06s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/2/check/trove-scenario-mariadb-single/78bc0ff/ : FAILURE in 28m 34s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/2/check/trove-scenario-mariadb-multi/285a95d/ : FAILURE in 36m 23s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/2/check/trove-scenario-mysql-single/fd28a99/ : FAILURE in 37m 07s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/2/check/trove-scenario-mysql-multi/b7d68f0/ : FAILURE in 41m 11s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/2/check/trove-scenario-postgresql-single/4fe7c30/ : FAILURE in 53m 18s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/2/check/trove-scenario-postgresql-multi/0e760cf/ : FAILURE in 41m 35s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/2/check/trove-scenario-pxc-single/4ccbdb8/ : FAILURE in 37m 42s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/2/check/trove-scenario-pxc-multi/08abad5/ : FAILURE in 40m 13s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/2/check/trove-scenario-redis-single/59fbd48/ : FAILURE in 40m 56s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/2/check/trove-scenario-redis-multi/603f8e8/ : FAILURE in 41m 53s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/2/check/trove-tox-fakemodetests/045e80a/ : SUCCESS in 6m 53s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/2/check/trove-tox-apiexamples/8b26190/ : SUCCESS in 6m 39s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/2/check/trove-tox-bandit-baseline/5a06731/ : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/2/check/openstack-tox-pylint/2a22b3b/ : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":2},{"id":"1c5bbb14ffe22e8f4991b1f342c53cd33eb3be17","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-15 02:33:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"65ae4574eb5155e89c744bd55f190da6d75559cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 03:20:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/652068/3/check/openstack-tox-cover/dd27884/cover/ : SUCCESS in 27m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/3/check/openstack-tox-lower-constraints/b4d3269/ : SUCCESS in 15m 03s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/3/check/openstack-tox-pep8/a6f76fe/ : SUCCESS in 10m 41s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/3/check/openstack-tox-py27/7447169/ : SUCCESS in 10m 49s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/3/check/openstack-tox-py35/81341df/ : SUCCESS in 12m 28s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/3/check/openstack-tox-py36/9dddd09/ : SUCCESS in 12m 58s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/3/check/openstack-tox-py37/82fb106/ : SUCCESS in 15m 53s\n- openstack-tox-docs http://logs.openstack.org/68/652068/3/check/openstack-tox-docs/23fef16/html/ : SUCCESS in 4m 49s\n- trove-functional-mysql http://logs.openstack.org/68/652068/3/check/trove-functional-mysql/25a8867/ : FAILURE in 34m 34s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/3/check/trove-scenario-mariadb-single/4030574/ : FAILURE in 32m 43s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/3/check/trove-scenario-mariadb-multi/1408869/ : FAILURE in 34m 22s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/3/check/trove-scenario-mysql-single/5875169/ : FAILURE in 40m 55s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/3/check/trove-scenario-mysql-multi/5116d99/ : FAILURE in 35m 06s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/3/check/trove-scenario-postgresql-single/f054128/ : FAILURE in 39m 39s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/3/check/trove-scenario-postgresql-multi/af397a0/ : FAILURE in 32m 54s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/3/check/trove-scenario-pxc-single/db81df7/ : FAILURE in 38m 51s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/3/check/trove-scenario-pxc-multi/75c0470/ : FAILURE in 33m 05s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/3/check/trove-scenario-redis-single/c9bb57e/ : FAILURE in 38m 07s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/3/check/trove-scenario-redis-multi/c17ad3a/ : FAILURE in 34m 42s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/3/check/trove-tox-fakemodetests/4f8e82a/ : SUCCESS in 5m 48s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/3/check/trove-tox-apiexamples/45f3b77/ : SUCCESS in 5m 21s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/3/check/trove-tox-bandit-baseline/c38af75/ : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/3/check/openstack-tox-pylint/78fcaeb/ : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":3},{"id":"c9e114889c58db033c6756b4618495694fc280dc","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-15 03:24:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3194321467a6bfdc18429c46cceb1ac74b01ecf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 04:27:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/652068/4/check/openstack-tox-cover/7821517/cover/ : SUCCESS in 19m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/4/check/openstack-tox-lower-constraints/9d7c78c/ : SUCCESS in 13m 18s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/4/check/openstack-tox-pep8/6eaaaad/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/4/check/openstack-tox-py27/035f559/ : SUCCESS in 11m 04s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/4/check/openstack-tox-py35/79250d4/ : SUCCESS in 14m 29s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/4/check/openstack-tox-py36/d39f219/ : SUCCESS in 12m 35s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/4/check/openstack-tox-py37/6f3d0e1/ : SUCCESS in 19m 04s\n- openstack-tox-docs http://logs.openstack.org/68/652068/4/check/openstack-tox-docs/f313e26/html/ : SUCCESS in 5m 26s\n- trove-functional-mysql http://logs.openstack.org/68/652068/4/check/trove-functional-mysql/75fb06e/ : FAILURE in 56m 01s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/4/check/trove-scenario-mariadb-single/29b7b68/ : FAILURE in 35m 01s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/4/check/trove-scenario-mariadb-multi/ddc9676/ : FAILURE in 50m 10s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/4/check/trove-scenario-mysql-single/39af27a/ : FAILURE in 43m 27s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/4/check/trove-scenario-mysql-multi/bc5de70/ : FAILURE in 51m 32s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/4/check/trove-scenario-postgresql-single/2aa42b8/ : FAILURE in 47m 10s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/4/check/trove-scenario-postgresql-multi/5abdc89/ : FAILURE in 46m 13s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/4/check/trove-scenario-pxc-single/ee842d9/ : FAILURE in 53m 34s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/4/check/trove-scenario-pxc-multi/66756a4/ : FAILURE in 51m 39s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/4/check/trove-scenario-redis-single/20ee4c5/ : FAILURE in 51m 27s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/4/check/trove-scenario-redis-multi/6cd4dec/ : FAILURE in 57m 22s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/4/check/trove-tox-fakemodetests/58cb01a/ : SUCCESS in 6m 07s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/4/check/trove-tox-apiexamples/392f115/ : SUCCESS in 5m 47s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/4/check/trove-tox-bandit-baseline/dc78e44/ : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/4/check/openstack-tox-pylint/490e275/ : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":4},{"id":"c839f75fc33b8ff7af201d11377404460afabc76","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-15 07:24:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4b8e548fe7976a4ab904db07ba358c7abc0e03ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 09:08:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/68/652068/5/check/openstack-tox-cover/1b91151/cover/ : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/5/check/openstack-tox-lower-constraints/394db3f/ : SUCCESS in 13m 52s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/5/check/openstack-tox-pep8/f0639d0/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/5/check/openstack-tox-py27/2a3973e/ : SUCCESS in 11m 56s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/5/check/openstack-tox-py35/b865ee4/ : SUCCESS in 13m 22s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/5/check/openstack-tox-py36/2e1cd4a/ : SUCCESS in 13m 25s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/5/check/openstack-tox-py37/0e9dfd4/ : SUCCESS in 18m 37s\n- openstack-tox-docs http://logs.openstack.org/68/652068/5/check/openstack-tox-docs/dec8b69/html/ : SUCCESS in 5m 16s\n- trove-functional-mysql http://logs.openstack.org/68/652068/5/check/trove-functional-mysql/1e7daf0/ : FAILURE in 1h 37m 44s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/5/check/trove-scenario-mariadb-single/388e109/ : FAILURE in 33m 38s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/5/check/trove-scenario-mariadb-multi/76dfff8/ : FAILURE in 1h 01m 41s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/5/check/trove-scenario-mysql-single/debef00/ : FAILURE in 55m 17s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/5/check/trove-scenario-mysql-multi/376d0e9/ : FAILURE in 48m 33s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/5/check/trove-scenario-postgresql-single/c24192b/ : FAILURE in 58m 28s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/5/check/trove-scenario-postgresql-multi/6459bc9/ : FAILURE in 59m 23s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/5/check/trove-scenario-pxc-single/39f8f83/ : FAILURE in 48m 14s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/5/check/trove-scenario-pxc-multi/c618975/ : FAILURE in 48m 32s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/5/check/trove-scenario-redis-single/f5946d1/ : FAILURE in 50m 34s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/5/check/trove-scenario-redis-multi/9f5dbe5/ : FAILURE in 58m 16s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/5/check/trove-tox-fakemodetests/800333a/ : SUCCESS in 5m 53s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/5/check/trove-tox-apiexamples/bd27208/ : SUCCESS in 5m 33s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/5/check/trove-tox-bandit-baseline/e3d242c/ : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/5/check/openstack-tox-pylint/167ed9a/ : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":5},{"id":"ce0938342a69da6355ade4943d00846f39c9d114","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-18 11:23:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9a15b33f15de945968286136516c44ad93fda63a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-18 12:02:24.000000000","message":"Patch Set 6:\n\nHi Bartosz, Kasper, I really need your opinion towards this patch, especially for the change related to the guest agent. Thanks!","accounts_in_message":[],"_revision_number":6},{"id":"de1489f4b54f6cb4896df1ac2e5e0c735471c2af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 12:57:33.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/6/check/requirements-check/43c0b91/ : FAILURE in 3m 52s\n- openstack-tox-cover http://logs.openstack.org/68/652068/6/check/openstack-tox-cover/5d79d74/cover/ : SUCCESS in 22m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/6/check/openstack-tox-lower-constraints/6c55386/ : SUCCESS in 13m 28s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/6/check/openstack-tox-pep8/62be616/ : FAILURE in 6m 29s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/6/check/openstack-tox-py27/e59e344/ : SUCCESS in 12m 21s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/6/check/openstack-tox-py35/f1e234e/ : SUCCESS in 14m 59s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/6/check/openstack-tox-py36/0639e02/ : SUCCESS in 12m 20s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/6/check/openstack-tox-py37/f667deb/ : SUCCESS in 18m 13s\n- openstack-tox-docs http://logs.openstack.org/68/652068/6/check/openstack-tox-docs/f5b1124/html/ : SUCCESS in 5m 40s\n- trove-functional-mysql http://logs.openstack.org/68/652068/6/check/trove-functional-mysql/9d0eecd/ : FAILURE in 1h 09m 12s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/6/check/trove-scenario-mariadb-single/622e263/ : FAILURE in 42m 27s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/6/check/trove-scenario-mariadb-multi/5100dc0/ : FAILURE in 50m 05s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/6/check/trove-scenario-mysql-single/b597a39/ : FAILURE in 48m 35s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/6/check/trove-scenario-mysql-multi/53cce69/ : FAILURE in 45m 32s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/6/check/trove-scenario-postgresql-single/79230bb/ : FAILURE in 48m 35s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/6/check/trove-scenario-postgresql-multi/af9b89c/ : FAILURE in 56m 03s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/6/check/trove-scenario-pxc-single/7c994e0/ : FAILURE in 1h 01m 41s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/6/check/trove-scenario-pxc-multi/1382661/ : FAILURE in 1h 04m 15s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/6/check/trove-scenario-redis-single/c91c772/ : FAILURE in 1h 00m 32s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/6/check/trove-scenario-redis-multi/a1535b8/ : FAILURE in 58m 02s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/6/check/trove-tox-fakemodetests/61c0d8a/ : SUCCESS in 6m 24s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/6/check/trove-tox-apiexamples/1cdc5d1/ : SUCCESS in 5m 51s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/6/check/trove-tox-bandit-baseline/8701c15/ : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/6/check/openstack-tox-pylint/c7bf8d1/ : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":6},{"id":"cd51a55f5b02d9cb1ef3e14ac967864ac2a8daca","author":{"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},"date":"2019-04-19 06:23:55.000000000","message":"Patch Set 6:\n\nLingxian, I will take a look after holidays around the middle of the next week.","accounts_in_message":[],"_revision_number":6},{"id":"834742cfd8380903bf78983dde6fa24c227ef599","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-19 08:49:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"92a6c1ad6538ac9757429ddfadedb6b02f9b1e63","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-19 08:51:27.000000000","message":"Patch Set 7:\n\n\u003e Lingxian, I will take a look after holidays around the middle of\n \u003e the next week.\n\nSure, enjoy your holiday :-)","accounts_in_message":[],"_revision_number":7},{"id":"63c4bd70ff5a83b582e6c1144f0492988c171cde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 12:18:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/7/check/requirements-check/568a389/ : FAILURE in 3m 31s\n- openstack-tox-cover http://logs.openstack.org/68/652068/7/check/openstack-tox-cover/b8d6a82/cover/ : SUCCESS in 18m 05s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/7/check/openstack-tox-lower-constraints/2f6ca15/ : SUCCESS in 15m 50s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/7/check/openstack-tox-pep8/91fd5bf/ : FAILURE in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/7/check/openstack-tox-py27/0033826/ : SUCCESS in 11m 23s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/7/check/openstack-tox-py35/49ef92a/ : SUCCESS in 14m 11s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/7/check/openstack-tox-py36/e36cf34/ : SUCCESS in 11m 51s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/7/check/openstack-tox-py37/62e4bd8/ : SUCCESS in 25m 04s\n- openstack-tox-docs http://logs.openstack.org/68/652068/7/check/openstack-tox-docs/2a9160c/html/ : SUCCESS in 4m 54s\n- trove-functional-mysql finger://ze03.openstack.org/df2df69a2e0d4e22823ce2158efb5da0 : RETRY_LIMIT in 40m 33s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/7/check/trove-scenario-mariadb-single/7dfb984/ : FAILURE in 46m 26s (non-voting)\n- trove-scenario-mariadb-multi finger://ze11.openstack.org/01b29cb24bab490d850a190b6f6d6444 : RETRY_LIMIT in 44m 29s (non-voting)\n- trove-scenario-mysql-single finger://ze03.openstack.org/3d65ff7749c2451cb3a0aa0f4c429661 : RETRY_LIMIT in 58m 27s\n- trove-scenario-mysql-multi finger://ze05.openstack.org/2a7d71a0f65a41c5be0b0805bb6dfb66 : RETRY_LIMIT in 52m 34s\n- trove-scenario-postgresql-single finger://ze01.openstack.org/7810644d98f24695a5602bac1e269dd3 : RETRY_LIMIT in 58m 08s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/7/check/trove-scenario-postgresql-multi/32ad9b5/ : FAILURE in 53m 43s (non-voting)\n- trove-scenario-pxc-single finger://ze02.openstack.org/1b64602e66b44d659b098d6c55b8afa6 : RETRY_LIMIT in 45m 32s (non-voting)\n- trove-scenario-pxc-multi finger://ze04.openstack.org/8d736cbf30d142b9aff866c903df5f07 : RETRY_LIMIT in 59m 05s (non-voting)\n- trove-scenario-redis-single finger://ze11.openstack.org/f3a43a84dc404bcf9ec0ba28668dd68d : RETRY_LIMIT in 51m 37s (non-voting)\n- trove-scenario-redis-multi finger://ze07.openstack.org/403e6bd96fa84cd0aac9a8d6a16d8977 : RETRY_LIMIT in 41m 27s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/7/check/trove-tox-fakemodetests/3184510/ : SUCCESS in 10m 23s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/7/check/trove-tox-apiexamples/b4b9447/ : SUCCESS in 5m 23s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/7/check/trove-tox-bandit-baseline/5369f61/ : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/7/check/openstack-tox-pylint/79e8b6d/ : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":7},{"id":"b5dc898e72af6579c6638da379447fb6c49f5f88","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-19 12:20:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b03dbbe1fa95d29878bc8f3d2c40ba4356ea73d8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-21 08:03:23.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"b99ef03611026a59bffe2bd7df027d8fbf3687a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 08:56:55.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/8/check/requirements-check/fb01d3f/ : FAILURE in 4m 16s\n- openstack-tox-cover http://logs.openstack.org/68/652068/8/check/openstack-tox-cover/fccb7bb/cover/ : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/8/check/openstack-tox-lower-constraints/2de96f2/ : SUCCESS in 12m 24s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/8/check/openstack-tox-pep8/64a635a/ : SUCCESS in 7m 26s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/8/check/openstack-tox-py27/5a19575/ : SUCCESS in 14m 24s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/8/check/openstack-tox-py35/3feccea/ : SUCCESS in 16m 56s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/8/check/openstack-tox-py36/32171c0/ : SUCCESS in 15m 53s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/8/check/openstack-tox-py37/9a60c04/ : SUCCESS in 17m 01s\n- openstack-tox-docs http://logs.openstack.org/68/652068/8/check/openstack-tox-docs/19731f7/html/ : SUCCESS in 8m 26s\n- trove-functional-mysql http://logs.openstack.org/68/652068/8/check/trove-functional-mysql/c3b6f5f/ : FAILURE in 32m 14s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/8/check/trove-scenario-mariadb-single/d4edaec/ : FAILURE in 47m 15s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/8/check/trove-scenario-mariadb-multi/b9752f3/ : FAILURE in 30m 53s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/8/check/trove-scenario-mysql-single/21f9a90/ : FAILURE in 30m 46s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/8/check/trove-scenario-mysql-multi/6faf50b/ : FAILURE in 36m 02s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/8/check/trove-scenario-postgresql-single/325b9ee/ : FAILURE in 30m 45s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/8/check/trove-scenario-postgresql-multi/0ac385d/ : FAILURE in 31m 45s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/8/check/trove-scenario-pxc-single/b469a3a/ : FAILURE in 32m 33s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/8/check/trove-scenario-pxc-multi/865862e/ : FAILURE in 33m 45s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/8/check/trove-scenario-redis-single/df34594/ : FAILURE in 33m 23s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/8/check/trove-scenario-redis-multi/6448500/ : FAILURE in 33m 13s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/8/check/trove-tox-fakemodetests/1294595/ : SUCCESS in 5m 40s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/8/check/trove-tox-apiexamples/c059f4d/ : SUCCESS in 7m 01s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/8/check/trove-tox-bandit-baseline/aabf174/ : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/8/check/openstack-tox-pylint/02fe4de/ : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":8},{"id":"e7fe0ed68e5588602b4a300001c2afae36b6a611","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-21 11:09:33.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"c467fdac90b8216bbb7d2307c934e411bd2d9af4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 12:02:26.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/8/check/requirements-check/11ee168/ : FAILURE in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/68/652068/8/check/openstack-tox-cover/5533bf6/cover/ : SUCCESS in 18m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/8/check/openstack-tox-lower-constraints/38455fa/ : SUCCESS in 13m 50s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/8/check/openstack-tox-pep8/c52f2c1/ : SUCCESS in 7m 53s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/8/check/openstack-tox-py27/0720326/ : SUCCESS in 12m 50s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/8/check/openstack-tox-py35/9c95447/ : SUCCESS in 13m 19s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/8/check/openstack-tox-py36/05defdd/ : SUCCESS in 14m 02s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/8/check/openstack-tox-py37/eea7cbe/ : SUCCESS in 15m 35s\n- openstack-tox-docs http://logs.openstack.org/68/652068/8/check/openstack-tox-docs/eb43747/html/ : SUCCESS in 7m 16s\n- trove-functional-mysql http://logs.openstack.org/68/652068/8/check/trove-functional-mysql/f447002/ : FAILURE in 30m 35s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/8/check/trove-scenario-mariadb-single/70db5ac/ : FAILURE in 44m 44s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/8/check/trove-scenario-mariadb-multi/28e1ac2/ : FAILURE in 33m 03s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/8/check/trove-scenario-mysql-single/7d95c2a/ : FAILURE in 35m 55s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/8/check/trove-scenario-mysql-multi/7544547/ : FAILURE in 26m 12s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/8/check/trove-scenario-postgresql-single/5bcf3e2/ : FAILURE in 31m 12s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/8/check/trove-scenario-postgresql-multi/e743c9e/ : FAILURE in 36m 50s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/8/check/trove-scenario-pxc-single/ed77e54/ : FAILURE in 23m 08s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/8/check/trove-scenario-pxc-multi/1400062/ : FAILURE in 34m 45s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/8/check/trove-scenario-redis-single/b0856d5/ : FAILURE in 39m 07s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/8/check/trove-scenario-redis-multi/f9ca9e4/ : FAILURE in 24m 28s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/8/check/trove-tox-fakemodetests/570175d/ : SUCCESS in 6m 29s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/8/check/trove-tox-apiexamples/a45263f/ : SUCCESS in 5m 55s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/8/check/trove-tox-bandit-baseline/1d4d631/ : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/8/check/openstack-tox-pylint/f4d9fdf/ : SUCCESS in 10m 50s","accounts_in_message":[],"_revision_number":8},{"id":"7c3221d1755353543f272186d2c1b995610c21cd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-22 06:31:48.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"ae148e86ff8dcd4b20da658e561c12a5eb502c91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:16:56.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/9/check/requirements-check/4197edf/ : FAILURE in 3m 27s\n- openstack-tox-cover http://logs.openstack.org/68/652068/9/check/openstack-tox-cover/5475a4a/cover/ : SUCCESS in 20m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/9/check/openstack-tox-lower-constraints/35f1179/ : SUCCESS in 12m 54s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/9/check/openstack-tox-pep8/0f8792c/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/9/check/openstack-tox-py27/1536a2e/ : SUCCESS in 10m 06s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/9/check/openstack-tox-py35/92256e6/ : SUCCESS in 13m 52s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/9/check/openstack-tox-py36/e346ab9/ : SUCCESS in 14m 16s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/9/check/openstack-tox-py37/120d825/ : SUCCESS in 19m 01s\n- openstack-tox-docs http://logs.openstack.org/68/652068/9/check/openstack-tox-docs/cf3c188/html/ : SUCCESS in 5m 04s\n- trove-functional-mysql http://logs.openstack.org/68/652068/9/check/trove-functional-mysql/061dcbe/ : FAILURE in 1h 36m 45s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/9/check/trove-scenario-mariadb-single/f69ef7e/ : FAILURE in 41m 36s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/9/check/trove-scenario-mariadb-multi/5f6489d/ : FAILURE in 48m 41s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/9/check/trove-scenario-mysql-single/f68f0ff/ : FAILURE in 1h 14m 00s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/9/check/trove-scenario-mysql-multi/88124f5/ : FAILURE in 55m 24s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/9/check/trove-scenario-postgresql-single/9bfad7d/ : FAILURE in 48m 31s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/9/check/trove-scenario-postgresql-multi/4593a79/ : FAILURE in 48m 50s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/9/check/trove-scenario-pxc-single/277af3c/ : FAILURE in 51m 24s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/9/check/trove-scenario-pxc-multi/be5e881/ : FAILURE in 1h 05m 48s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/9/check/trove-scenario-redis-single/cd7569e/ : FAILURE in 1h 05m 56s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/9/check/trove-scenario-redis-multi/60cb09c/ : FAILURE in 1h 03m 02s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/9/check/trove-tox-fakemodetests/5d91e78/ : SUCCESS in 5m 28s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/9/check/trove-tox-apiexamples/bc1965b/ : SUCCESS in 5m 05s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/9/check/trove-tox-bandit-baseline/cdd0256/ : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/9/check/openstack-tox-pylint/5cb1ecb/ : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":9},{"id":"deb228c1df7b8bf1735fc499ab2239c3cc5bc8b9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-23 12:37:56.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c75d91206869363e7925c45d1240b1a5f3dc5de6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 15:40:42.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/10/check/requirements-check/f2d06fd/ : FAILURE in 5m 26s\n- openstack-tox-cover http://logs.openstack.org/68/652068/10/check/openstack-tox-cover/3c4cb4d/ : FAILURE in 18m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/10/check/openstack-tox-lower-constraints/010ccfe/ : FAILURE in 13m 53s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/10/check/openstack-tox-pep8/5b221e8/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/10/check/openstack-tox-py27/ba42de0/ : FAILURE in 12m 07s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/10/check/openstack-tox-py35/0bf6904/ : FAILURE in 15m 02s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/10/check/openstack-tox-py36/1ff6959/ : FAILURE in 13m 35s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/10/check/openstack-tox-py37/214b23c/ : FAILURE in 19m 10s\n- openstack-tox-docs http://logs.openstack.org/68/652068/10/check/openstack-tox-docs/dcc6184/html/ : SUCCESS in 7m 36s\n- trove-functional-mysql finger://ze09.openstack.org/af1232b27fac43e89713437adee3740d : RETRY_LIMIT in 58m 49s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/10/check/trove-scenario-mariadb-single/29706b0/ : FAILURE in 39m 58s (non-voting)\n- trove-scenario-mariadb-multi finger://ze10.openstack.org/f8f832e2d2164c12ac7c7825b6df58d2 : RETRY_LIMIT in 54m 39s (non-voting)\n- trove-scenario-mysql-single finger://ze09.openstack.org/f3feccffdad14d9ab01742777a69f937 : RETRY_LIMIT in 50m 28s\n- trove-scenario-mysql-multi finger://ze02.openstack.org/067361844ddb407c89c8889d76f24c1b : RETRY_LIMIT in 52m 24s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/10/check/trove-scenario-postgresql-single/2faa3e5/ : FAILURE in 53m 20s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/10/check/trove-scenario-postgresql-multi/192a543/ : POST_FAILURE in 45m 03s (non-voting)\n- trove-scenario-pxc-single finger://ze03.openstack.org/e82e493ff6604b19a3fb4ecfd2fd293d : RETRY_LIMIT in 53m 52s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/10/check/trove-scenario-pxc-multi/3f78a23/ : FAILURE in 47m 12s (non-voting)\n- trove-scenario-redis-single finger://ze07.openstack.org/35a947da9f574f88b00acb53cc726761 : RETRY_LIMIT in 40m 38s (non-voting)\n- trove-scenario-redis-multi finger://ze03.openstack.org/3cea608ebe2d47aca408b4163ddcbde8 : RETRY_LIMIT in 40m 39s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/10/check/trove-tox-fakemodetests/89cffa4/ : SUCCESS in 10m 16s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/10/check/trove-tox-apiexamples/5b36f19/ : SUCCESS in 7m 45s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/10/check/trove-tox-bandit-baseline/53da7bc/ : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/10/check/openstack-tox-pylint/2b6594a/ : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":10},{"id":"f91dcf4bc804857c05f90f67bc9f1136fe1fbf42","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-23 23:13:25.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"75baecbc17770dd0b892c84defacd686a8dfc8c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 01:47:15.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/11/check/requirements-check/b75f4b7/ : SUCCESS in 3m 09s\n- openstack-tox-cover http://logs.openstack.org/68/652068/11/check/openstack-tox-cover/771cde0/cover/ : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/11/check/openstack-tox-lower-constraints/8facc9a/ : SUCCESS in 11m 44s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/11/check/openstack-tox-pep8/fc7a16f/ : SUCCESS in 5m 58s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/11/check/openstack-tox-py27/a647464/ : SUCCESS in 14m 00s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/11/check/openstack-tox-py35/169b21e/ : SUCCESS in 16m 43s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/11/check/openstack-tox-py36/89007ad/ : SUCCESS in 14m 56s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/11/check/openstack-tox-py37/31c4003/ : POST_FAILURE in 23m 09s\n- openstack-tox-docs http://logs.openstack.org/68/652068/11/check/openstack-tox-docs/ca8c722/html/ : SUCCESS in 5m 37s\n- trove-functional-mysql http://logs.openstack.org/68/652068/11/check/trove-functional-mysql/6962a32/ : FAILURE in 2h 26m 09s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/11/check/trove-scenario-mariadb-single/e78ceaf/ : FAILURE in 40m 22s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/11/check/trove-scenario-mariadb-multi/bbc41c7/ : FAILURE in 49m 43s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/11/check/trove-scenario-mysql-single/8fcfb8b/ : FAILURE in 1h 21m 20s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/11/check/trove-scenario-mysql-multi/40e98c3/ : FAILURE in 1h 00m 14s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/11/check/trove-scenario-postgresql-single/a711457/ : FAILURE in 47m 08s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/11/check/trove-scenario-postgresql-multi/a1c0560/ : FAILURE in 48m 17s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/11/check/trove-scenario-pxc-single/53fe2c3/ : POST_FAILURE in 44m 44s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/11/check/trove-scenario-pxc-multi/55bc5d3/ : FAILURE in 49m 03s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/11/check/trove-scenario-redis-single/b667e69/ : FAILURE in 54m 36s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/11/check/trove-scenario-redis-multi/267b3ec/ : FAILURE in 1h 00m 40s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/11/check/trove-tox-fakemodetests/59e8f3e/ : SUCCESS in 6m 03s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/11/check/trove-tox-apiexamples/3591843/ : SUCCESS in 7m 10s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/11/check/trove-tox-bandit-baseline/3158e06/ : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/11/check/openstack-tox-pylint/26ef235/ : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":11},{"id":"21500e71c340bb77357cabce6fa141045f01af58","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-25 21:33:41.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"e24cfd0f779600f6ad7d54e2a38940cfd118f331","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 23:55:21.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/12/check/requirements-check/e169249/ : SUCCESS in 3m 19s\n- openstack-tox-cover http://logs.openstack.org/68/652068/12/check/openstack-tox-cover/de0de36/cover/ : SUCCESS in 22m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/12/check/openstack-tox-lower-constraints/b4a2c37/ : SUCCESS in 12m 04s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/12/check/openstack-tox-pep8/efabce1/ : SUCCESS in 5m 58s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/12/check/openstack-tox-py27/fbd7866/ : SUCCESS in 12m 31s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/12/check/openstack-tox-py35/c56de27/ : SUCCESS in 14m 39s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/12/check/openstack-tox-py36/d6f37f3/ : SUCCESS in 12m 13s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/12/check/openstack-tox-py37/1c5b459/ : SUCCESS in 18m 09s\n- openstack-tox-docs http://logs.openstack.org/68/652068/12/check/openstack-tox-docs/28dfe3f/html/ : SUCCESS in 6m 49s\n- trove-functional-mysql http://logs.openstack.org/68/652068/12/check/trove-functional-mysql/a23e702/ : SUCCESS in 2h 05m 10s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/12/check/trove-scenario-mariadb-single/3b07968/ : FAILURE in 47m 35s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/12/check/trove-scenario-mariadb-multi/de5ebd7/ : FAILURE in 43m 43s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/12/check/trove-scenario-mysql-single/8169c9f/ : FAILURE in 1h 31m 36s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/12/check/trove-scenario-mysql-multi/c0c4784/ : FAILURE in 1h 05m 34s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/12/check/trove-scenario-postgresql-single/14d4eac/ : FAILURE in 1h 03m 14s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/12/check/trove-scenario-postgresql-multi/98bbe2e/ : FAILURE in 44m 20s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/12/check/trove-scenario-pxc-single/ffc63c0/ : FAILURE in 46m 44s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/12/check/trove-scenario-pxc-multi/f88cf5b/ : FAILURE in 58m 55s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/12/check/trove-scenario-redis-single/a53cb9b/ : FAILURE in 1h 10m 40s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/12/check/trove-scenario-redis-multi/7a54c4f/ : FAILURE in 1h 12m 02s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/12/check/trove-tox-fakemodetests/f5d0836/ : SUCCESS in 8m 42s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/12/check/trove-tox-apiexamples/e9eb819/ : SUCCESS in 9m 10s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/12/check/trove-tox-bandit-baseline/34663c5/ : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/12/check/openstack-tox-pylint/f43fc38/ : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":12},{"id":"e8638aed4d0df428aee39aec5f6302e7ba764e48","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-26 02:46:47.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"72d7124db2bacc1c988d3af21beab1870dfcebb3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-26 02:46:55.000000000","message":"Patch Set 13: Workflow-1","accounts_in_message":[],"_revision_number":13},{"id":"549543c5c82d49d8c70f5c04a74e672c11af7f1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 05:01:52.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/13/check/requirements-check/112d4c8/ : SUCCESS in 3m 13s\n- openstack-tox-cover http://logs.openstack.org/68/652068/13/check/openstack-tox-cover/af526b8/cover/ : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/13/check/openstack-tox-lower-constraints/1291242/ : SUCCESS in 12m 36s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/13/check/openstack-tox-pep8/40a54e6/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/13/check/openstack-tox-py27/4422f81/ : SUCCESS in 13m 24s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/13/check/openstack-tox-py35/52b3828/ : SUCCESS in 14m 04s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/13/check/openstack-tox-py36/35e24df/ : SUCCESS in 12m 41s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/13/check/openstack-tox-py37/bff7944/ : SUCCESS in 17m 04s\n- openstack-tox-docs http://logs.openstack.org/68/652068/13/check/openstack-tox-docs/cccfa07/html/ : SUCCESS in 5m 03s\n- trove-functional-mysql http://logs.openstack.org/68/652068/13/check/trove-functional-mysql/38704cf/ : FAILURE in 2h 09m 38s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/13/check/trove-scenario-mariadb-single/0b7689b/ : FAILURE in 40m 19s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/13/check/trove-scenario-mariadb-multi/4f1a85e/ : FAILURE in 46m 18s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/13/check/trove-scenario-mysql-single/60c1c9a/ : FAILURE in 1h 19m 12s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/13/check/trove-scenario-mysql-multi/3490936/ : FAILURE in 52m 25s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/13/check/trove-scenario-postgresql-single/2838b34/ : FAILURE in 50m 03s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/13/check/trove-scenario-postgresql-multi/b9b3ced/ : FAILURE in 50m 37s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/13/check/trove-scenario-pxc-single/fa5e2f9/ : FAILURE in 50m 45s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/13/check/trove-scenario-pxc-multi/bdf1cd1/ : FAILURE in 52m 53s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/13/check/trove-scenario-redis-single/7b214e4/ : FAILURE in 57m 09s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/13/check/trove-scenario-redis-multi/e820ca4/ : FAILURE in 1h 16m 35s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/13/check/trove-tox-fakemodetests/a435973/ : SUCCESS in 7m 44s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/13/check/trove-tox-apiexamples/04c9377/ : SUCCESS in 4m 56s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/13/check/trove-tox-bandit-baseline/fe0533b/ : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/13/check/openstack-tox-pylint/844a9a5/ : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":13},{"id":"cc656907ef26609a8c276abc45b9e794fb79146b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-27 22:51:18.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d5fdf995321b41f3f03e0c26556d86945980b3e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 23:47:01.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/14/check/requirements-check/297e29b/ : SUCCESS in 4m 54s\n- openstack-tox-cover http://logs.openstack.org/68/652068/14/check/openstack-tox-cover/594812c/cover/ : SUCCESS in 17m 47s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/14/check/openstack-tox-lower-constraints/06bfa14/ : SUCCESS in 12m 35s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/14/check/openstack-tox-pep8/50d3f90/ : FAILURE in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/14/check/openstack-tox-py27/bdae5c6/ : SUCCESS in 12m 58s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/14/check/openstack-tox-py35/e7ff8b6/ : SUCCESS in 14m 26s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/14/check/openstack-tox-py36/c3b9fd1/ : SUCCESS in 13m 52s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/14/check/openstack-tox-py37/1fc8bc6/ : SUCCESS in 17m 23s\n- openstack-tox-docs http://logs.openstack.org/68/652068/14/check/openstack-tox-docs/7ee2eb5/html/ : SUCCESS in 6m 52s\n- trove-functional-mysql http://logs.openstack.org/68/652068/14/check/trove-functional-mysql/971dcef/ : FAILURE in 34m 37s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/14/check/trove-scenario-mariadb-single/6d7dd7b/ : FAILURE in 39m 39s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/14/check/trove-scenario-mariadb-multi/3162710/ : FAILURE in 49m 26s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/14/check/trove-scenario-mysql-single/e062b94/ : FAILURE in 42m 17s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/14/check/trove-scenario-mysql-multi/fa09f52/ : FAILURE in 43m 47s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/14/check/trove-scenario-postgresql-single/7861bc3/ : FAILURE in 45m 29s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/14/check/trove-scenario-postgresql-multi/c9aa304/ : FAILURE in 46m 06s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/14/check/trove-scenario-pxc-single/8c6f019/ : FAILURE in 45m 56s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/14/check/trove-scenario-pxc-multi/8e8d0aa/ : FAILURE in 48m 13s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/14/check/trove-scenario-redis-single/b58b6d9/ : FAILURE in 50m 35s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/14/check/trove-scenario-redis-multi/f673d7b/ : FAILURE in 50m 02s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/14/check/trove-tox-fakemodetests/76c4359/ : SUCCESS in 5m 19s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/14/check/trove-tox-apiexamples/85154ff/ : SUCCESS in 5m 08s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/14/check/trove-tox-bandit-baseline/87f67d8/ : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/14/check/openstack-tox-pylint/61770de/ : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":14},{"id":"7cdfbab9f41fc0161b9dd2a77c41f7addf95d3e2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-28 05:35:46.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"74bec6cb8ab5857bb50ddf5a990459216a577a76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 06:45:31.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/14/check/requirements-check/8b88486/ : SUCCESS in 3m 32s\n- openstack-tox-cover http://logs.openstack.org/68/652068/14/check/openstack-tox-cover/5ec521e/cover/ : SUCCESS in 17m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/14/check/openstack-tox-lower-constraints/fe4fe5e/ : SUCCESS in 12m 26s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/14/check/openstack-tox-pep8/9a3db88/ : FAILURE in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/14/check/openstack-tox-py27/a2f12a6/ : SUCCESS in 12m 03s\n- openstack-tox-py35 http://logs.openstack.org/68/652068/14/check/openstack-tox-py35/88ff7ad/ : SUCCESS in 13m 50s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/14/check/openstack-tox-py36/288fa84/ : SUCCESS in 13m 51s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/14/check/openstack-tox-py37/38a9bc4/ : SUCCESS in 18m 09s\n- openstack-tox-docs http://logs.openstack.org/68/652068/14/check/openstack-tox-docs/f61dadd/html/ : SUCCESS in 5m 21s\n- trove-functional-mysql http://logs.openstack.org/68/652068/14/check/trove-functional-mysql/01f08ec/ : FAILURE in 54m 30s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/14/check/trove-scenario-mariadb-single/fd5bfe4/ : FAILURE in 52m 03s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/14/check/trove-scenario-mariadb-multi/b789d31/ : FAILURE in 40m 17s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/14/check/trove-scenario-mysql-single/d9c2bb6/ : FAILURE in 52m 39s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/14/check/trove-scenario-mysql-multi/7da72cf/ : FAILURE in 35m 10s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/14/check/trove-scenario-postgresql-single/17690ff/ : FAILURE in 28m 27s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/14/check/trove-scenario-postgresql-multi/3b67d50/ : FAILURE in 49m 23s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/14/check/trove-scenario-pxc-single/034c7e0/ : FAILURE in 44m 44s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/14/check/trove-scenario-pxc-multi/9d70a85/ : FAILURE in 44m 48s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/14/check/trove-scenario-redis-single/894e0f4/ : FAILURE in 1h 02m 01s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/14/check/trove-scenario-redis-multi/4c1f2a5/ : FAILURE in 1h 00m 30s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/14/check/trove-tox-fakemodetests/a60ce8b/ : SUCCESS in 5m 59s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/14/check/trove-tox-apiexamples/148f39c/ : SUCCESS in 7m 12s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/14/check/trove-tox-bandit-baseline/60aac5a/ : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/14/check/openstack-tox-pylint/005bf41/ : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":14},{"id":"a50cc0e2a129a18e3090bf7767afb2e0f8dc2f16","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-29 02:51:23.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"6b8b80d04a8ba2234cdc38ae0f3bb599d0e2914d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 03:49:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/15/check/requirements-check/1985b76/ : SUCCESS in 3m 29s\n- openstack-tox-cover http://logs.openstack.org/68/652068/15/check/openstack-tox-cover/e5cbd6e/cover/ : SUCCESS in 22m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/15/check/openstack-tox-lower-constraints/db66239/ : SUCCESS in 10m 43s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/15/check/openstack-tox-pep8/1c306fd/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/15/check/openstack-tox-py27/fed45c3/ : SUCCESS in 10m 48s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/15/check/openstack-tox-py36/a8e17da/ : SUCCESS in 10m 57s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/15/check/openstack-tox-py37/3317125/ : SUCCESS in 17m 59s\n- openstack-tox-docs http://logs.openstack.org/68/652068/15/check/openstack-tox-docs/100d130/html/ : SUCCESS in 4m 56s\n- trove-functional-mysql http://logs.openstack.org/68/652068/15/check/trove-functional-mysql/9336b1c/ : FAILURE in 47m 04s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/15/check/trove-scenario-mariadb-single/ab8773d/ : FAILURE in 42m 27s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/15/check/trove-scenario-mariadb-multi/f3b08fb/ : FAILURE in 44m 17s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/15/check/trove-scenario-mysql-single/b35cfff/ : FAILURE in 38m 08s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/15/check/trove-scenario-mysql-multi/aea5a2e/ : FAILURE in 40m 20s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/15/check/trove-scenario-postgresql-single/4f57b67/ : FAILURE in 44m 49s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/15/check/trove-scenario-postgresql-multi/89ece5b/ : FAILURE in 43m 32s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/15/check/trove-scenario-pxc-single/008fa06/ : FAILURE in 47m 16s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/15/check/trove-scenario-pxc-multi/2bb4625/ : FAILURE in 42m 57s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/15/check/trove-scenario-redis-single/2bf39bd/ : FAILURE in 52m 53s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/15/check/trove-scenario-redis-multi/d6bb6e9/ : FAILURE in 47m 45s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/15/check/trove-tox-fakemodetests/da43cf3/ : SUCCESS in 5m 59s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/15/check/trove-tox-apiexamples/eae3abd/ : SUCCESS in 5m 21s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/15/check/trove-tox-bandit-baseline/788ec3a/ : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/15/check/openstack-tox-pylint/a37fd33/ : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":15},{"id":"cbd8546b1cdeb06ec96e4658a66a6f6b52b0742e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-08 13:08:16.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"e7ed818e2cab23e52961661f7325eff80a87893e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 15:26:03.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/16/check/requirements-check/27044b3/ : SUCCESS in 3m 19s\n- openstack-tox-cover http://logs.openstack.org/68/652068/16/check/openstack-tox-cover/f68c764/cover/ : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/16/check/openstack-tox-lower-constraints/f3c022c/ : SUCCESS in 11m 56s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/16/check/openstack-tox-pep8/b10a514/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/16/check/openstack-tox-py27/da05138/ : SUCCESS in 11m 01s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/16/check/openstack-tox-py36/a61e750/ : SUCCESS in 12m 37s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/16/check/openstack-tox-py37/8d70e72/ : SUCCESS in 16m 30s\n- openstack-tox-docs http://logs.openstack.org/68/652068/16/check/openstack-tox-docs/fa921e5/html/ : SUCCESS in 4m 55s\n- trove-functional-mysql http://logs.openstack.org/68/652068/16/check/trove-functional-mysql/46ef3b7/ : SUCCESS in 2h 12m 48s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/16/check/trove-scenario-mariadb-single/13fad05/ : FAILURE in 1h 01m 07s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/16/check/trove-scenario-mariadb-multi/2a19f5a/ : FAILURE in 1h 25m 32s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/16/check/trove-scenario-mysql-single/4fcb523/ : FAILURE in 2h 08m 07s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/16/check/trove-scenario-mysql-multi/71378ae/ : SUCCESS in 2h 01m 40s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/16/check/trove-scenario-postgresql-single/ee1c259/ : FAILURE in 1h 25m 23s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/16/check/trove-scenario-postgresql-multi/ab416ca/ : FAILURE in 1h 23m 20s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/16/check/trove-scenario-pxc-single/b342161/ : FAILURE in 1h 24m 17s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/16/check/trove-scenario-pxc-multi/2b08480/ : FAILURE in 1h 23m 26s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/16/check/trove-scenario-redis-single/d3d0113/ : FAILURE in 1h 53m 41s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/16/check/trove-scenario-redis-multi/b42bf8f/ : FAILURE in 1h 42m 47s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/16/check/trove-tox-fakemodetests/ee860b4/ : SUCCESS in 5m 08s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/16/check/trove-tox-apiexamples/713deef/ : SUCCESS in 4m 47s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/16/check/trove-tox-bandit-baseline/499bd89/ : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/16/check/openstack-tox-pylint/0d53a23/ : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":16},{"id":"5e1628e32ea0731dc39ceab87db9a4ed7c02cb6c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-09 08:11:59.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"7c15ddf7a82ed13d11ea894aa572a76728b7b144","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 10:17:35.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/17/check/requirements-check/cd7c352/ : SUCCESS in 5m 31s\n- openstack-tox-cover http://logs.openstack.org/68/652068/17/check/openstack-tox-cover/8d1271b/cover/ : SUCCESS in 17m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/17/check/openstack-tox-lower-constraints/97a8173/ : SUCCESS in 12m 25s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/17/check/openstack-tox-pep8/977ec96/ : SUCCESS in 6m 54s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/17/check/openstack-tox-py27/5a3f80d/ : SUCCESS in 12m 22s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/17/check/openstack-tox-py36/291bcfe/ : SUCCESS in 12m 52s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/17/check/openstack-tox-py37/1ede13e/ : SUCCESS in 17m 38s\n- openstack-tox-docs http://logs.openstack.org/68/652068/17/check/openstack-tox-docs/ff1474d/html/ : SUCCESS in 6m 37s\n- trove-functional-mysql http://logs.openstack.org/68/652068/17/check/trove-functional-mysql/5e39480/ : SUCCESS in 2h 00m 30s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/17/check/trove-scenario-mariadb-single/e7d43b7/ : FAILURE in 41m 33s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/17/check/trove-scenario-mariadb-multi/347c091/ : FAILURE in 40m 26s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/17/check/trove-scenario-mysql-single/011a0ec/ : FAILURE in 1h 50m 51s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/17/check/trove-scenario-mysql-multi/cfcc926/ : SUCCESS in 1h 38m 21s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/17/check/trove-scenario-postgresql-single/6b0907a/ : FAILURE in 41m 01s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/17/check/trove-scenario-postgresql-multi/abeaa4d/ : FAILURE in 39m 16s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/17/check/trove-scenario-pxc-single/00e6405/ : FAILURE in 41m 27s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/17/check/trove-scenario-pxc-multi/aaad1cd/ : FAILURE in 41m 32s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/17/check/trove-scenario-redis-single/2cbed2a/ : FAILURE in 1h 04m 07s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/17/check/trove-scenario-redis-multi/008fa34/ : FAILURE in 1h 04m 19s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/17/check/trove-tox-fakemodetests/e8c0820/ : SUCCESS in 7m 21s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/17/check/trove-tox-apiexamples/029cf32/ : SUCCESS in 5m 07s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/17/check/trove-tox-bandit-baseline/561c53e/ : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/17/check/openstack-tox-pylint/a1929c4/ : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":17},{"id":"85a063a72b8631174cf7140b76718dd6e177033b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-10 22:23:57.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"66d7a16a35ec896d28381d8de18960023472c208","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 00:22:01.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/18/check/requirements-check/e6ceefd/ : SUCCESS in 3m 55s\n- openstack-tox-cover http://logs.openstack.org/68/652068/18/check/openstack-tox-cover/4ff15cf/cover/ : SUCCESS in 19m 58s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/18/check/openstack-tox-lower-constraints/b5ef53a/ : SUCCESS in 11m 52s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/18/check/openstack-tox-pep8/612fce5/ : SUCCESS in 5m 17s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/18/check/openstack-tox-py27/27823d3/ : SUCCESS in 12m 02s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/18/check/openstack-tox-py36/a765e6f/ : SUCCESS in 12m 53s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/18/check/openstack-tox-py37/667a657/ : SUCCESS in 15m 05s\n- openstack-tox-docs http://logs.openstack.org/68/652068/18/check/openstack-tox-docs/d8f3030/html/ : SUCCESS in 5m 16s\n- trove-functional-mysql http://logs.openstack.org/68/652068/18/check/trove-functional-mysql/6dc3c07/ : SUCCESS in 1h 56m 08s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/18/check/trove-scenario-mariadb-single/2a40a96/ : FAILURE in 37m 42s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/18/check/trove-scenario-mariadb-multi/b351325/ : FAILURE in 42m 37s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/18/check/trove-scenario-mysql-single/41194a9/ : FAILURE in 1h 34m 41s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/18/check/trove-scenario-mysql-multi/7248926/ : SUCCESS in 1h 50m 01s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/18/check/trove-scenario-postgresql-single/6f4bf52/ : FAILURE in 39m 11s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/18/check/trove-scenario-postgresql-multi/bb46ebe/ : FAILURE in 49m 14s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/18/check/trove-scenario-pxc-single/3dc15d3/ : FAILURE in 44m 00s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/18/check/trove-scenario-pxc-multi/0176de5/ : FAILURE in 44m 36s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/18/check/trove-scenario-redis-single/76feb52/ : FAILURE in 1h 07m 24s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/18/check/trove-scenario-redis-multi/7b4a295/ : FAILURE in 1h 11m 18s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/18/check/trove-tox-fakemodetests/495377d/ : SUCCESS in 6m 35s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/18/check/trove-tox-apiexamples/902bfbe/ : SUCCESS in 7m 42s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/18/check/trove-tox-bandit-baseline/c46977e/ : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/18/check/openstack-tox-pylint/c46220c/ : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":18},{"id":"52123855877f14b99c163a0a7507308f71c486f9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-13 12:37:00.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"b3ffbabaf60a2884f69af7f8973923a1e002c765","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 14:34:47.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/19/check/requirements-check/66c16ac/ : FAILURE in 3m 27s\n- openstack-tox-cover http://logs.openstack.org/68/652068/19/check/openstack-tox-cover/bd9d428/cover/ : SUCCESS in 15m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/19/check/openstack-tox-lower-constraints/c987f58/ : SUCCESS in 12m 51s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/19/check/openstack-tox-pep8/49db0cb/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/19/check/openstack-tox-py27/21e647f/ : SUCCESS in 11m 47s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/19/check/openstack-tox-py36/ba58813/ : SUCCESS in 12m 11s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/19/check/openstack-tox-py37/fc76399/ : SUCCESS in 17m 48s\n- openstack-tox-docs http://logs.openstack.org/68/652068/19/check/openstack-tox-docs/6be2a7f/html/ : SUCCESS in 5m 13s\n- trove-functional-mysql http://logs.openstack.org/68/652068/19/check/trove-functional-mysql/331f917/ : FAILURE in 28m 54s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/19/check/trove-scenario-mariadb-single/f6a4e34/ : FAILURE in 49m 38s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/19/check/trove-scenario-mariadb-multi/c4c689b/ : FAILURE in 50m 35s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/19/check/trove-scenario-mysql-single/373d4bb/ : FAILURE in 1h 51m 23s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/19/check/trove-scenario-mysql-multi/d53c9fb/ : FAILURE in 25m 45s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/19/check/trove-scenario-postgresql-single/89e9804/ : FAILURE in 44m 25s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/19/check/trove-scenario-postgresql-multi/d756ea4/ : FAILURE in 44m 29s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/19/check/trove-scenario-pxc-single/23d1c86/ : FAILURE in 47m 15s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/19/check/trove-scenario-pxc-multi/c8571d8/ : FAILURE in 50m 39s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/19/check/trove-scenario-redis-single/9e6a371/ : FAILURE in 1h 09m 29s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/19/check/trove-scenario-redis-multi/c3d374e/ : FAILURE in 1h 05m 00s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/19/check/trove-tox-fakemodetests/a2eff0e/ : SUCCESS in 5m 28s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/19/check/trove-tox-apiexamples/38020b8/ : SUCCESS in 4m 49s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/19/check/trove-tox-bandit-baseline/51c257a/ : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/19/check/openstack-tox-pylint/6282b2a/ : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":19},{"id":"3f865f10fc5bf0c8471cee8a326daa5302600f95","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-13 22:57:41.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"44a47c8d8525d3f8fb4f18d3ab9670f440afd643","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-13 22:58:04.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"40363a01bb2a467963392de4d3486f14d6c31710","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:56:45.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/20/check/build-openstack-api-ref/1f47d50/html/ : SUCCESS in 5m 02s\n- requirements-check http://logs.openstack.org/68/652068/20/check/requirements-check/6f70a32/ : FAILURE in 3m 26s\n- openstack-tox-cover http://logs.openstack.org/68/652068/20/check/openstack-tox-cover/b15c674/cover/ : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/20/check/openstack-tox-lower-constraints/1218bb5/ : SUCCESS in 14m 08s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/20/check/openstack-tox-pep8/fa58229/ : SUCCESS in 6m 55s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/20/check/openstack-tox-py27/9d48088/ : SUCCESS in 10m 58s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/20/check/openstack-tox-py36/6e15f2e/ : SUCCESS in 14m 37s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/20/check/openstack-tox-py37/23e66cf/ : SUCCESS in 18m 16s\n- openstack-tox-docs http://logs.openstack.org/68/652068/20/check/openstack-tox-docs/c24ee41/html/ : SUCCESS in 5m 59s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/20/check/build-openstack-releasenotes/8482cd7/html/ : SUCCESS in 6m 03s\n- trove-functional-mysql http://logs.openstack.org/68/652068/20/check/trove-functional-mysql/fa91ac7/ : SUCCESS in 1h 53m 13s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/20/check/trove-scenario-mariadb-single/0794eef/ : FAILURE in 40m 46s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/20/check/trove-scenario-mariadb-multi/033d1a1/ : FAILURE in 39m 41s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/20/check/trove-scenario-mysql-single/eb2e341/ : FAILURE in 1h 33m 27s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/20/check/trove-scenario-mysql-multi/8b1e1cc/ : SUCCESS in 1h 28m 02s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/20/check/trove-scenario-postgresql-single/dfa3eb8/ : FAILURE in 40m 39s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/20/check/trove-scenario-postgresql-multi/c036f0f/ : FAILURE in 41m 40s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/20/check/trove-scenario-pxc-single/d0b96a9/ : FAILURE in 39m 56s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/20/check/trove-scenario-pxc-multi/e066500/ : FAILURE in 39m 11s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/20/check/trove-scenario-redis-single/0157af9/ : FAILURE in 1h 05m 07s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/20/check/trove-scenario-redis-multi/41fb67e/ : FAILURE in 1h 01m 13s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/20/check/trove-tox-fakemodetests/e7219c4/ : SUCCESS in 5m 17s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/20/check/trove-tox-apiexamples/2ac5c80/ : SUCCESS in 6m 10s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/20/check/trove-tox-bandit-baseline/5216528/ : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/20/check/openstack-tox-pylint/5862fb1/ : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":20},{"id":"46772edf9a5c10aa2d2a0f4cec611de66d1917f5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-14 02:59:20.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"36b9cdb1bf66817066082eb7f76daf5b41a585a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 05:13:36.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/21/check/build-openstack-api-ref/437edda/html/ : SUCCESS in 4m 43s\n- requirements-check http://logs.openstack.org/68/652068/21/check/requirements-check/f99ff5c/ : SUCCESS in 3m 09s\n- openstack-tox-cover http://logs.openstack.org/68/652068/21/check/openstack-tox-cover/b7dff82/cover/ : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/21/check/openstack-tox-lower-constraints/ec481d2/ : SUCCESS in 11m 15s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/21/check/openstack-tox-pep8/0766c0a/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/21/check/openstack-tox-py27/fcab997/ : SUCCESS in 11m 21s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/21/check/openstack-tox-py36/49d359a/ : SUCCESS in 13m 57s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/21/check/openstack-tox-py37/9fea55f/ : SUCCESS in 18m 21s\n- openstack-tox-docs http://logs.openstack.org/68/652068/21/check/openstack-tox-docs/34a7b6b/html/ : SUCCESS in 5m 50s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/21/check/build-openstack-releasenotes/2afe25f/html/ : SUCCESS in 5m 36s\n- trove-functional-mysql http://logs.openstack.org/68/652068/21/check/trove-functional-mysql/d7f6798/ : SUCCESS in 2h 08m 44s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/21/check/trove-scenario-mariadb-single/4ab8b28/ : FAILURE in 38m 42s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/21/check/trove-scenario-mariadb-multi/1f9e866/ : FAILURE in 33m 57s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/21/check/trove-scenario-mysql-single/16161df/ : SUCCESS in 1h 38m 49s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/21/check/trove-scenario-mysql-multi/194b8c3/ : SUCCESS in 1h 32m 09s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/21/check/trove-scenario-postgresql-single/a860e7c/ : FAILURE in 37m 58s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/21/check/trove-scenario-postgresql-multi/5ddf930/ : FAILURE in 38m 43s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/21/check/trove-scenario-pxc-single/b8f5b48/ : FAILURE in 32m 41s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/21/check/trove-scenario-pxc-multi/f82bc78/ : FAILURE in 45m 03s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/21/check/trove-scenario-redis-single/368e790/ : FAILURE in 1h 02m 01s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/21/check/trove-scenario-redis-multi/03e7c4c/ : FAILURE in 1h 48m 59s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/21/check/trove-tox-fakemodetests/d124ecf/ : SUCCESS in 5m 59s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/21/check/trove-tox-apiexamples/4af85a1/ : SUCCESS in 4m 33s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/21/check/trove-tox-bandit-baseline/f4f0f2f/ : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/21/check/openstack-tox-pylint/eb28b86/ : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":21},{"id":"395d973ba0a3e1a9a6676cd3b0b1c779be8463b6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-14 06:57:02.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"65e56324f708de1fbcc7b6f56bac4e6034fb8fe4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 09:18:06.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/22/check/requirements-check/8806dae/ : SUCCESS in 4m 35s\n- openstack-tox-cover http://logs.openstack.org/68/652068/22/check/openstack-tox-cover/17bdf57/cover/ : SUCCESS in 15m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/22/check/openstack-tox-lower-constraints/83b6a04/ : SUCCESS in 11m 50s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/22/check/openstack-tox-pep8/e236228/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/22/check/openstack-tox-py27/d96287b/ : SUCCESS in 14m 04s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/22/check/openstack-tox-py36/0e154af/ : SUCCESS in 13m 46s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/22/check/openstack-tox-py37/7ff5074/ : SUCCESS in 16m 10s\n- openstack-tox-docs http://logs.openstack.org/68/652068/22/check/openstack-tox-docs/c1037d4/html/ : SUCCESS in 6m 12s\n- trove-functional-mysql http://logs.openstack.org/68/652068/22/check/trove-functional-mysql/41fcbd9/ : FAILURE in 1h 58m 32s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/22/check/trove-scenario-mariadb-single/3767429/ : FAILURE in 41m 13s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-mariadb-multi/bc3acd8/ : FAILURE in 42m 26s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/22/check/trove-scenario-mysql-single/f91a121/ : FAILURE in 2h 17m 27s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-mysql-multi/b11f978/ : SUCCESS in 1h 38m 51s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/22/check/trove-scenario-postgresql-single/08ed1f0/ : FAILURE in 42m 40s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-postgresql-multi/4a45a84/ : FAILURE in 40m 48s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/22/check/trove-scenario-pxc-single/fccc34e/ : FAILURE in 43m 08s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-pxc-multi/ccb9cd6/ : FAILURE in 43m 26s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/22/check/trove-scenario-redis-single/a033d33/ : FAILURE in 55m 11s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-redis-multi/12094d4/ : FAILURE in 1h 01m 55s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/22/check/trove-tox-fakemodetests/d573348/ : SUCCESS in 5m 28s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/22/check/trove-tox-apiexamples/3a047c3/ : SUCCESS in 6m 02s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/22/check/trove-tox-bandit-baseline/4993ee5/ : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/22/check/openstack-tox-pylint/3d07654/ : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":22},{"id":"d86d437a93e5ea607e1aeb4d604a6f1eaac3cd83","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-14 11:36:24.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"4eb9be27ef00096b23a4a03c0a3e5a65d59233f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 13:24:42.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/22/check/requirements-check/e95822b/ : SUCCESS in 4m 28s\n- openstack-tox-cover http://logs.openstack.org/68/652068/22/check/openstack-tox-cover/6e163de/cover/ : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/22/check/openstack-tox-lower-constraints/3eccff2/ : SUCCESS in 10m 48s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/22/check/openstack-tox-pep8/c2fd846/ : SUCCESS in 7m 26s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/22/check/openstack-tox-py27/b423ebe/ : SUCCESS in 11m 00s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/22/check/openstack-tox-py36/d460713/ : SUCCESS in 12m 23s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/22/check/openstack-tox-py37/a7f791e/ : SUCCESS in 17m 55s\n- openstack-tox-docs http://logs.openstack.org/68/652068/22/check/openstack-tox-docs/549ed0a/html/ : SUCCESS in 4m 49s\n- trove-functional-mysql http://logs.openstack.org/68/652068/22/check/trove-functional-mysql/16d9873/ : FAILURE in 39m 18s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/22/check/trove-scenario-mariadb-single/8dc2593/ : FAILURE in 37m 21s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-mariadb-multi/2ba9b51/ : FAILURE in 42m 57s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/22/check/trove-scenario-mysql-single/6cba1c8/ : FAILURE in 1h 35m 32s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-mysql-multi/4f97d83/ : SUCCESS in 1h 40m 57s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/22/check/trove-scenario-postgresql-single/a8ca7a7/ : FAILURE in 36m 14s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-postgresql-multi/523748a/ : FAILURE in 39m 42s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/22/check/trove-scenario-pxc-single/f21ecc1/ : FAILURE in 42m 15s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-pxc-multi/ec45d18/ : FAILURE in 39m 59s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/22/check/trove-scenario-redis-single/3354e39/ : FAILURE in 1h 01m 51s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-redis-multi/e212118/ : FAILURE in 1h 03m 50s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/22/check/trove-tox-fakemodetests/a753541/ : SUCCESS in 6m 15s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/22/check/trove-tox-apiexamples/85c5f14/ : SUCCESS in 5m 03s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/22/check/trove-tox-bandit-baseline/afd6f96/ : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/22/check/openstack-tox-pylint/596abb1/ : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":22},{"id":"b9f68257aeebb6ecf7ee3d010954e25660e537dd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-14 22:33:15.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"1efc0773b2a1011de4a4b887f0ce135a11b69f9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 00:39:20.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/22/check/requirements-check/fdc403a/ : SUCCESS in 3m 20s\n- openstack-tox-cover http://logs.openstack.org/68/652068/22/check/openstack-tox-cover/ffa99b7/cover/ : SUCCESS in 20m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/22/check/openstack-tox-lower-constraints/d68df6a/ : SUCCESS in 11m 35s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/22/check/openstack-tox-pep8/b401d0d/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/22/check/openstack-tox-py27/72ed017/ : SUCCESS in 10m 52s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/22/check/openstack-tox-py36/0d12489/ : SUCCESS in 12m 14s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/22/check/openstack-tox-py37/35580ea/ : SUCCESS in 17m 22s\n- openstack-tox-docs http://logs.openstack.org/68/652068/22/check/openstack-tox-docs/c8e15ce/html/ : SUCCESS in 4m 30s\n- trove-functional-mysql http://logs.openstack.org/68/652068/22/check/trove-functional-mysql/21504bc/ : SUCCESS in 2h 00m 25s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/22/check/trove-scenario-mariadb-single/1d71f66/ : FAILURE in 40m 17s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-mariadb-multi/a71b391/ : FAILURE in 39m 36s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/22/check/trove-scenario-mysql-single/bcc52c9/ : FAILURE in 1h 53m 02s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-mysql-multi/4404ff0/ : SUCCESS in 1h 30m 36s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/22/check/trove-scenario-postgresql-single/0305b89/ : FAILURE in 41m 21s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-postgresql-multi/7fff575/ : FAILURE in 44m 14s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/22/check/trove-scenario-pxc-single/84ca618/ : FAILURE in 39m 52s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-pxc-multi/f82b912/ : FAILURE in 47m 57s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/22/check/trove-scenario-redis-single/9cf48b5/ : FAILURE in 55m 33s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-redis-multi/8af4f71/ : FAILURE in 54m 21s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/22/check/trove-tox-fakemodetests/9ffb6e0/ : SUCCESS in 5m 01s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/22/check/trove-tox-apiexamples/02e2e70/ : SUCCESS in 4m 54s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/22/check/trove-tox-bandit-baseline/2ad81bb/ : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/22/check/openstack-tox-pylint/2a2ac49/ : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":22},{"id":"6fc33ee20baf1e7142fa05c97dd7d3b8f84ca49d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-15 01:52:19.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"3f08a1d33c97cdfabf522e34e1216215d01e6960","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 04:00:38.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/22/check/requirements-check/747b60d/ : SUCCESS in 3m 15s\n- openstack-tox-cover http://logs.openstack.org/68/652068/22/check/openstack-tox-cover/5e0d9ea/cover/ : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/22/check/openstack-tox-lower-constraints/0a989d8/ : SUCCESS in 12m 53s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/22/check/openstack-tox-pep8/5df37fa/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/22/check/openstack-tox-py27/c89a0fe/ : SUCCESS in 11m 00s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/22/check/openstack-tox-py36/9ed1f29/ : SUCCESS in 12m 35s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/22/check/openstack-tox-py37/7b4c4f7/ : SUCCESS in 17m 46s\n- openstack-tox-docs http://logs.openstack.org/68/652068/22/check/openstack-tox-docs/aff8e28/html/ : SUCCESS in 6m 10s\n- trove-functional-mysql http://logs.openstack.org/68/652068/22/check/trove-functional-mysql/ec3b559/ : FAILURE in 2h 01m 21s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/22/check/trove-scenario-mariadb-single/7c27293/ : FAILURE in 42m 51s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-mariadb-multi/0121051/ : FAILURE in 37m 02s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/22/check/trove-scenario-mysql-single/7bda7ea/ : FAILURE in 1h 38m 15s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-mysql-multi/3337d4c/ : FAILURE in 1h 32m 03s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/22/check/trove-scenario-postgresql-single/1ae30e6/ : FAILURE in 38m 23s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-postgresql-multi/c3e932f/ : FAILURE in 39m 52s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/22/check/trove-scenario-pxc-single/f8d79f8/ : FAILURE in 37m 29s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-pxc-multi/33358bd/ : FAILURE in 26m 13s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/22/check/trove-scenario-redis-single/4b1fd9a/ : FAILURE in 1h 01m 02s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/22/check/trove-scenario-redis-multi/290b585/ : FAILURE in 1h 06m 52s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/22/check/trove-tox-fakemodetests/eae51f9/ : SUCCESS in 5m 35s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/22/check/trove-tox-apiexamples/b195774/ : SUCCESS in 4m 51s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/22/check/trove-tox-bandit-baseline/8d151ab/ : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/22/check/openstack-tox-pylint/7b7c6b8/ : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":22},{"id":"ed1a46f9aabe683920f95fbe43221c1850f9a0b9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-17 01:17:53.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"97a01f483b5eeef6e55c9fdd2a7b51ee7d09ff1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 03:39:51.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/23/check/requirements-check/0068331/ : SUCCESS in 4m 33s\n- openstack-tox-cover http://logs.openstack.org/68/652068/23/check/openstack-tox-cover/921c58b/cover/ : SUCCESS in 16m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/23/check/openstack-tox-lower-constraints/2bbe70f/ : SUCCESS in 12m 44s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/23/check/openstack-tox-pep8/7be1a64/ : SUCCESS in 6m 07s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/23/check/openstack-tox-py27/71ad4a9/ : SUCCESS in 12m 11s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/23/check/openstack-tox-py36/2303b53/ : SUCCESS in 13m 49s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/23/check/openstack-tox-py37/6c2f4b7/ : SUCCESS in 16m 57s\n- openstack-tox-docs http://logs.openstack.org/68/652068/23/check/openstack-tox-docs/8f8becc/html/ : SUCCESS in 4m 37s\n- trove-functional-mysql http://logs.openstack.org/68/652068/23/check/trove-functional-mysql/06edc56/ : FAILURE in 2h 09m 23s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/23/check/trove-scenario-mariadb-single/15ea04f/ : FAILURE in 49m 46s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/23/check/trove-scenario-mariadb-multi/52ab89a/ : FAILURE in 38m 32s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/23/check/trove-scenario-mysql-single/ff40cf9/ : SUCCESS in 1h 40m 43s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/23/check/trove-scenario-mysql-multi/689250f/ : SUCCESS in 1h 34m 32s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/23/check/trove-scenario-postgresql-single/d939830/ : FAILURE in 39m 56s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/23/check/trove-scenario-postgresql-multi/ed7c1d8/ : FAILURE in 39m 50s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/23/check/trove-scenario-pxc-single/776ea9c/ : FAILURE in 45m 31s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/23/check/trove-scenario-pxc-multi/4197f61/ : FAILURE in 38m 56s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/23/check/trove-scenario-redis-single/f15fef2/ : FAILURE in 53m 11s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/23/check/trove-scenario-redis-multi/b819c31/ : FAILURE in 1h 00m 48s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/23/check/trove-tox-fakemodetests/e688795/ : SUCCESS in 7m 09s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/23/check/trove-tox-apiexamples/b7bb93c/ : SUCCESS in 4m 41s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/23/check/trove-tox-bandit-baseline/b7daab1/ : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/23/check/openstack-tox-pylint/3ccb48c/ : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":23},{"id":"bf75bd6ed1c7bec99e8d72c54c725c1f3aae0b4b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-17 21:59:13.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"e5d790af01d5ce3519323eaa7a4261b83cef11e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 00:08:16.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/24/check/requirements-check/79079fb/ : SUCCESS in 3m 59s\n- openstack-tox-cover http://logs.openstack.org/68/652068/24/check/openstack-tox-cover/a3bfece/cover/ : SUCCESS in 15m 58s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/24/check/openstack-tox-lower-constraints/e106050/ : SUCCESS in 12m 11s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/24/check/openstack-tox-pep8/acc0eee/ : SUCCESS in 6m 04s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/24/check/openstack-tox-py27/7d56336/ : SUCCESS in 10m 15s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/24/check/openstack-tox-py36/a3d42ea/ : SUCCESS in 12m 20s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/24/check/openstack-tox-py37/e9fc829/ : SUCCESS in 17m 32s\n- openstack-tox-docs http://logs.openstack.org/68/652068/24/check/openstack-tox-docs/ec679a1/html/ : SUCCESS in 6m 14s\n- trove-functional-mysql http://logs.openstack.org/68/652068/24/check/trove-functional-mysql/05c88ca/ : FAILURE in 2h 03m 51s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/24/check/trove-scenario-mariadb-single/789f981/ : FAILURE in 40m 19s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/24/check/trove-scenario-mariadb-multi/102813e/ : FAILURE in 42m 38s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/24/check/trove-scenario-mysql-single/77da8c1/ : SUCCESS in 1h 37m 11s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/24/check/trove-scenario-mysql-multi/9742046/ : SUCCESS in 1h 36m 21s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/24/check/trove-scenario-postgresql-single/c382cc2/ : FAILURE in 42m 12s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/24/check/trove-scenario-postgresql-multi/6117ecd/ : FAILURE in 37m 54s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/24/check/trove-scenario-pxc-single/8ea1157/ : FAILURE in 48m 07s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/24/check/trove-scenario-pxc-multi/6fc3aac/ : FAILURE in 38m 14s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/24/check/trove-scenario-redis-single/7a70198/ : FAILURE in 59m 39s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/24/check/trove-scenario-redis-multi/dde75c0/ : FAILURE in 1h 02m 44s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/24/check/trove-tox-fakemodetests/3df8543/ : SUCCESS in 6m 40s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/24/check/trove-tox-apiexamples/58b5596/ : SUCCESS in 4m 59s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/24/check/trove-tox-bandit-baseline/cc9c385/ : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/24/check/openstack-tox-pylint/fe825a5/ : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":24},{"id":"19465264b1d65539bea545ce0f8240145f4590f1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-18 11:30:58.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"ef36a4c2daa69bc0801ac9f9c972543dafcdc045","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 13:16:55.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/24/check/requirements-check/dc38731/ : SUCCESS in 2m 59s\n- openstack-tox-cover http://logs.openstack.org/68/652068/24/check/openstack-tox-cover/0209678/cover/ : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/24/check/openstack-tox-lower-constraints/0c0ece3/ : SUCCESS in 11m 46s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/24/check/openstack-tox-pep8/032a532/ : SUCCESS in 7m 05s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/24/check/openstack-tox-py27/039b745/ : SUCCESS in 11m 49s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/24/check/openstack-tox-py36/9376b41/ : SUCCESS in 11m 04s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/24/check/openstack-tox-py37/b8fa85f/ : SUCCESS in 16m 52s\n- openstack-tox-docs http://logs.openstack.org/68/652068/24/check/openstack-tox-docs/d1265c8/html/ : SUCCESS in 4m 12s\n- trove-functional-mysql http://logs.openstack.org/68/652068/24/check/trove-functional-mysql/bca1384/ : FAILURE in 1h 45m 24s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/24/check/trove-scenario-mariadb-single/8b8febf/ : FAILURE in 42m 29s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/24/check/trove-scenario-mariadb-multi/c4c91bf/ : FAILURE in 1h 10m 55s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/24/check/trove-scenario-mysql-single/22d5d1a/ : FAILURE in 1h 36m 24s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/24/check/trove-scenario-mysql-multi/4a6e6c3/ : SUCCESS in 1h 18m 49s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/24/check/trove-scenario-postgresql-single/1a0e9bb/ : FAILURE in 40m 35s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/24/check/trove-scenario-postgresql-multi/d57f7a1/ : FAILURE in 40m 38s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/24/check/trove-scenario-pxc-single/ee1fbc8/ : FAILURE in 36m 31s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/24/check/trove-scenario-pxc-multi/f7bbd2a/ : FAILURE in 29m 52s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/24/check/trove-scenario-redis-single/a0705c5/ : FAILURE in 1h 01m 37s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/24/check/trove-scenario-redis-multi/a526eab/ : FAILURE in 45m 04s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/24/check/trove-tox-fakemodetests/5abf90a/ : SUCCESS in 4m 58s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/24/check/trove-tox-apiexamples/c1d488a/ : SUCCESS in 4m 38s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/24/check/trove-tox-bandit-baseline/1e54bf6/ : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/24/check/openstack-tox-pylint/b0fe680/ : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":24},{"id":"1132186dfc39e292afb0aeac8f39388af4565de4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-19 10:48:44.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"69c818a6fe25a8f9dad78dec954bf9ac7880d3d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 13:03:41.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/25/check/requirements-check/e1c9fe6/ : SUCCESS in 3m 42s\n- openstack-tox-cover http://logs.openstack.org/68/652068/25/check/openstack-tox-cover/1d73a64/cover/ : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/25/check/openstack-tox-lower-constraints/0cb0d59/ : SUCCESS in 11m 54s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/25/check/openstack-tox-pep8/6ea33cf/ : SUCCESS in 6m 10s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/25/check/openstack-tox-py27/5fb3440/ : SUCCESS in 10m 51s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/25/check/openstack-tox-py36/01ae4e1/ : SUCCESS in 10m 18s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/25/check/openstack-tox-py37/083a00b/ : SUCCESS in 17m 27s\n- openstack-tox-docs http://logs.openstack.org/68/652068/25/check/openstack-tox-docs/e48af1e/html/ : SUCCESS in 4m 26s\n- trove-functional-mysql http://logs.openstack.org/68/652068/25/check/trove-functional-mysql/424f97a/ : FAILURE in 2h 03m 20s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/25/check/trove-scenario-mariadb-single/dfa6cc7/ : FAILURE in 34m 37s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/25/check/trove-scenario-mariadb-multi/bb5a40c/ : FAILURE in 35m 59s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/25/check/trove-scenario-mysql-single/6e76ba3/ : SUCCESS in 1h 42m 23s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/25/check/trove-scenario-mysql-multi/a370395/ : SUCCESS in 1h 17m 48s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/25/check/trove-scenario-postgresql-single/e7842f0/ : FAILURE in 30m 42s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/25/check/trove-scenario-postgresql-multi/8c427e8/ : FAILURE in 35m 49s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/25/check/trove-scenario-pxc-single/26b6de8/ : FAILURE in 39m 59s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/25/check/trove-scenario-pxc-multi/3b03294/ : FAILURE in 38m 34s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/25/check/trove-scenario-redis-single/730df0a/ : FAILURE in 59m 19s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/25/check/trove-scenario-redis-multi/a5ad6b6/ : FAILURE in 57m 24s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/25/check/trove-tox-fakemodetests/204e8c9/ : SUCCESS in 7m 19s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/25/check/trove-tox-apiexamples/63c9c48/ : SUCCESS in 4m 39s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/25/check/trove-tox-bandit-baseline/e3f14a7/ : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/25/check/openstack-tox-pylint/73216e8/ : SUCCESS in 11m 12s","accounts_in_message":[],"_revision_number":25},{"id":"b17200e7c3ebe755b981960f3c33cca6291f84ec","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-20 02:21:04.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"ba70fb4325cc724f2acc93fc4cf87db5b81f8367","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:27:31.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/26/check/requirements-check/0380d07/ : SUCCESS in 3m 25s\n- openstack-tox-cover http://logs.openstack.org/68/652068/26/check/openstack-tox-cover/2f62f7b/cover/ : SUCCESS in 17m 29s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/26/check/openstack-tox-lower-constraints/b1e1fc6/ : SUCCESS in 13m 18s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/26/check/openstack-tox-pep8/b696d2c/ : SUCCESS in 6m 19s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/26/check/openstack-tox-py27/e0a6f79/ : SUCCESS in 10m 36s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/26/check/openstack-tox-py36/1d70193/ : SUCCESS in 12m 10s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/26/check/openstack-tox-py37/dc33d1d/ : SUCCESS in 15m 10s\n- openstack-tox-docs http://logs.openstack.org/68/652068/26/check/openstack-tox-docs/d000116/html/ : SUCCESS in 4m 42s\n- trove-functional-mysql http://logs.openstack.org/68/652068/26/check/trove-functional-mysql/f3fc6d0/ : FAILURE in 2h 01m 32s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/26/check/trove-scenario-mariadb-single/37d9c1e/ : FAILURE in 32m 08s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/26/check/trove-scenario-mariadb-multi/3b6b925/ : FAILURE in 39m 50s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/26/check/trove-scenario-mysql-single/11d59c3/ : SUCCESS in 1h 29m 32s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/26/check/trove-scenario-mysql-multi/55a0215/ : SUCCESS in 1h 28m 08s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/26/check/trove-scenario-postgresql-single/37554e1/ : FAILURE in 42m 05s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/26/check/trove-scenario-postgresql-multi/a795651/ : FAILURE in 38m 27s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/26/check/trove-scenario-pxc-single/f604f96/ : FAILURE in 37m 24s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/26/check/trove-scenario-pxc-multi/fa78b88/ : FAILURE in 37m 46s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/26/check/trove-scenario-redis-single/6a5c2ac/ : FAILURE in 51m 58s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/26/check/trove-scenario-redis-multi/f5fca51/ : FAILURE in 58m 22s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/26/check/trove-tox-fakemodetests/3dbfd59/ : SUCCESS in 5m 10s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/26/check/trove-tox-apiexamples/d265a54/ : SUCCESS in 6m 18s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/26/check/trove-tox-bandit-baseline/6541e6f/ : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/26/check/openstack-tox-pylint/223ce5c/ : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":26},{"id":"aa53d86412f0e15bb4e36a279270833047617806","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-20 11:49:15.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"2e8ca8499313e003e160851548d47a42d9bdcb06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 14:17:42.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/27/check/requirements-check/b30d517/ : SUCCESS in 3m 27s\n- openstack-tox-cover http://logs.openstack.org/68/652068/27/check/openstack-tox-cover/072fdea/cover/ : SUCCESS in 18m 56s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/27/check/openstack-tox-lower-constraints/ed6ec42/ : SUCCESS in 12m 06s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/27/check/openstack-tox-pep8/ef96cc7/ : FAILURE in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/27/check/openstack-tox-py27/da43b1e/ : SUCCESS in 11m 23s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/27/check/openstack-tox-py36/95a4469/ : SUCCESS in 12m 48s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/27/check/openstack-tox-py37/dcdbccd/ : SUCCESS in 16m 51s\n- openstack-tox-docs http://logs.openstack.org/68/652068/27/check/openstack-tox-docs/f590e3f/html/ : SUCCESS in 4m 56s\n- trove-functional-mysql http://logs.openstack.org/68/652068/27/check/trove-functional-mysql/691b9cf/ : FAILURE in 2h 04m 31s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/27/check/trove-scenario-mariadb-single/0d262bc/ : FAILURE in 36m 44s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/27/check/trove-scenario-mariadb-multi/318e1fc/ : FAILURE in 38m 08s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/27/check/trove-scenario-mysql-single/dc06e9a/ : SUCCESS in 2h 05m 09s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/27/check/trove-scenario-mysql-multi/0395990/ : SUCCESS in 2h 16m 45s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/27/check/trove-scenario-postgresql-single/9c631b5/ : FAILURE in 40m 24s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/27/check/trove-scenario-postgresql-multi/adc2c7b/ : FAILURE in 43m 02s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/27/check/trove-scenario-pxc-single/757eb74/ : FAILURE in 43m 37s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/27/check/trove-scenario-pxc-multi/875d257/ : FAILURE in 37m 37s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/27/check/trove-scenario-redis-single/432327e/ : FAILURE in 1h 03m 36s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/27/check/trove-scenario-redis-multi/0adf6de/ : FAILURE in 1h 15m 51s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/27/check/trove-tox-fakemodetests/e11dd71/ : SUCCESS in 5m 34s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/27/check/trove-tox-apiexamples/093c094/ : SUCCESS in 5m 27s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/27/check/trove-tox-bandit-baseline/b78ed5f/ : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/27/check/openstack-tox-pylint/30f9c64/ : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":27},{"id":"015237a5234ddca66c2b93fd5eb24fdfec91c8c8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-21 02:07:52.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"9225abf9d0bcda8f4a142826bb9d10ef6019f0e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 03:00:02.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/68/652068/28/check/requirements-check/d229e3d/ : SUCCESS in 3m 35s\n- openstack-tox-cover http://logs.openstack.org/68/652068/28/check/openstack-tox-cover/06fdbe4/ : FAILURE in 16m 29s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/28/check/openstack-tox-lower-constraints/a6506a4/ : SUCCESS in 12m 01s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/28/check/openstack-tox-pep8/0e856f7/ : FAILURE in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/28/check/openstack-tox-py27/bae54d2/ : SUCCESS in 10m 46s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/28/check/openstack-tox-py36/c4cb671/ : SUCCESS in 11m 37s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/28/check/openstack-tox-py37/e71e4b3/ : SUCCESS in 17m 06s\n- openstack-tox-docs http://logs.openstack.org/68/652068/28/check/openstack-tox-docs/0239a76/html/ : SUCCESS in 4m 58s\n- trove-functional-mysql http://logs.openstack.org/68/652068/28/check/trove-functional-mysql/7c3cd58/ : FAILURE in 43m 07s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/28/check/trove-scenario-mariadb-single/e702b69/ : FAILURE in 37m 16s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/28/check/trove-scenario-mariadb-multi/6262536/ : FAILURE in 43m 15s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/28/check/trove-scenario-mysql-single/3ad80f1/ : FAILURE in 43m 24s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/28/check/trove-scenario-mysql-multi/895f6fc/ : FAILURE in 46m 06s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/28/check/trove-scenario-postgresql-single/945181b/ : FAILURE in 38m 05s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/28/check/trove-scenario-postgresql-multi/6687ce9/ : FAILURE in 38m 50s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/28/check/trove-scenario-pxc-single/a57dbc3/ : FAILURE in 38m 00s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/28/check/trove-scenario-pxc-multi/87d464a/ : FAILURE in 39m 50s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/28/check/trove-scenario-redis-single/ae934f0/ : FAILURE in 42m 16s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/28/check/trove-scenario-redis-multi/40b082a/ : FAILURE in 45m 33s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/28/check/trove-tox-fakemodetests/0c36d30/ : FAILURE in 4m 47s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/28/check/trove-tox-apiexamples/3c74ada/ : SUCCESS in 4m 47s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/28/check/trove-tox-bandit-baseline/48b6580/ : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/28/check/openstack-tox-pylint/31651a2/ : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":28},{"id":"0e6f2d75b3ff54e2870ddb64e590efc57d56341d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-21 04:21:53.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"360daec20a1e4db025cbbb4c909f9d87e30a3277","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 06:26:49.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/29/check/build-openstack-api-ref/71066a3/html/ : SUCCESS in 5m 42s\n- requirements-check http://logs.openstack.org/68/652068/29/check/requirements-check/f86a442/ : SUCCESS in 3m 23s\n- openstack-tox-cover http://logs.openstack.org/68/652068/29/check/openstack-tox-cover/1922b0b/cover/ : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/29/check/openstack-tox-lower-constraints/1efd9b4/ : SUCCESS in 11m 28s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/29/check/openstack-tox-pep8/1f6e256/ : FAILURE in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/29/check/openstack-tox-py27/cad22f6/ : SUCCESS in 10m 29s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/29/check/openstack-tox-py36/a30492e/ : SUCCESS in 12m 12s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/29/check/openstack-tox-py37/16ebb37/ : SUCCESS in 16m 07s\n- openstack-tox-docs http://logs.openstack.org/68/652068/29/check/openstack-tox-docs/9ab3257/html/ : SUCCESS in 4m 55s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/29/check/build-openstack-releasenotes/84da8d8/html/ : SUCCESS in 6m 30s\n- trove-functional-mysql http://logs.openstack.org/68/652068/29/check/trove-functional-mysql/aeead33/ : FAILURE in 2h 04m 19s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/29/check/trove-scenario-mariadb-single/84c0d10/ : FAILURE in 36m 31s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/29/check/trove-scenario-mariadb-multi/a098509/ : FAILURE in 37m 26s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/29/check/trove-scenario-mysql-single/60050ba/ : FAILURE in 1h 58m 17s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/29/check/trove-scenario-mysql-multi/6266ff3/ : SUCCESS in 1h 37m 26s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/29/check/trove-scenario-postgresql-single/ea14191/ : FAILURE in 41m 03s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/29/check/trove-scenario-postgresql-multi/295ff1a/ : FAILURE in 36m 04s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/29/check/trove-scenario-pxc-single/a8c4938/ : FAILURE in 37m 36s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/29/check/trove-scenario-pxc-multi/05d2bdd/ : FAILURE in 41m 54s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/29/check/trove-scenario-redis-single/a7cb3bb/ : FAILURE in 1h 00m 25s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/29/check/trove-scenario-redis-multi/a526624/ : FAILURE in 57m 30s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/29/check/trove-tox-fakemodetests/055176b/ : SUCCESS in 9m 15s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/29/check/trove-tox-apiexamples/e3f392b/ : SUCCESS in 7m 20s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/29/check/trove-tox-bandit-baseline/b3945f4/ : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/29/check/openstack-tox-pylint/8659086/ : SUCCESS in 9m 45s","accounts_in_message":[],"_revision_number":29},{"id":"ff3310e3ea26dff122da2abde4507aeef5197666","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-21 08:48:46.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"46bf60545b7fd89132152de782b6297f036f7f5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 11:15:57.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/30/check/build-openstack-api-ref/ac8a3d4/html/ : SUCCESS in 5m 31s\n- requirements-check http://logs.openstack.org/68/652068/30/check/requirements-check/329c5cf/ : SUCCESS in 3m 28s\n- openstack-tox-cover http://logs.openstack.org/68/652068/30/check/openstack-tox-cover/ea0d88e/cover/ : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/30/check/openstack-tox-lower-constraints/9d36994/ : SUCCESS in 13m 48s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/30/check/openstack-tox-pep8/db7f613/ : SUCCESS in 8m 07s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/30/check/openstack-tox-py27/e912d76/ : SUCCESS in 10m 58s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/30/check/openstack-tox-py36/3c0effc/ : SUCCESS in 14m 46s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/30/check/openstack-tox-py37/9e8206b/ : SUCCESS in 18m 54s\n- openstack-tox-docs http://logs.openstack.org/68/652068/30/check/openstack-tox-docs/1e9a39c/html/ : SUCCESS in 4m 38s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/30/check/build-openstack-releasenotes/9d4b18d/html/ : SUCCESS in 6m 22s\n- trove-functional-mysql http://logs.openstack.org/68/652068/30/check/trove-functional-mysql/34b2732/ : FAILURE in 2h 18m 46s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/30/check/trove-scenario-mariadb-single/9a61ae2/ : FAILURE in 35m 11s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/30/check/trove-scenario-mariadb-multi/8d59f2e/ : FAILURE in 33m 43s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/30/check/trove-scenario-mysql-single/454e066/ : FAILURE in 1h 56m 40s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/30/check/trove-scenario-mysql-multi/668e76e/ : SUCCESS in 1h 30m 28s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/30/check/trove-scenario-postgresql-single/b2fd687/ : FAILURE in 39m 56s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/30/check/trove-scenario-postgresql-multi/668f5ae/ : FAILURE in 41m 40s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/30/check/trove-scenario-pxc-single/beb63ad/ : FAILURE in 51m 31s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/30/check/trove-scenario-pxc-multi/5709133/ : FAILURE in 41m 01s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/30/check/trove-scenario-redis-single/72499fe/ : FAILURE in 1h 06m 48s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/30/check/trove-scenario-redis-multi/55393dd/ : FAILURE in 1h 06m 56s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/30/check/trove-tox-fakemodetests/8ac687f/ : SUCCESS in 5m 20s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/30/check/trove-tox-apiexamples/004f23f/ : SUCCESS in 4m 55s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/30/check/trove-tox-bandit-baseline/ece5944/ : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/30/check/openstack-tox-pylint/3feda66/ : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":30},{"id":"4fa7e6eee4a1dd4fcd34ebe264e99bedf945becd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-27 12:11:35.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"c08a2e3549403f416f8b7317d0ee1839a38b00df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 15:03:54.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/31/check/build-openstack-api-ref/e16d608/ : FAILURE in 8m 29s\n- requirements-check http://logs.openstack.org/68/652068/31/check/requirements-check/6fea506/ : SUCCESS in 7m 26s\n- openstack-tox-cover http://logs.openstack.org/68/652068/31/check/openstack-tox-cover/aa81df2/cover/ : SUCCESS in 27m 03s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/31/check/openstack-tox-lower-constraints/4da287e/ : SUCCESS in 18m 03s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/31/check/openstack-tox-pep8/2ee76ad/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/31/check/openstack-tox-py27/d446258/ : SUCCESS in 13m 03s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/31/check/openstack-tox-py36/e8fbe56/ : SUCCESS in 13m 59s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/31/check/openstack-tox-py37/dc8e8ba/ : SUCCESS in 18m 24s\n- openstack-tox-docs http://logs.openstack.org/68/652068/31/check/openstack-tox-docs/61b1c5b/html/ : SUCCESS in 6m 04s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/31/check/build-openstack-releasenotes/c72b2af/html/ : SUCCESS in 6m 50s\n- trove-functional-mysql http://logs.openstack.org/68/652068/31/check/trove-functional-mysql/1c38e2c/ : FAILURE in 2h 32m 06s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/31/check/trove-scenario-mariadb-single/e79e771/ : FAILURE in 38m 52s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/31/check/trove-scenario-mariadb-multi/1c9ba24/ : FAILURE in 48m 20s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/31/check/trove-scenario-mysql-single/8c65176/ : FAILURE in 1h 35m 06s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/31/check/trove-scenario-mysql-multi/470ad51/ : FAILURE in 1h 04m 24s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/31/check/trove-scenario-postgresql-single/434172a/ : FAILURE in 38m 10s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/31/check/trove-scenario-postgresql-multi/cdae516/ : FAILURE in 40m 43s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/31/check/trove-scenario-pxc-single/8b65372/ : FAILURE in 47m 56s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/31/check/trove-scenario-pxc-multi/2003f02/ : FAILURE in 43m 43s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/31/check/trove-scenario-redis-single/b3d48a2/ : FAILURE in 1h 09m 10s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/31/check/trove-scenario-redis-multi/c66b101/ : FAILURE in 54m 58s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/31/check/trove-tox-fakemodetests/1f389ef/ : SUCCESS in 5m 23s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/31/check/trove-tox-apiexamples/0877e40/ : SUCCESS in 5m 59s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/31/check/trove-tox-bandit-baseline/4d37006/ : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/31/check/openstack-tox-pylint/da8dc37/ : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":31},{"id":"109afa73ca1286d1a83e659d290d13f967ddefdf","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-27 23:49:09.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.","accounts_in_message":[],"_revision_number":32},{"id":"7d35fe695689f828e4f3a0c8f4c4b3c8f34fb186","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 01:54:00.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/32/check/build-openstack-api-ref/4271ada/html/ : SUCCESS in 4m 41s\n- requirements-check http://logs.openstack.org/68/652068/32/check/requirements-check/8133225/ : SUCCESS in 3m 17s\n- openstack-tox-cover http://logs.openstack.org/68/652068/32/check/openstack-tox-cover/6305905/cover/ : SUCCESS in 18m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/32/check/openstack-tox-lower-constraints/8f392af/ : SUCCESS in 12m 19s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/32/check/openstack-tox-pep8/cc17554/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/32/check/openstack-tox-py27/98e6da8/ : SUCCESS in 11m 32s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/32/check/openstack-tox-py36/338bc80/ : SUCCESS in 12m 26s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/32/check/openstack-tox-py37/ab4772d/ : SUCCESS in 17m 53s\n- openstack-tox-docs http://logs.openstack.org/68/652068/32/check/openstack-tox-docs/be94a75/html/ : SUCCESS in 4m 29s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/32/check/build-openstack-releasenotes/6a8ca69/html/ : SUCCESS in 6m 02s\n- trove-functional-mysql http://logs.openstack.org/68/652068/32/check/trove-functional-mysql/2659a7e/ : FAILURE in 2h 03m 42s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/32/check/trove-scenario-mariadb-single/3aa886f/ : FAILURE in 38m 31s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/32/check/trove-scenario-mariadb-multi/702cc01/ : FAILURE in 34m 16s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/32/check/trove-scenario-mysql-single/8edaeee/ : FAILURE in 1h 38m 47s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/32/check/trove-scenario-mysql-multi/ad0f41b/ : FAILURE in 1h 12m 17s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/32/check/trove-scenario-postgresql-single/630b1f0/ : FAILURE in 38m 59s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/32/check/trove-scenario-postgresql-multi/ef16228/ : FAILURE in 33m 34s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/32/check/trove-scenario-pxc-single/569c06d/ : FAILURE in 39m 00s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/32/check/trove-scenario-pxc-multi/1e6673b/ : FAILURE in 34m 19s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/32/check/trove-scenario-redis-single/a5d7a52/ : FAILURE in 1h 10m 00s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/32/check/trove-scenario-redis-multi/f637cfd/ : FAILURE in 46m 04s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/32/check/trove-tox-fakemodetests/6432731/ : SUCCESS in 5m 05s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/32/check/trove-tox-apiexamples/a3b35a8/ : SUCCESS in 4m 47s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/32/check/trove-tox-bandit-baseline/abebf23/ : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/32/check/openstack-tox-pylint/989b6c5/ : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":32},{"id":"9eb21899e687b357f5ba325a752fe533a255a4bc","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-05-28 08:09:04.000000000","message":"Patch Set 32: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"1c1fa066bece41270a403b01dca706f60bd4cf1a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-28 08:50:37.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"88ce95d90b2389108365e652ff027a18f9f4d077","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-05-28 09:10:27.000000000","message":"Patch Set 32:\n\n\u003e (1 comment)\n\nWhen user is root , `~` mens `/root`, right? \nwhy do change to hard code `/root`？ Hard code means the guest agent only run as root user.\nIMO, trove-guestagent  run with non-root user is for security requirement similar with other openstack project services, the high permission operations should do with root helper.","accounts_in_message":[],"_revision_number":32},{"id":"fe12b6d7565980452b36128d8339a575a4612f4f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-01 10:07:14.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"1bdf3e7e5862b8b7c04ca6cd116c0a5a2843dc04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 12:34:14.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/33/check/build-openstack-api-ref/eed0e09/html/ : SUCCESS in 4m 48s\n- requirements-check http://logs.openstack.org/68/652068/33/check/requirements-check/d83acd2/ : SUCCESS in 3m 02s\n- openstack-tox-cover http://logs.openstack.org/68/652068/33/check/openstack-tox-cover/eb79a00/cover/ : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/33/check/openstack-tox-lower-constraints/b05d127/ : SUCCESS in 12m 01s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/33/check/openstack-tox-pep8/c01b270/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/33/check/openstack-tox-py27/a0afdcd/ : SUCCESS in 11m 04s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/33/check/openstack-tox-py36/aaf2043/ : SUCCESS in 12m 00s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/33/check/openstack-tox-py37/7aa0cf9/ : SUCCESS in 17m 19s\n- openstack-tox-docs http://logs.openstack.org/68/652068/33/check/openstack-tox-docs/ecc52d9/html/ : SUCCESS in 4m 31s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/33/check/build-openstack-releasenotes/20cde7f/html/ : SUCCESS in 5m 56s\n- trove-functional-mysql http://logs.openstack.org/68/652068/33/check/trove-functional-mysql/e0d992d/ : FAILURE in 2h 20m 38s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/33/check/trove-scenario-mariadb-single/452dda2/ : FAILURE in 41m 08s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/33/check/trove-scenario-mariadb-multi/bd43698/ : FAILURE in 46m 13s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/33/check/trove-scenario-mysql-single/d032200/ : FAILURE in 1h 36m 13s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/33/check/trove-scenario-mysql-multi/9333aa1/ : FAILURE in 1h 04m 49s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/33/check/trove-scenario-postgresql-single/a792c6f/ : FAILURE in 38m 39s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/33/check/trove-scenario-postgresql-multi/30bec49/ : FAILURE in 37m 32s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/33/check/trove-scenario-pxc-single/49cd2b3/ : FAILURE in 39m 09s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/33/check/trove-scenario-pxc-multi/9beecec/ : FAILURE in 38m 58s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/33/check/trove-scenario-redis-single/4ee555d/ : FAILURE in 1h 01m 41s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/33/check/trove-scenario-redis-multi/5693350/ : FAILURE in 1h 00m 30s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/33/check/trove-tox-fakemodetests/dc99e33/ : SUCCESS in 5m 14s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/33/check/trove-tox-apiexamples/694d22a/ : SUCCESS in 4m 52s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/33/check/trove-tox-bandit-baseline/f086328/ : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/33/check/openstack-tox-pylint/30e9de3/ : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":33},{"id":"4403da44703f87062d93c504a8818f4255090f96","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-03 12:40:23.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"ba1bec1edb6bbf53c142e054444d9c2c6fd6a134","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:50:39.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/34/check/build-openstack-api-ref/30adc6c/ : POST_FAILURE in 3m 33s\n- requirements-check http://logs.openstack.org/68/652068/34/check/requirements-check/0927002/ : SUCCESS in 3m 31s\n- openstack-tox-cover http://logs.openstack.org/68/652068/34/check/openstack-tox-cover/0302893/cover/ : SUCCESS in 19m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/34/check/openstack-tox-lower-constraints/7690877/ : SUCCESS in 12m 41s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/34/check/openstack-tox-pep8/133349a/ : FAILURE in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/34/check/openstack-tox-py27/a7ea279/ : SUCCESS in 10m 57s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/34/check/openstack-tox-py36/02abf80/ : FAILURE in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/34/check/openstack-tox-py37/fd14896/ : SUCCESS in 17m 58s\n- openstack-tox-docs http://logs.openstack.org/68/652068/34/check/openstack-tox-docs/34504a7/html/ : SUCCESS in 5m 28s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/34/check/build-openstack-releasenotes/a24927e/html/ : SUCCESS in 6m 10s\n- trove-functional-mysql http://logs.openstack.org/68/652068/34/check/trove-functional-mysql/4ff9c40/ : FAILURE in 2h 48m 22s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/34/check/trove-scenario-mariadb-single/1925aa6/ : FAILURE in 1h 19m 55s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/34/check/trove-scenario-mariadb-multi/c5d38a5/ : FAILURE in 13m 10s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/34/check/trove-scenario-mysql-single/802d22a/ : FAILURE in 2h 08m 22s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/34/check/trove-scenario-mysql-multi/9c295fe/ : SUCCESS in 2h 19m 17s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/34/check/trove-scenario-postgresql-single/288348b/ : FAILURE in 1h 16m 56s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/34/check/trove-scenario-postgresql-multi/6516e50/ : FAILURE in 42m 38s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/34/check/trove-scenario-pxc-single/10acba6/ : FAILURE in 31m 22s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/34/check/trove-scenario-pxc-multi/3b2bdcf/ : FAILURE in 1h 22m 28s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/34/check/trove-scenario-redis-single/28a84d6/ : FAILURE in 1h 15m 17s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/34/check/trove-scenario-redis-multi/9f464c1/ : FAILURE in 11m 35s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/34/check/trove-tox-fakemodetests/6ebf83a/ : SUCCESS in 5m 34s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/34/check/trove-tox-apiexamples/28064b1/ : SUCCESS in 6m 04s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/34/check/trove-tox-bandit-baseline/dca3412/ : FAILURE in 3m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/34/check/openstack-tox-pylint/07f1359/ : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":34},{"id":"c7fa2c0a79c59e9d671c7d5c61aeb5dfc72344c7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-05 03:54:34.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"f91a22952135f94c25c3ef329fced4a2ca3096b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 06:20:06.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/35/check/build-openstack-api-ref/05464d3/html/ : SUCCESS in 5m 23s\n- requirements-check http://logs.openstack.org/68/652068/35/check/requirements-check/3d3bc06/ : SUCCESS in 5m 39s\n- openstack-tox-cover http://logs.openstack.org/68/652068/35/check/openstack-tox-cover/b64c356/ : FAILURE in 17m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/35/check/openstack-tox-lower-constraints/d2d8368/ : SUCCESS in 11m 54s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/35/check/openstack-tox-pep8/a35e6a6/ : FAILURE in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/35/check/openstack-tox-py27/3b4d8ce/ : SUCCESS in 11m 24s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/35/check/openstack-tox-py36/a2e85e2/ : SUCCESS in 11m 48s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/35/check/openstack-tox-py37/6f65d89/ : SUCCESS in 15m 44s\n- openstack-tox-docs http://logs.openstack.org/68/652068/35/check/openstack-tox-docs/6b8b26b/html/ : SUCCESS in 4m 28s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/35/check/build-openstack-releasenotes/01df527/html/ : SUCCESS in 7m 28s\n- trove-functional-mysql http://logs.openstack.org/68/652068/35/check/trove-functional-mysql/eb9eb22/ : FAILURE in 2h 21m 05s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/35/check/trove-scenario-mariadb-single/e84af8c/ : FAILURE in 38m 56s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/35/check/trove-scenario-mariadb-multi/7064f7c/ : FAILURE in 42m 21s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/35/check/trove-scenario-mysql-single/6344108/ : SUCCESS in 1h 40m 12s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/35/check/trove-scenario-mysql-multi/8962869/ : SUCCESS in 1h 27m 46s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/35/check/trove-scenario-postgresql-single/d75616a/ : FAILURE in 40m 37s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/35/check/trove-scenario-postgresql-multi/7d3d7ce/ : FAILURE in 35m 54s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/35/check/trove-scenario-pxc-single/20f0e3e/ : FAILURE in 43m 47s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/35/check/trove-scenario-pxc-multi/95e1c57/ : FAILURE in 36m 32s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/35/check/trove-scenario-redis-single/1241620/ : FAILURE in 1h 04m 03s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/35/check/trove-scenario-redis-multi/180da7c/ : FAILURE in 58m 52s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/35/check/trove-tox-fakemodetests/f4f84eb/ : FAILURE in 5m 00s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/35/check/trove-tox-apiexamples/31df805/ : SUCCESS in 5m 33s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/35/check/trove-tox-bandit-baseline/bd11ad3/ : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/35/check/openstack-tox-pylint/3a940e0/ : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":35},{"id":"395fcf4de26f54fe1f1f9987f8ddcd1347ebd830","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-05 09:20:45.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"863153df220bf3b91af9523abfeaac4a5772741b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-05 09:23:51.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"7936a579d5baa8bef97a867d4db608090168cc86","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-05 11:48:56.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"934632ece1a43c514ba74cac01bda395eae19fc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 14:46:46.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/38/check/build-openstack-api-ref/a7d5df5/html/ : SUCCESS in 4m 24s\n- requirements-check http://logs.openstack.org/68/652068/38/check/requirements-check/36c4bdc/ : SUCCESS in 3m 32s\n- openstack-tox-cover http://logs.openstack.org/68/652068/38/check/openstack-tox-cover/1eeb624/cover/ : SUCCESS in 18m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/38/check/openstack-tox-lower-constraints/de1c714/ : SUCCESS in 12m 29s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/38/check/openstack-tox-pep8/deb115c/ : SUCCESS in 6m 00s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/38/check/openstack-tox-py27/d092f15/ : SUCCESS in 11m 14s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/38/check/openstack-tox-py36/f24a688/ : SUCCESS in 12m 40s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/38/check/openstack-tox-py37/f30008c/ : SUCCESS in 17m 05s\n- openstack-tox-docs http://logs.openstack.org/68/652068/38/check/openstack-tox-docs/3fc0286/html/ : SUCCESS in 6m 08s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/38/check/build-openstack-releasenotes/e8562c8/html/ : SUCCESS in 6m 37s\n- trove-functional-mysql http://logs.openstack.org/68/652068/38/check/trove-functional-mysql/8df2002/ : FAILURE in 2h 41m 23s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/38/check/trove-scenario-mariadb-single/9f8995c/ : FAILURE in 37m 48s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/38/check/trove-scenario-mariadb-multi/7a28767/ : FAILURE in 36m 49s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/38/check/trove-scenario-mysql-single/92315f0/ : SUCCESS in 1h 36m 20s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/38/check/trove-scenario-mysql-multi/bb3df67/ : SUCCESS in 1h 25m 33s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/38/check/trove-scenario-postgresql-single/ea580e1/ : FAILURE in 36m 01s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/38/check/trove-scenario-postgresql-multi/3e4d27e/ : FAILURE in 39m 37s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/38/check/trove-scenario-pxc-single/9f6eb4b/ : FAILURE in 49m 42s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/38/check/trove-scenario-pxc-multi/c193926/ : FAILURE in 41m 18s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/38/check/trove-scenario-redis-single/b76a421/ : FAILURE in 1h 04m 24s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/38/check/trove-scenario-redis-multi/11c2d2c/ : FAILURE in 1h 00m 01s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/38/check/trove-tox-fakemodetests/74eddcb/ : SUCCESS in 6m 22s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/38/check/trove-tox-apiexamples/d94d093/ : SUCCESS in 8m 07s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/38/check/trove-tox-bandit-baseline/5f96853/ : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/38/check/openstack-tox-pylint/5a3ea25/ : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":38},{"id":"7a53d1c910d008279b19b3373f50a93728ee00a6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-05 23:59:35.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"5af00de20ad080c11693012df52a3de4a8f5fd34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 03:02:37.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/39/check/build-openstack-api-ref/2457e79/html/ : SUCCESS in 6m 18s\n- requirements-check http://logs.openstack.org/68/652068/39/check/requirements-check/a8b1556/ : SUCCESS in 3m 28s\n- openstack-tox-cover http://logs.openstack.org/68/652068/39/check/openstack-tox-cover/7ed9f53/ : FAILURE in 17m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/39/check/openstack-tox-lower-constraints/9d13b40/ : SUCCESS in 12m 37s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/39/check/openstack-tox-pep8/a0378a5/ : FAILURE in 5m 57s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/39/check/openstack-tox-py27/f4fcb40/ : SUCCESS in 11m 12s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/39/check/openstack-tox-py36/1cc96d8/ : SUCCESS in 12m 29s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/39/check/openstack-tox-py37/236cb96/ : SUCCESS in 22m 08s\n- openstack-tox-docs http://logs.openstack.org/68/652068/39/check/openstack-tox-docs/b5243e3/html/ : SUCCESS in 6m 25s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/39/check/build-openstack-releasenotes/e8ba345/html/ : SUCCESS in 7m 14s\n- trove-functional-mysql http://logs.openstack.org/68/652068/39/check/trove-functional-mysql/61cc972/ : FAILURE in 2h 32m 07s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/39/check/trove-scenario-mariadb-single/c2051bf/ : FAILURE in 41m 51s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/39/check/trove-scenario-mariadb-multi/4c362a5/ : FAILURE in 1h 29m 28s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/39/check/trove-scenario-mysql-single/f00cf33/ : FAILURE in 2h 57m 07s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/39/check/trove-scenario-mysql-multi/b316667/ : FAILURE in 1h 29m 59s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/39/check/trove-scenario-postgresql-single/43bb2e3/ : FAILURE in 1h 02m 47s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/39/check/trove-scenario-postgresql-multi/ea82105/ : FAILURE in 1h 14m 52s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/39/check/trove-scenario-pxc-single/f1857bf/ : FAILURE in 1h 31m 37s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/39/check/trove-scenario-pxc-multi/bd3547a/ : FAILURE in 45m 18s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/39/check/trove-scenario-redis-single/58fa46a/ : FAILURE in 1h 07m 59s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/39/check/trove-scenario-redis-multi/248ecf2/ : FAILURE in 1h 28m 07s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/39/check/trove-tox-fakemodetests/7d09b05/ : FAILURE in 6m 34s\n- trove-tox-apiexamples http://logs.openstack.org/68/652068/39/check/trove-tox-apiexamples/2fd3766/ : FAILURE in 5m 28s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/39/check/trove-tox-bandit-baseline/f8a14a5/ : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/39/check/openstack-tox-pylint/7c5f706/ : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":39},{"id":"6a5f8f7ba742c4c0eb9b2612989c336754f58901","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-06 09:29:45.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"33852b7010d2d9189997e84c65bf0614f5c6fe6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 12:28:34.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/40/check/build-openstack-api-ref/9d66400/html/ : SUCCESS in 5m 09s\n- requirements-check http://logs.openstack.org/68/652068/40/check/requirements-check/302a41e/ : SUCCESS in 4m 22s\n- openstack-tox-cover http://logs.openstack.org/68/652068/40/check/openstack-tox-cover/542f50f/cover/ : SUCCESS in 18m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/40/check/openstack-tox-lower-constraints/7a3b4a3/ : SUCCESS in 13m 02s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/40/check/openstack-tox-pep8/45da7b5/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/40/check/openstack-tox-py27/3f04976/ : SUCCESS in 11m 42s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/40/check/openstack-tox-py36/65182fd/ : SUCCESS in 11m 56s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/40/check/openstack-tox-py37/2fe4236/ : SUCCESS in 16m 59s\n- openstack-tox-docs http://logs.openstack.org/68/652068/40/check/openstack-tox-docs/cc1f63f/html/ : SUCCESS in 5m 12s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/40/check/build-openstack-releasenotes/89a63b5/html/ : SUCCESS in 6m 32s\n- trove-functional-mysql http://logs.openstack.org/68/652068/40/check/trove-functional-mysql/815592d/ : FAILURE in 2h 51m 42s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/40/check/trove-scenario-mariadb-single/0cec0c3/ : FAILURE in 1h 06m 07s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/40/check/trove-scenario-mariadb-multi/3ca63a3/ : FAILURE in 56m 48s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/40/check/trove-scenario-mysql-single/a3b019e/ : SUCCESS in 2h 04m 11s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/40/check/trove-scenario-mysql-multi/b971025/ : SUCCESS in 1h 35m 38s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/40/check/trove-scenario-postgresql-single/9a18cff/ : FAILURE in 1h 08m 34s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/40/check/trove-scenario-postgresql-multi/3ec831d/ : FAILURE in 1h 08m 28s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/40/check/trove-scenario-pxc-single/5317277/ : FAILURE in 59m 08s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/40/check/trove-scenario-pxc-multi/16c2579/ : FAILURE in 53m 30s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/40/check/trove-scenario-redis-single/a097b6d/ : FAILURE in 1h 45m 27s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/40/check/trove-scenario-redis-multi/7e25fce/ : FAILURE in 1h 21m 54s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/40/check/trove-tox-fakemodetests/3d31da3/ : SUCCESS in 7m 32s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/40/check/trove-tox-bandit-baseline/5fd84f1/ : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/40/check/openstack-tox-pylint/ec0ff91/ : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":40},{"id":"cd70f377500213a94f8c0d4c2a31772e19d17e6d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-06 22:08:15.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"a8c6664620d5ab59e0445ed8a4b39628d34f3021","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 23:14:35.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/41/check/build-openstack-api-ref/0a83661/html/ : SUCCESS in 6m 31s\n- requirements-check http://logs.openstack.org/68/652068/41/check/requirements-check/9aab6aa/ : SUCCESS in 3m 14s\n- openstack-tox-cover http://logs.openstack.org/68/652068/41/check/openstack-tox-cover/a90e139/cover/ : SUCCESS in 21m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/41/check/openstack-tox-lower-constraints/be4e7cb/ : SUCCESS in 12m 30s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/41/check/openstack-tox-pep8/daeb35e/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/41/check/openstack-tox-py27/4d6b3c9/ : SUCCESS in 12m 03s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/41/check/openstack-tox-py36/6ae6df2/ : SUCCESS in 12m 41s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/41/check/openstack-tox-py37/985bfc9/ : SUCCESS in 17m 41s\n- openstack-tox-docs http://logs.openstack.org/68/652068/41/check/openstack-tox-docs/2910f59/html/ : SUCCESS in 5m 22s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/41/check/build-openstack-releasenotes/4269b6e/html/ : SUCCESS in 6m 22s\n- trove-functional-mysql http://logs.openstack.org/68/652068/41/check/trove-functional-mysql/4596a17/ : FAILURE in 30m 15s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/41/check/trove-scenario-mariadb-single/dfedc11/ : FAILURE in 38m 22s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/41/check/trove-scenario-mariadb-multi/e9e82b5/ : FAILURE in 42m 27s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/41/check/trove-scenario-mysql-single/e8068f6/ : FAILURE in 33m 25s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/41/check/trove-scenario-mysql-multi/16e205e/ : FAILURE in 45m 38s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/41/check/trove-scenario-postgresql-single/ee2195e/ : FAILURE in 35m 18s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/41/check/trove-scenario-postgresql-multi/38071de/ : FAILURE in 38m 23s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/41/check/trove-scenario-pxc-single/660fb7b/ : FAILURE in 34m 39s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/41/check/trove-scenario-pxc-multi/8443212/ : FAILURE in 34m 25s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/41/check/trove-scenario-redis-single/b053af1/ : FAILURE in 31m 39s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/41/check/trove-scenario-redis-multi/944c14e/ : FAILURE in 36m 20s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/41/check/trove-tox-fakemodetests/0ccfe4a/ : SUCCESS in 5m 50s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/41/check/trove-tox-bandit-baseline/869583f/ : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/41/check/openstack-tox-pylint/a176478/ : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":41},{"id":"f657120fdce3147d57214ea3cb444f5f240b91c7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-06 23:18:33.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"2973b062649077da3abc4df6956918d816ae15f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:06:44.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/41/check/build-openstack-api-ref/77e1964/html/ : SUCCESS in 5m 37s\n- requirements-check http://logs.openstack.org/68/652068/41/check/requirements-check/d0879df/ : SUCCESS in 4m 07s\n- openstack-tox-cover http://logs.openstack.org/68/652068/41/check/openstack-tox-cover/f0bc413/cover/ : SUCCESS in 17m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/41/check/openstack-tox-lower-constraints/98de435/ : SUCCESS in 13m 21s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/41/check/openstack-tox-pep8/48e693f/ : SUCCESS in 6m 27s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/41/check/openstack-tox-py27/19672dc/ : SUCCESS in 18m 27s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/41/check/openstack-tox-py36/adcba3e/ : SUCCESS in 12m 34s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/41/check/openstack-tox-py37/78d80ba/ : SUCCESS in 18m 16s\n- openstack-tox-docs http://logs.openstack.org/68/652068/41/check/openstack-tox-docs/6924def/html/ : SUCCESS in 5m 00s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/41/check/build-openstack-releasenotes/e953adb/html/ : SUCCESS in 5m 50s\n- trove-functional-mysql http://logs.openstack.org/68/652068/41/check/trove-functional-mysql/ba8d799/ : FAILURE in 30m 25s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/41/check/trove-scenario-mariadb-single/d9b2c0c/ : FAILURE in 35m 57s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/41/check/trove-scenario-mariadb-multi/6166f69/ : FAILURE in 37m 17s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/41/check/trove-scenario-mysql-single/400acfd/ : FAILURE in 30m 20s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/41/check/trove-scenario-mysql-multi/97a9b15/ : FAILURE in 34m 59s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/41/check/trove-scenario-postgresql-single/e287389/ : FAILURE in 37m 44s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/41/check/trove-scenario-postgresql-multi/f224293/ : FAILURE in 34m 53s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/41/check/trove-scenario-pxc-single/bd2237c/ : FAILURE in 36m 36s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/41/check/trove-scenario-pxc-multi/9ec76da/ : FAILURE in 34m 01s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/41/check/trove-scenario-redis-single/8b8522d/ : FAILURE in 34m 35s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/41/check/trove-scenario-redis-multi/106da72/ : FAILURE in 41m 14s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/41/check/trove-tox-fakemodetests/556b63e/ : SUCCESS in 5m 32s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/41/check/trove-tox-bandit-baseline/fa7e315/ : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/41/check/openstack-tox-pylint/1f517cf/ : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":41},{"id":"a340b9e538535569fbfdc7c953268326657f3057","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-07 02:11:01.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"6cd72141e4653492b8fbd06b191140f1cf170125","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 03:03:34.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/42/check/build-openstack-api-ref/39f69f8/html/ : SUCCESS in 5m 04s\n- requirements-check http://logs.openstack.org/68/652068/42/check/requirements-check/2fd4953/ : SUCCESS in 4m 02s\n- openstack-tox-cover http://logs.openstack.org/68/652068/42/check/openstack-tox-cover/ed2d14f/cover/ : SUCCESS in 19m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/42/check/openstack-tox-lower-constraints/d76b2f7/ : SUCCESS in 12m 19s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/42/check/openstack-tox-pep8/4d35f29/ : SUCCESS in 6m 59s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/42/check/openstack-tox-py27/e86760f/ : SUCCESS in 12m 26s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/42/check/openstack-tox-py36/a3726c8/ : SUCCESS in 18m 11s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/42/check/openstack-tox-py37/938ffec/ : SUCCESS in 15m 40s\n- openstack-tox-docs http://logs.openstack.org/68/652068/42/check/openstack-tox-docs/3370188/html/ : SUCCESS in 4m 43s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/42/check/build-openstack-releasenotes/1c778c8/html/ : SUCCESS in 6m 38s\n- trove-functional-mysql http://logs.openstack.org/68/652068/42/check/trove-functional-mysql/ac929eb/ : FAILURE in 30m 28s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/42/check/trove-scenario-mariadb-single/2ad08e0/ : FAILURE in 36m 50s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/42/check/trove-scenario-mariadb-multi/3beac9d/ : FAILURE in 38m 23s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/42/check/trove-scenario-mysql-single/4b479d6/ : FAILURE in 30m 49s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/42/check/trove-scenario-mysql-multi/b607180/ : FAILURE in 30m 37s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/42/check/trove-scenario-postgresql-single/78d71eb/ : FAILURE in 33m 02s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/42/check/trove-scenario-postgresql-multi/004cc37/ : FAILURE in 36m 57s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/42/check/trove-scenario-pxc-single/c94004d/ : FAILURE in 35m 49s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/42/check/trove-scenario-pxc-multi/eb83e32/ : FAILURE in 35m 02s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/42/check/trove-scenario-redis-single/733c556/ : FAILURE in 34m 59s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/42/check/trove-scenario-redis-multi/52ab1ad/ : FAILURE in 35m 49s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/42/check/trove-tox-fakemodetests/48152cb/ : SUCCESS in 6m 20s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/42/check/trove-tox-bandit-baseline/cd0da5c/ : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/42/check/openstack-tox-pylint/7bece84/ : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":42},{"id":"da531561415a734931829055220c3d5abdf7a260","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-07 04:00:30.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"1fd42f75f1a3acda05e00251beefdde83aa7a96e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 06:44:12.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/43/check/build-openstack-api-ref/867ec99/html/ : SUCCESS in 4m 56s\n- requirements-check http://logs.openstack.org/68/652068/43/check/requirements-check/fd2df74/ : SUCCESS in 4m 42s\n- openstack-tox-cover http://logs.openstack.org/68/652068/43/check/openstack-tox-cover/24eb2b4/cover/ : SUCCESS in 15m 36s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/43/check/openstack-tox-lower-constraints/db3ebdb/ : SUCCESS in 12m 18s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/43/check/openstack-tox-pep8/074615c/ : SUCCESS in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/43/check/openstack-tox-py27/cc060bd/ : SUCCESS in 10m 41s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/43/check/openstack-tox-py36/e86129d/ : SUCCESS in 13m 57s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/43/check/openstack-tox-py37/93d20fe/ : SUCCESS in 17m 12s\n- openstack-tox-docs http://logs.openstack.org/68/652068/43/check/openstack-tox-docs/167767c/html/ : SUCCESS in 5m 58s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/43/check/build-openstack-releasenotes/7624ac0/html/ : SUCCESS in 5m 52s\n- trove-functional-mysql http://logs.openstack.org/68/652068/43/check/trove-functional-mysql/3efac3d/ : FAILURE in 2h 38m 22s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/43/check/trove-scenario-mariadb-single/c779e4c/ : FAILURE in 39m 40s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/43/check/trove-scenario-mariadb-multi/21ac2c9/ : FAILURE in 31m 46s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/43/check/trove-scenario-mysql-single/d270165/ : SUCCESS in 1h 56m 39s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/43/check/trove-scenario-mysql-multi/e6b9500/ : SUCCESS in 1h 35m 33s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/43/check/trove-scenario-postgresql-single/2ed1251/ : FAILURE in 30m 44s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/43/check/trove-scenario-postgresql-multi/b2c6c0d/ : FAILURE in 40m 33s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/43/check/trove-scenario-pxc-single/97fadf3/ : FAILURE in 40m 24s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/43/check/trove-scenario-pxc-multi/36f5307/ : FAILURE in 30m 57s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/43/check/trove-scenario-redis-single/3c59d6a/ : FAILURE in 30m 32s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/43/check/trove-scenario-redis-multi/ae7f72e/ : FAILURE in 40m 27s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/43/check/trove-tox-fakemodetests/ea81cc5/ : SUCCESS in 6m 00s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/43/check/trove-tox-bandit-baseline/e6c6f27/ : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/43/check/openstack-tox-pylint/ede65ac/ : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":43},{"id":"14865a9dae4b31664a5658bd3f18f3cc57b06a45","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-07 09:16:15.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"7fdc38a646fcaabde3ae9e4c46edf377f28922ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:27:08.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/44/check/build-openstack-api-ref/af043aa/html/ : SUCCESS in 5m 00s\n- requirements-check http://logs.openstack.org/68/652068/44/check/requirements-check/1304e69/ : SUCCESS in 3m 16s\n- openstack-tox-cover http://logs.openstack.org/68/652068/44/check/openstack-tox-cover/af7a15f/cover/ : SUCCESS in 18m 18s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/44/check/openstack-tox-lower-constraints/d6ff30f/ : SUCCESS in 12m 30s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/44/check/openstack-tox-pep8/3fd2e1b/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/44/check/openstack-tox-py27/a3ac779/ : SUCCESS in 10m 39s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/44/check/openstack-tox-py36/faac5ec/ : SUCCESS in 12m 09s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/44/check/openstack-tox-py37/85bea27/ : SUCCESS in 16m 45s\n- openstack-tox-docs http://logs.openstack.org/68/652068/44/check/openstack-tox-docs/eb55235/html/ : SUCCESS in 5m 33s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/44/check/build-openstack-releasenotes/2e556d5/html/ : SUCCESS in 6m 12s\n- trove-functional-mysql http://logs.openstack.org/68/652068/44/check/trove-functional-mysql/74822d8/ : FAILURE in 2h 05m 11s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/44/check/trove-scenario-mariadb-single/c5cd924/ : FAILURE in 59m 13s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/44/check/trove-scenario-mariadb-multi/4de3483/ : FAILURE in 59m 54s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/44/check/trove-scenario-mysql-single/78a53df/ : SUCCESS in 1h 58m 38s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/44/check/trove-scenario-mysql-multi/64a8585/ : SUCCESS in 1h 37m 38s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/44/check/trove-scenario-postgresql-single/463451d/ : FAILURE in 39m 13s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/44/check/trove-scenario-postgresql-multi/a273533/ : FAILURE in 39m 00s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/44/check/trove-scenario-pxc-single/eecaea5/ : FAILURE in 38m 53s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/44/check/trove-scenario-pxc-multi/7f6a134/ : FAILURE in 37m 43s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/44/check/trove-scenario-redis-single/a41172b/ : FAILURE in 1h 04m 30s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/44/check/trove-scenario-redis-multi/c7bc086/ : FAILURE in 1h 06m 12s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/44/check/trove-tox-fakemodetests/389effe/ : SUCCESS in 5m 53s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/44/check/trove-tox-bandit-baseline/c85407d/ : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/44/check/openstack-tox-pylint/35b8c9c/ : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":44},{"id":"7210b6e6a06ebb039ba66e1a021ecc67bcd85a87","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-07 12:46:34.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"111d9068e192b2b1da4ce7e76775cb64478f46d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:43:13.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/45/check/build-openstack-api-ref/01c2eaa/html/ : SUCCESS in 4m 55s\n- requirements-check http://logs.openstack.org/68/652068/45/check/requirements-check/ab985b7/ : SUCCESS in 3m 11s\n- openstack-tox-cover http://logs.openstack.org/68/652068/45/check/openstack-tox-cover/0c7ae54/cover/ : SUCCESS in 19m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/45/check/openstack-tox-lower-constraints/9a6324f/ : SUCCESS in 14m 11s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/45/check/openstack-tox-pep8/d68eb84/ : SUCCESS in 7m 20s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/45/check/openstack-tox-py27/321464c/ : SUCCESS in 10m 41s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/45/check/openstack-tox-py36/bc9e512/ : SUCCESS in 12m 19s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/45/check/openstack-tox-py37/27b376c/ : SUCCESS in 17m 32s\n- openstack-tox-docs http://logs.openstack.org/68/652068/45/check/openstack-tox-docs/d2d1418/html/ : SUCCESS in 4m 29s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/45/check/build-openstack-releasenotes/68113a2/html/ : SUCCESS in 7m 37s\n- trove-functional-mysql http://logs.openstack.org/68/652068/45/check/trove-functional-mysql/85621d2/ : SUCCESS in 1h 55m 13s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/45/check/trove-scenario-mariadb-single/b4642f7/ : FAILURE in 37m 40s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/45/check/trove-scenario-mariadb-multi/8c6b21a/ : FAILURE in 45m 23s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/45/check/trove-scenario-mysql-single/487f02f/ : SUCCESS in 1h 36m 49s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/45/check/trove-scenario-mysql-multi/6f27536/ : SUCCESS in 1h 35m 58s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/45/check/trove-scenario-postgresql-single/80fe7de/ : FAILURE in 44m 28s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/45/check/trove-scenario-postgresql-multi/cc04bd1/ : FAILURE in 33m 02s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/45/check/trove-scenario-pxc-single/c2e9f4a/ : FAILURE in 42m 44s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/45/check/trove-scenario-pxc-multi/91004b3/ : FAILURE in 36m 31s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/45/check/trove-scenario-redis-single/6b3f2ff/ : FAILURE in 1h 15m 20s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/45/check/trove-scenario-redis-multi/1d48632/ : FAILURE in 1h 02m 41s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/45/check/trove-tox-fakemodetests/ed57e4c/ : SUCCESS in 5m 56s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/45/check/trove-tox-bandit-baseline/a1a336b/ : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/45/check/openstack-tox-pylint/3f8f1e9/ : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":45},{"id":"acc870b60f2c1feac7fc6efe9d413db171b3753e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-08 03:29:47.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"a72f51f27dfffb29f58b5b1e6a7cdcfd7e66b23b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 05:38:16.000000000","message":"Patch Set 45:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/45/check/build-openstack-api-ref/7c9b0e9/html/ : SUCCESS in 5m 15s\n- requirements-check http://logs.openstack.org/68/652068/45/check/requirements-check/b79330e/ : SUCCESS in 3m 22s\n- openstack-tox-cover http://logs.openstack.org/68/652068/45/check/openstack-tox-cover/4d31a4a/cover/ : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/45/check/openstack-tox-lower-constraints/9e539ed/ : SUCCESS in 12m 36s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/45/check/openstack-tox-pep8/ff0a3dc/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/45/check/openstack-tox-py27/7295cd7/ : SUCCESS in 12m 15s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/45/check/openstack-tox-py36/0199db4/ : SUCCESS in 19m 11s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/45/check/openstack-tox-py37/63fa3df/ : SUCCESS in 18m 18s\n- openstack-tox-docs http://logs.openstack.org/68/652068/45/check/openstack-tox-docs/945263e/html/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/45/check/build-openstack-releasenotes/d19e5f6/html/ : SUCCESS in 6m 15s\n- trove-functional-mysql http://logs.openstack.org/68/652068/45/check/trove-functional-mysql/7269d28/ : SUCCESS in 2h 05m 20s\n- trove-scenario-mariadb-single http://logs.openstack.org/68/652068/45/check/trove-scenario-mariadb-single/223698f/ : FAILURE in 1h 08m 27s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/68/652068/45/check/trove-scenario-mariadb-multi/98acf26/ : FAILURE in 45m 18s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/45/check/trove-scenario-mysql-single/d599097/ : SUCCESS in 1h 39m 56s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/45/check/trove-scenario-mysql-multi/ef1d490/ : SUCCESS in 1h 36m 35s\n- trove-scenario-postgresql-single http://logs.openstack.org/68/652068/45/check/trove-scenario-postgresql-single/69e7317/ : FAILURE in 34m 50s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/68/652068/45/check/trove-scenario-postgresql-multi/e48904c/ : FAILURE in 41m 13s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/68/652068/45/check/trove-scenario-pxc-single/b2beafe/ : FAILURE in 48m 12s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/68/652068/45/check/trove-scenario-pxc-multi/df3740e/ : FAILURE in 40m 58s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/68/652068/45/check/trove-scenario-redis-single/b5f99cd/ : FAILURE in 1h 04m 05s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/68/652068/45/check/trove-scenario-redis-multi/22405d9/ : FAILURE in 1h 03m 33s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/45/check/trove-tox-fakemodetests/f324a3c/ : SUCCESS in 8m 06s\n- trove-tox-bandit-baseline http://logs.openstack.org/68/652068/45/check/trove-tox-bandit-baseline/913c1b1/ : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/68/652068/45/check/openstack-tox-pylint/861a3b0/ : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":45},{"id":"2280014f512beca5c3e7bd8b7c0207a3b5cf9051","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-08 05:46:35.000000000","message":"Patch Set 45: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":45},{"id":"659d67c351cc5faaa53ddd1cc4dd0f93fd6d643d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 05:46:50.000000000","message":"Patch Set 45: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":45},{"id":"6301796d5d3dd841878bcf95fe1d99178d862b69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 07:57:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":45},{"id":"6259776a0e373f8bdf9cdb3e4fa55c9e3ecc18b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 07:57:46.000000000","message":"Patch Set 45: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/68/652068/45/gate/build-openstack-api-ref/fa48383/html/ : SUCCESS in 5m 07s\n- requirements-check http://logs.openstack.org/68/652068/45/gate/requirements-check/45f5ce9/ : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/652068/45/gate/openstack-tox-lower-constraints/c4e13cd/ : SUCCESS in 11m 01s\n- openstack-tox-pep8 http://logs.openstack.org/68/652068/45/gate/openstack-tox-pep8/1861ecb/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/68/652068/45/gate/openstack-tox-py27/2bab500/ : SUCCESS in 11m 46s\n- openstack-tox-py36 http://logs.openstack.org/68/652068/45/gate/openstack-tox-py36/27c4623/ : SUCCESS in 13m 02s\n- openstack-tox-py37 http://logs.openstack.org/68/652068/45/gate/openstack-tox-py37/c0fadc6/ : SUCCESS in 17m 38s\n- openstack-tox-docs http://logs.openstack.org/68/652068/45/gate/openstack-tox-docs/55610ec/html/ : SUCCESS in 5m 11s\n- build-openstack-releasenotes http://logs.openstack.org/68/652068/45/gate/build-openstack-releasenotes/9ab98ab/html/ : SUCCESS in 6m 55s\n- trove-functional-mysql http://logs.openstack.org/68/652068/45/gate/trove-functional-mysql/c711d82/ : SUCCESS in 2h 04m 42s\n- trove-scenario-mysql-single http://logs.openstack.org/68/652068/45/gate/trove-scenario-mysql-single/6ba1889/ : SUCCESS in 1h 35m 45s\n- trove-scenario-mysql-multi http://logs.openstack.org/68/652068/45/gate/trove-scenario-mysql-multi/333afa6/ : SUCCESS in 1h 57m 54s\n- trove-tox-fakemodetests http://logs.openstack.org/68/652068/45/gate/trove-tox-fakemodetests/7955275/ : SUCCESS in 5m 25s\n- openstack-tox-pylint http://logs.openstack.org/68/652068/45/gate/openstack-tox-pylint/e23afd0/ : SUCCESS in 9m 45s","accounts_in_message":[],"_revision_number":45}],"current_revision_number":45,"current_revision":"2e052b02620f3b25aac409e7afa42e11f5bd0bd8","revisions":{"4bb257def19708126090d8a3ef2a0fdbf128aae6":{"kind":"REWORK","_number":1,"created":"2019-04-12 13:12:12.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/1"}}},"commit":{"parents":[{"commit":"f4e17a34b91c99db363092c3ee93dca6da239f37","subject":"Merge \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4e17a34b91c99db363092c3ee93dca6da239f37"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-12 07:11:22.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-12 13:11:56.000000000","tz":720},"subject":"[WIP] Use service tenant deployment model by default in DevStack","message":"[WIP] Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the service tenant rather than the normal user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure approach for\nthe cloud provider, either private or public.\n\nDevStack should always follow the best practice with the production\ndeployment.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4bb257def19708126090d8a3ef2a0fdbf128aae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4bb257def19708126090d8a3ef2a0fdbf128aae6"}]},"branch":"refs/heads/master"},"7db8de3aa17fecb93518d1995bdb4a87af502925":{"kind":"REWORK","_number":2,"created":"2019-04-13 11:38:00.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/2"}}},"commit":{"parents":[{"commit":"f4e17a34b91c99db363092c3ee93dca6da239f37","subject":"Merge \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4e17a34b91c99db363092c3ee93dca6da239f37"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-12 07:11:22.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-13 11:37:47.000000000","tz":720},"subject":"[WIP] Use service tenant deployment model by default in DevStack","message":"[WIP] Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the service tenant rather than the normal user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure approach for\nthe cloud provider, either private or public.\n\nDevStack should always follow the best practice with the production\ndeployment.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7db8de3aa17fecb93518d1995bdb4a87af502925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7db8de3aa17fecb93518d1995bdb4a87af502925"}]},"branch":"refs/heads/master"},"7a7ba707bb3d10f3fe7c1e2e04fc2022067d391e":{"kind":"REWORK","_number":3,"created":"2019-04-15 02:33:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/3"}}},"commit":{"parents":[{"commit":"f4e17a34b91c99db363092c3ee93dca6da239f37","subject":"Merge \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4e17a34b91c99db363092c3ee93dca6da239f37"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-12 07:11:22.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-15 02:32:48.000000000","tz":720},"subject":"[WIP] Use service tenant deployment model by default in DevStack","message":"[WIP] Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the service tenant rather than the normal user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure approach for\nthe cloud provider, either private or public.\n\nDevStack should always follow the best practice with the production\ndeployment.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a7ba707bb3d10f3fe7c1e2e04fc2022067d391e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a7ba707bb3d10f3fe7c1e2e04fc2022067d391e"}]},"branch":"refs/heads/master"},"e968b7b5bd6eb29eceff96284721d2aca4e1ec3a":{"kind":"REWORK","_number":4,"created":"2019-04-15 03:24:50.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/4"}}},"commit":{"parents":[{"commit":"f4e17a34b91c99db363092c3ee93dca6da239f37","subject":"Merge \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4e17a34b91c99db363092c3ee93dca6da239f37"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-12 07:11:22.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-15 03:24:38.000000000","tz":720},"subject":"[WIP] Use service tenant deployment model by default in DevStack","message":"[WIP] Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the service tenant rather than the normal user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure approach for\nthe cloud provider, either private or public.\n\nDevStack should always follow the best practice with the production\ndeployment.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e968b7b5bd6eb29eceff96284721d2aca4e1ec3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e968b7b5bd6eb29eceff96284721d2aca4e1ec3a"}]},"branch":"refs/heads/master"},"554b092c2831ab1e2098844e0dd8d04ffe3b85ec":{"kind":"REWORK","_number":5,"created":"2019-04-15 07:24:53.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/5"}}},"commit":{"parents":[{"commit":"f4e17a34b91c99db363092c3ee93dca6da239f37","subject":"Merge \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4e17a34b91c99db363092c3ee93dca6da239f37"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-12 07:11:22.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-15 07:17:56.000000000","tz":720},"subject":"[WIP] Use service tenant deployment model by default in DevStack","message":"[WIP] Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the service tenant rather than the normal user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure approach for\nthe cloud provider, either private or public.\n\nDevStack should always follow the best practice with the production\ndeployment.\n\nChanges included in this patch:\n- Add \u0027admin\u0027 role to Trove service user.\n- Create a Trove management network for Trove service user.\n- Disable IPv6 subnet creation, I didn\u0027t see Trove could support IPv6\n  although the IPv6 subnet was created during DevStack installation.\n- Converge all the config options into a single config file.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/554b092c2831ab1e2098844e0dd8d04ffe3b85ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/554b092c2831ab1e2098844e0dd8d04ffe3b85ec"}]},"branch":"refs/heads/master"},"3d6791a748a4fc5053d2f22d5271193d8e05ee3a":{"kind":"REWORK","_number":6,"created":"2019-04-18 11:23:16.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/6"}}},"commit":{"parents":[{"commit":"7f7ae90ee9b869d62c559f30daf50589250ab811","subject":"Fix cloudinit mariadb scenario test error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f7ae90ee9b869d62c559f30daf50589250ab811"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:23:12.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3d6791a748a4fc5053d2f22d5271193d8e05ee3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3d6791a748a4fc5053d2f22d5271193d8e05ee3a"}]},"branch":"refs/heads/master"},"7d879cbcd1f663a8405e204af12bdb4cbfe8e1f9":{"kind":"REWORK","_number":7,"created":"2019-04-19 08:49:36.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/7"}}},"commit":{"parents":[{"commit":"7f7ae90ee9b869d62c559f30daf50589250ab811","subject":"Fix cloudinit mariadb scenario test error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f7ae90ee9b869d62c559f30daf50589250ab811"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-19 08:48:52.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d879cbcd1f663a8405e204af12bdb4cbfe8e1f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d879cbcd1f663a8405e204af12bdb4cbfe8e1f9"}]},"branch":"refs/heads/master"},"33844bdd2101477bf8f75a9e7f96589d93cf2e89":{"kind":"REWORK","_number":8,"created":"2019-04-19 12:20:04.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/8"}}},"commit":{"parents":[{"commit":"7f7ae90ee9b869d62c559f30daf50589250ab811","subject":"Fix cloudinit mariadb scenario test error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7f7ae90ee9b869d62c559f30daf50589250ab811"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-19 12:09:12.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33844bdd2101477bf8f75a9e7f96589d93cf2e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33844bdd2101477bf8f75a9e7f96589d93cf2e89"}]},"branch":"refs/heads/master"},"c0dca13f2ebd459c305d393274e81c4abd474097":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-04-22 06:31:48.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/9"}}},"commit":{"parents":[{"commit":"eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c","subject":"Use opendev.org instead of git.openstack.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-22 06:31:48.000000000","tz":0},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c0dca13f2ebd459c305d393274e81c4abd474097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c0dca13f2ebd459c305d393274e81c4abd474097"}]},"branch":"refs/heads/master"},"d06896628cfd0101f50bdf2d143a395db231e32f":{"kind":"REWORK","_number":10,"created":"2019-04-23 12:37:56.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/10"}}},"commit":{"parents":[{"commit":"eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c","subject":"Use opendev.org instead of git.openstack.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-23 12:37:36.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d06896628cfd0101f50bdf2d143a395db231e32f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d06896628cfd0101f50bdf2d143a395db231e32f"}]},"branch":"refs/heads/master"},"5ad8703d170f07bdaeba56d60ab93b4e62c633a8":{"kind":"REWORK","_number":11,"created":"2019-04-23 23:13:25.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/11"}}},"commit":{"parents":[{"commit":"eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c","subject":"Use opendev.org instead of git.openstack.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-23 23:13:09.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5ad8703d170f07bdaeba56d60ab93b4e62c633a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5ad8703d170f07bdaeba56d60ab93b4e62c633a8"}]},"branch":"refs/heads/master"},"67acde5880879d470180c650dab36e2797749dd5":{"kind":"REWORK","_number":12,"created":"2019-04-25 21:33:41.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/12"}}},"commit":{"parents":[{"commit":"eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c","subject":"Use opendev.org instead of git.openstack.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-25 21:33:29.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/67acde5880879d470180c650dab36e2797749dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/67acde5880879d470180c650dab36e2797749dd5"}]},"branch":"refs/heads/master"},"b73178498b7802bca07ac24c5d3af07458953c4d":{"kind":"REWORK","_number":13,"created":"2019-04-26 02:46:47.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/13"}}},"commit":{"parents":[{"commit":"eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c","subject":"Use opendev.org instead of git.openstack.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-26 02:46:36.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b73178498b7802bca07ac24c5d3af07458953c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b73178498b7802bca07ac24c5d3af07458953c4d"}]},"branch":"refs/heads/master"},"27e58b69a7853320cda781c43f95d3ecf832b8a9":{"kind":"REWORK","_number":14,"created":"2019-04-27 22:51:18.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/14"}}},"commit":{"parents":[{"commit":"eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c","subject":"Use opendev.org instead of git.openstack.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-27 12:50:50.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/27e58b69a7853320cda781c43f95d3ecf832b8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/27e58b69a7853320cda781c43f95d3ecf832b8a9"}]},"branch":"refs/heads/master"},"a83c3a23c72028cc19b2ab541859dcfcb987ae9c":{"kind":"REWORK","_number":15,"created":"2019-04-29 02:51:23.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/15"}}},"commit":{"parents":[{"commit":"eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c","subject":"Use opendev.org instead of git.openstack.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-29 02:51:01.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a83c3a23c72028cc19b2ab541859dcfcb987ae9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a83c3a23c72028cc19b2ab541859dcfcb987ae9c"}]},"branch":"refs/heads/master"},"7ba3d0824914f702244c7fb17d444bb26bdcb6c2":{"kind":"REWORK","_number":16,"created":"2019-05-08 13:08:16.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/16"}}},"commit":{"parents":[{"commit":"81c31ca6063ebe700e2dbcc0e30795104143ef3a","subject":"Merge \"Instead of deprecated keystone cli in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81c31ca6063ebe700e2dbcc0e30795104143ef3a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-08 13:08:05.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ba3d0824914f702244c7fb17d444bb26bdcb6c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ba3d0824914f702244c7fb17d444bb26bdcb6c2"}]},"branch":"refs/heads/master"},"19664b78d406ac9e1fe9c97fb9bb9dc2a2d987a4":{"kind":"REWORK","_number":17,"created":"2019-05-09 08:11:59.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/17"}}},"commit":{"parents":[{"commit":"81c31ca6063ebe700e2dbcc0e30795104143ef3a","subject":"Merge \"Instead of deprecated keystone cli in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81c31ca6063ebe700e2dbcc0e30795104143ef3a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-09 08:11:45.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/19664b78d406ac9e1fe9c97fb9bb9dc2a2d987a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/19664b78d406ac9e1fe9c97fb9bb9dc2a2d987a4"}]},"branch":"refs/heads/master"},"4e4ff5aefb7baa7c17303db4bdfbf1692d0fbf1c":{"kind":"REWORK","_number":18,"created":"2019-05-10 22:23:57.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/18"}}},"commit":{"parents":[{"commit":"81c31ca6063ebe700e2dbcc0e30795104143ef3a","subject":"Merge \"Instead of deprecated keystone cli in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81c31ca6063ebe700e2dbcc0e30795104143ef3a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-10 22:23:48.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e4ff5aefb7baa7c17303db4bdfbf1692d0fbf1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e4ff5aefb7baa7c17303db4bdfbf1692d0fbf1c"}]},"branch":"refs/heads/master"},"875ef4a642f0e75a906c65ad065227a955ee4ff1":{"kind":"REWORK","_number":19,"created":"2019-05-13 12:37:00.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/19"}}},"commit":{"parents":[{"commit":"81c31ca6063ebe700e2dbcc0e30795104143ef3a","subject":"Merge \"Instead of deprecated keystone cli in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81c31ca6063ebe700e2dbcc0e30795104143ef3a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-13 12:36:51.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/875ef4a642f0e75a906c65ad065227a955ee4ff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/875ef4a642f0e75a906c65ad065227a955ee4ff1"}]},"branch":"refs/heads/master"},"c4483c73e05cf85be120db3aed712760ed73c12f":{"kind":"REWORK","_number":20,"created":"2019-05-13 22:58:04.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/20"}}},"commit":{"parents":[{"commit":"81c31ca6063ebe700e2dbcc0e30795104143ef3a","subject":"Merge \"Instead of deprecated keystone cli in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81c31ca6063ebe700e2dbcc0e30795104143ef3a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-13 22:57:53.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4483c73e05cf85be120db3aed712760ed73c12f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4483c73e05cf85be120db3aed712760ed73c12f"}]},"branch":"refs/heads/master"},"2e85399ef51d3d1a922d63a628603a87111bd356":{"kind":"REWORK","_number":21,"created":"2019-05-14 02:59:20.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/21"}}},"commit":{"parents":[{"commit":"81c31ca6063ebe700e2dbcc0e30795104143ef3a","subject":"Merge \"Instead of deprecated keystone cli in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81c31ca6063ebe700e2dbcc0e30795104143ef3a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-14 02:59:09.000000000","tz":720},"subject":"Use service tenant deployment model by default in DevStack","message":"Use service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it\u0027s recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- The guest agent service is running as the root user because it needs\n  privilege to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e85399ef51d3d1a922d63a628603a87111bd356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e85399ef51d3d1a922d63a628603a87111bd356"}]},"branch":"refs/heads/master"},"0ba94aada53da4b01ad3b7bd36a3b7853d1e784f":{"kind":"REWORK","_number":22,"created":"2019-05-14 06:57:02.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/22"}}},"commit":{"parents":[{"commit":"8da960b8eb845bae3437fe31824909da344406a6","subject":"Merge \"Fix tox debug mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8da960b8eb845bae3437fe31824909da344406a6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-14 06:56:12.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0ba94aada53da4b01ad3b7bd36a3b7853d1e784f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0ba94aada53da4b01ad3b7bd36a3b7853d1e784f"}]},"branch":"refs/heads/master"},"12e8f2842063cffbee2e5073beea1fbae3591119":{"kind":"REWORK","_number":23,"created":"2019-05-17 01:17:53.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/23"}}},"commit":{"parents":[{"commit":"8da960b8eb845bae3437fe31824909da344406a6","subject":"Merge \"Fix tox debug mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8da960b8eb845bae3437fe31824909da344406a6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-16 23:03:43.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/12e8f2842063cffbee2e5073beea1fbae3591119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/12e8f2842063cffbee2e5073beea1fbae3591119"}]},"branch":"refs/heads/master"},"d54e3eda90744769b9a1437c4bab5e3d162fd1e3":{"kind":"REWORK","_number":24,"created":"2019-05-17 21:59:13.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/24"}}},"commit":{"parents":[{"commit":"8da960b8eb845bae3437fe31824909da344406a6","subject":"Merge \"Fix tox debug mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8da960b8eb845bae3437fe31824909da344406a6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-17 21:59:01.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d54e3eda90744769b9a1437c4bab5e3d162fd1e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d54e3eda90744769b9a1437c4bab5e3d162fd1e3"}]},"branch":"refs/heads/master"},"7a8b3fe910ab4f8336b6273070e184b3a0fe52b2":{"kind":"REWORK","_number":25,"created":"2019-05-19 10:48:44.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/25"}}},"commit":{"parents":[{"commit":"8da960b8eb845bae3437fe31824909da344406a6","subject":"Merge \"Fix tox debug mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8da960b8eb845bae3437fe31824909da344406a6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-19 10:48:31.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a8b3fe910ab4f8336b6273070e184b3a0fe52b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a8b3fe910ab4f8336b6273070e184b3a0fe52b2"}]},"branch":"refs/heads/master"},"af6146ab237e9b7a7159e38f047dabe63e5480de":{"kind":"REWORK","_number":26,"created":"2019-05-20 02:21:04.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/26"}}},"commit":{"parents":[{"commit":"8da960b8eb845bae3437fe31824909da344406a6","subject":"Merge \"Fix tox debug mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8da960b8eb845bae3437fe31824909da344406a6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-20 02:20:55.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af6146ab237e9b7a7159e38f047dabe63e5480de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af6146ab237e9b7a7159e38f047dabe63e5480de"}]},"branch":"refs/heads/master"},"96f43bb9ec5221c94fe67cdb49440056d542eb62":{"kind":"REWORK","_number":27,"created":"2019-05-20 11:49:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/27"}}},"commit":{"parents":[{"commit":"8da960b8eb845bae3437fe31824909da344406a6","subject":"Merge \"Fix tox debug mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8da960b8eb845bae3437fe31824909da344406a6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-20 11:49:02.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/96f43bb9ec5221c94fe67cdb49440056d542eb62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/96f43bb9ec5221c94fe67cdb49440056d542eb62"}]},"branch":"refs/heads/master"},"70453c9ee67ed31a55e10bae2f9f6d351a035684":{"kind":"REWORK","_number":28,"created":"2019-05-21 02:07:52.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/28"}}},"commit":{"parents":[{"commit":"8da960b8eb845bae3437fe31824909da344406a6","subject":"Merge \"Fix tox debug mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8da960b8eb845bae3437fe31824909da344406a6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-21 02:07:41.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/70453c9ee67ed31a55e10bae2f9f6d351a035684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/70453c9ee67ed31a55e10bae2f9f6d351a035684"}]},"branch":"refs/heads/master"},"b129812741d6b468ea6db8d55f92fd480d6ae89f":{"kind":"REWORK","_number":29,"created":"2019-05-21 04:21:53.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/29"}}},"commit":{"parents":[{"commit":"8da960b8eb845bae3437fe31824909da344406a6","subject":"Merge \"Fix tox debug mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8da960b8eb845bae3437fe31824909da344406a6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-21 04:21:43.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b129812741d6b468ea6db8d55f92fd480d6ae89f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b129812741d6b468ea6db8d55f92fd480d6ae89f"}]},"branch":"refs/heads/master"},"4e1e752b0455304c8f1dd7cacaa18e6889dc6d46":{"kind":"REWORK","_number":30,"created":"2019-05-21 08:48:46.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/30"}}},"commit":{"parents":[{"commit":"8da960b8eb845bae3437fe31824909da344406a6","subject":"Merge \"Fix tox debug mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8da960b8eb845bae3437fe31824909da344406a6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-21 08:48:33.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change.\n- Adding a new config option `management_gateway` for guest agent in\n  order to config the source based policy routing.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e1e752b0455304c8f1dd7cacaa18e6889dc6d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e1e752b0455304c8f1dd7cacaa18e6889dc6d46"}]},"branch":"refs/heads/master"},"8de479fd811aac505ba4a47f86bf00828de2ae18":{"kind":"REWORK","_number":31,"created":"2019-05-27 12:11:35.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/31"}}},"commit":{"parents":[{"commit":"8da960b8eb845bae3437fe31824909da344406a6","subject":"Merge \"Fix tox debug mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8da960b8eb845bae3437fe31824909da344406a6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-27 12:10:14.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change in the future.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8de479fd811aac505ba4a47f86bf00828de2ae18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8de479fd811aac505ba4a47f86bf00828de2ae18"}]},"branch":"refs/heads/master"},"91f652d52f2c4ef9b1f0c02ad002f4cfa295a583":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2019-05-27 23:49:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/32"}}},"commit":{"parents":[{"commit":"c37ab5f34b6bd1939a4121be684a56ae755dcdd4","subject":"docs: fix build failure on html_last_updated_fmt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c37ab5f34b6bd1939a4121be684a56ae755dcdd4"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-27 22:49:01.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- The guest agent service is running as the root user because it needs\n  privileges to do the network change in the future.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91f652d52f2c4ef9b1f0c02ad002f4cfa295a583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91f652d52f2c4ef9b1f0c02ad002f4cfa295a583"}]},"branch":"refs/heads/master"},"7791505bddbdd88dc763e4ca15da00d130191cee":{"kind":"REWORK","_number":33,"created":"2019-06-01 10:07:14.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/33"}}},"commit":{"parents":[{"commit":"c37ab5f34b6bd1939a4121be684a56ae755dcdd4","subject":"docs: fix build failure on html_last_updated_fmt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c37ab5f34b6bd1939a4121be684a56ae755dcdd4"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-01 10:05:32.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7791505bddbdd88dc763e4ca15da00d130191cee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7791505bddbdd88dc763e4ca15da00d130191cee"}]},"branch":"refs/heads/master"},"26e69397674357925c0b9bbd574ea13e8523d750":{"kind":"REWORK","_number":34,"created":"2019-06-03 12:40:23.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/34"}}},"commit":{"parents":[{"commit":"c37ab5f34b6bd1939a4121be684a56ae755dcdd4","subject":"docs: fix build failure on html_last_updated_fmt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c37ab5f34b6bd1939a4121be684a56ae755dcdd4"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-03 05:46:11.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26e69397674357925c0b9bbd574ea13e8523d750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26e69397674357925c0b9bbd574ea13e8523d750"}]},"branch":"refs/heads/master"},"68521e6b73ff4d062433a410fabf78784e4f08f3":{"kind":"REWORK","_number":35,"created":"2019-06-05 03:54:34.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/35"}}},"commit":{"parents":[{"commit":"73978ae595e393f6b0604be209f53d4051a50c48","subject":"Merge \"Add releasenotes for Redis upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73978ae595e393f6b0604be209f53d4051a50c48"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-05 03:54:22.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/68521e6b73ff4d062433a410fabf78784e4f08f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/68521e6b73ff4d062433a410fabf78784e4f08f3"}]},"branch":"refs/heads/master"},"59acb098b16ef389c8eac3d82a474dc4d2559da8":{"kind":"REWORK","_number":36,"created":"2019-06-05 09:20:45.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/36"}}},"commit":{"parents":[{"commit":"73978ae595e393f6b0604be209f53d4051a50c48","subject":"Merge \"Add releasenotes for Redis upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73978ae595e393f6b0604be209f53d4051a50c48"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-05 09:20:21.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59acb098b16ef389c8eac3d82a474dc4d2559da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59acb098b16ef389c8eac3d82a474dc4d2559da8"}]},"branch":"refs/heads/master"},"477903afb200d17efc704514ac7c5aa26e24c346":{"kind":"REWORK","_number":37,"created":"2019-06-05 09:23:51.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/37"}}},"commit":{"parents":[{"commit":"73978ae595e393f6b0604be209f53d4051a50c48","subject":"Merge \"Add releasenotes for Redis upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73978ae595e393f6b0604be209f53d4051a50c48"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-05 09:23:41.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/477903afb200d17efc704514ac7c5aa26e24c346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/477903afb200d17efc704514ac7c5aa26e24c346"}]},"branch":"refs/heads/master"},"c767f5b6b6e4b99e1c1e781727a74bb1880e0c6c":{"kind":"REWORK","_number":38,"created":"2019-06-05 11:48:56.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/38"}}},"commit":{"parents":[{"commit":"73978ae595e393f6b0604be209f53d4051a50c48","subject":"Merge \"Add releasenotes for Redis upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73978ae595e393f6b0604be209f53d4051a50c48"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-05 10:02:27.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c767f5b6b6e4b99e1c1e781727a74bb1880e0c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c767f5b6b6e4b99e1c1e781727a74bb1880e0c6c"}]},"branch":"refs/heads/master"},"a9839ad795b340c2c986b40ac372f8c358d3ecd2":{"kind":"REWORK","_number":39,"created":"2019-06-05 23:59:35.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/39"}}},"commit":{"parents":[{"commit":"73978ae595e393f6b0604be209f53d4051a50c48","subject":"Merge \"Add releasenotes for Redis upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73978ae595e393f6b0604be209f53d4051a50c48"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-04-18 11:17:39.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-05 23:59:20.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a9839ad795b340c2c986b40ac372f8c358d3ecd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a9839ad795b340c2c986b40ac372f8c358d3ecd2"}]},"branch":"refs/heads/master"},"92d6c8fa82561f4f2003c720fb6df824f1c496bf":{"kind":"REWORK","_number":40,"created":"2019-06-06 09:29:45.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/40"}}},"commit":{"parents":[{"commit":"37e8dedac27a4567c02c78c4fb7d45b1d50e717c","subject":"Remove the trove-tox-apiexamples CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/37e8dedac27a4567c02c78c4fb7d45b1d50e717c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-06 09:27:16.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-06 09:29:17.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/92d6c8fa82561f4f2003c720fb6df824f1c496bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/92d6c8fa82561f4f2003c720fb6df824f1c496bf"}]},"branch":"refs/heads/master"},"a0fd96985a5f25b3ac718c8d9656b13a135d8dc0":{"kind":"REWORK","_number":41,"created":"2019-06-06 22:08:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/41"}}},"commit":{"parents":[{"commit":"37e8dedac27a4567c02c78c4fb7d45b1d50e717c","subject":"Remove the trove-tox-apiexamples CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/37e8dedac27a4567c02c78c4fb7d45b1d50e717c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-06 09:27:16.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-06 22:06:46.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a0fd96985a5f25b3ac718c8d9656b13a135d8dc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a0fd96985a5f25b3ac718c8d9656b13a135d8dc0"}]},"branch":"refs/heads/master"},"5a8fca437dc0b5393863189a2ad659499516dc5d":{"kind":"REWORK","_number":42,"created":"2019-06-07 02:11:01.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/42"}}},"commit":{"parents":[{"commit":"37e8dedac27a4567c02c78c4fb7d45b1d50e717c","subject":"Remove the trove-tox-apiexamples CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/37e8dedac27a4567c02c78c4fb7d45b1d50e717c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-06 09:27:16.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-07 02:10:48.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a8fca437dc0b5393863189a2ad659499516dc5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a8fca437dc0b5393863189a2ad659499516dc5d"}]},"branch":"refs/heads/master"},"7a665907cd1a6530a89f39f0d05f1795cec53067":{"kind":"REWORK","_number":43,"created":"2019-06-07 04:00:30.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/43"}}},"commit":{"parents":[{"commit":"37e8dedac27a4567c02c78c4fb7d45b1d50e717c","subject":"Remove the trove-tox-apiexamples CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/37e8dedac27a4567c02c78c4fb7d45b1d50e717c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-06 09:27:16.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-07 04:00:17.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a665907cd1a6530a89f39f0d05f1795cec53067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a665907cd1a6530a89f39f0d05f1795cec53067"}]},"branch":"refs/heads/master"},"79e4d74b410caa8d8a79b251b6ef713dcb613c50":{"kind":"REWORK","_number":44,"created":"2019-06-07 09:16:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/44"}}},"commit":{"parents":[{"commit":"37e8dedac27a4567c02c78c4fb7d45b1d50e717c","subject":"Remove the trove-tox-apiexamples CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/37e8dedac27a4567c02c78c4fb7d45b1d50e717c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-06 09:27:16.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-07 07:30:20.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/79e4d74b410caa8d8a79b251b6ef713dcb613c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/79e4d74b410caa8d8a79b251b6ef713dcb613c50"}]},"branch":"refs/heads/master"},"2e052b02620f3b25aac409e7afa42e11f5bd0bd8":{"kind":"REWORK","_number":45,"created":"2019-06-07 12:46:34.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/652068/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/652068/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/652068/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/652068/45"}}},"commit":{"parents":[{"commit":"37e8dedac27a4567c02c78c4fb7d45b1d50e717c","subject":"Remove the trove-tox-apiexamples CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/37e8dedac27a4567c02c78c4fb7d45b1d50e717c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-06 09:27:16.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-07 12:46:21.000000000","tz":720},"subject":"Enable service tenant deployment model by default in DevStack","message":"Enable service tenant deployment model by default in DevStack\n\nThe service tenant deployment model means Trove creates most of the\nresources(vm, volume, security group, etc.) relating to a database\ninstance in the Trove service tenant rather than the end user.\n\nWith this deployment model, most of the related resources behind the\nscenes are invisible to the user, which is a more secure deployment\nmodel for either private or public cloud provider.\n\nDevStack should follow this model as it will be recommended for Trove\ndeployment in production.\n\nChanges included in this patch that are necessary in order to make that\nhappen:\n\n- Add \u0027admin\u0027 role to Trove service user(username: trove, project:\n  service) in DevStack.\n- Create Trove management network resources for Trove service user in\n  DevStack.\n- Enable Trove remote client configuration by default in DevStack.\n- Mainly use alt_demo user in alt_demo project for integration tests,\n  config trove user as the admin role user in integration tests.\n- Disable the module related tests(module_groups) for now because of no\n  use cases but need effort to fix all the failed tests in the service\n  tenant model.\n\nStory: #2005445\nTask: #30489\nChange-Id: I2efb69d3d50344914a875b773f62a227dba2ccaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e052b02620f3b25aac409e7afa42e11f5bd0bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e052b02620f3b25aac409e7afa42e11f5bd0bd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
