)]}'
{"id":"openstack%2Ftrove~652456","triplet_id":"openstack%2Ftrove~master~I0d0a3f62d652853963dc4fe204a413eec2bc1830","project":"openstack/trove","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"I0d0a3f62d652853963dc4fe204a413eec2bc1830","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-04-15 01:20:23.000000000","updated":"2019-04-28 23:59:15.000000000","submitted":"2019-04-28 23:59:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652456-1556495955051-7785bd57","meta_rev_id":"2a600d290bb6b148b92f57c83b495e4322a0a4aa","_number":652456,"virtual_id_number":652456,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},{"value":0,"_account_id":25113,"name":"Fan Zhang","email":"zh.f@outlook.com","username":"fanzhang"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-04-28 23:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},{"value":0,"_account_id":25113,"name":"Fan Zhang","email":"zh.f@outlook.com","username":"fanzhang"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-04-28 22:03:48.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},{"value":0,"_account_id":25113,"name":"Fan Zhang","email":"zh.f@outlook.com","username":"fanzhang"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-04-28 22:03:48.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},{"value":0,"_account_id":25113,"name":"Fan Zhang","email":"zh.f@outlook.com","username":"fanzhang"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25113,"name":"Fan Zhang","email":"zh.f@outlook.com","username":"fanzhang"},{"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"},{"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 05:47:26.000000000","updated_by":{"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"},"reviewer":{"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"},"state":"REVIEWER"},{"updated":"2019-04-17 05:47:39.000000000","updated_by":{"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},"reviewer":{"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},"state":"REVIEWER"},{"updated":"2019-04-22 02:48:14.000000000","updated_by":{"_account_id":25113,"name":"Fan Zhang","email":"zh.f@outlook.com","username":"fanzhang"},"reviewer":{"_account_id":25113,"name":"Fan Zhang","email":"zh.f@outlook.com","username":"fanzhang"},"state":"REVIEWER"},{"updated":"2019-04-22 22:59:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-04-28 01:39:40.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-04-28 22:03:48.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-04-28 23:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44cee7088911fe0c7bebed3fd1f040c6616fc6e5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-15 01:20:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1402d71110a1d1fba93fa480b86dc17e64a0cad6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 03:14:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/56/652456/1/check/test-release-openstack/ab6d4eb/ : SUCCESS in 3m 48s\n- openstack-tox-cover http://logs.openstack.org/56/652456/1/check/openstack-tox-cover/5594467/cover/ : SUCCESS in 17m 00s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652456/1/check/openstack-tox-lower-constraints/6e22f4a/ : SUCCESS in 11m 36s\n- openstack-tox-pep8 http://logs.openstack.org/56/652456/1/check/openstack-tox-pep8/588af11/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/56/652456/1/check/openstack-tox-py27/0b7f6fe/ : SUCCESS in 10m 26s\n- openstack-tox-py36 http://logs.openstack.org/56/652456/1/check/openstack-tox-py36/33a2062/ : SUCCESS in 12m 35s\n- openstack-tox-py37 http://logs.openstack.org/56/652456/1/check/openstack-tox-py37/432d690/ : SUCCESS in 16m 00s\n- openstack-tox-docs http://logs.openstack.org/56/652456/1/check/openstack-tox-docs/58b15ee/html/ : SUCCESS in 5m 36s\n- build-openstack-releasenotes http://logs.openstack.org/56/652456/1/check/build-openstack-releasenotes/bf28035/html/ : SUCCESS in 6m 21s\n- trove-functional-mysql http://logs.openstack.org/56/652456/1/check/trove-functional-mysql/4f5f840/ : SUCCESS in 1h 50m 33s\n- trove-scenario-mariadb-single http://logs.openstack.org/56/652456/1/check/trove-scenario-mariadb-single/9e6a9ef/ : FAILURE in 32m 12s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-mariadb-multi/27b00dc/ : FAILURE in 43m 06s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/56/652456/1/check/trove-scenario-mysql-single/284a06a/ : SUCCESS in 1h 43m 32s\n- trove-scenario-mysql-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-mysql-multi/739810e/ : SUCCESS in 1h 28m 44s\n- trove-scenario-postgresql-single http://logs.openstack.org/56/652456/1/check/trove-scenario-postgresql-single/1cfe333/ : FAILURE in 44m 31s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-postgresql-multi/e6a1260/ : FAILURE in 42m 55s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/56/652456/1/check/trove-scenario-pxc-single/2e7902b/ : FAILURE in 44m 52s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-pxc-multi/47895c2/ : FAILURE in 43m 22s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/56/652456/1/check/trove-scenario-redis-single/c2e0267/ : FAILURE in 1h 09m 30s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-redis-multi/cff25c6/ : FAILURE in 1h 05m 05s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/56/652456/1/check/trove-tox-fakemodetests/38c9306/ : SUCCESS in 5m 54s\n- trove-tox-apiexamples http://logs.openstack.org/56/652456/1/check/trove-tox-apiexamples/ed35055/ : SUCCESS in 5m 32s\n- trove-tox-bandit-baseline http://logs.openstack.org/56/652456/1/check/trove-tox-bandit-baseline/6443689/ : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/56/652456/1/check/openstack-tox-pylint/e135a7a/ : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"819ace2ed1c0503d7619f6eaf563436b2ffdd0e5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-15 07:26:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"52491684927926a02a398c14b48fb7c0eced5e29","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-19 10:01:52.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c94dd9cfb6dfeccc4cfb0f89317369b13edc39e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 10:02:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf2550e6a14a5f37defc042f0865cfc3612af6ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 13:28:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/56/652456/1/gate/test-release-openstack/6477a85/ : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652456/1/gate/openstack-tox-lower-constraints/6871c2e/ : SUCCESS in 12m 34s\n- openstack-tox-pep8 http://logs.openstack.org/56/652456/1/gate/openstack-tox-pep8/ac806c5/ : SUCCESS in 5m 48s\n- openstack-tox-py27 http://logs.openstack.org/56/652456/1/gate/openstack-tox-py27/730aa2c/ : SUCCESS in 11m 16s\n- openstack-tox-py36 http://logs.openstack.org/56/652456/1/gate/openstack-tox-py36/88cef43/ : SUCCESS in 12m 56s\n- openstack-tox-py37 http://logs.openstack.org/56/652456/1/gate/openstack-tox-py37/b4450d3/ : SUCCESS in 17m 38s\n- openstack-tox-docs http://logs.openstack.org/56/652456/1/gate/openstack-tox-docs/269d339/html/ : SUCCESS in 5m 00s\n- build-openstack-releasenotes http://logs.openstack.org/56/652456/1/gate/build-openstack-releasenotes/cfbe833/html/ : SUCCESS in 6m 31s\n- trove-functional-mysql finger://ze05.openstack.org/8dba9e8166b94ca6b330209a188f847c : RETRY_LIMIT in 1h 06m 01s\n- trove-scenario-mysql-single http://logs.openstack.org/56/652456/1/gate/trove-scenario-mysql-single/bc1239f/ : FAILURE in 44m 50s\n- trove-scenario-mysql-multi http://logs.openstack.org/56/652456/1/gate/trove-scenario-mysql-multi/c52dd9c/ : FAILURE in 44m 56s\n- trove-tox-fakemodetests http://logs.openstack.org/56/652456/1/gate/trove-tox-fakemodetests/a0eac53/ : SUCCESS in 5m 50s\n- trove-tox-apiexamples http://logs.openstack.org/56/652456/1/gate/trove-tox-apiexamples/1d545fa/ : SUCCESS in 5m 25s\n- openstack-tox-pylint http://logs.openstack.org/56/652456/1/gate/openstack-tox-pylint/aee85d3/ : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2c524882c6e0fd2f7be3325cb89394d22efe2065","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-20 05:38:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"da6f1a46ac3d67f4803db3dadf7ea0feba74d65d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 06:29:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/56/652456/1/check/test-release-openstack/bbd2542/ : SUCCESS in 2m 55s\n- openstack-tox-cover http://logs.openstack.org/56/652456/1/check/openstack-tox-cover/3ecaeef/cover/ : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652456/1/check/openstack-tox-lower-constraints/0f65d81/ : SUCCESS in 12m 20s\n- openstack-tox-pep8 http://logs.openstack.org/56/652456/1/check/openstack-tox-pep8/6bf330e/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/56/652456/1/check/openstack-tox-py27/1013708/ : SUCCESS in 11m 02s\n- openstack-tox-py36 http://logs.openstack.org/56/652456/1/check/openstack-tox-py36/0437e94/ : SUCCESS in 12m 15s\n- openstack-tox-py37 http://logs.openstack.org/56/652456/1/check/openstack-tox-py37/f24ef79/ : SUCCESS in 17m 46s\n- openstack-tox-docs http://logs.openstack.org/56/652456/1/check/openstack-tox-docs/f2e7bdf/html/ : SUCCESS in 4m 52s\n- build-openstack-releasenotes http://logs.openstack.org/56/652456/1/check/build-openstack-releasenotes/20725a1/html/ : SUCCESS in 6m 12s\n- trove-functional-mysql http://logs.openstack.org/56/652456/1/check/trove-functional-mysql/9931c6f/ : FAILURE in 39m 14s\n- trove-scenario-mariadb-single http://logs.openstack.org/56/652456/1/check/trove-scenario-mariadb-single/3c06524/ : FAILURE in 44m 31s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-mariadb-multi/2569af6/ : FAILURE in 35m 25s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/56/652456/1/check/trove-scenario-mysql-single/36154e6/ : FAILURE in 32m 27s\n- trove-scenario-mysql-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-mysql-multi/f1430c6/ : FAILURE in 31m 53s\n- trove-scenario-postgresql-single http://logs.openstack.org/56/652456/1/check/trove-scenario-postgresql-single/8923d60/ : FAILURE in 40m 19s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-postgresql-multi/eb72e7a/ : FAILURE in 38m 08s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/56/652456/1/check/trove-scenario-pxc-single/7e39d7c/ : FAILURE in 38m 03s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-pxc-multi/246e7c8/ : FAILURE in 38m 31s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/56/652456/1/check/trove-scenario-redis-single/c91f5c1/ : FAILURE in 32m 38s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-redis-multi/8449855/ : FAILURE in 32m 17s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/56/652456/1/check/trove-tox-fakemodetests/89c8337/ : SUCCESS in 5m 44s\n- trove-tox-apiexamples http://logs.openstack.org/56/652456/1/check/trove-tox-apiexamples/d2c59d6/ : SUCCESS in 4m 58s\n- trove-tox-bandit-baseline http://logs.openstack.org/56/652456/1/check/trove-tox-bandit-baseline/57cb30d/ : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/56/652456/1/check/openstack-tox-pylint/7006c4a/ : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c8d7f7ce30f3d60a22fbd3ddf526997c2fb027bd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-21 08:05:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b01731e8e0a4258fed564b966a117778c88e0c42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 08:51:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/56/652456/1/check/test-release-openstack/9c349ef/ : SUCCESS in 4m 21s\n- openstack-tox-cover http://logs.openstack.org/56/652456/1/check/openstack-tox-cover/65f8918/cover/ : SUCCESS in 17m 41s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652456/1/check/openstack-tox-lower-constraints/3b8b59c/ : SUCCESS in 12m 14s\n- openstack-tox-pep8 http://logs.openstack.org/56/652456/1/check/openstack-tox-pep8/fb8f5d6/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/56/652456/1/check/openstack-tox-py27/ea2960b/ : SUCCESS in 12m 01s\n- openstack-tox-py36 http://logs.openstack.org/56/652456/1/check/openstack-tox-py36/7ad0ef7/ : SUCCESS in 12m 12s\n- openstack-tox-py37 http://logs.openstack.org/56/652456/1/check/openstack-tox-py37/cdad639/ : SUCCESS in 17m 33s\n- openstack-tox-docs http://logs.openstack.org/56/652456/1/check/openstack-tox-docs/821762f/html/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes http://logs.openstack.org/56/652456/1/check/build-openstack-releasenotes/20a16ec/html/ : SUCCESS in 6m 20s\n- trove-functional-mysql http://logs.openstack.org/56/652456/1/check/trove-functional-mysql/76119c4/ : FAILURE in 32m 26s\n- trove-scenario-mariadb-single http://logs.openstack.org/56/652456/1/check/trove-scenario-mariadb-single/9bff25c/ : FAILURE in 41m 08s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-mariadb-multi/bc38c81/ : FAILURE in 33m 01s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/56/652456/1/check/trove-scenario-mysql-single/9b6f513/ : FAILURE in 32m 38s\n- trove-scenario-mysql-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-mysql-multi/7f88a7c/ : FAILURE in 32m 04s\n- trove-scenario-postgresql-single http://logs.openstack.org/56/652456/1/check/trove-scenario-postgresql-single/303a71c/ : FAILURE in 28m 06s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-postgresql-multi/481f79d/ : FAILURE in 33m 39s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/56/652456/1/check/trove-scenario-pxc-single/d9b5a47/ : FAILURE in 32m 37s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-pxc-multi/547d45d/ : FAILURE in 32m 57s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/56/652456/1/check/trove-scenario-redis-single/8dbb4df/ : FAILURE in 30m 04s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-redis-multi/792b955/ : FAILURE in 31m 16s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/56/652456/1/check/trove-tox-fakemodetests/95f6d3b/ : SUCCESS in 5m 26s\n- trove-tox-apiexamples http://logs.openstack.org/56/652456/1/check/trove-tox-apiexamples/99f7192/ : SUCCESS in 5m 22s\n- trove-tox-bandit-baseline http://logs.openstack.org/56/652456/1/check/trove-tox-bandit-baseline/1a4698c/ : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/56/652456/1/check/openstack-tox-pylint/0a014ea/ : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":1},{"id":"96c557f897f5bbb7dec5c3ee614be44e7d0958d5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-21 10:36:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4199407775049c7a493f0401b98b605755162f3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 11:24:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/56/652456/1/check/test-release-openstack/b38ca9c/ : SUCCESS in 3m 17s\n- openstack-tox-cover http://logs.openstack.org/56/652456/1/check/openstack-tox-cover/7112c62/cover/ : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652456/1/check/openstack-tox-lower-constraints/615beae/ : SUCCESS in 12m 49s\n- openstack-tox-pep8 http://logs.openstack.org/56/652456/1/check/openstack-tox-pep8/8deb54a/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/56/652456/1/check/openstack-tox-py27/28e607c/ : SUCCESS in 11m 14s\n- openstack-tox-py36 http://logs.openstack.org/56/652456/1/check/openstack-tox-py36/7e600fe/ : SUCCESS in 12m 50s\n- openstack-tox-py37 http://logs.openstack.org/56/652456/1/check/openstack-tox-py37/5cbe486/ : SUCCESS in 17m 47s\n- openstack-tox-docs http://logs.openstack.org/56/652456/1/check/openstack-tox-docs/8ffa4e9/html/ : SUCCESS in 5m 01s\n- build-openstack-releasenotes http://logs.openstack.org/56/652456/1/check/build-openstack-releasenotes/efb730a/html/ : SUCCESS in 6m 28s\n- trove-functional-mysql http://logs.openstack.org/56/652456/1/check/trove-functional-mysql/0e80998/ : FAILURE in 32m 25s\n- trove-scenario-mariadb-single http://logs.openstack.org/56/652456/1/check/trove-scenario-mariadb-single/ba53369/ : FAILURE in 41m 38s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-mariadb-multi/07827c2/ : FAILURE in 32m 15s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/56/652456/1/check/trove-scenario-mysql-single/ae7b697/ : FAILURE in 33m 15s\n- trove-scenario-mysql-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-mysql-multi/a2f61e7/ : FAILURE in 32m 44s\n- trove-scenario-postgresql-single http://logs.openstack.org/56/652456/1/check/trove-scenario-postgresql-single/b97daf5/ : FAILURE in 39m 29s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-postgresql-multi/a606427/ : FAILURE in 32m 24s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/56/652456/1/check/trove-scenario-pxc-single/f1e0f06/ : FAILURE in 31m 51s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-pxc-multi/8464fbe/ : FAILURE in 35m 25s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/56/652456/1/check/trove-scenario-redis-single/33c8570/ : FAILURE in 33m 55s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/56/652456/1/check/trove-scenario-redis-multi/f14c157/ : FAILURE in 35m 42s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/56/652456/1/check/trove-tox-fakemodetests/59e4d45/ : SUCCESS in 5m 39s\n- trove-tox-apiexamples http://logs.openstack.org/56/652456/1/check/trove-tox-apiexamples/e765687/ : SUCCESS in 5m 20s\n- trove-tox-bandit-baseline http://logs.openstack.org/56/652456/1/check/trove-tox-bandit-baseline/54c7ae1/ : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/56/652456/1/check/openstack-tox-pylint/fa5edc1/ : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":1},{"id":"df6df337b25b745958497c815aaa8b554d468511","author":{"_account_id":25113,"name":"Fan Zhang","email":"zh.f@outlook.com","username":"fanzhang"},"date":"2019-04-22 02:48:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"530d635ef46ffbc5473fd263c70373b53b611981","author":{"_account_id":25113,"name":"Fan Zhang","email":"zh.f@outlook.com","username":"fanzhang"},"date":"2019-04-22 02:48:14.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2653fd6c0e99882fb6ef55f95cc18d482e0add38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 03:38:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/56/652456/2/check/test-release-openstack/bf3de5d/ : SUCCESS in 3m 16s\n- openstack-tox-cover http://logs.openstack.org/56/652456/2/check/openstack-tox-cover/eb04690/cover/ : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652456/2/check/openstack-tox-lower-constraints/1639168/ : SUCCESS in 12m 27s\n- openstack-tox-pep8 http://logs.openstack.org/56/652456/2/check/openstack-tox-pep8/bb1de25/ : SUCCESS in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/56/652456/2/check/openstack-tox-py27/6750853/ : SUCCESS in 11m 14s\n- openstack-tox-py36 http://logs.openstack.org/56/652456/2/check/openstack-tox-py36/bdff817/ : SUCCESS in 12m 38s\n- openstack-tox-py37 http://logs.openstack.org/56/652456/2/check/openstack-tox-py37/3a4a1e9/ : SUCCESS in 17m 50s\n- openstack-tox-docs http://logs.openstack.org/56/652456/2/check/openstack-tox-docs/ad037f3/html/ : SUCCESS in 5m 28s\n- build-openstack-releasenotes http://logs.openstack.org/56/652456/2/check/build-openstack-releasenotes/2cdf403/html/ : SUCCESS in 6m 43s\n- trove-functional-mysql http://logs.openstack.org/56/652456/2/check/trove-functional-mysql/c2d4c13/ : FAILURE in 32m 49s\n- trove-scenario-mariadb-single http://logs.openstack.org/56/652456/2/check/trove-scenario-mariadb-single/8694550/ : FAILURE in 44m 24s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/56/652456/2/check/trove-scenario-mariadb-multi/593d51d/ : FAILURE in 33m 10s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/56/652456/2/check/trove-scenario-mysql-single/3d85a71/ : FAILURE in 34m 20s\n- trove-scenario-mysql-multi http://logs.openstack.org/56/652456/2/check/trove-scenario-mysql-multi/87dfb5f/ : FAILURE in 36m 23s\n- trove-scenario-postgresql-single http://logs.openstack.org/56/652456/2/check/trove-scenario-postgresql-single/85bf09a/ : FAILURE in 32m 11s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/56/652456/2/check/trove-scenario-postgresql-multi/bd74a27/ : FAILURE in 36m 46s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/56/652456/2/check/trove-scenario-pxc-single/09a388e/ : FAILURE in 35m 01s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/56/652456/2/check/trove-scenario-pxc-multi/2e2a98b/ : FAILURE in 34m 58s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/56/652456/2/check/trove-scenario-redis-single/bb76c6e/ : FAILURE in 32m 30s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/56/652456/2/check/trove-scenario-redis-multi/3f5a6fd/ : FAILURE in 32m 55s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/56/652456/2/check/trove-tox-fakemodetests/86180ff/ : SUCCESS in 5m 46s\n- trove-tox-apiexamples http://logs.openstack.org/56/652456/2/check/trove-tox-apiexamples/a73187b/ : SUCCESS in 5m 29s\n- trove-tox-bandit-baseline http://logs.openstack.org/56/652456/2/check/trove-tox-bandit-baseline/955f065/ : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/56/652456/2/check/openstack-tox-pylint/b4eae3a/ : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6e6bf487bdd4c6c8ddf36730a220badf058f4c7a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-22 06:07:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"39023dbfa7d2876385b24d6f6c22cb221d3a1f70","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-22 06:18:51.000000000","message":"Patch Set 2: -Code-Review\n\n@zhongshengping, thanks for helping, but please stop rechecking until https://review.opendev.org/#/c/654231/ is merged.","accounts_in_message":[],"_revision_number":2},{"id":"a7d63aff8a7003dfad5f982601209451ca79e361","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-22 06:32:48.000000000","message":"Patch Set 2:\n\nplease rebase","accounts_in_message":[],"_revision_number":2},{"id":"f0339b448914563e2158c14f591eb8163647d792","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-22 06:33:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3c25ffe538074344415bf04bcbb5c210a4d2fcad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:55:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/56/652456/3/check/test-release-openstack/eb6c697/ : SUCCESS in 3m 04s\n- openstack-tox-cover http://logs.openstack.org/56/652456/3/check/openstack-tox-cover/a2aca2c/cover/ : SUCCESS in 18m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652456/3/check/openstack-tox-lower-constraints/f8425f6/ : SUCCESS in 13m 06s\n- openstack-tox-pep8 http://logs.openstack.org/56/652456/3/check/openstack-tox-pep8/a0f16a4/ : SUCCESS in 7m 28s\n- openstack-tox-py27 http://logs.openstack.org/56/652456/3/check/openstack-tox-py27/c17b544/ : SUCCESS in 12m 17s\n- openstack-tox-py36 http://logs.openstack.org/56/652456/3/check/openstack-tox-py36/45305d3/ : SUCCESS in 13m 23s\n- openstack-tox-py37 http://logs.openstack.org/56/652456/3/check/openstack-tox-py37/daf81e2/ : SUCCESS in 17m 53s\n- openstack-tox-docs http://logs.openstack.org/56/652456/3/check/openstack-tox-docs/88a78b7/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/56/652456/3/check/build-openstack-releasenotes/3d34e49/html/ : SUCCESS in 7m 25s\n- trove-functional-mysql http://logs.openstack.org/56/652456/3/check/trove-functional-mysql/ae1a9c6/ : SUCCESS in 1h 57m 11s\n- trove-scenario-mariadb-single http://logs.openstack.org/56/652456/3/check/trove-scenario-mariadb-single/9e32a11/ : FAILURE in 38m 55s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/56/652456/3/check/trove-scenario-mariadb-multi/ed86f97/ : FAILURE in 48m 19s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/56/652456/3/check/trove-scenario-mysql-single/4193522/ : SUCCESS in 1h 57m 16s\n- trove-scenario-mysql-multi http://logs.openstack.org/56/652456/3/check/trove-scenario-mysql-multi/085c571/ : SUCCESS in 1h 42m 47s\n- trove-scenario-postgresql-single http://logs.openstack.org/56/652456/3/check/trove-scenario-postgresql-single/261d740/ : FAILURE in 48m 53s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/56/652456/3/check/trove-scenario-postgresql-multi/c3c1a7a/ : FAILURE in 49m 13s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/56/652456/3/check/trove-scenario-pxc-single/5daf8db/ : FAILURE in 48m 10s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/56/652456/3/check/trove-scenario-pxc-multi/4be8c23/ : FAILURE in 48m 54s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/56/652456/3/check/trove-scenario-redis-single/5172d12/ : FAILURE in 1h 05m 18s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/56/652456/3/check/trove-scenario-redis-multi/96bc589/ : FAILURE in 1h 11m 03s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/56/652456/3/check/trove-tox-fakemodetests/cffb004/ : SUCCESS in 5m 40s\n- trove-tox-apiexamples http://logs.openstack.org/56/652456/3/check/trove-tox-apiexamples/4abc0cf/ : SUCCESS in 5m 52s\n- trove-tox-bandit-baseline http://logs.openstack.org/56/652456/3/check/trove-tox-bandit-baseline/23314c6/ : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/56/652456/3/check/openstack-tox-pylint/736cb78/ : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":3},{"id":"de1e3ae480ae8c866583b1bdd6581886e0b38c48","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-22 11:37:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"90e2372244f2f7c461a7f31a06074b3ee5c4f16f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-22 22:59:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"24f887df6c46545c380051d3127c04aee6c67c35","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-04-28 01:39:40.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"066cc4880d6372a3933b276a48361e35617036e4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-28 05:37:56.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e579ba3989ab5c0c3d8f748977171c79b9689ccd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 07:28:24.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"27f5015f2c3a4eae7dd3e202668996c8bd4582cd","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 07:37:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1245df6c5458db9c105e7dff95b3849dace5fd24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 09:38:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/56/652456/4/check/test-release-openstack/dff4787/ : SUCCESS in 3m 23s\n- openstack-tox-cover http://logs.openstack.org/56/652456/4/check/openstack-tox-cover/fc771b7/cover/ : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652456/4/check/openstack-tox-lower-constraints/75e3581/ : SUCCESS in 12m 33s\n- openstack-tox-pep8 http://logs.openstack.org/56/652456/4/check/openstack-tox-pep8/f39d255/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/56/652456/4/check/openstack-tox-py27/b390a3e/ : SUCCESS in 10m 11s\n- openstack-tox-py36 http://logs.openstack.org/56/652456/4/check/openstack-tox-py36/60d392d/ : SUCCESS in 12m 24s\n- openstack-tox-py37 http://logs.openstack.org/56/652456/4/check/openstack-tox-py37/1d49af9/ : SUCCESS in 15m 13s\n- openstack-tox-docs http://logs.openstack.org/56/652456/4/check/openstack-tox-docs/f7c182a/html/ : SUCCESS in 5m 20s\n- build-openstack-releasenotes http://logs.openstack.org/56/652456/4/check/build-openstack-releasenotes/e12eef6/html/ : SUCCESS in 6m 24s\n- trove-functional-mysql http://logs.openstack.org/56/652456/4/check/trove-functional-mysql/48ae103/ : SUCCESS in 1h 58m 52s\n- trove-scenario-mariadb-single http://logs.openstack.org/56/652456/4/check/trove-scenario-mariadb-single/5d881d8/ : FAILURE in 32m 21s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/56/652456/4/check/trove-scenario-mariadb-multi/98f1424/ : FAILURE in 48m 13s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/56/652456/4/check/trove-scenario-mysql-single/f99bfcf/ : SUCCESS in 1h 52m 51s\n- trove-scenario-mysql-multi http://logs.openstack.org/56/652456/4/check/trove-scenario-mysql-multi/e556bc4/ : SUCCESS in 1h 22m 11s\n- trove-scenario-postgresql-single http://logs.openstack.org/56/652456/4/check/trove-scenario-postgresql-single/b18bfe6/ : FAILURE in 48m 12s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/56/652456/4/check/trove-scenario-postgresql-multi/e70a8a8/ : FAILURE in 45m 50s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/56/652456/4/check/trove-scenario-pxc-single/4614526/ : FAILURE in 28m 56s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/56/652456/4/check/trove-scenario-pxc-multi/656045f/ : FAILURE in 47m 50s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/56/652456/4/check/trove-scenario-redis-single/791d982/ : FAILURE in 1h 17m 22s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/56/652456/4/check/trove-scenario-redis-multi/83175df/ : FAILURE in 1h 06m 48s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/56/652456/4/check/trove-tox-fakemodetests/ff6539e/ : SUCCESS in 6m 02s\n- trove-tox-apiexamples http://logs.openstack.org/56/652456/4/check/trove-tox-apiexamples/4200582/ : SUCCESS in 4m 57s\n- trove-tox-bandit-baseline http://logs.openstack.org/56/652456/4/check/trove-tox-bandit-baseline/ddad8dd/ : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/56/652456/4/check/openstack-tox-pylint/b38ba38/ : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":4},{"id":"0703beacd247ac424c2864aefbdb1759cd3018d8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-04-28 22:03:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"afe01f047223bc53e4ee5860f1df759e496c5898","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 22:04:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8e09934b5e446e96d9f8bff1d55e83f016f2be82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 23:59:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/56/652456/4/gate/test-release-openstack/3066e3c/ : SUCCESS in 3m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652456/4/gate/openstack-tox-lower-constraints/32aebd7/ : SUCCESS in 12m 05s\n- openstack-tox-pep8 http://logs.openstack.org/56/652456/4/gate/openstack-tox-pep8/c94fdd5/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/56/652456/4/gate/openstack-tox-py27/2ff796c/ : SUCCESS in 10m 04s\n- openstack-tox-py36 http://logs.openstack.org/56/652456/4/gate/openstack-tox-py36/1cc1d83/ : SUCCESS in 10m 21s\n- openstack-tox-py37 http://logs.openstack.org/56/652456/4/gate/openstack-tox-py37/7de731d/ : SUCCESS in 16m 51s\n- openstack-tox-docs http://logs.openstack.org/56/652456/4/gate/openstack-tox-docs/19f2703/html/ : SUCCESS in 4m 35s\n- build-openstack-releasenotes http://logs.openstack.org/56/652456/4/gate/build-openstack-releasenotes/8b2e464/html/ : SUCCESS in 5m 55s\n- trove-functional-mysql http://logs.openstack.org/56/652456/4/gate/trove-functional-mysql/d198c9f/ : SUCCESS in 1h 50m 23s\n- trove-scenario-mysql-single http://logs.openstack.org/56/652456/4/gate/trove-scenario-mysql-single/8597d19/ : SUCCESS in 1h 27m 47s\n- trove-scenario-mysql-multi http://logs.openstack.org/56/652456/4/gate/trove-scenario-mysql-multi/7356ea5/ : SUCCESS in 1h 14m 08s\n- trove-tox-fakemodetests http://logs.openstack.org/56/652456/4/gate/trove-tox-fakemodetests/edb2a9a/ : SUCCESS in 5m 21s\n- trove-tox-apiexamples http://logs.openstack.org/56/652456/4/gate/trove-tox-apiexamples/cec94ee/ : SUCCESS in 5m 04s\n- openstack-tox-pylint http://logs.openstack.org/56/652456/4/gate/openstack-tox-pylint/d09b0cb/ : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":4},{"id":"9b8b6702411b71fea415a5a06ecd48ac20798336","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 23:59:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4e63540245da0a910d6e69fb4a0cd56c45ec46a4","revisions":{"989701c090481fd974f506c83b534eb7628dcd41":{"kind":"REWORK","_number":1,"created":"2019-04-15 01:20:23.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/56/652456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/652456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/652456/1"}}},"commit":{"parents":[{"commit":"f4e17a34b91c99db363092c3ee93dca6da239f37","subject":"Merge \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4e17a34b91c99db363092c3ee93dca6da239f37"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-15 01:19:57.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-15 01:19:57.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I0d0a3f62d652853963dc4fe204a413eec2bc1830\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/989701c090481fd974f506c83b534eb7628dcd41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/989701c090481fd974f506c83b534eb7628dcd41"}]},"branch":"refs/heads/master"},"5676db56d8db0a1f1cea1df69bc9feb4e20eef7c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-22 02:48:12.000000000","uploader":{"_account_id":25113,"name":"Fan Zhang","email":"zh.f@outlook.com","username":"fanzhang"},"ref":"refs/changes/56/652456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/652456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/652456/2"}}},"commit":{"parents":[{"commit":"13711bd589c3474da4a89ab2dc805ee26275e70c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/13711bd589c3474da4a89ab2dc805ee26275e70c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-15 01:19:57.000000000","tz":480},"committer":{"name":"Fan Zhang","email":"zh.f@outlook.com","date":"2019-04-22 02:48:12.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I0d0a3f62d652853963dc4fe204a413eec2bc1830\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5676db56d8db0a1f1cea1df69bc9feb4e20eef7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5676db56d8db0a1f1cea1df69bc9feb4e20eef7c"}]},"branch":"refs/heads/master"},"9a06700421d34490b2fed8b7e7bd739fea14af9f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-22 06:33:26.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/56/652456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/652456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/652456/3"}}},"commit":{"parents":[{"commit":"eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c","subject":"Use opendev.org instead of git.openstack.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eafd303e0a781eaaf10acdc4d208e49fbf0f4c2c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-15 01:19:57.000000000","tz":480},"committer":{"name":"zhongshengping","email":"chdzsp@163.com","date":"2019-04-22 06:33:26.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I0d0a3f62d652853963dc4fe204a413eec2bc1830\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9a06700421d34490b2fed8b7e7bd739fea14af9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9a06700421d34490b2fed8b7e7bd739fea14af9f"}]},"branch":"refs/heads/master"},"4e63540245da0a910d6e69fb4a0cd56c45ec46a4":{"kind":"REWORK","_number":4,"created":"2019-04-28 07:37:17.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/56/652456/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/652456/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/652456/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/652456/4"}}},"commit":{"parents":[{"commit":"b53872099dbfb5ca644818f0e762e84706defe47","subject":"Merge \"Move to opendev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b53872099dbfb5ca644818f0e762e84706defe47"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-15 01:19:57.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:36:48.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I0d0a3f62d652853963dc4fe204a413eec2bc1830\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e63540245da0a910d6e69fb4a0cd56c45ec46a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e63540245da0a910d6e69fb4a0cd56c45ec46a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
