)]}'
{"id":"openstack%2Ftrove~657724","triplet_id":"openstack%2Ftrove~master~I5f2cabc35415cac3c4e81afa9084cbd8677ccd22","project":"openstack/trove","branch":"master","topic":"fix_tox_debug_err","hashtags":[],"change_id":"I5f2cabc35415cac3c4e81afa9084cbd8677ccd22","subject":"Fix tox debug mode","status":"MERGED","created":"2019-05-08 02:26:31.000000000","updated":"2019-05-14 06:53:34.000000000","submitted":"2019-05-14 06:53:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657724-1557816814681-49d5f157","meta_rev_id":"8c3245e0a09a8658bc32f24f537a7fb1d08ae7d3","_number":657724,"virtual_id_number":657724,"owner":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"value":0,"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},{"value":2,"date":"2019-05-14 06:53:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"value":0,"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-13 22:58:48.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"value":0,"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-13 22:58:48.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"value":0,"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"},{"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-08 06:26:02.000000000","updated_by":{"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},"reviewer":{"_account_id":28695,"name":"Bartosz Zurkowski","email":"b.zurkowski@samsung.com","username":"b.zurkowski"},"state":"REVIEWER"},{"updated":"2019-05-08 06:26:27.000000000","updated_by":{"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"},"reviewer":{"_account_id":28671,"name":"Marcin Piwowarczyk","email":"m.piwowarczy@samsung.com","username":"m.piwowarczy"},"state":"REVIEWER"},{"updated":"2019-05-13 22:58:48.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-05-14 06:53:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a7610779e4417eba034713d1bf71ab659442b56","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-05-08 02:26:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d23b36a85fb747406bbf8163f8aabc9804191b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 04:35:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/1/check/build-openstack-api-ref/aee8637/html/ : SUCCESS in 9m 22s\n- requirements-check http://logs.openstack.org/24/657724/1/check/requirements-check/c0616d6/ : FAILURE in 3m 49s\n- openstack-tox-cover http://logs.openstack.org/24/657724/1/check/openstack-tox-cover/c30aa4c/cover/ : SUCCESS in 24m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/1/check/openstack-tox-lower-constraints/c7d7a1f/ : SUCCESS in 19m 30s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/1/check/openstack-tox-pep8/598c82a/ : SUCCESS in 16m 09s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/1/check/openstack-tox-py27/37dda2e/ : SUCCESS in 18m 07s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/1/check/openstack-tox-py36/ad12de1/ : SUCCESS in 15m 25s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/1/check/openstack-tox-py37/278debb/ : SUCCESS in 18m 37s\n- openstack-tox-docs http://logs.openstack.org/24/657724/1/check/openstack-tox-docs/24b65c7/html/ : SUCCESS in 10m 10s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/1/check/build-openstack-releasenotes/f1d6cf1/html/ : SUCCESS in 11m 30s\n- trove-functional-mysql http://logs.openstack.org/24/657724/1/check/trove-functional-mysql/c365a2d/ : FAILURE in 42m 20s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/1/check/trove-scenario-mariadb-single/55fa30b/ : FAILURE in 56m 58s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/1/check/trove-scenario-mariadb-multi/286c660/ : FAILURE in 55m 12s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/1/check/trove-scenario-mysql-single/a8480f3/ : SUCCESS in 2h 05m 38s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/1/check/trove-scenario-mysql-multi/f9e524a/ : SUCCESS in 1h 53m 15s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/1/check/trove-scenario-postgresql-single/f23ebe8/ : FAILURE in 50m 23s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/1/check/trove-scenario-postgresql-multi/a3c3b08/ : FAILURE in 57m 59s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/1/check/trove-scenario-pxc-single/c0f6057/ : FAILURE in 30m 33s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/1/check/trove-scenario-pxc-multi/38395d1/ : FAILURE in 37m 50s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/1/check/trove-scenario-redis-single/57a4248/ : FAILURE in 1h 35m 11s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/1/check/trove-scenario-redis-multi/712beed/ : FAILURE in 1h 19m 20s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/1/check/trove-tox-fakemodetests/9f58b6a/ : SUCCESS in 11m 26s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/1/check/trove-tox-apiexamples/98c8239/ : SUCCESS in 10m 29s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/1/check/openstack-tox-pylint/7417570/ : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"79dc5a6336c26e507e0907a8efdaeb4702dcd7df","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-05-08 06:08:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"add79ac6cf5c11dfff82875f508868390403c2a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 08:43:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/2/check/build-openstack-api-ref/693f23f/html/ : SUCCESS in 10m 15s\n- requirements-check http://logs.openstack.org/24/657724/2/check/requirements-check/4c0030f/ : FAILURE in 3m 43s\n- openstack-tox-cover http://logs.openstack.org/24/657724/2/check/openstack-tox-cover/470fd78/cover/ : SUCCESS in 20m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/2/check/openstack-tox-lower-constraints/9ab7143/ : SUCCESS in 17m 55s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/2/check/openstack-tox-pep8/5c1161e/ : SUCCESS in 10m 52s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/2/check/openstack-tox-py27/33c4249/ : SUCCESS in 17m 52s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/2/check/openstack-tox-py36/0515325/ : SUCCESS in 19m 09s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/2/check/openstack-tox-py37/ebf6235/ : SUCCESS in 17m 50s\n- openstack-tox-docs http://logs.openstack.org/24/657724/2/check/openstack-tox-docs/97d0098/html/ : SUCCESS in 10m 37s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/2/check/build-openstack-releasenotes/db36363/html/ : SUCCESS in 11m 02s\n- trove-functional-mysql http://logs.openstack.org/24/657724/2/check/trove-functional-mysql/b066681/ : SUCCESS in 2h 31m 18s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/2/check/trove-scenario-mariadb-single/354c24e/ : FAILURE in 44m 27s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/2/check/trove-scenario-mariadb-multi/2711cf4/ : FAILURE in 1h 17m 28s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/2/check/trove-scenario-mysql-single/7bd4c88/ : SUCCESS in 2h 10m 11s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/2/check/trove-scenario-mysql-multi/6afc429/ : SUCCESS in 1h 51m 29s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/2/check/trove-scenario-postgresql-single/d4cdec4/ : FAILURE in 1h 04m 57s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/2/check/trove-scenario-postgresql-multi/22ee129/ : FAILURE in 57m 56s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/2/check/trove-scenario-pxc-single/3737132/ : FAILURE in 1h 12m 16s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/2/check/trove-scenario-pxc-multi/f924f36/ : FAILURE in 1h 17m 22s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/2/check/trove-scenario-redis-single/7d5315c/ : FAILURE in 1h 32m 21s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/2/check/trove-scenario-redis-multi/27d6943/ : FAILURE in 1h 27m 02s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/2/check/trove-tox-fakemodetests/d72bad2/ : SUCCESS in 10m 51s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/2/check/trove-tox-apiexamples/ee40ce4/ : SUCCESS in 10m 56s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/2/check/openstack-tox-pylint/0c02ded/ : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a439eeee03c6b7ef353a4faf09755bbe0621b904","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-05-10 03:08:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"15065e97ee3934b042f49d69178c77929772b15c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 04:34:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/3/check/build-openstack-api-ref/d4bd4b8/html/ : SUCCESS in 5m 21s\n- requirements-check http://logs.openstack.org/24/657724/3/check/requirements-check/64e1a0f/ : SUCCESS in 3m 38s\n- openstack-tox-cover http://logs.openstack.org/24/657724/3/check/openstack-tox-cover/7306b34/cover/ : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/3/check/openstack-tox-lower-constraints/90dcaa7/ : SUCCESS in 11m 26s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/3/check/openstack-tox-pep8/7ba69f3/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/3/check/openstack-tox-py27/6c00d97/ : SUCCESS in 11m 09s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/3/check/openstack-tox-py36/fb18803/ : SUCCESS in 12m 21s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/3/check/openstack-tox-py37/0dfcf6b/ : SUCCESS in 17m 56s\n- openstack-tox-docs http://logs.openstack.org/24/657724/3/check/openstack-tox-docs/125cda9/html/ : SUCCESS in 4m 45s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/3/check/build-openstack-releasenotes/aff45b8/html/ : SUCCESS in 7m 44s\n- trove-functional-mysql http://logs.openstack.org/24/657724/3/check/trove-functional-mysql/1080b63/ : FAILURE in 31m 11s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/3/check/trove-scenario-mariadb-single/7f6db5f/ : FAILURE in 1h 24m 02s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-mariadb-multi/5160d82/ : FAILURE in 1h 08m 21s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/3/check/trove-scenario-mysql-single/ea8540f/ : FAILURE in 36m 32s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-mysql-multi/3086aa6/ : FAILURE in 29m 59s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/3/check/trove-scenario-postgresql-single/6221dbf/ : FAILURE in 37m 11s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-postgresql-multi/153a84c/ : FAILURE in 43m 50s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/3/check/trove-scenario-pxc-single/4141bad/ : FAILURE in 44m 56s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-pxc-multi/59396b8/ : FAILURE in 36m 50s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/3/check/trove-scenario-redis-single/e66d4b6/ : FAILURE in 1h 09m 00s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-redis-multi/db61301/ : FAILURE in 57m 30s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/3/check/trove-tox-fakemodetests/d1338b7/ : SUCCESS in 5m 15s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/3/check/trove-tox-apiexamples/0de4ae2/ : SUCCESS in 7m 19s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/3/check/openstack-tox-pylint/78ce141/ : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":3},{"id":"ff4538d30d3aaa43bd2571663f55ac62ce7b51fc","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-05-10 06:02:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"240e3d71e5e512dc9a4fbc9156a4e04f90a45fb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 08:00:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/3/check/build-openstack-api-ref/2ad1acb/html/ : SUCCESS in 7m 04s\n- requirements-check http://logs.openstack.org/24/657724/3/check/requirements-check/6ca345a/ : SUCCESS in 3m 40s\n- openstack-tox-cover http://logs.openstack.org/24/657724/3/check/openstack-tox-cover/ffc37c6/cover/ : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/3/check/openstack-tox-lower-constraints/7fc25ce/ : SUCCESS in 13m 31s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/3/check/openstack-tox-pep8/4a43747/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/3/check/openstack-tox-py27/ec62f57/ : SUCCESS in 13m 24s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/3/check/openstack-tox-py36/54d21d7/ : SUCCESS in 14m 02s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/3/check/openstack-tox-py37/181f205/ : SUCCESS in 18m 12s\n- openstack-tox-docs http://logs.openstack.org/24/657724/3/check/openstack-tox-docs/4c7d303/html/ : SUCCESS in 5m 44s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/3/check/build-openstack-releasenotes/5c0dffe/html/ : SUCCESS in 5m 51s\n- trove-functional-mysql http://logs.openstack.org/24/657724/3/check/trove-functional-mysql/a5e6824/ : FAILURE in 1h 25m 14s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/3/check/trove-scenario-mariadb-single/d6afdd7/ : FAILURE in 1h 43m 58s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-mariadb-multi/e9a4d13/ : FAILURE in 58m 30s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/3/check/trove-scenario-mysql-single/a1a7f80/ : FAILURE in 35m 31s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-mysql-multi/602ee44/ : FAILURE in 38m 02s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/3/check/trove-scenario-postgresql-single/42635eb/ : FAILURE in 45m 05s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-postgresql-multi/23fb860/ : FAILURE in 41m 07s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/3/check/trove-scenario-pxc-single/afa5123/ : FAILURE in 41m 27s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-pxc-multi/8696afe/ : FAILURE in 46m 37s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/3/check/trove-scenario-redis-single/dbbd527/ : FAILURE in 1h 17m 19s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-redis-multi/ea5f14e/ : FAILURE in 1h 07m 43s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/3/check/trove-tox-fakemodetests/60b350f/ : SUCCESS in 5m 41s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/3/check/trove-tox-apiexamples/205c6c8/ : SUCCESS in 5m 02s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/3/check/openstack-tox-pylint/79b9ba9/ : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":3},{"id":"78de262aad33278cf9f5b8cf76c3a78e029b905e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-10 08:48:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a7ce8bb555a21413707925507441e122548bd3c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 10:01:27.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/3/check/build-openstack-api-ref/aaa4a24/html/ : SUCCESS in 4m 41s\n- requirements-check http://logs.openstack.org/24/657724/3/check/requirements-check/c2f065f/ : SUCCESS in 3m 00s\n- openstack-tox-cover http://logs.openstack.org/24/657724/3/check/openstack-tox-cover/5f5f1c0/cover/ : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/3/check/openstack-tox-lower-constraints/1c6f3f1/ : SUCCESS in 11m 53s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/3/check/openstack-tox-pep8/b68c8a2/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/3/check/openstack-tox-py27/2b45dc0/ : SUCCESS in 11m 00s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/3/check/openstack-tox-py36/da6e58f/ : SUCCESS in 12m 18s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/3/check/openstack-tox-py37/79047c0/ : SUCCESS in 18m 14s\n- openstack-tox-docs http://logs.openstack.org/24/657724/3/check/openstack-tox-docs/41d1fde/html/ : SUCCESS in 4m 45s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/3/check/build-openstack-releasenotes/a3e8279/html/ : SUCCESS in 6m 06s\n- trove-functional-mysql http://logs.openstack.org/24/657724/3/check/trove-functional-mysql/86c00aa/ : FAILURE in 31m 23s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/3/check/trove-scenario-mariadb-single/2a5efe7/ : FAILURE in 38m 24s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-mariadb-multi/633f7dc/ : FAILURE in 38m 48s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/3/check/trove-scenario-mysql-single/1b470d6/ : FAILURE in 32m 28s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-mysql-multi/ee1d2d2/ : FAILURE in 38m 06s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/3/check/trove-scenario-postgresql-single/e3a7131/ : FAILURE in 42m 06s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-postgresql-multi/37115dd/ : FAILURE in 39m 14s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/3/check/trove-scenario-pxc-single/73b441d/ : FAILURE in 39m 23s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-pxc-multi/afee63c/ : FAILURE in 39m 29s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/3/check/trove-scenario-redis-single/d68b08b/ : FAILURE in 1h 08m 05s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-redis-multi/464b5a0/ : FAILURE in 59m 41s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/3/check/trove-tox-fakemodetests/0948bee/ : SUCCESS in 5m 09s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/3/check/trove-tox-apiexamples/efc8bdb/ : SUCCESS in 5m 08s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/3/check/openstack-tox-pylint/b0a9dc0/ : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":3},{"id":"77959ff9fb46495c9b8ebbcd9e769171b861642f","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-05-11 11:59:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a1a55e43e7c8f25c8cc71c381d32e6c1555f1fbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 13:55:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/3/check/build-openstack-api-ref/4b1e8ef/html/ : SUCCESS in 7m 07s\n- requirements-check http://logs.openstack.org/24/657724/3/check/requirements-check/7dd9655/ : SUCCESS in 3m 54s\n- openstack-tox-cover http://logs.openstack.org/24/657724/3/check/openstack-tox-cover/8c847eb/cover/ : SUCCESS in 16m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/3/check/openstack-tox-lower-constraints/0ec9844/ : SUCCESS in 12m 09s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/3/check/openstack-tox-pep8/0a71f65/ : SUCCESS in 6m 27s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/3/check/openstack-tox-py27/8f27c8e/ : SUCCESS in 11m 47s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/3/check/openstack-tox-py36/22204ed/ : SUCCESS in 12m 47s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/3/check/openstack-tox-py37/96db9bd/ : SUCCESS in 16m 40s\n- openstack-tox-docs http://logs.openstack.org/24/657724/3/check/openstack-tox-docs/e957a99/html/ : SUCCESS in 7m 50s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/3/check/build-openstack-releasenotes/5453e00/html/ : SUCCESS in 5m 59s\n- trove-functional-mysql http://logs.openstack.org/24/657724/3/check/trove-functional-mysql/5503822/ : FAILURE in 43m 25s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/3/check/trove-scenario-mariadb-single/4c67507/ : FAILURE in 33m 56s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-mariadb-multi/345d01d/ : FAILURE in 1h 07m 38s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/3/check/trove-scenario-mysql-single/fa7382e/ : SUCCESS in 1h 43m 42s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-mysql-multi/5a1f541/ : SUCCESS in 1h 15m 53s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/3/check/trove-scenario-postgresql-single/af39d94/ : FAILURE in 29m 47s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-postgresql-multi/c002f6b/ : FAILURE in 39m 39s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/3/check/trove-scenario-pxc-single/51fbcb4/ : FAILURE in 39m 03s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-pxc-multi/e2655ed/ : FAILURE in 43m 44s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/3/check/trove-scenario-redis-single/3249968/ : FAILURE in 1h 10m 35s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-redis-multi/52b1295/ : FAILURE in 52m 38s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/3/check/trove-tox-fakemodetests/fecd8c4/ : SUCCESS in 6m 14s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/3/check/trove-tox-apiexamples/834b168/ : SUCCESS in 4m 53s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/3/check/openstack-tox-pylint/b301bf2/ : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":3},{"id":"73e6637088233eef11f56dfe4496d83761551f33","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-05-13 01:23:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fc01569aa64ced7527028b64e14be07ef678d29d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 03:17:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/3/check/build-openstack-api-ref/b0e830c/html/ : SUCCESS in 4m 46s\n- requirements-check http://logs.openstack.org/24/657724/3/check/requirements-check/7717698/ : FAILURE in 3m 11s\n- openstack-tox-cover http://logs.openstack.org/24/657724/3/check/openstack-tox-cover/ff7a87e/cover/ : SUCCESS in 18m 29s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/3/check/openstack-tox-lower-constraints/1f27b96/ : SUCCESS in 11m 04s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/3/check/openstack-tox-pep8/f0dbfc5/ : SUCCESS in 6m 16s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/3/check/openstack-tox-py27/2939882/ : SUCCESS in 10m 55s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/3/check/openstack-tox-py36/d822ec3/ : SUCCESS in 11m 34s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/3/check/openstack-tox-py37/6358c6c/ : SUCCESS in 16m 13s\n- openstack-tox-docs http://logs.openstack.org/24/657724/3/check/openstack-tox-docs/dcc60d9/html/ : SUCCESS in 5m 35s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/3/check/build-openstack-releasenotes/6dc139f/html/ : SUCCESS in 5m 49s\n- trove-functional-mysql http://logs.openstack.org/24/657724/3/check/trove-functional-mysql/bab4dd5/ : SUCCESS in 1h 41m 07s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/3/check/trove-scenario-mariadb-single/419c22b/ : FAILURE in 37m 08s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-mariadb-multi/a0e8d81/ : FAILURE in 39m 11s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/3/check/trove-scenario-mysql-single/6445e62/ : SUCCESS in 1h 52m 04s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-mysql-multi/fb8d249/ : SUCCESS in 1h 30m 42s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/3/check/trove-scenario-postgresql-single/5e4366c/ : FAILURE in 37m 44s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-postgresql-multi/4bd0fa8/ : FAILURE in 39m 11s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/3/check/trove-scenario-pxc-single/e9bd0f0/ : FAILURE in 39m 07s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-pxc-multi/f0933fb/ : FAILURE in 32m 26s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/3/check/trove-scenario-redis-single/e3074fc/ : FAILURE in 53m 03s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/3/check/trove-scenario-redis-multi/db03c4f/ : FAILURE in 57m 55s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/3/check/trove-tox-fakemodetests/61b2283/ : SUCCESS in 5m 14s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/3/check/trove-tox-apiexamples/caf4dd0/ : SUCCESS in 4m 32s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/3/check/openstack-tox-pylint/24f9765/ : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":3},{"id":"fecaa6d634fc13d6e5ffe3e25e16b075a018a9c6","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-05-13 06:28:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"df240b77e4332f39252f413e3be182962c5aba33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 08:25:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/4/check/build-openstack-api-ref/804faf6/html/ : SUCCESS in 5m 25s\n- requirements-check http://logs.openstack.org/24/657724/4/check/requirements-check/2a1e03f/ : FAILURE in 3m 33s\n- openstack-tox-cover http://logs.openstack.org/24/657724/4/check/openstack-tox-cover/ba86199/cover/ : SUCCESS in 19m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/4/check/openstack-tox-lower-constraints/990e1a5/ : SUCCESS in 12m 56s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/4/check/openstack-tox-pep8/63e5ef7/ : SUCCESS in 5m 21s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/4/check/openstack-tox-py27/b682c80/ : SUCCESS in 10m 53s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/4/check/openstack-tox-py36/306aafc/ : SUCCESS in 12m 36s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/4/check/openstack-tox-py37/ce068a1/ : SUCCESS in 17m 44s\n- openstack-tox-docs http://logs.openstack.org/24/657724/4/check/openstack-tox-docs/8889db6/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/4/check/build-openstack-releasenotes/f12dade/html/ : SUCCESS in 7m 01s\n- trove-functional-mysql http://logs.openstack.org/24/657724/4/check/trove-functional-mysql/f0b88ac/ : SUCCESS in 1h 51m 45s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/4/check/trove-scenario-mariadb-single/a33d21e/ : FAILURE in 36m 05s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/4/check/trove-scenario-mariadb-multi/4aa28e7/ : FAILURE in 37m 51s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/4/check/trove-scenario-mysql-single/2b138b1/ : SUCCESS in 1h 49m 01s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/4/check/trove-scenario-mysql-multi/5e0f22f/ : FAILURE in 36m 46s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/4/check/trove-scenario-postgresql-single/e721b31/ : FAILURE in 41m 19s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/4/check/trove-scenario-postgresql-multi/2207901/ : FAILURE in 40m 21s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/4/check/trove-scenario-pxc-single/634f8c3/ : FAILURE in 42m 10s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/4/check/trove-scenario-pxc-multi/a72d2e9/ : FAILURE in 45m 50s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/4/check/trove-scenario-redis-single/3fdd2eb/ : FAILURE in 1h 05m 20s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/4/check/trove-scenario-redis-multi/7a7f579/ : FAILURE in 58m 27s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/4/check/trove-tox-fakemodetests/c9fe0cb/ : SUCCESS in 5m 25s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/4/check/trove-tox-apiexamples/0b828fe/ : SUCCESS in 5m 19s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/4/check/openstack-tox-pylint/6b7b806/ : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":4},{"id":"0a39ecbef784098e5aa6f2cd6ffb26153cda67f0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-13 09:53:31.000000000","message":"Patch Set 4: Code-Review-1\n\nyou should fix the CI failure instead of keeping recheck","accounts_in_message":[],"_revision_number":4},{"id":"ebf6d0487e4fb9d81d2b2ad3c571df8eed7e8730","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-05-13 10:28:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a9dd5f6b1dc8da904a0c38f70169f21056ce4658","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 12:31:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/5/check/build-openstack-api-ref/886fa81/html/ : SUCCESS in 5m 50s\n- requirements-check http://logs.openstack.org/24/657724/5/check/requirements-check/0f053b6/ : SUCCESS in 4m 05s\n- openstack-tox-cover http://logs.openstack.org/24/657724/5/check/openstack-tox-cover/da53bad/cover/ : SUCCESS in 16m 56s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/5/check/openstack-tox-lower-constraints/1f1b76a/ : SUCCESS in 11m 31s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/5/check/openstack-tox-pep8/ee9c659/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/5/check/openstack-tox-py27/c80c366/ : SUCCESS in 10m 54s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/5/check/openstack-tox-py36/4230014/ : SUCCESS in 12m 28s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/5/check/openstack-tox-py37/0d79a84/ : SUCCESS in 31m 17s\n- openstack-tox-docs http://logs.openstack.org/24/657724/5/check/openstack-tox-docs/1c6aade/html/ : SUCCESS in 4m 43s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/5/check/build-openstack-releasenotes/6963340/html/ : SUCCESS in 6m 03s\n- trove-functional-mysql http://logs.openstack.org/24/657724/5/check/trove-functional-mysql/e1f2698/ : SUCCESS in 2h 01m 58s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/5/check/trove-scenario-mariadb-single/7d2f77b/ : FAILURE in 42m 14s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-mariadb-multi/87ce8f6/ : FAILURE in 43m 30s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/5/check/trove-scenario-mysql-single/8985f59/ : SUCCESS in 1h 54m 32s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-mysql-multi/9d02088/ : FAILURE in 27m 54s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/5/check/trove-scenario-postgresql-single/d575ece/ : FAILURE in 43m 19s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-postgresql-multi/917ddc2/ : FAILURE in 38m 25s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/5/check/trove-scenario-pxc-single/cbff0b8/ : FAILURE in 43m 14s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-pxc-multi/3e04558/ : FAILURE in 40m 55s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/5/check/trove-scenario-redis-single/94199e0/ : FAILURE in 1h 12m 19s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-redis-multi/40ca240/ : FAILURE in 1h 03m 58s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/5/check/trove-tox-fakemodetests/3466d8d/ : SUCCESS in 5m 26s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/5/check/trove-tox-apiexamples/130b561/ : SUCCESS in 16m 29s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/5/check/openstack-tox-pylint/7035547/ : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":5},{"id":"094eaf607afb876c6b21b04bbe47349fed0692e5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-13 12:46:07.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"68297ee280c14d4fd8fb5ba26f527469fc9e62cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 14:58:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/5/check/build-openstack-api-ref/4b3d77e/html/ : SUCCESS in 8m 38s\n- requirements-check http://logs.openstack.org/24/657724/5/check/requirements-check/f3088d7/ : SUCCESS in 3m 55s\n- openstack-tox-cover http://logs.openstack.org/24/657724/5/check/openstack-tox-cover/04e425f/cover/ : SUCCESS in 16m 22s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/5/check/openstack-tox-lower-constraints/7237f42/ : SUCCESS in 11m 19s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/5/check/openstack-tox-pep8/9b45baf/ : SUCCESS in 7m 43s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/5/check/openstack-tox-py27/ac4b026/ : SUCCESS in 16m 05s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/5/check/openstack-tox-py36/4b74384/ : SUCCESS in 18m 59s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/5/check/openstack-tox-py37/01959c7/ : SUCCESS in 23m 16s\n- openstack-tox-docs http://logs.openstack.org/24/657724/5/check/openstack-tox-docs/5413c8e/html/ : SUCCESS in 7m 22s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/5/check/build-openstack-releasenotes/76c0552/html/ : SUCCESS in 9m 37s\n- trove-functional-mysql http://logs.openstack.org/24/657724/5/check/trove-functional-mysql/066e109/ : SUCCESS in 2h 10m 47s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/5/check/trove-scenario-mariadb-single/579980e/ : FAILURE in 44m 15s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-mariadb-multi/577af76/ : FAILURE in 47m 35s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/5/check/trove-scenario-mysql-single/09da92e/ : SUCCESS in 2h 04m 25s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-mysql-multi/e16dd17/ : SUCCESS in 2h 08m 00s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/5/check/trove-scenario-postgresql-single/d08f3de/ : FAILURE in 49m 12s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-postgresql-multi/cf31d98/ : FAILURE in 1h 00m 13s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/5/check/trove-scenario-pxc-single/df03ecf/ : FAILURE in 59m 42s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-pxc-multi/1ec88cf/ : FAILURE in 54m 13s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/5/check/trove-scenario-redis-single/65d2e68/ : FAILURE in 1h 29m 51s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-redis-multi/228ba1b/ : FAILURE in 1h 17m 24s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/5/check/trove-tox-fakemodetests/8247195/ : SUCCESS in 5m 45s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/5/check/trove-tox-apiexamples/5071698/ : SUCCESS in 4m 55s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/5/check/openstack-tox-pylint/6b71ef0/ : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":5},{"id":"26abeeb10036614cf4f829490d88999dd895a00f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-13 22:58:48.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nthanks!","accounts_in_message":[],"_revision_number":5},{"id":"ce737ce6d57661e9a34b483b42ac18f3561674b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 22:59:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cbbc8064e6621dbb03e38ceb0be92042097f8945","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:48:30.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/5/gate/build-openstack-api-ref/519741d/html/ : SUCCESS in 5m 13s\n- requirements-check http://logs.openstack.org/24/657724/5/gate/requirements-check/b68c7c8/ : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/5/gate/openstack-tox-lower-constraints/50aa828/ : SUCCESS in 12m 38s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/5/gate/openstack-tox-pep8/b724e3b/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/5/gate/openstack-tox-py27/2243b6a/ : SUCCESS in 10m 22s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/5/gate/openstack-tox-py36/572d655/ : SUCCESS in 11m 16s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/5/gate/openstack-tox-py37/83ea7c4/ : SUCCESS in 15m 28s\n- openstack-tox-docs http://logs.openstack.org/24/657724/5/gate/openstack-tox-docs/18600c8/html/ : SUCCESS in 5m 45s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/5/gate/build-openstack-releasenotes/f77f0f5/html/ : SUCCESS in 6m 25s\n- trove-functional-mysql http://logs.openstack.org/24/657724/5/gate/trove-functional-mysql/a11806b/ : FAILURE in 39m 34s\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/5/gate/trove-scenario-mysql-single/a47691f/ : FAILURE in 29m 19s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/5/gate/trove-scenario-mysql-multi/56ab4d8/ : SUCCESS in 1h 47m 28s\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/5/gate/trove-tox-fakemodetests/f972f20/ : SUCCESS in 5m 42s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/5/gate/trove-tox-apiexamples/9826d16/ : SUCCESS in 4m 29s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/5/gate/openstack-tox-pylint/d860bad/ : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":5},{"id":"21959c2c50aad564e6b621366e89e2ce449504c8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-14 02:43:42.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"51326a6ae1b904e72d365d614a176c2a8e8de465","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-14 02:44:03.000000000","message":"Patch Set 5:\n\nI don\u0027t know why Jenkins doesn\u0027t like this patch.","accounts_in_message":[],"_revision_number":5},{"id":"1c6dc335564a439a65ca6c6ad3c285f4260c5007","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 04:39:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/5/check/build-openstack-api-ref/c1929b9/html/ : SUCCESS in 4m 11s\n- requirements-check http://logs.openstack.org/24/657724/5/check/requirements-check/81a5453/ : SUCCESS in 3m 21s\n- openstack-tox-cover http://logs.openstack.org/24/657724/5/check/openstack-tox-cover/a261e0a/cover/ : SUCCESS in 16m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/5/check/openstack-tox-lower-constraints/d63b103/ : SUCCESS in 10m 59s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/5/check/openstack-tox-pep8/1787ab8/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/5/check/openstack-tox-py27/f62680a/ : SUCCESS in 10m 11s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/5/check/openstack-tox-py36/4d59a8d/ : SUCCESS in 11m 06s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/5/check/openstack-tox-py37/a90832a/ : SUCCESS in 18m 07s\n- openstack-tox-docs http://logs.openstack.org/24/657724/5/check/openstack-tox-docs/bdfd0c7/html/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/5/check/build-openstack-releasenotes/4900020/html/ : SUCCESS in 5m 53s\n- trove-functional-mysql http://logs.openstack.org/24/657724/5/check/trove-functional-mysql/4dc3bfe/ : SUCCESS in 1h 54m 41s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/657724/5/check/trove-scenario-mariadb-single/230b54d/ : FAILURE in 39m 51s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-mariadb-multi/d6370cf/ : FAILURE in 43m 02s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/5/check/trove-scenario-mysql-single/fe702dc/ : SUCCESS in 1h 39m 41s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-mysql-multi/8dc16d6/ : SUCCESS in 1h 39m 37s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/657724/5/check/trove-scenario-postgresql-single/50b46b0/ : FAILURE in 40m 57s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-postgresql-multi/fbddfa3/ : FAILURE in 34m 46s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/657724/5/check/trove-scenario-pxc-single/4683ef7/ : FAILURE in 39m 49s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-pxc-multi/a7f241f/ : FAILURE in 40m 00s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/657724/5/check/trove-scenario-redis-single/5c61ea8/ : FAILURE in 56m 06s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/657724/5/check/trove-scenario-redis-multi/15385ff/ : FAILURE in 1h 04m 39s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/5/check/trove-tox-fakemodetests/1fbcac1/ : SUCCESS in 5m 57s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/5/check/trove-tox-apiexamples/ec63c74/ : SUCCESS in 5m 02s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/5/check/openstack-tox-pylint/adf0a32/ : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":5},{"id":"f4ea79ddc72f9e950adef1e85de4f7d277d35875","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 04:39:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d80443bf924b70bed7074736c5680549265095d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 06:53:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"7ec83b0dae20ec8efb13458b2df07b73c88c3e54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 06:53:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/24/657724/5/gate/build-openstack-api-ref/b84430f/html/ : SUCCESS in 4m 54s\n- requirements-check http://logs.openstack.org/24/657724/5/gate/requirements-check/5500af5/ : SUCCESS in 2m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/657724/5/gate/openstack-tox-lower-constraints/ce7fa49/ : SUCCESS in 12m 07s\n- openstack-tox-pep8 http://logs.openstack.org/24/657724/5/gate/openstack-tox-pep8/44ce4a9/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/24/657724/5/gate/openstack-tox-py27/31d02ea/ : SUCCESS in 11m 10s\n- openstack-tox-py36 http://logs.openstack.org/24/657724/5/gate/openstack-tox-py36/a5661d4/ : SUCCESS in 11m 44s\n- openstack-tox-py37 http://logs.openstack.org/24/657724/5/gate/openstack-tox-py37/ca71141/ : SUCCESS in 17m 58s\n- openstack-tox-docs http://logs.openstack.org/24/657724/5/gate/openstack-tox-docs/05af446/html/ : SUCCESS in 4m 30s\n- build-openstack-releasenotes http://logs.openstack.org/24/657724/5/gate/build-openstack-releasenotes/f84cfc0/html/ : SUCCESS in 5m 39s\n- trove-functional-mysql http://logs.openstack.org/24/657724/5/gate/trove-functional-mysql/69b8241/ : SUCCESS in 2h 08m 48s\n- trove-scenario-mysql-single http://logs.openstack.org/24/657724/5/gate/trove-scenario-mysql-single/8560ffa/ : SUCCESS in 1h 48m 47s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/657724/5/gate/trove-scenario-mysql-multi/96ea216/ : SUCCESS in 1h 37m 59s\n- trove-tox-fakemodetests http://logs.openstack.org/24/657724/5/gate/trove-tox-fakemodetests/540ac89/ : SUCCESS in 5m 22s\n- trove-tox-apiexamples http://logs.openstack.org/24/657724/5/gate/trove-tox-apiexamples/b51fccc/ : SUCCESS in 4m 26s\n- openstack-tox-pylint http://logs.openstack.org/24/657724/5/gate/openstack-tox-pylint/4ed7d16/ : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"85d2b17c49cd5f7e878b32f8e58217923c6c2aa2","revisions":{"d1296156f0cc00481f71f1709f9988db01c9348a":{"kind":"REWORK","_number":1,"created":"2019-05-08 02:26:31.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/657724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/657724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/657724/1"}}},"commit":{"parents":[{"commit":"4e63540245da0a910d6e69fb4a0cd56c45ec46a4","subject":"Dropping the py35 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e63540245da0a910d6e69fb4a0cd56c45ec46a4"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-05-08 02:16:24.000000000","tz":0},"committer":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-05-08 02:16:24.000000000","tz":0},"subject":"Fix tox debug mode","message":"Fix tox debug mode\n\nadd oslotest in test-requirement.txt for\nfixing tox debug mode failed.\n\nChange-Id: I5f2cabc35415cac3c4e81afa9084cbd8677ccd22\nStory: #2005615\nTask: #30862\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d1296156f0cc00481f71f1709f9988db01c9348a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d1296156f0cc00481f71f1709f9988db01c9348a"}]},"branch":"refs/heads/master"},"eecb72ca76286002bfa5795b27bab4935590aa78":{"kind":"REWORK","_number":2,"created":"2019-05-08 06:08:38.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/657724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/657724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/657724/2"}}},"commit":{"parents":[{"commit":"4e63540245da0a910d6e69fb4a0cd56c45ec46a4","subject":"Dropping the py35 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e63540245da0a910d6e69fb4a0cd56c45ec46a4"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-05-08 02:16:24.000000000","tz":0},"committer":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-05-08 06:07:40.000000000","tz":0},"subject":"Fix tox debug mode","message":"Fix tox debug mode\n\nadd oslotest in test-requirement.txt for\nfixing tox debug mode failed.\n\nChange-Id: I5f2cabc35415cac3c4e81afa9084cbd8677ccd22\nStory: #2005615\nTask: #30862\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eecb72ca76286002bfa5795b27bab4935590aa78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eecb72ca76286002bfa5795b27bab4935590aa78"}]},"branch":"refs/heads/master"},"ba74eb5740d2d1ccc41eedb116053387dd80be5d":{"kind":"REWORK","_number":3,"created":"2019-05-10 03:08:14.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/657724/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/657724/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/657724/3"}}},"commit":{"parents":[{"commit":"4e63540245da0a910d6e69fb4a0cd56c45ec46a4","subject":"Dropping the py35 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e63540245da0a910d6e69fb4a0cd56c45ec46a4"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-05-08 02:16:24.000000000","tz":0},"committer":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-05-10 02:52:40.000000000","tz":0},"subject":"Fix tox debug mode","message":"Fix tox debug mode\n\n- add oslotest in test-requirement.txt for\nfixing tox debug mode failed.\n- Update jsonschema following the global-requirments.txt\nto succeed zuul requirements-check job.\n(See: https://review.openstack.org/649789)\n\nChange-Id: I5f2cabc35415cac3c4e81afa9084cbd8677ccd22\nStory: #2005615\nTask: #30862\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba74eb5740d2d1ccc41eedb116053387dd80be5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba74eb5740d2d1ccc41eedb116053387dd80be5d"}]},"branch":"refs/heads/master"},"d7d8e4ae6cda920a90b1017f1eb31e51f5308eca":{"kind":"REWORK","_number":4,"created":"2019-05-13 06:28:55.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/657724/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/657724/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/657724/4"}}},"commit":{"parents":[{"commit":"e1a027067cab5476385cec2327efa24b75c68d58","subject":"Merge \"Pass kwargs to exception to get better format of error message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1a027067cab5476385cec2327efa24b75c68d58"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-05-08 02:16:24.000000000","tz":0},"committer":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-05-13 06:23:58.000000000","tz":0},"subject":"Fix tox debug mode","message":"Fix tox debug mode\n\n- add oslotest in test-requirement.txt for\nfixing tox debug mode failed.\n- Update jsonschema/sphinx following the global-requirments.txt\nto pass zuul requirements-check job.\n(See: https://review.openstack.org/649789\nhttps://review.opendev.org/#/c/657890/)\n\nChange-Id: I5f2cabc35415cac3c4e81afa9084cbd8677ccd22\nStory: #2005615\nTask: #30862\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d7d8e4ae6cda920a90b1017f1eb31e51f5308eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d7d8e4ae6cda920a90b1017f1eb31e51f5308eca"}]},"branch":"refs/heads/master"},"85d2b17c49cd5f7e878b32f8e58217923c6c2aa2":{"kind":"REWORK","_number":5,"created":"2019-05-13 10:28:17.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/657724/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/657724/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/657724/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/657724/5"}}},"commit":{"parents":[{"commit":"e1a027067cab5476385cec2327efa24b75c68d58","subject":"Merge \"Pass kwargs to exception to get better format of error message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1a027067cab5476385cec2327efa24b75c68d58"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-05-08 02:16:24.000000000","tz":0},"committer":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-05-13 10:27:00.000000000","tz":0},"subject":"Fix tox debug mode","message":"Fix tox debug mode\n\n- add oslotest in test-requirement.txt for\nfixing tox debug mode failed.\n- Update jsonschema/sphinx following the global-requirments.txt\nto pass zuul requirements-check job.\n(See: https://review.openstack.org/649789\nhttps://review.opendev.org/#/c/657890/)\n\nChange-Id: I5f2cabc35415cac3c4e81afa9084cbd8677ccd22\nStory: #2005615\nTask: #30862\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/85d2b17c49cd5f7e878b32f8e58217923c6c2aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/85d2b17c49cd5f7e878b32f8e58217923c6c2aa2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
