)]}'
{"id":"openstack%2Ftrove~661656","triplet_id":"openstack%2Ftrove~master~Ie760abeb4a3ea8666da25480a8461812a67514eb","project":"openstack/trove","branch":"master","topic":"fix-doc","hashtags":[],"change_id":"Ie760abeb4a3ea8666da25480a8461812a67514eb","subject":"docs: fix build failure on html_last_updated_fmt","status":"MERGED","created":"2019-05-27 21:14:22.000000000","updated":"2019-06-01 10:17:09.000000000","submitted":"2019-06-01 10:17:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661656-1559384229027-4f1731f4","meta_rev_id":"1a56c678993cda6e78fa851a463894527696efc8","_number":661656,"virtual_id_number":661656,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-06-01 10:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-06-01 10:07:03.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-06-01 10:07:03.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-01 10:17:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ca72e47c620ae7d7cfa1cbb1f139023b16e24c6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-27 21:14:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b6b426ab4a75450816f1c33ccf016af3657ab7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 21:21:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/56/661656/1/check/openstack-tox-pep8/350f72d/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/56/661656/1/check/openstack-tox-docs/374f11f/html/ : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f0e3d5a9aa5a24619241e14894be881fbf695a81","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-05-27 21:43:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"23edf2e88f665ea8daa8ce9c3aece78324249248","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 21:53:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/56/661656/2/check/build-openstack-api-ref/d6a6f36/html/ : SUCCESS in 6m 44s\n- openstack-tox-pep8 http://logs.openstack.org/56/661656/2/check/openstack-tox-pep8/00c1da9/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/56/661656/2/check/openstack-tox-docs/e65e0a1/html/ : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"0eb4da4caea5c30753b46f4624a40df6ec096b06","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-01 10:07:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"266cdf5067b3e2fc9654bcbb7557d29d74e4ee5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 10:07:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"92e85db828ce89e38821af07541624c17ccb1159","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 10:17:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/56/661656/2/gate/build-openstack-api-ref/967883f/html/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/56/661656/2/gate/openstack-tox-pep8/db400b7/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/56/661656/2/gate/openstack-tox-docs/c438635/html/ : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"37db8971a2f4d4548bedcefba7e307d5271d0d52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 10:17:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c37ab5f34b6bd1939a4121be684a56ae755dcdd4","revisions":{"0fa889a4a85b2a79b214e66a2252b8854b7acec8":{"kind":"REWORK","_number":1,"created":"2019-05-27 21:14:22.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/56/661656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/661656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/661656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/661656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/661656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/661656/1"}}},"commit":{"parents":[{"commit":"a038b356fd657992ff3f8e681dfc0b6e53741d83","subject":"Merge \"Add Redis datastore upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a038b356fd657992ff3f8e681dfc0b6e53741d83"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-27 21:14:10.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-27 21:14:10.000000000","tz":720},"subject":"docs: fix build failure on html_last_updated_fmt","message":"docs: fix build failure on html_last_updated_fmt\n\nChange-Id: Ie760abeb4a3ea8666da25480a8461812a67514eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0fa889a4a85b2a79b214e66a2252b8854b7acec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0fa889a4a85b2a79b214e66a2252b8854b7acec8"}]},"branch":"refs/heads/master"},"c37ab5f34b6bd1939a4121be684a56ae755dcdd4":{"kind":"REWORK","_number":2,"created":"2019-05-27 21:43:23.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/56/661656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/661656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/661656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/661656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/661656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/661656/2"}}},"commit":{"parents":[{"commit":"a038b356fd657992ff3f8e681dfc0b6e53741d83","subject":"Merge \"Add Redis datastore upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a038b356fd657992ff3f8e681dfc0b6e53741d83"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-27 21:14:10.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-05-27 21:42:08.000000000","tz":720},"subject":"docs: fix build failure on html_last_updated_fmt","message":"docs: fix build failure on html_last_updated_fmt\n\nError message in CI job build-openstack-api-ref:\nThe config value `html_last_updated_fmt\u0027 has type `bytes\u0027; expected\n`str\u0027.\n\nChange-Id: Ie760abeb4a3ea8666da25480a8461812a67514eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c37ab5f34b6bd1939a4121be684a56ae755dcdd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c37ab5f34b6bd1939a4121be684a56ae755dcdd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
