)]}'
{"id":"openstack%2Ftrove~662924","triplet_id":"openstack%2Ftrove~master~I403b2cbd49e86393f1307c2e27f218ee4d3af2b7","project":"openstack/trove","branch":"master","topic":"bp/volume-data-snapshot-backup-type","hashtags":[],"change_id":"I403b2cbd49e86393f1307c2e27f218ee4d3af2b7","subject":"Add volume-snapshot backup type","status":"ABANDONED","created":"2019-06-04 03:19:41.000000000","updated":"2024-08-19 03:34:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4036b1e83749b796e76c1cd459a39c62c8a37425","_number":662924,"virtual_id_number":662924,"owner":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-20 06:57:11.000000000","_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":-1,"date":"2019-09-17 09:41:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"value":-1,"date":"2019-09-30 00:22:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 09:41:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-30 00:22:59.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"3d55360f773583021f24410b3e3b3d22d366ab30","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-06-04 03:19:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29966f6f278cf26b2145380b06b089636e05904d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 05:21:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/24/662924/1/check/openstack-tox-cover/4e770c6/ : FAILURE in 15m 06s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/662924/1/check/openstack-tox-lower-constraints/95140c1/ : SUCCESS in 11m 56s\n- openstack-tox-pep8 http://logs.openstack.org/24/662924/1/check/openstack-tox-pep8/fc3eda3/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/24/662924/1/check/openstack-tox-py27/f9a2d7b/ : SUCCESS in 11m 49s\n- openstack-tox-py36 http://logs.openstack.org/24/662924/1/check/openstack-tox-py36/145fa4c/ : SUCCESS in 12m 47s\n- openstack-tox-py37 http://logs.openstack.org/24/662924/1/check/openstack-tox-py37/ffa3a61/ : SUCCESS in 21m 05s\n- openstack-tox-docs http://logs.openstack.org/24/662924/1/check/openstack-tox-docs/b0b6076/html/ : SUCCESS in 6m 02s\n- trove-functional-mysql http://logs.openstack.org/24/662924/1/check/trove-functional-mysql/164e0e6/ : SUCCESS in 1h 49m 51s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/662924/1/check/trove-scenario-mariadb-single/7471c1d/ : FAILURE in 37m 35s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/662924/1/check/trove-scenario-mariadb-multi/4508899/ : FAILURE in 38m 33s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/662924/1/check/trove-scenario-mysql-single/2e59265/ : SUCCESS in 1h 53m 05s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/662924/1/check/trove-scenario-mysql-multi/e80417f/ : SUCCESS in 1h 41m 20s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/662924/1/check/trove-scenario-postgresql-single/81505fd/ : FAILURE in 48m 26s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/662924/1/check/trove-scenario-postgresql-multi/21e7ce9/ : FAILURE in 40m 21s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/662924/1/check/trove-scenario-pxc-single/19ed532/ : FAILURE in 38m 39s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/662924/1/check/trove-scenario-pxc-multi/5ae3814/ : FAILURE in 40m 35s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/662924/1/check/trove-scenario-redis-single/a6ac5b2/ : FAILURE in 1h 11m 41s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/662924/1/check/trove-scenario-redis-multi/d844f59/ : FAILURE in 1h 07m 33s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/662924/1/check/trove-tox-fakemodetests/c75445d/ : FAILURE in 5m 21s\n- trove-tox-apiexamples http://logs.openstack.org/24/662924/1/check/trove-tox-apiexamples/58c535c/ : FAILURE in 5m 00s\n- trove-tox-bandit-baseline http://logs.openstack.org/24/662924/1/check/trove-tox-bandit-baseline/3011bea/ : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/24/662924/1/check/openstack-tox-pylint/f4d9824/ : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fa6db6e5f295da81f2481b75729ca4538535701b","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-06-04 10:10:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a08c899f07313f7cc907af466fd4bbba0fa73085","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 12:51:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/24/662924/2/check/openstack-tox-cover/c96e486/cover/ : SUCCESS in 19m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/662924/2/check/openstack-tox-lower-constraints/d0ed1f3/ : SUCCESS in 13m 44s\n- openstack-tox-pep8 http://logs.openstack.org/24/662924/2/check/openstack-tox-pep8/94cdf22/ : SUCCESS in 7m 09s\n- openstack-tox-py27 http://logs.openstack.org/24/662924/2/check/openstack-tox-py27/24dda3b/ : SUCCESS in 11m 26s\n- openstack-tox-py36 http://logs.openstack.org/24/662924/2/check/openstack-tox-py36/f1be2f0/ : SUCCESS in 12m 53s\n- openstack-tox-py37 http://logs.openstack.org/24/662924/2/check/openstack-tox-py37/cf1394d/ : SUCCESS in 19m 30s\n- openstack-tox-docs http://logs.openstack.org/24/662924/2/check/openstack-tox-docs/576dd2e/html/ : SUCCESS in 4m 46s\n- trove-functional-mysql http://logs.openstack.org/24/662924/2/check/trove-functional-mysql/77d4b0d/ : SUCCESS in 2h 02m 20s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/662924/2/check/trove-scenario-mariadb-single/655a607/ : FAILURE in 1h 20m 33s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/662924/2/check/trove-scenario-mariadb-multi/d613dec/ : FAILURE in 1h 09m 47s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/662924/2/check/trove-scenario-mysql-single/73cf22b/ : SUCCESS in 2h 34m 25s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/662924/2/check/trove-scenario-mysql-multi/85328eb/ : SUCCESS in 2h 09m 10s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/662924/2/check/trove-scenario-postgresql-single/e48740e/ : FAILURE in 1h 08m 01s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/662924/2/check/trove-scenario-postgresql-multi/4f9c0a3/ : FAILURE in 1h 06m 20s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/662924/2/check/trove-scenario-pxc-single/3303d7d/ : FAILURE in 1h 16m 10s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/662924/2/check/trove-scenario-pxc-multi/9809d24/ : FAILURE in 1h 15m 54s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/662924/2/check/trove-scenario-redis-single/b772704/ : FAILURE in 1h 34m 04s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/662924/2/check/trove-scenario-redis-multi/9929bbf/ : FAILURE in 1h 24m 41s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/662924/2/check/trove-tox-fakemodetests/42def75/ : SUCCESS in 6m 16s\n- trove-tox-apiexamples http://logs.openstack.org/24/662924/2/check/trove-tox-apiexamples/89922b6/ : SUCCESS in 4m 51s\n- trove-tox-bandit-baseline http://logs.openstack.org/24/662924/2/check/trove-tox-bandit-baseline/d915a76/ : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/24/662924/2/check/openstack-tox-pylint/9488aaa/ : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":2},{"id":"aabf3a4f785fc8149b232d7e3e15717777856338","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-05 04:14:02.000000000","message":"Patch Set 2:\n\nHi renminmin, thanks for this patch, currently my bandwidth has been fully allocated to https://review.opendev.org/652068, i am going to work on that full time these days. Hopefully it could be finished within this week.\n\nConsidering the tramendous change this patch brings, I will definitely take a look after the service tenant model task is finished, because that\u0027s the highest priority in T dev cycle. We have plenty of work to do after that.\n\nThanks for your patience and sorry for the delay.","accounts_in_message":[],"_revision_number":2},{"id":"e75f04ec37ba8fc338fcacd263f9270bfee3ba4a","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-07-04 08:56:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1e664472284ed98adc5b87c20b0025bc0d4ed5a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 09:45:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/24/662924/3/check/openstack-tox-cover/25638b6/ : RETRY_LIMIT in 2m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/662924/3/check/openstack-tox-lower-constraints/5776076/ : RETRY_LIMIT in 2m 09s\n- openstack-tox-pep8 http://logs.openstack.org/24/662924/3/check/openstack-tox-pep8/87cd0ed/ : RETRY_LIMIT in 1m 56s\n- openstack-tox-py27 http://logs.openstack.org/24/662924/3/check/openstack-tox-py27/29de333/ : RETRY_LIMIT in 2m 05s\n- openstack-tox-py36 http://logs.openstack.org/24/662924/3/check/openstack-tox-py36/5027860/ : RETRY_LIMIT in 1m 49s\n- openstack-tox-py37 http://logs.openstack.org/24/662924/3/check/openstack-tox-py37/75755b2/ : RETRY_LIMIT in 2m 13s\n- openstack-tox-docs http://logs.openstack.org/24/662924/3/check/openstack-tox-docs/3909efa/html/ : SUCCESS in 3m 34s\n- trove-functional-mysql http://logs.openstack.org/24/662924/3/check/trove-functional-mysql/caf6328/ : FAILURE in 44m 38s\n- trove-scenario-mariadb-single http://logs.openstack.org/24/662924/3/check/trove-scenario-mariadb-single/f719659/ : FAILURE in 41m 41s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/24/662924/3/check/trove-scenario-mariadb-multi/7ba2449/ : FAILURE in 30m 32s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/24/662924/3/check/trove-scenario-mysql-single/024900c/ : FAILURE in 41m 23s\n- trove-scenario-mysql-multi http://logs.openstack.org/24/662924/3/check/trove-scenario-mysql-multi/dca4fab/ : FAILURE in 43m 32s\n- trove-scenario-postgresql-single http://logs.openstack.org/24/662924/3/check/trove-scenario-postgresql-single/7d15d17/ : FAILURE in 35m 31s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/24/662924/3/check/trove-scenario-postgresql-multi/a1206a3/ : FAILURE in 34m 51s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/24/662924/3/check/trove-scenario-pxc-single/21e375d/ : FAILURE in 33m 33s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/24/662924/3/check/trove-scenario-pxc-multi/81d31b1/ : FAILURE in 34m 03s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/24/662924/3/check/trove-scenario-redis-single/070d74a/ : FAILURE in 36m 34s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/24/662924/3/check/trove-scenario-redis-multi/5afc675/ : FAILURE in 33m 38s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/24/662924/3/check/trove-tox-fakemodetests/f6c43fd/ : RETRY_LIMIT in 1m 47s\n- trove-tox-bandit-baseline http://logs.openstack.org/24/662924/3/check/trove-tox-bandit-baseline/658dacc/ : RETRY_LIMIT in 2m 45s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/24/662924/3/check/openstack-tox-pylint/3cb0972/ : RETRY_LIMIT in 2m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e3225fa92ca4c157727f662e7bf97bcd96ebf016","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-14 22:10:38.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b6adc88116568aab3f5c6728bbbe18a9b74b55d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 00:37:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2efbdf9399148b586b4d80da4282cd9 : SUCCESS in 18m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2891b225e48c4c61a0f8c81d6edc5df5 : SUCCESS in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713a8d7e334940d983ad4d7b6d755dd7 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bd61b17c946d4100a87463b5ca7a5556 : SUCCESS in 11m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c61c81e2a5b4c93b41426d28de2d8d6 : SUCCESS in 12m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b15aca1b3318464fa27dd155d9823357 : SUCCESS in 20m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd1d16154254bb5b7c18ba342730e73 : SUCCESS in 6m 08s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/52440fe27f784a228c74302ed664b8d0 : FAILURE in 2h 06m 56s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/2d891abe6cae426faeb91a75324d4c9c : FAILURE in 42m 35s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/1b93878a1f534f4f84be1c07d83128f3 : FAILURE in 1h 07m 56s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5cead6d5aa9b42a9a2b81292dca0f75e : SUCCESS in 1h 51m 51s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2ab6de9dde4c4287a1844c090990466b : SUCCESS in 1h 34m 23s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/9cfbd1ea16e74d16870a00c29a459a0a : FAILURE in 40m 50s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/c0af7606bf3046e895b2e716c4a4e295 : FAILURE in 42m 15s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/99d3b065c0f240d1add860bca4b33839 : FAILURE in 49m 49s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/13c3a5f9c33b483fa52755a63612347f : FAILURE in 39m 19s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/45e67c66a3aa4336b68957a7f6a4d28d : FAILURE in 1h 17m 03s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/ea3e94593cb74a0ab8b8c220a8d481d4 : SUCCESS in 1h 26m 50s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/8dea1ce53ad647f88d968f16a535fd32 : SUCCESS in 5m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1b03fadb44f4721bb4c35e4e75c9522 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9b3a5a84f7e40a78e1343fba7c76141 : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":4},{"id":"a41be4a450857dd8fe28fbbd84cf1e3e3778e5a6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-24 10:38:03.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"49b7d104de72d540c722ab522a2d82e74b470737","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 12:48:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/821c4b9fa4e84246b545f636940cfaaa : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9ad13b4d4a54726b1639ecd4e6c66a6 : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2022a3b05b0e44dc8ce61e672b9151ff : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/339dd44ccfc84343a5ab276485960909 : SUCCESS in 11m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba952939cfa44e06a7d1dc2b26a4659b : SUCCESS in 13m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02c36c7851fa4c8796b644cf6577ef06 : SUCCESS in 17m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6640092ef6b4e4f9a4e7ab21672824a : SUCCESS in 5m 17s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/a30a47b9b83b4d60bec78f378c54d866 : SUCCESS in 2h 05m 45s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4904952a3d174c60b1810ae1404e5e02 : FAILURE in 36m 58s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/0e4b19fe96b14dda89b79254d63b6482 : FAILURE in 36m 17s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/fe9f1484cc4e4ada880581721ad77f24 : SUCCESS in 1h 41m 32s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fbacfa1dd4a1498eabaed16816645107 : SUCCESS in 1h 35m 58s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/fc4acdac21a44fecb02b2594bc8fd9b6 : FAILURE in 38m 55s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/879feb6c21824a80bc34615f4d03b728 : FAILURE in 37m 15s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/42b247d9457f46fd9ceff1613492d254 : FAILURE in 35m 57s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/3f99736928a04f88a212345309bc67b1 : FAILURE in 49m 51s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/9c382d370aa640498038242a4aa474dc : FAILURE in 1h 12m 17s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/8f862b02adaa481a90041bb47479e26d : SUCCESS in 1h 07m 30s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/598eedbaf1814616bab606795e723994 : SUCCESS in 6m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/48464191ffd647e6b18b4fee68fc450c : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/246e8c733c8f40fdab7fc6040d8acfdc : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":4},{"id":"f53fb4f0417f6bbcf1198a1282ae4792f188239f","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-09-02 14:45:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6ceff3233aba726b1cd88d1737ce1ec4c560fb2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 17:09:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1c7e2cae92c4d9da8603abbca48a1ca : FAILURE in 18m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd58a57c408540c19a702d4ee90e51c0 : FAILURE in 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e41508d48c1f42b990da9304a7766791 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/075e9f768bd74e01a36753ae0db6c23c : POST_FAILURE in 10m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/147e09079ed9410ea84fb61d738f9741 : FAILURE in 13m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bbd25abec3ac4e998364316041b25167 : FAILURE in 16m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9995b5d952034e968cd0303e008746aa : SUCCESS in 5m 16s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/80dd9e0519b64931bbe083ec65f58a35 : FAILURE in 2h 17m 48s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f3e943c94b7244c9b4bd9e0d9b443ee1 : FAILURE in 48m 17s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d73efea57e244d4bad8461c5013c7288 : FAILURE in 39m 51s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/92b5e963ee2c4e2cbdc095a9386f58b9 : SUCCESS in 1h 56m 37s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/dd691ae6a41944ad9f73aab304a34574 : FAILURE in 1h 07m 45s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/9378048a65e2433fae53acdc740bc063 : FAILURE in 38m 11s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/0312a8d0c7074acdb4bf14a9189e88a4 : FAILURE in 48m 02s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/be3650c2655b488798e6609fc6c2b807 : FAILURE in 43m 54s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/72689eb5c7994657ada5bb3a4c9a1485 : FAILURE in 53m 48s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/c43538b35f74434d98c5dc708cee3b13 : FAILURE in 47m 48s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/b8d42361b93145a38accc1fc11dda017 : FAILURE in 42m 30s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/b4d40f06b5ad43938fcc089211a7da50 : SUCCESS in 6m 24s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/292f2f97c96e4751aead46ebd7e11ae1 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ef4fcb0e586b4ea19e52854f5c11bb2e : FAILURE in 8m 46s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/460a682f102840d88e1d74c0cb0d64af : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":5},{"id":"f560160a5f018d65ccab1b5ca4986a78c6874b7f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-03 02:32:26.000000000","message":"Patch Set 5:\n\nHi @renminmin, thanks for your patience for the review. I just started reviewing the snapshot type backup patch, the whole idea looks OK to me, but in order to make the review process faster and easier, could you please split the patches into separate ones? e.g. one patch for support of creating snapshot type backup, one for creating instance using snapshot type backup, etc. We will see if we could make some of them merged before the end of this dev cycle. Thanks again!","accounts_in_message":[],"_revision_number":5},{"id":"ee585fd75b4debfef3f26c794a38adfd5f751dd5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-03 02:33:48.000000000","message":"Patch Set 5:\n\nI created two separate tasks in https://storyboard.openstack.org/#!/story/2005819, feel free to add more.","accounts_in_message":[],"_revision_number":5},{"id":"3a8bc44718f19bd72e24c6c4d133753d3ce5ce80","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-09-16 07:56:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bd79eee802862fbf003586d1a11dd117e55e1c75","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-09-16 08:06:34.000000000","message":"Patch Set 6:\n\n\u003e I created two separate tasks in https://storyboard.openstack.org/#!/story/2005819,\n \u003e feel free to add more.\n\nSure. I will continue to work on this story.","accounts_in_message":[],"_revision_number":6},{"id":"6fc081e34bddcf59b56a9ea068e2b9f1cd5922c8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-16 08:11:00.000000000","message":"Patch Set 6:\n\n\u003e \u003e I created two separate tasks in https://storyboard.openstack.org/#!/story/2005819,\n \u003e \u003e feel free to add more.\n \u003e \n \u003e Sure. I will continue to work on this story.\n\nThanks, I also added two tasks relating to the doc.","accounts_in_message":[],"_revision_number":6},{"id":"22998d5fdb6dd99d1c9664b9e83137baa8800d88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 08:55:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2b0492284ed4301914d4db40d7a8871 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4d368bbe16f420592db7ed92926f384 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35da0abd220a4127b231f435e9272be3 : SUCCESS in 7m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4be00e77fc0c4fa984d7500360c535cd : SUCCESS in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3c7551b8d224e16962e8e4cf0a2db32 : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/386c7bf9fea34345b67ce9c318435c05 : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe4514391434332bebbb3c1d9efb258 : SUCCESS in 7m 18s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/fc40f1b6ed26416db0ab88f3aacdae14 : FAILURE in 32m 38s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f1805ebf47cb48c4a361bf2673068c32 : FAILURE in 36m 46s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/5ef9c4ccc6394f55b2a0d7d7685571f6 : FAILURE in 36m 18s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8fe6b8e44b2c489889ea864d2409e27d : FAILURE in 38m 06s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/48d3e85ec4284927911ff7f78de235a0 : FAILURE in 25m 13s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/c883bd7c547146f7bf8650ef6363f44d : FAILURE in 53m 38s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/4792da53db2a4055ac514a916a4c0239 : FAILURE in 34m 45s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/869e669680d6424fa98ca4ee66b34769 : FAILURE in 34m 38s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/d4c9d85c52aa4f0aae3db204560625f5 : FAILURE in 27m 45s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/e08f4cfd1bfb46f0a6eb52c86eb865fc : FAILURE in 30m 02s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/f864b0598948440faf6ef46aaf20eb41 : FAILURE in 38m 58s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/ceec081356be45d98304474da82accee : SUCCESS in 4m 55s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de7dc941b6bd4995b88af3e5795ad3e1 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1dcf40aea2734dc585ec3b1261dbe20b : SUCCESS in 8m 55s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3bf5206a7d74468a940a35c7bd27605 : FAILURE in 29m 57s","accounts_in_message":[],"_revision_number":6},{"id":"bb5be90acfaa0a6dc1bce674404b7f8372f21653","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-09-17 02:05:51.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"fea985f68f4fe6fd1c8b803b70a0274ed6f2a5b8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 02:07:28.000000000","message":"Patch Set 6:\n\nyou could rebase on this patch https://review.opendev.org/#/c/682510/","accounts_in_message":[],"_revision_number":6},{"id":"f0b915299d6a5cb49d7b3f471cadcce21c327f76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 02:46:21.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47404c02385e4611a66b3c4d366d08a1 : FAILURE in 14m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8dd2a94497248c3bca092ec0c05bcf6 : FAILURE in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8b68ca9a7fa4b19af3e8010290e82f7 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/356d519fa022486d9817054e1ae89831 : FAILURE in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad8617cda9f9464e830ee73f3bcc6cf3 : FAILURE in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ec9f3a614d643d0a296096b01a60e6e : FAILURE in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ea664d51dc40d587ab2347ebfd6a01 : SUCCESS in 5m 56s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/317f3aa878c44ef0b6c304a69be2e299 : FAILURE in 31m 59s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f52b475e42e84a37bbe7bbba1b676290 : FAILURE in 30m 32s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/f8986143af1549c5ba86e21df5b56174 : FAILURE in 31m 30s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/bfbd72f0d7574ab881d3834a3ff7a406 : FAILURE in 33m 54s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/dd878dc2ba834980accd7f2714821984 : FAILURE in 32m 04s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/e794e4f3df33470790c1eff1213d5df7 : FAILURE in 31m 29s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/209c60d656ce405aa41e2972278f1b46 : FAILURE in 32m 55s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/f4f9e4ee2a844af98e738b52e7221b3b : FAILURE in 32m 38s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/f45669ea7ea24d87bb9cf20d37f984ed : FAILURE in 32m 27s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/1ee3c1cabc824b70837bc23507095b5c : FAILURE in 32m 54s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/3707ae23d2834be38d12c753b4eae8e3 : FAILURE in 29m 25s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/97b95ec9e7de42e4a949034e173a2188 : SUCCESS in 6m 48s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba471b5da74545848c73e9134cbe7dc2 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9feb2f26cfef4d3f83e0580f567707e4 : FAILURE in 8m 24s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31b9356de3d24ec5bbf9c4218c7a8e99 : FAILURE in 32m 32s","accounts_in_message":[],"_revision_number":6},{"id":"55a0ee6c9fc4fa32a5818968542daa9530f4fe88","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-09-17 07:36:52.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"4e000344e15b44189df4ab354d7acb5a1a5b7cde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 09:41:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60a518c8eff44199a8d08e3dc818267b : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fed66ae2920548c4ba604986a844d15a : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a80c5a2d4e1648c7b8b1b2144b9809c1 : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a10a54acad2843d29bc2e3861c758207 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fe382e6ea3c4a4daac0f26e97dac826 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4eb150e3a613420dbad8da5d02351e73 : SUCCESS in 13m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d3ea02eab74cc195e49a756e8fedd4 : SUCCESS in 8m 28s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/79139296bc8a49b78ca08ae0051d2379 : FAILURE in 1h 34m 27s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/9b5d365b30434ce0abcbe00bfa4aff2e : FAILURE in 1h 31m 11s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7db40b94b876477f9e11cab1430ead23 : FAILURE in 1h 02m 31s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/1c79519bcaea4df79c57060133f83339 : SUCCESS in 1h 56m 30s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/33523ec60db441b49a816a818ce65745 : SUCCESS in 1h 50m 52s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d92f807dd3ff43cfbf17ad75f94e6bb5 : FAILURE in 50m 40s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/46589c9880064886b0574769b428862f : FAILURE in 50m 25s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/7e342c16ce614218988836db5d42a1db : FAILURE in 44m 57s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/533fb54e8012409faf8669e4e0e9c2f3 : FAILURE in 43m 28s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/d33f476f766448b386ad3057f038107f : FAILURE in 40m 45s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/1f7450d08f3e4f979425848a849d243b : FAILURE in 55m 30s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/ba5646d1135e497cb45a065791cc01b6 : SUCCESS in 5m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6ddf5c6fc6db4a9d9510c4e1b638f3a9 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7e46bd4258f14ca5acbee288522f42ff : SUCCESS in 7m 43s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d71528724ec248f091357f57de3c62a6 : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":7},{"id":"5e3b282bf5bf0551de5b41b57fced52b4f9f8f2c","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-09-17 09:49:44.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"73484bce8ac97bd67f55416454cdd0099c1ed407","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-17 10:44:27.000000000","message":"Patch Set 7:\n\nHi @renminmin, just FYI, we are at RC release stage, after stable/train branch is cut, I\u0027ll review the series of your patches asap. Thanks for contributing to Trove (and thanks for splitting the previous \u0027big\u0027 patch).","accounts_in_message":[],"_revision_number":7},{"id":"c5fc3ac13b3cb05d478d1395e1904cfc5548c2a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:08:28.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc720b52e04444ad9bd168b78784c4f6 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0462add074040a89bcc4b7e5146c9b3 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb1130566f374f97a526086aa063b52c : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5f80aa3d42e4c55a1760f90c0e852c5 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/393c14f847ff4a009a9673bed160dcee : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4464d7c5f2b74294a99aa715b30c8304 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbdd81af43154fe296c060b06eaa7ce6 : SUCCESS in 7m 30s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/883cbe2986d141168588714698dd284a : FAILURE in 1h 40m 51s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/91d63b7f1bd04025a9349a04c31152fc : FAILURE in 1h 17m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/b3bf946817ea40d48ae8a6f7eddbc44e : FAILURE in 50m 50s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/1ef2be5ea55d49759a81286258419348 : FAILURE in 1h 59m 46s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2431f5aa939d43329589a6b5777b2d54 : RETRY_LIMIT in 5m 32s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/5d4f29b7b7af4c4b9f923f3e1b994cc9 : FAILURE in 44m 56s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/a14fc088add6470ba01143e6c327d564 : FAILURE in 41m 20s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/10984b3e320a4f7bbf5e773261a9b386 : FAILURE in 44m 17s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/88321425c71840f69ae60d9ddf495921 : FAILURE in 36m 05s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/7f9e683d6c514231a4c7e8c6b887e450 : FAILURE in 53m 29s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/e73fd8d07d3947e3abe918945b9c9286 : FAILURE in 45m 39s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/36523f5ff1344cc58437b53a2497ca50 : SUCCESS in 6m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b158370fdca426f9d861e658e4cddfc : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/961d56a777114622a303838054748ef2 : SUCCESS in 7m 02s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0a047612b14480ba5536349a2b8fdd0 : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":7},{"id":"9929842473ae436d901ab44626c84cd62f85bcde","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-09-18 01:47:59.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"bb952e5b4da11af90321932f485dbfba31413e6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:17:46.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7e8495e8fef4fd0b553c5c2c2daeec4 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b060ba21b8a415ca72bcb21825258d3 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b26a802c5b754aacb0e76731d5ce14f9 : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61125f18a534405498b86b2b983b7b52 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c6e28b20faf44a1a7eba1099b282f5b : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57638029390e4062a40af7bd281c35de : SUCCESS in 13m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c354accd6f46659efd14faab17292a : SUCCESS in 5m 23s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d96a2054561b471b83d9d1fe853554c8 : SUCCESS in 2h 48m 23s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/66742648422f43efb9693f49cb047a94 : FAILURE in 1h 14m 23s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7d6b6cc106074181ad02d85a422acaed : FAILURE in 50m 21s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/81cc745e7a1143709d122632d080d7b9 : SUCCESS in 1h 56m 53s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1106896561544c719a28e49551176350 : FAILURE in 1h 33m 45s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/09dd798e54e946d6bf4a982843e0aad4 : FAILURE in 45m 57s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/fabd5f5642654766bb88c920c96a20e4 : FAILURE in 42m 29s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/352f0d5ef95848bd82e199be1bdbcd26 : FAILURE in 46m 00s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/4558985668004a3986dbfb0969f17cf0 : FAILURE in 50m 38s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/eefb0537192c4ace9a80aacb8be6418c : FAILURE in 46m 36s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/2d582c5b34c84c90889d897f0230edc0 : FAILURE in 44m 44s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/d480d8bcde9349d69067850a218c1ad6 : SUCCESS in 5m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1a38e466ae114db780098d7003a06921 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4534a3c1423745cfa98559aac1eccebe : SUCCESS in 9m 22s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f85da5ced8d846b59c1190bc829ae938 : SUCCESS in 35m 00s","accounts_in_message":[],"_revision_number":7},{"id":"f6d70fc82a04cc8e957b651edbbdac872489672b","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-09-19 09:38:52.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"dee589d1a89b5a97520197ed4663f23fa8e1cf1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 12:43:24.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc55a3fb90984cb0a137a6af80ba7e48 : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63a0fa2144bd4cd58b00858e435267b7 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40cd834d128443cdaee7d0fada5f9b8e : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4066823b34ad460fa39a3179a0e09996 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce6eb7ccfd7a4c07b264401862144279 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47b881c4d931455b9c2a269d1634cff1 : SUCCESS in 12m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390b7bf98c924f2aa15c7b9bf246d818 : SUCCESS in 4m 44s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c501b78bed5e4297a2f241db1452498c : SUCCESS in 2h 54m 18s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/67e428a246d44150bac5bdda077bf739 : FAILURE in 1h 33m 04s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/bf95d53c8f664bd499812f8134a09422 : FAILURE in 1h 04m 13s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/dcc0aa3b55cc4d799b26f28010540c6b : FAILURE in 1h 45m 20s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/15970535c839477cab4651f408c90dac : FAILURE in 1h 52m 07s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/7835f09889e244de979dd516769c295f : FAILURE in 1h 39m 03s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/59482006229a491dacdacfd1e6907e7b : FAILURE in 44m 44s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/cfb3044835e64cefa6baa7e7d4166237 : FAILURE in 54m 46s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/c89c0ad6702d4ab2af19618ae587e4cd : FAILURE in 59m 25s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/c2c3157dba1e4cae902f11c6f0653fc7 : FAILURE in 47m 20s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/e8147bc0075143a5af99eeea92c086d0 : FAILURE in 49m 34s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/2a87edeaf5434901bf559d99057d6e10 : SUCCESS in 4m 58s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5b2976dde1bc487aab311e9cce5bfbbd : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b20c2f1a955a40eabd6a33ade3048898 : SUCCESS in 8m 45s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34c28bccb77643fca883a3c0b02b524c : SUCCESS in 35m 53s","accounts_in_message":[],"_revision_number":7},{"id":"d85052b712dada2d37110e98d5a62c2e8d26fc32","author":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"date":"2019-09-20 06:57:11.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"10d358b5334ca990687685d1e39d9c62945349e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 09:59:06.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5955559ae6b41fe9425d61d3dc060ad : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1828fb6b6185463e9823af9faf4f8c7d : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ef8b99240f4303a6357a5402bea93b : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/802dfe0c015548cf990e9dd8d5c99ffe : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d189591ae70148e482a872dd87b10222 : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31be441f15a6482a971acbc3a2f7b3bd : SUCCESS in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74145ef77984480daaf84de28241d305 : SUCCESS in 5m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a6cd8b92fa424c3ca6ec6bb3f5e848a4 : SUCCESS in 10m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/901cd63d9acd4099bc1204e7faa0ef9d : SUCCESS in 6m 57s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/6ff96bb5ba754a68b5d9fe3602e2dd24 : SUCCESS in 5m 34s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8f9c6c44724d42f886b1bc36ab3cc210 : SUCCESS in 2h 26m 55s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/157dc18731bb4a188b899393ff11f54e : SUCCESS in 1h 59m 50s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/973dee34e2124d50b2c26d9c5eecd01b : FAILURE in 1h 12m 21s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/99ff436df6414f72beda6826120600ee : FAILURE in 47m 27s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/154c0ff04d1242aeb252c041b72128eb : FAILURE in 45m 55s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/7d8978aaea9c4797b045c0d36ae214b0 : FAILURE in 1h 15m 40s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/85718977b43a4406a90c8f95e380d15d : FAILURE in 1h 07m 56s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/63852fff74bc4b4ea51a1224924f370e : FAILURE in 47m 14s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/c8a2714c00ac479ba85a3ad2685ce178 : FAILURE in 59m 05s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/223b0a6c8a004f41b10be251d6e43591 : SUCCESS in 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3c2646ce9a5aa34079a15a608472ae10b2bc2426","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-30 00:22:59.000000000","message":"Patch Set 7: Code-Review-1\n\nHi @renminmin, I can see there is a new parameter named \u0027snapshot\u0027 is added as the backup type, but IMHO, how the backup is created should be transparent to the end users but only configurable by the cloud admin. I\u0027m wondering if it really matters for the end user to choose the backup type, given the user can not see the cinder volumes in the service tenant deployment model? Thoughts?","accounts_in_message":[],"_revision_number":7},{"id":"98e4e2cbc2f1d6d7ebf760f1ae1199f69621d91c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-30 00:34:14.000000000","message":"Patch Set 7:\n\nSo I\u0027d suggest we add a new file in \u0027trove.common.strategies.storage\u0027 named \u0027cinder.py\u0027 and add the volume management logic accordingly. As a result, the cloud admin only needs to config something like:\n\n```\nstorage_strategy \u003d CinderStorage\nstorage_namespace \u003d trove.common.strategies.storage.cinder\n```","accounts_in_message":[],"_revision_number":7},{"id":"4036b1e83749b796e76c1cd459a39c62c8a37425","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-19 03:34:31.000000000","message":"Abandoned\n\nno long support. and may be duplicated with: https://review.opendev.org/c/openstack/trove/+/903440","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"704cf73bc3951eee5c93290ac7a72f39568f8401","revisions":{"de0eef4e41a5f80055d5dd3d61a8806932242f26":{"kind":"REWORK","_number":1,"created":"2019-06-04 03:19:41.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/662924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/662924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/662924/1"}}},"commit":{"parents":[{"commit":"c37ab5f34b6bd1939a4121be684a56ae755dcdd4","subject":"docs: fix build failure on html_last_updated_fmt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c37ab5f34b6bd1939a4121be684a56ae755dcdd4"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-06-03 10:37:05.000000000","tz":0},"committer":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-06-04 03:18:39.000000000","tz":0},"subject":"Add volume-snapshot backup type","message":"Add volume-snapshot backup type\n\n- Support instance create database backup by attached-volume\n  snapshot.\n  Request:\n  POST v1/{project_id}/backups\n  {\n    \"backup\": {\n      \"description\": \"My Volume Snapshot Backup\",\n      \"snapshhot\": 1,\n      \"instance\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n      \"name\": \"snapshot\"\n    }\n  }\n\n- Support instance create from snapshot backup\n\n- Support instance create replica by attached-volume clone type.\n  Master instacen generate snapshot type backup, then create replica\n  instance from generated snapshot backup.\n  Request:\n\n    POST v1/{project_id}/backups\n      {\n        \"instance\": {\n          \"flavorRef\": 1,\n          \"name\": \"json_rack_instance\",\n          \"replica_of\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n          \"replica_count\": 1,\n          \"volume\": {\n            \"size\": 2\n            \"type\": volume_type\n            \"clone\": 1\n          }\n        }\n      }\n\nStory: #2005819\nTask: #33565\n\nImplements: blueprint volume-data-snapshot-backup-type\nChange-Id: I403b2cbd49e86393f1307c2e27f218ee4d3af2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/de0eef4e41a5f80055d5dd3d61a8806932242f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/de0eef4e41a5f80055d5dd3d61a8806932242f26"}]},"branch":"refs/heads/master"},"aadacb03d105ca2f8fe4abeeb4ab1a68c7ae9920":{"kind":"REWORK","_number":2,"created":"2019-06-04 10:10:40.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/662924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/662924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/662924/2"}}},"commit":{"parents":[{"commit":"c37ab5f34b6bd1939a4121be684a56ae755dcdd4","subject":"docs: fix build failure on html_last_updated_fmt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c37ab5f34b6bd1939a4121be684a56ae755dcdd4"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-06-03 10:37:05.000000000","tz":0},"committer":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-06-04 10:09:26.000000000","tz":0},"subject":"Add volume-snapshot backup type","message":"Add volume-snapshot backup type\n\n- Support instance create database backup by attached-volume\n  snapshot.\n  Request:\n  POST v1/{project_id}/backups\n  {\n    \"backup\": {\n      \"description\": \"My Volume Snapshot Backup\",\n      \"snapshhot\": 1,\n      \"instance\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n      \"name\": \"snapshot\"\n    }\n  }\n\n- Support instance create from snapshot backup\n\n- Support instance create replica by attached-volume clone type.\n  Master instacen generate snapshot type backup, then create replica\n  instance from generated snapshot backup.\n  Request:\n\n    POST v1/{project_id}/backups\n      {\n        \"instance\": {\n          \"flavorRef\": 1,\n          \"name\": \"json_rack_instance\",\n          \"replica_of\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n          \"replica_count\": 1,\n          \"volume\": {\n            \"size\": 2\n            \"type\": volume_type\n            \"clone\": 1\n          }\n        }\n      }\n\nStory: #2005819\nTask: #33565\n\nImplements: blueprint volume-data-snapshot-backup-type\nChange-Id: I403b2cbd49e86393f1307c2e27f218ee4d3af2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aadacb03d105ca2f8fe4abeeb4ab1a68c7ae9920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aadacb03d105ca2f8fe4abeeb4ab1a68c7ae9920"}]},"branch":"refs/heads/master"},"9a20f188ef41e2853e4cfafea6702d1a94dc387c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-04 08:56:27.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/662924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/662924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/662924/3"}}},"commit":{"parents":[{"commit":"f72015179ea921c8264a77390359d638e0bd8727","subject":"Merge \"Update the outdated content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f72015179ea921c8264a77390359d638e0bd8727"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-06-03 10:37:05.000000000","tz":0},"committer":{"name":"renminmin","email":"rmm0811@gmail.com","date":"2019-07-04 08:56:27.000000000","tz":0},"subject":"Add volume-snapshot backup type","message":"Add volume-snapshot backup type\n\n- Support instance create database backup by attached-volume\n  snapshot.\n  Request:\n  POST v1/{project_id}/backups\n  {\n    \"backup\": {\n      \"description\": \"My Volume Snapshot Backup\",\n      \"snapshhot\": 1,\n      \"instance\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n      \"name\": \"snapshot\"\n    }\n  }\n\n- Support instance create from snapshot backup\n\n- Support instance create replica by attached-volume clone type.\n  Master instacen generate snapshot type backup, then create replica\n  instance from generated snapshot backup.\n  Request:\n\n    POST v1/{project_id}/backups\n      {\n        \"instance\": {\n          \"flavorRef\": 1,\n          \"name\": \"json_rack_instance\",\n          \"replica_of\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n          \"replica_count\": 1,\n          \"volume\": {\n            \"size\": 2\n            \"type\": volume_type\n            \"clone\": 1\n          }\n        }\n      }\n\nStory: #2005819\nTask: #33565\n\nImplements: blueprint volume-data-snapshot-backup-type\nChange-Id: I403b2cbd49e86393f1307c2e27f218ee4d3af2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9a20f188ef41e2853e4cfafea6702d1a94dc387c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9a20f188ef41e2853e4cfafea6702d1a94dc387c"}]},"branch":"refs/heads/master"},"08d031eda221ec4056319a35bc495336769492a6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-14 22:10:38.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/24/662924/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/662924/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/662924/4"}}},"commit":{"parents":[{"commit":"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1","subject":"Fix Trove CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-06-03 10:37:05.000000000","tz":0},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-14 22:10:38.000000000","tz":0},"subject":"Add volume-snapshot backup type","message":"Add volume-snapshot backup type\n\n- Support instance create database backup by attached-volume\n  snapshot.\n  Request:\n  POST v1/{project_id}/backups\n  {\n    \"backup\": {\n      \"description\": \"My Volume Snapshot Backup\",\n      \"snapshhot\": 1,\n      \"instance\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n      \"name\": \"snapshot\"\n    }\n  }\n\n- Support instance create from snapshot backup\n\n- Support instance create replica by attached-volume clone type.\n  Master instacen generate snapshot type backup, then create replica\n  instance from generated snapshot backup.\n  Request:\n\n    POST v1/{project_id}/backups\n      {\n        \"instance\": {\n          \"flavorRef\": 1,\n          \"name\": \"json_rack_instance\",\n          \"replica_of\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n          \"replica_count\": 1,\n          \"volume\": {\n            \"size\": 2\n            \"type\": volume_type\n            \"clone\": 1\n          }\n        }\n      }\n\nStory: #2005819\nTask: #33565\n\nImplements: blueprint volume-data-snapshot-backup-type\nChange-Id: I403b2cbd49e86393f1307c2e27f218ee4d3af2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/08d031eda221ec4056319a35bc495336769492a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/08d031eda221ec4056319a35bc495336769492a6"}]},"branch":"refs/heads/master"},"d88113dbe59ff4f96658f05ac9f89234bd82240c":{"kind":"REWORK","_number":5,"created":"2019-09-02 14:45:34.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/662924/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/662924/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/662924/5"}}},"commit":{"parents":[{"commit":"ba6e526c602bf9cf428cda41b64bf5a90f62a6a5","subject":"Merge \"Support management security group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba6e526c602bf9cf428cda41b64bf5a90f62a6a5"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-06-03 10:37:05.000000000","tz":0},"committer":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-09-02 14:43:37.000000000","tz":0},"subject":"Add volume-snapshot backup type","message":"Add volume-snapshot backup type\n\n- Support instance create database backup by attached-volume\n  snapshot.\n  Request:\n  POST v1/{project_id}/backups\n  {\n    \"backup\": {\n      \"description\": \"My Volume Snapshot Backup\",\n      \"snapshhot\": 1,\n      \"instance\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n      \"name\": \"snapshot\"\n    }\n  }\n\n- Support instance create from snapshot backup\n\n- Support instance create replica by attached-volume clone type.\n  Master instacen generate snapshot type backup, then create replica\n  instance from generated snapshot backup.\n  Request:\n\n    POST v1/{project_id}/backups\n      {\n        \"instance\": {\n          \"flavorRef\": 1,\n          \"name\": \"json_rack_instance\",\n          \"replica_of\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n          \"replica_count\": 1,\n          \"volume\": {\n            \"size\": 2\n            \"type\": volume_type\n            \"clone\": 1\n          }\n        }\n      }\n\nStory: #2005819\nTask: #33565\n\nImplements: blueprint volume-data-snapshot-backup-type\nChange-Id: I403b2cbd49e86393f1307c2e27f218ee4d3af2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d88113dbe59ff4f96658f05ac9f89234bd82240c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d88113dbe59ff4f96658f05ac9f89234bd82240c"}]},"branch":"refs/heads/master"},"24e6b47bc3822fd42f8fe7b14b8017d36f86899e":{"kind":"REWORK","_number":6,"created":"2019-09-16 07:56:06.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/662924/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/662924/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/662924/6"}}},"commit":{"parents":[{"commit":"b43248be7981d994e69b0f708e5168eea8448c2d","subject":"[doc] Create instance in user guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b43248be7981d994e69b0f708e5168eea8448c2d"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-09-16 07:52:10.000000000","tz":0},"committer":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-09-16 07:52:10.000000000","tz":0},"subject":"Add volume-snapshot backup type","message":"Add volume-snapshot backup type\n\nSupport instance create database backup by attached-volume\nsnapshot.\n\nRequest:\n  POST v1/{project_id}/backups\n  {\n    \"backup\": {\n      \"description\": \"My Volume Snapshot Backup\",\n      \"snapshhot\": 1,\n      \"instance\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n      \"name\": \"snapshot\"\n    }\n  }\n\nStory: #2005819\nTask: #33565\n\nImplements: blueprint volume-data-snapshot-backup-type\nChange-Id: I403b2cbd49e86393f1307c2e27f218ee4d3af2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24e6b47bc3822fd42f8fe7b14b8017d36f86899e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24e6b47bc3822fd42f8fe7b14b8017d36f86899e"}]},"branch":"refs/heads/master"},"704cf73bc3951eee5c93290ac7a72f39568f8401":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-09-17 07:36:52.000000000","uploader":{"_account_id":8037,"name":"renminmin","email":"rmm0811@gmail.com","username":"renminmin"},"ref":"refs/changes/24/662924/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/662924/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/662924/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/662924/7"}}},"commit":{"parents":[{"commit":"f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1","subject":"Fix Trove CI failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f771e939ebe23a0e0e7da20d87b4256f5e3d2bc1"}]}],"author":{"name":"Minmin Ren","email":"renmm6@chinaunicom.cn","date":"2019-09-16 07:52:10.000000000","tz":0},"committer":{"name":"renminmin","email":"rmm0811@gmail.com","date":"2019-09-17 07:36:52.000000000","tz":0},"subject":"Add volume-snapshot backup type","message":"Add volume-snapshot backup type\n\nSupport instance create database backup by attached-volume\nsnapshot.\n\nRequest:\n  POST v1/{project_id}/backups\n  {\n    \"backup\": {\n      \"description\": \"My Volume Snapshot Backup\",\n      \"snapshhot\": 1,\n      \"instance\": \"44b277eb-39be-4921-be31-3d61b43651d7\",\n      \"name\": \"snapshot\"\n    }\n  }\n\nStory: #2005819\nTask: #33565\n\nImplements: blueprint volume-data-snapshot-backup-type\nChange-Id: I403b2cbd49e86393f1307c2e27f218ee4d3af2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/704cf73bc3951eee5c93290ac7a72f39568f8401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/704cf73bc3951eee5c93290ac7a72f39568f8401"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
