)]}'
{"id":"openstack%2Ftrove~664154","triplet_id":"openstack%2Ftrove~master~Ia0d954237dd3c9c2b2ebedf440fa3c656da30777","project":"openstack/trove","branch":"master","topic":"trovestack","hashtags":[],"change_id":"Ia0d954237dd3c9c2b2ebedf440fa3c656da30777","subject":"trovestack guide","status":"MERGED","created":"2019-06-09 09:21:25.000000000","updated":"2019-06-11 02:33:23.000000000","submitted":"2019-06-11 02:33:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664154-1560220403944-8e310be5","meta_rev_id":"95b3367dc197034afe820ba43c000c7b1463a0e9","_number":664154,"virtual_id_number":664154,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-11 02:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 00:28:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 00:28:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 02:33:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4037d0b6e819268e0aa3b63751bb837a7c56624e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-09 09:21:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19d4a814368224e1f16376fcaa583becb38cf4b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 10:42:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/54/664154/1/check/openstack-tox-cover/11d654b/ : FAILURE in 14m 58s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/54/664154/1/check/openstack-tox-lower-constraints/93be7bb/ : FAILURE in 13m 44s\n- openstack-tox-pep8 http://logs.openstack.org/54/664154/1/check/openstack-tox-pep8/b4d7fa8/ : SUCCESS in 7m 21s\n- openstack-tox-py27 http://logs.openstack.org/54/664154/1/check/openstack-tox-py27/5f75087/ : FAILURE in 9m 45s\n- openstack-tox-py36 http://logs.openstack.org/54/664154/1/check/openstack-tox-py36/b31ef00/ : FAILURE in 13m 06s\n- openstack-tox-py37 http://logs.openstack.org/54/664154/1/check/openstack-tox-py37/457a075/ : FAILURE in 16m 22s\n- openstack-tox-docs http://logs.openstack.org/54/664154/1/check/openstack-tox-docs/088959e/html/ : SUCCESS in 7m 39s\n- trove-functional-mysql http://logs.openstack.org/54/664154/1/check/trove-functional-mysql/246a34f/ : FAILURE in 27m 01s\n- trove-scenario-mariadb-single http://logs.openstack.org/54/664154/1/check/trove-scenario-mariadb-single/fa0d656/ : FAILURE in 46m 51s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/54/664154/1/check/trove-scenario-mariadb-multi/62cbd0e/ : FAILURE in 23m 16s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/54/664154/1/check/trove-scenario-mysql-single/6e2260d/ : FAILURE in 29m 50s\n- trove-scenario-mysql-multi http://logs.openstack.org/54/664154/1/check/trove-scenario-mysql-multi/de42f2f/ : FAILURE in 28m 04s\n- trove-scenario-postgresql-single http://logs.openstack.org/54/664154/1/check/trove-scenario-postgresql-single/43dfbc8/ : FAILURE in 31m 09s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/54/664154/1/check/trove-scenario-postgresql-multi/efd74d8/ : FAILURE in 26m 33s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/54/664154/1/check/trove-scenario-pxc-single/db0a7d3/ : FAILURE in 29m 43s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/54/664154/1/check/trove-scenario-pxc-multi/989bf50/ : FAILURE in 23m 20s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/54/664154/1/check/trove-scenario-redis-single/6ba6b37/ : FAILURE in 30m 33s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/54/664154/1/check/trove-scenario-redis-multi/9dcfa01/ : FAILURE in 26m 48s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/54/664154/1/check/trove-tox-fakemodetests/6717682/ : FAILURE in 5m 35s\n- trove-tox-bandit-baseline http://logs.openstack.org/54/664154/1/check/trove-tox-bandit-baseline/694f6bd/ : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/54/664154/1/check/openstack-tox-pylint/f9e6e2e/ : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":1},{"id":"38d61a13f06eb4134b36f76922881b8697f7442f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-09 22:02:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"939ad3ace61baa129e581b0bdcbd614192465a31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 22:56:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/54/664154/2/check/openstack-tox-cover/be931e7/ : FAILURE in 15m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/54/664154/2/check/openstack-tox-lower-constraints/e477cc8/ : FAILURE in 15m 50s\n- openstack-tox-pep8 http://logs.openstack.org/54/664154/2/check/openstack-tox-pep8/f445f54/ : SUCCESS in 6m 22s\n- openstack-tox-py27 http://logs.openstack.org/54/664154/2/check/openstack-tox-py27/b9181f0/ : FAILURE in 17m 04s\n- openstack-tox-py36 http://logs.openstack.org/54/664154/2/check/openstack-tox-py36/9fd7dfb/ : FAILURE in 17m 22s\n- openstack-tox-py37 http://logs.openstack.org/54/664154/2/check/openstack-tox-py37/5aff727/ : FAILURE in 21m 17s\n- openstack-tox-docs http://logs.openstack.org/54/664154/2/check/openstack-tox-docs/1b5ae76/html/ : SUCCESS in 8m 45s\n- trove-functional-mysql http://logs.openstack.org/54/664154/2/check/trove-functional-mysql/f45e8c0/ : FAILURE in 37m 53s\n- trove-scenario-mariadb-single http://logs.openstack.org/54/664154/2/check/trove-scenario-mariadb-single/a242a53/ : FAILURE in 30m 08s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/54/664154/2/check/trove-scenario-mariadb-multi/0a7b080/ : FAILURE in 30m 09s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/54/664154/2/check/trove-scenario-mysql-single/58fefc6/ : FAILURE in 36m 27s\n- trove-scenario-mysql-multi http://logs.openstack.org/54/664154/2/check/trove-scenario-mysql-multi/a8368e7/ : FAILURE in 35m 31s\n- trove-scenario-postgresql-single http://logs.openstack.org/54/664154/2/check/trove-scenario-postgresql-single/23e4c30/ : FAILURE in 35m 00s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/54/664154/2/check/trove-scenario-postgresql-multi/2246181/ : FAILURE in 37m 16s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/54/664154/2/check/trove-scenario-pxc-single/46f5d28/ : FAILURE in 28m 17s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/54/664154/2/check/trove-scenario-pxc-multi/31d785d/ : FAILURE in 29m 37s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/54/664154/2/check/trove-scenario-redis-single/f09c898/ : FAILURE in 31m 34s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/54/664154/2/check/trove-scenario-redis-multi/21c17c0/ : FAILURE in 38m 40s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/54/664154/2/check/trove-tox-fakemodetests/f4c66e1/ : FAILURE in 6m 34s\n- trove-tox-bandit-baseline http://logs.openstack.org/54/664154/2/check/trove-tox-bandit-baseline/4d02a03/ : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/54/664154/2/check/openstack-tox-pylint/24929c0/ : SUCCESS in 10m 50s","accounts_in_message":[],"_revision_number":2},{"id":"36f0390b398b644eff4925032d9b34f78eb29c42","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-09 23:03:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"07ab971a8ece4de1d7ad1d35a3bf6316c0d65be5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 23:47:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/54/664154/3/check/openstack-tox-cover/7cc0eb1/ : FAILURE in 17m 04s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/54/664154/3/check/openstack-tox-lower-constraints/3b600d0/ : FAILURE in 12m 51s\n- openstack-tox-pep8 http://logs.openstack.org/54/664154/3/check/openstack-tox-pep8/668de6a/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/54/664154/3/check/openstack-tox-py27/ad2a7ca/ : FAILURE in 12m 22s\n- openstack-tox-py36 http://logs.openstack.org/54/664154/3/check/openstack-tox-py36/a450e2c/ : FAILURE in 10m 54s\n- openstack-tox-py37 http://logs.openstack.org/54/664154/3/check/openstack-tox-py37/7f64444/ : FAILURE in 16m 03s\n- openstack-tox-docs http://logs.openstack.org/54/664154/3/check/openstack-tox-docs/96294b4/html/ : SUCCESS in 6m 09s\n- trove-functional-mysql http://logs.openstack.org/54/664154/3/check/trove-functional-mysql/0fafd42/ : FAILURE in 32m 15s\n- trove-scenario-mariadb-single http://logs.openstack.org/54/664154/3/check/trove-scenario-mariadb-single/ba2f2f3/ : FAILURE in 33m 49s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/54/664154/3/check/trove-scenario-mariadb-multi/5c3feae/ : FAILURE in 27m 00s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/54/664154/3/check/trove-scenario-mysql-single/5e5ad55/ : FAILURE in 33m 44s\n- trove-scenario-mysql-multi http://logs.openstack.org/54/664154/3/check/trove-scenario-mysql-multi/128b68d/ : FAILURE in 26m 26s\n- trove-scenario-postgresql-single http://logs.openstack.org/54/664154/3/check/trove-scenario-postgresql-single/065a273/ : FAILURE in 34m 30s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/54/664154/3/check/trove-scenario-postgresql-multi/9327b0f/ : FAILURE in 26m 38s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/54/664154/3/check/trove-scenario-pxc-single/0f304d5/ : FAILURE in 34m 57s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/54/664154/3/check/trove-scenario-pxc-multi/efae316/ : FAILURE in 33m 34s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/54/664154/3/check/trove-scenario-redis-single/a915c82/ : FAILURE in 23m 30s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/54/664154/3/check/trove-scenario-redis-multi/7027482/ : FAILURE in 30m 01s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/54/664154/3/check/trove-tox-fakemodetests/a1ca07f/ : FAILURE in 6m 27s\n- trove-tox-bandit-baseline http://logs.openstack.org/54/664154/3/check/trove-tox-bandit-baseline/f41be05/ : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/54/664154/3/check/openstack-tox-pylint/fa48253/ : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":3},{"id":"86d64ea793677f449a4ec44f3e0067dedc3fca31","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-10 09:12:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b20b082b47b3cf9c549f007eccd3585a0942af3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 11:33:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/54/664154/4/check/openstack-tox-cover/908a525/cover/ : SUCCESS in 16m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/54/664154/4/check/openstack-tox-lower-constraints/1778f1a/ : SUCCESS in 12m 26s\n- openstack-tox-pep8 http://logs.openstack.org/54/664154/4/check/openstack-tox-pep8/c6277ae/ : SUCCESS in 6m 16s\n- openstack-tox-py27 http://logs.openstack.org/54/664154/4/check/openstack-tox-py27/eff1ed6/ : SUCCESS in 12m 05s\n- openstack-tox-py36 http://logs.openstack.org/54/664154/4/check/openstack-tox-py36/280f6c6/ : SUCCESS in 12m 46s\n- openstack-tox-py37 http://logs.openstack.org/54/664154/4/check/openstack-tox-py37/6e97df5/ : SUCCESS in 18m 12s\n- openstack-tox-docs http://logs.openstack.org/54/664154/4/check/openstack-tox-docs/a12dffb/html/ : SUCCESS in 6m 21s\n- trove-functional-mysql http://logs.openstack.org/54/664154/4/check/trove-functional-mysql/7bdd33a/ : SUCCESS in 2h 05m 27s\n- trove-scenario-mariadb-single http://logs.openstack.org/54/664154/4/check/trove-scenario-mariadb-single/38a54ef/ : FAILURE in 41m 47s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/54/664154/4/check/trove-scenario-mariadb-multi/e19cce7/ : FAILURE in 40m 46s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/54/664154/4/check/trove-scenario-mysql-single/c60ec19/ : SUCCESS in 1h 37m 19s\n- trove-scenario-mysql-multi http://logs.openstack.org/54/664154/4/check/trove-scenario-mysql-multi/d72dabe/ : SUCCESS in 1h 35m 34s\n- trove-scenario-postgresql-single http://logs.openstack.org/54/664154/4/check/trove-scenario-postgresql-single/447fc5c/ : FAILURE in 38m 13s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/54/664154/4/check/trove-scenario-postgresql-multi/e9303e5/ : FAILURE in 42m 06s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/54/664154/4/check/trove-scenario-pxc-single/cf3e43b/ : FAILURE in 52m 07s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/54/664154/4/check/trove-scenario-pxc-multi/699756e/ : FAILURE in 49m 08s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/54/664154/4/check/trove-scenario-redis-single/7852cd7/ : FAILURE in 1h 15m 01s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/54/664154/4/check/trove-scenario-redis-multi/a1ea0f2/ : FAILURE in 59m 17s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/54/664154/4/check/trove-tox-fakemodetests/cf5dfa0/ : SUCCESS in 8m 44s\n- trove-tox-bandit-baseline http://logs.openstack.org/54/664154/4/check/trove-tox-bandit-baseline/a1c7aae/ : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/54/664154/4/check/openstack-tox-pylint/951bcab/ : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":4},{"id":"5be1a63e53a4696d509246a7bd3474088537dc99","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-10 12:09:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9d09fbb3cc3f290122b4277584ec65adc789e50f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 14:40:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/54/664154/5/check/openstack-tox-cover/fd984f3/cover/ : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/54/664154/5/check/openstack-tox-lower-constraints/eedef4b/ : SUCCESS in 11m 06s\n- openstack-tox-pep8 http://logs.openstack.org/54/664154/5/check/openstack-tox-pep8/7aa0760/ : SUCCESS in 6m 51s\n- openstack-tox-py27 http://logs.openstack.org/54/664154/5/check/openstack-tox-py27/b28114a/ : SUCCESS in 12m 20s\n- openstack-tox-py36 http://logs.openstack.org/54/664154/5/check/openstack-tox-py36/0fc1e47/ : SUCCESS in 11m 40s\n- openstack-tox-py37 http://logs.openstack.org/54/664154/5/check/openstack-tox-py37/39b2f8b/ : SUCCESS in 18m 07s\n- openstack-tox-docs http://logs.openstack.org/54/664154/5/check/openstack-tox-docs/e1890e8/html/ : SUCCESS in 5m 37s\n- trove-functional-mysql http://logs.openstack.org/54/664154/5/check/trove-functional-mysql/c711c9c/ : FAILURE in 2h 03m 02s\n- trove-scenario-mariadb-single http://logs.openstack.org/54/664154/5/check/trove-scenario-mariadb-single/b1b2db1/ : FAILURE in 38m 28s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/54/664154/5/check/trove-scenario-mariadb-multi/660b939/ : FAILURE in 38m 06s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/54/664154/5/check/trove-scenario-mysql-single/2156b67/ : SUCCESS in 1h 36m 28s\n- trove-scenario-mysql-multi http://logs.openstack.org/54/664154/5/check/trove-scenario-mysql-multi/946f497/ : SUCCESS in 1h 40m 50s\n- trove-scenario-postgresql-single http://logs.openstack.org/54/664154/5/check/trove-scenario-postgresql-single/c4229ec/ : FAILURE in 38m 05s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/54/664154/5/check/trove-scenario-postgresql-multi/a542be7/ : FAILURE in 36m 05s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/54/664154/5/check/trove-scenario-pxc-single/d1f8187/ : FAILURE in 41m 16s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/54/664154/5/check/trove-scenario-pxc-multi/ed2d28c/ : FAILURE in 41m 16s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/54/664154/5/check/trove-scenario-redis-single/57fd6dd/ : FAILURE in 1h 03m 26s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/54/664154/5/check/trove-scenario-redis-multi/33df836/ : FAILURE in 59m 07s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/54/664154/5/check/trove-tox-fakemodetests/97a9904/ : SUCCESS in 5m 47s\n- trove-tox-bandit-baseline http://logs.openstack.org/54/664154/5/check/trove-tox-bandit-baseline/0063aa0/ : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/54/664154/5/check/openstack-tox-pylint/d34043a/ : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":5},{"id":"ad01967a3e6943a8bfc0b7e7236549958581b06e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-10 21:40:15.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"76c7572a31d62cce41062f07f06c35f115bf172b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 23:53:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/54/664154/5/check/openstack-tox-cover/ae68753/cover/ : SUCCESS in 19m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/54/664154/5/check/openstack-tox-lower-constraints/4a85d08/ : SUCCESS in 12m 35s\n- openstack-tox-pep8 http://logs.openstack.org/54/664154/5/check/openstack-tox-pep8/dd856fb/ : SUCCESS in 7m 49s\n- openstack-tox-py27 http://logs.openstack.org/54/664154/5/check/openstack-tox-py27/0020a92/ : SUCCESS in 10m 49s\n- openstack-tox-py36 http://logs.openstack.org/54/664154/5/check/openstack-tox-py36/7ebf542/ : SUCCESS in 13m 01s\n- openstack-tox-py37 http://logs.openstack.org/54/664154/5/check/openstack-tox-py37/6ee38d1/ : SUCCESS in 18m 51s\n- openstack-tox-docs http://logs.openstack.org/54/664154/5/check/openstack-tox-docs/ff11440/html/ : SUCCESS in 5m 34s\n- trove-functional-mysql http://logs.openstack.org/54/664154/5/check/trove-functional-mysql/e1d9127/ : SUCCESS in 2h 05m 52s\n- trove-scenario-mariadb-single http://logs.openstack.org/54/664154/5/check/trove-scenario-mariadb-single/6993c20/ : FAILURE in 42m 43s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/54/664154/5/check/trove-scenario-mariadb-multi/7fe548a/ : FAILURE in 41m 07s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/54/664154/5/check/trove-scenario-mysql-single/ec17e70/ : SUCCESS in 1h 44m 58s\n- trove-scenario-mysql-multi http://logs.openstack.org/54/664154/5/check/trove-scenario-mysql-multi/b9518a4/ : SUCCESS in 1h 40m 50s\n- trove-scenario-postgresql-single http://logs.openstack.org/54/664154/5/check/trove-scenario-postgresql-single/78b5c33/ : FAILURE in 42m 11s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/54/664154/5/check/trove-scenario-postgresql-multi/3ed2736/ : FAILURE in 44m 57s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/54/664154/5/check/trove-scenario-pxc-single/d5468b5/ : FAILURE in 43m 33s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/54/664154/5/check/trove-scenario-pxc-multi/5f2369b/ : FAILURE in 42m 36s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/54/664154/5/check/trove-scenario-redis-single/e0c34a0/ : FAILURE in 1h 07m 32s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/54/664154/5/check/trove-scenario-redis-multi/1c1fc05/ : FAILURE in 1h 04m 21s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/54/664154/5/check/trove-tox-fakemodetests/98fe4a6/ : SUCCESS in 6m 11s\n- trove-tox-bandit-baseline http://logs.openstack.org/54/664154/5/check/trove-tox-bandit-baseline/13994fa/ : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/54/664154/5/check/openstack-tox-pylint/6527f68/ : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":5},{"id":"4d3cbc72721184b5e921096192d3fb3ff946ae65","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-11 00:28:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2379ef0fb5330a3a6296693e1af110d1c26b7aea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 00:28:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"69642dae74ae34bbae50e7dc4598413b5fb883d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 02:33:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/54/664154/5/gate/openstack-tox-lower-constraints/1850696/ : SUCCESS in 12m 24s\n- openstack-tox-pep8 http://logs.openstack.org/54/664154/5/gate/openstack-tox-pep8/069d0c9/ : SUCCESS in 5m 57s\n- openstack-tox-py27 http://logs.openstack.org/54/664154/5/gate/openstack-tox-py27/fe96d4f/ : SUCCESS in 10m 58s\n- openstack-tox-py36 http://logs.openstack.org/54/664154/5/gate/openstack-tox-py36/467c228/ : SUCCESS in 14m 55s\n- openstack-tox-py37 http://logs.openstack.org/54/664154/5/gate/openstack-tox-py37/d951096/ : SUCCESS in 17m 56s\n- openstack-tox-docs http://logs.openstack.org/54/664154/5/gate/openstack-tox-docs/b52d0c6/html/ : SUCCESS in 5m 17s\n- trove-functional-mysql http://logs.openstack.org/54/664154/5/gate/trove-functional-mysql/5c3073b/ : SUCCESS in 1h 49m 49s\n- trove-scenario-mysql-single http://logs.openstack.org/54/664154/5/gate/trove-scenario-mysql-single/083d866/ : SUCCESS in 1h 52m 13s\n- trove-scenario-mysql-multi http://logs.openstack.org/54/664154/5/gate/trove-scenario-mysql-multi/49e0b71/ : SUCCESS in 1h 36m 46s\n- trove-tox-fakemodetests http://logs.openstack.org/54/664154/5/gate/trove-tox-fakemodetests/f2f2e80/ : SUCCESS in 5m 34s\n- openstack-tox-pylint http://logs.openstack.org/54/664154/5/gate/openstack-tox-pylint/fff03b0/ : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":5},{"id":"e3b9be3103ab544b444f3d375f40d2d72684f7f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 02:33:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"49a11dd9c7f8537e0a38f13733962533523050f1","revisions":{"3aaf118d91bcb4f08e8869babe7e2f136759f138":{"kind":"REWORK","_number":1,"created":"2019-06-09 09:21:25.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/54/664154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/664154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/664154/1"}}},"commit":{"parents":[{"commit":"4f4a87a771b3646fcd4c3cfcf561cbeccefbcd19","subject":"Merge \"Enable service tenant deployment model by default in DevStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4f4a87a771b3646fcd4c3cfcf561cbeccefbcd19"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-09 00:53:26.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-09 09:21:07.000000000","tz":720},"subject":"trovestack guide","message":"trovestack guide\n\nChange-Id: Ia0d954237dd3c9c2b2ebedf440fa3c656da30777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3aaf118d91bcb4f08e8869babe7e2f136759f138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3aaf118d91bcb4f08e8869babe7e2f136759f138"}]},"branch":"refs/heads/master"},"bc40cd9c2d3b8373c7069834ecea546868c7ed2e":{"kind":"REWORK","_number":2,"created":"2019-06-09 22:02:02.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/54/664154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/664154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/664154/2"}}},"commit":{"parents":[{"commit":"4f4a87a771b3646fcd4c3cfcf561cbeccefbcd19","subject":"Merge \"Enable service tenant deployment model by default in DevStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4f4a87a771b3646fcd4c3cfcf561cbeccefbcd19"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-09 00:53:26.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-09 22:01:53.000000000","tz":720},"subject":"trovestack guide","message":"trovestack guide\n\nChange-Id: Ia0d954237dd3c9c2b2ebedf440fa3c656da30777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bc40cd9c2d3b8373c7069834ecea546868c7ed2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bc40cd9c2d3b8373c7069834ecea546868c7ed2e"}]},"branch":"refs/heads/master"},"d32ba285af40318855ecdebcb967cc5c59c0eeaf":{"kind":"REWORK","_number":3,"created":"2019-06-09 23:03:32.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/54/664154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/664154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/664154/3"}}},"commit":{"parents":[{"commit":"4f4a87a771b3646fcd4c3cfcf561cbeccefbcd19","subject":"Merge \"Enable service tenant deployment model by default in DevStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4f4a87a771b3646fcd4c3cfcf561cbeccefbcd19"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-09 00:53:26.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-09 23:03:26.000000000","tz":720},"subject":"trovestack guide","message":"trovestack guide\n\nChange-Id: Ia0d954237dd3c9c2b2ebedf440fa3c656da30777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d32ba285af40318855ecdebcb967cc5c59c0eeaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d32ba285af40318855ecdebcb967cc5c59c0eeaf"}]},"branch":"refs/heads/master"},"db71685d59d67f4a41b9443f0b73c9512c8f7481":{"kind":"REWORK","_number":4,"created":"2019-06-10 09:12:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/54/664154/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/664154/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/664154/4"}}},"commit":{"parents":[{"commit":"4f4a87a771b3646fcd4c3cfcf561cbeccefbcd19","subject":"Merge \"Enable service tenant deployment model by default in DevStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4f4a87a771b3646fcd4c3cfcf561cbeccefbcd19"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-09 00:53:26.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-10 09:11:03.000000000","tz":720},"subject":"trovestack guide","message":"trovestack guide\n\nAlso fixed the migration script issue after upgrading SQLAlchemy\nversion from 1.2.19 to 1.3.4\n\nChange-Id: Ia0d954237dd3c9c2b2ebedf440fa3c656da30777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db71685d59d67f4a41b9443f0b73c9512c8f7481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db71685d59d67f4a41b9443f0b73c9512c8f7481"}]},"branch":"refs/heads/master"},"49a11dd9c7f8537e0a38f13733962533523050f1":{"kind":"REWORK","_number":5,"created":"2019-06-10 12:09:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/54/664154/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/664154/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/664154/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/664154/5"}}},"commit":{"parents":[{"commit":"4f4a87a771b3646fcd4c3cfcf561cbeccefbcd19","subject":"Merge \"Enable service tenant deployment model by default in DevStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4f4a87a771b3646fcd4c3cfcf561cbeccefbcd19"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-09 00:53:26.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-10 12:09:42.000000000","tz":720},"subject":"trovestack guide","message":"trovestack guide\n\nAlso fixed the migration script issue after upgrading SQLAlchemy\nversion from 1.2.19 to 1.3.4\n\nChange-Id: Ia0d954237dd3c9c2b2ebedf440fa3c656da30777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49a11dd9c7f8537e0a38f13733962533523050f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49a11dd9c7f8537e0a38f13733962533523050f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
