)]}'
{"id":"openstack%2Ftrove~665905","triplet_id":"openstack%2Ftrove~master~Ifb1b972be570c07e2e183f1c8a9bafa3ad090a86","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"Ifb1b972be570c07e2e183f1c8a9bafa3ad090a86","subject":"Remove mysql query cache settings","status":"MERGED","created":"2019-06-18 08:23:35.000000000","updated":"2020-10-02 12:21:33.000000000","submitted":"2020-10-02 12:19:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665905-1601641174922-9c0a1e66","meta_rev_id":"d6ab9e13258147dd2bbaf3d6b988c895aa089ab4","_number":665905,"virtual_id_number":665905,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-02 01:30:12.000000000","_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 12:19:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":2,"date":"2020-10-02 10:02:32.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2020-10-02 10:02:32.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 10:02:32.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-10-02 12:19:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c62dcd0900710499ddd7342103148a702a34235","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-18 08:23:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5521df9e8d80865a671e167f5ea0745f5de59c16","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-18 08:54:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8fb376300de5af6197a860fdf20829ae67a3c751","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-18 10:00:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d95d263e2df57e661a4afd342f97cb2fb873076","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 13:33:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/05/665905/3/check/openstack-tox-cover/b6b2a51/cover/ : SUCCESS in 18m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/05/665905/3/check/openstack-tox-lower-constraints/3d7f284/ : SUCCESS in 12m 53s\n- openstack-tox-pep8 http://logs.openstack.org/05/665905/3/check/openstack-tox-pep8/8ec0dc7/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/05/665905/3/check/openstack-tox-py27/62a72ea/ : SUCCESS in 10m 54s\n- openstack-tox-py36 http://logs.openstack.org/05/665905/3/check/openstack-tox-py36/f5f962b/ : SUCCESS in 13m 25s\n- openstack-tox-py37 http://logs.openstack.org/05/665905/3/check/openstack-tox-py37/aaae207/ : SUCCESS in 19m 03s\n- openstack-tox-docs http://logs.openstack.org/05/665905/3/check/openstack-tox-docs/bbe5954/html/ : SUCCESS in 4m 51s\n- trove-functional-mysql http://logs.openstack.org/05/665905/3/check/trove-functional-mysql/894704b/ : SUCCESS in 2h 20m 21s\n- trove-scenario-mariadb-single http://logs.openstack.org/05/665905/3/check/trove-scenario-mariadb-single/982639f/ : FAILURE in 38m 45s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/05/665905/3/check/trove-scenario-mariadb-multi/3de0cb9/ : FAILURE in 42m 22s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/05/665905/3/check/trove-scenario-mysql-single/ae938f7/ : SUCCESS in 1h 44m 08s\n- trove-scenario-mysql-multi http://logs.openstack.org/05/665905/3/check/trove-scenario-mysql-multi/cc68695/ : SUCCESS in 1h 25m 57s\n- trove-scenario-postgresql-single http://logs.openstack.org/05/665905/3/check/trove-scenario-postgresql-single/f71a471/ : FAILURE in 43m 35s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/05/665905/3/check/trove-scenario-postgresql-multi/831a100/ : FAILURE in 32m 22s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/05/665905/3/check/trove-scenario-pxc-single/790186a/ : FAILURE in 40m 20s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/05/665905/3/check/trove-scenario-pxc-multi/6d3a8ac/ : FAILURE in 42m 29s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/05/665905/3/check/trove-scenario-redis-single/2596f34/ : FAILURE in 1h 09m 13s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/05/665905/3/check/trove-scenario-redis-multi/16dd527/ : FAILURE in 1h 01m 55s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/05/665905/3/check/trove-tox-fakemodetests/ec63baa/ : SUCCESS in 5m 45s\n- trove-tox-bandit-baseline http://logs.openstack.org/05/665905/3/check/trove-tox-bandit-baseline/9ea2b06/ : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/05/665905/3/check/openstack-tox-pylint/adae77c/ : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":3},{"id":"eea098199ff60ea2ca4c6f9fd216a7d2a7a3552d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-24 06:33:23.000000000","message":"Patch Set 3: Code-Review-1\n\nhmm...the query cache will continue to be supported for the life time of MySQL 5.7 which is supported by Trove.","accounts_in_message":[],"_revision_number":3},{"id":"f402b39e859e894043285b6715510f2ff56bb02a","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-02 01:28:46.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e4dec2f71edffaf3c6f042e7f90e882b20387e6b","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-02 01:30:12.000000000","message":"Patch Set 4:\n\nWonder if this patch could be revisited, it\u0027s currently not possible to use mysql 8 with trove with this patch.\n\nquery cache is not recommended to be used and if users require it they could set it with config?\n\nI don\u0027t think trove should set a deprecated option by default?","accounts_in_message":[],"_revision_number":4},{"id":"f199b2d65da6cb36b8d9474e8d068c0fcbec948c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 03:48:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db26631ef8ae464c91a8e671e2bb6ca0 : FAILURE in 6m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6622c9ed6d06471e9dd4b8180ec2f067 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09c8272971c43b29fb9c00eb1f86e0b : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1a1e7ec08254e449fef668d6f6787c1 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c885188313d149d7b6a4a56fb249920b : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2037cdbf77024dd38992805fc2cf2b87 : SUCCESS in 5m 18s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a3e8e8fae65945c290a1cc6602218dfb : SUCCESS in 5m 26s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d8571f502f9a4b9d85d36677701fbb01 : TIMED_OUT in 2h 12m 44s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e3c1a355d3e46418a0057880c86250e : SUCCESS in 1h 07m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"81d33d31499a6d4b82674b2f285d7d4533684bda","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-02 10:02:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3ded24dda7470f31282f1b033e1052ac6fe5ad21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 10:02:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b6bb800dee14fa9563787f4eff537393f1a288f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:19:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/157b06eeb7fa413ea07c4c0cb89e732a : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af18b3c69654e29b0e4eb3da97fffe6 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33aee2483459401687f7969d9db94876 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d962fefb259447bbb1b2c2f67502e77 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc1317fb4424c818e9ff6fda7ce3aee : SUCCESS in 6m 06s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5aa31a73e6a9487db494da9aaad2e486 : FAILURE in 2h 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"061d5a5cb3a4b4a80fb0b0b84f3c145e65cc7a28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:19:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d6ab9e13258147dd2bbaf3d6b988c895aa089ab4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:21:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f882aed88994639b0c8b61fe51b3d97 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"608c740edb897211c8d4059980894f84f6667a19","revisions":{"1ef36c8f686477afb460310b3574e4b528880b23":{"kind":"REWORK","_number":1,"created":"2019-06-18 08:23:35.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/05/665905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/665905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/665905/1"}}},"commit":{"parents":[{"commit":"49a11dd9c7f8537e0a38f13733962533523050f1","subject":"trovestack guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49a11dd9c7f8537e0a38f13733962533523050f1"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:02:32.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:22:49.000000000","tz":600},"subject":"Remove mysql query cache settings","message":"Remove mysql query cache settings\n\nQuery cache is deprecated in 5.7 and removed in 8.0 [1]\n\n[1] https://dev.mysql.com/doc/refman/5.7/en/query-cache.html\n\nChange-Id: Ifb1b972be570c07e2e183f1c8a9bafa3ad090a86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1ef36c8f686477afb460310b3574e4b528880b23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1ef36c8f686477afb460310b3574e4b528880b23"}]},"branch":"refs/heads/master"},"84833366aeb78ece8c37435f31e3ca86ab64c563":{"kind":"REWORK","_number":2,"created":"2019-06-18 08:54:02.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/05/665905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/665905/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/665905/2"}}},"commit":{"parents":[{"commit":"49a11dd9c7f8537e0a38f13733962533523050f1","subject":"trovestack guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49a11dd9c7f8537e0a38f13733962533523050f1"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:02:32.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:53:35.000000000","tz":600},"subject":"Remove mysql query cache settings","message":"Remove mysql query cache settings\n\nQuery cache is deprecated in 5.7 and removed in 8.0 [1]\n\n[1] https://dev.mysql.com/doc/refman/5.7/en/query-cache.html\n\nChange-Id: Ifb1b972be570c07e2e183f1c8a9bafa3ad090a86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/84833366aeb78ece8c37435f31e3ca86ab64c563"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/84833366aeb78ece8c37435f31e3ca86ab64c563"}]},"branch":"refs/heads/master"},"d6aaafea7e2b3d5b40ce055b33434cf71200e217":{"kind":"REWORK","_number":3,"created":"2019-06-18 10:00:55.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/05/665905/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/665905/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/665905/3"}}},"commit":{"parents":[{"commit":"49a11dd9c7f8537e0a38f13733962533523050f1","subject":"trovestack guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49a11dd9c7f8537e0a38f13733962533523050f1"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:02:32.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 10:00:44.000000000","tz":600},"subject":"Remove mysql query cache settings","message":"Remove mysql query cache settings\n\nQuery cache is deprecated in 5.7 and removed in 8.0 [1]\n\n[1] https://dev.mysql.com/doc/refman/5.7/en/query-cache.html\n\nChange-Id: Ifb1b972be570c07e2e183f1c8a9bafa3ad090a86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6aaafea7e2b3d5b40ce055b33434cf71200e217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6aaafea7e2b3d5b40ce055b33434cf71200e217"}]},"branch":"refs/heads/master"},"608c740edb897211c8d4059980894f84f6667a19":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-10-02 01:28:46.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/05/665905/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/665905/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/665905/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/665905/4"}}},"commit":{"parents":[{"commit":"af83edd1c44d0991344a5f6b6746d039c1515554","subject":"Merge \"Handle case where no networks are selected at create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af83edd1c44d0991344a5f6b6746d039c1515554"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:02:32.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-02 01:28:46.000000000","tz":0},"subject":"Remove mysql query cache settings","message":"Remove mysql query cache settings\n\nQuery cache is deprecated in 5.7 and removed in 8.0 [1]\n\n[1] https://dev.mysql.com/doc/refman/5.7/en/query-cache.html\n\nChange-Id: Ifb1b972be570c07e2e183f1c8a9bafa3ad090a86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/608c740edb897211c8d4059980894f84f6667a19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/608c740edb897211c8d4059980894f84f6667a19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
