)]}'
{"id":"openstack%2Ftrove~665909","triplet_id":"openstack%2Ftrove~master~I5080de7b4b190053f1bcc1c4c815992c9a5889b7","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I5080de7b4b190053f1bcc1c4c815992c9a5889b7","subject":"Support new mysql8.0 utf8mb4 character and collation sets","status":"MERGED","created":"2019-06-18 08:34:13.000000000","updated":"2019-08-25 00:01:20.000000000","submitted":"2019-08-24 23:59:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665909-1566691199966-91ed003a","meta_rev_id":"8edea7d2c52285e12cb1884d3016ff337fe4bf54","_number":665909,"virtual_id_number":665909,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-15 06:22:42.000000000","_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":2,"date":"2019-08-24 23:59:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"recommended":{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2019-08-14 22:10:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-24 22:03:35.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-24 22:03:35.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 03:04:45.000000000","updated_by":{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},"reviewer":{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},"state":"REVIEWER"},{"updated":"2019-08-24 22:03:35.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-08-24 23:59:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac0649cb6e40b24613c15691f0b942ece38fab52","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-18 08:34:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2362ec3951b7109f57540680977a937b407cd80a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 11:26:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/09/665909/1/check/openstack-tox-cover/1f72dbc/cover/ : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/09/665909/1/check/openstack-tox-lower-constraints/095ca08/ : SUCCESS in 13m 24s\n- openstack-tox-pep8 http://logs.openstack.org/09/665909/1/check/openstack-tox-pep8/993f2e9/ : SUCCESS in 7m 40s\n- openstack-tox-py27 http://logs.openstack.org/09/665909/1/check/openstack-tox-py27/742eabe/ : SUCCESS in 11m 48s\n- openstack-tox-py36 http://logs.openstack.org/09/665909/1/check/openstack-tox-py36/12270b5/ : SUCCESS in 13m 14s\n- openstack-tox-py37 http://logs.openstack.org/09/665909/1/check/openstack-tox-py37/322999d/ : SUCCESS in 17m 56s\n- openstack-tox-docs http://logs.openstack.org/09/665909/1/check/openstack-tox-docs/9991d2f/html/ : SUCCESS in 5m 19s\n- trove-functional-mysql http://logs.openstack.org/09/665909/1/check/trove-functional-mysql/9006e87/ : SUCCESS in 2h 25m 36s\n- trove-scenario-mariadb-single http://logs.openstack.org/09/665909/1/check/trove-scenario-mariadb-single/d2cbb6c/ : FAILURE in 38m 29s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-mariadb-multi/8095687/ : FAILURE in 38m 56s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/09/665909/1/check/trove-scenario-mysql-single/406505e/ : SUCCESS in 1h 47m 03s\n- trove-scenario-mysql-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-mysql-multi/73f2a3f/ : SUCCESS in 1h 35m 41s\n- trove-scenario-postgresql-single http://logs.openstack.org/09/665909/1/check/trove-scenario-postgresql-single/9a99f5a/ : FAILURE in 42m 02s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-postgresql-multi/779ea1b/ : FAILURE in 45m 44s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/09/665909/1/check/trove-scenario-pxc-single/030d20c/ : FAILURE in 39m 33s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-pxc-multi/73bdfb2/ : FAILURE in 35m 45s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/09/665909/1/check/trove-scenario-redis-single/1a84ae7/ : FAILURE in 1h 09m 32s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-redis-multi/e64b175/ : FAILURE in 58m 36s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/09/665909/1/check/trove-tox-fakemodetests/68ee86a/ : SUCCESS in 6m 24s\n- trove-tox-bandit-baseline http://logs.openstack.org/09/665909/1/check/trove-tox-bandit-baseline/add5583/ : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/09/665909/1/check/openstack-tox-pylint/e35e683/ : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7feb8ecea9c6d6065070bc31f62c5bb04c398745","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-24 06:34:06.000000000","message":"Patch Set 1:\n\nHi Sam, I\u0027m wondering that are you already using Trove to deploy mysql 8?","accounts_in_message":[],"_revision_number":1},{"id":"f23e3e1b246bf05b7c5d9c3bbc1619a39b343b1d","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-24 09:29:47.000000000","message":"Patch Set 1:\n\n\u003e Hi Sam, I\u0027m wondering that are you already using Trove to deploy\n \u003e mysql 8?\n\nYes I have mysql8 and postgres 11.3 working under a bionic image. Still a couple more patches required to push upstream","accounts_in_message":[],"_revision_number":1},{"id":"99b62f8069b3c8c05d836180cd07ee551fb50eef","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-25 10:55:21.000000000","message":"Patch Set 1:\n\n\u003e \u003e Hi Sam, I\u0027m wondering that are you already using Trove to deploy\n \u003e \u003e mysql 8?\n \u003e \n \u003e Yes I have mysql8 and postgres 11.3 working under a bionic image.\n \u003e Still a couple more patches required to push upstream\n\nAre you using the elements for disk-image-create command to build the mysql 8 and postgres 11.3 guest image? Have you made any change?","accounts_in_message":[],"_revision_number":1},{"id":"59969f2133c40744e950ce32c54b06ab9c9e722f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-25 10:55:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f3e3c417b0909bb29bdec4b6fbe2853dfde06bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 10:55:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0dc97633b88dde5abfa7ef640543e78d151c63f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 12:01:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/09/665909/1/gate/openstack-tox-lower-constraints/2181e4d/ : RETRY_LIMIT in 2m 45s\n- openstack-tox-pep8 http://logs.openstack.org/09/665909/1/gate/openstack-tox-pep8/f869885/ : RETRY_LIMIT in 2m 11s\n- openstack-tox-py27 http://logs.openstack.org/09/665909/1/gate/openstack-tox-py27/fb54fd7/ : RETRY_LIMIT in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/09/665909/1/gate/openstack-tox-py36/24aa50e/ : RETRY_LIMIT in 2m 35s\n- openstack-tox-py37 http://logs.openstack.org/09/665909/1/gate/openstack-tox-py37/c17e4ae/ : RETRY_LIMIT in 2m 16s\n- openstack-tox-docs http://logs.openstack.org/09/665909/1/gate/openstack-tox-docs/94dc9dc/html/ : SUCCESS in 3m 16s\n- trove-functional-mysql http://logs.openstack.org/09/665909/1/gate/trove-functional-mysql/2d435ff/ : FAILURE in 38m 34s\n- trove-scenario-mysql-single http://logs.openstack.org/09/665909/1/gate/trove-scenario-mysql-single/5efaebf/ : FAILURE in 41m 22s\n- trove-scenario-mysql-multi http://logs.openstack.org/09/665909/1/gate/trove-scenario-mysql-multi/7483b00/ : FAILURE in 43m 39s\n- trove-tox-fakemodetests http://logs.openstack.org/09/665909/1/gate/trove-tox-fakemodetests/ae66d66/ : RETRY_LIMIT in 2m 05s\n- openstack-tox-pylint http://logs.openstack.org/09/665909/1/gate/openstack-tox-pylint/fc556d6/ : RETRY_LIMIT in 1m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f4e13d89c7958ecb326ebadf0108bef5d9b3b3db","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-25 12:16:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2207ca91bce0b88c54ec1cb41f74e3034a9b6cb7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-25 12:31:35.000000000","message":"Patch Set 1:\n\nNote: recheck  will not help here, you need a bindep.txt file that installs the missing binaries, see \n\nhttp://logs.openstack.org/09/665909/1/gate/openstack-tox-pylint/fc556d6/job-output.txt.gz#_2019-06-25_11_35_33_163427\n\nMore details at http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html","accounts_in_message":[],"_revision_number":1},{"id":"1c49c7a158f6833a90a992bca3f0444706f69ac9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 17:53:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/09/665909/1/check/openstack-tox-cover/e35427f/ : RETRY_LIMIT in 2m 05s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/09/665909/1/check/openstack-tox-lower-constraints/c629f3f/ : RETRY_LIMIT in 2m 03s\n- openstack-tox-pep8 http://logs.openstack.org/09/665909/1/check/openstack-tox-pep8/76163b6/ : RETRY_LIMIT in 1m 54s\n- openstack-tox-py27 http://logs.openstack.org/09/665909/1/check/openstack-tox-py27/337e8a5/ : RETRY_LIMIT in 1m 54s\n- openstack-tox-py36 http://logs.openstack.org/09/665909/1/check/openstack-tox-py36/65b5b0d/ : RETRY_LIMIT in 1m 43s\n- openstack-tox-py37 http://logs.openstack.org/09/665909/1/check/openstack-tox-py37/b86cb8d/ : RETRY_LIMIT in 2m 09s\n- openstack-tox-docs http://logs.openstack.org/09/665909/1/check/openstack-tox-docs/ee9cd73/html/ : SUCCESS in 3m 58s\n- trove-functional-mysql http://logs.openstack.org/09/665909/1/check/trove-functional-mysql/071f610/ : FAILURE in 40m 10s\n- trove-scenario-mariadb-single http://logs.openstack.org/09/665909/1/check/trove-scenario-mariadb-single/5bed88e/ : FAILURE in 30m 25s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-mariadb-multi/e7abea2/ : FAILURE in 33m 17s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/09/665909/1/check/trove-scenario-mysql-single/00d1add/ : FAILURE in 44m 01s\n- trove-scenario-mysql-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-mysql-multi/4952dc7/ : FAILURE in 46m 01s\n- trove-scenario-postgresql-single http://logs.openstack.org/09/665909/1/check/trove-scenario-postgresql-single/0139eb0/ : FAILURE in 40m 44s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-postgresql-multi/6958c55/ : FAILURE in 31m 10s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/09/665909/1/check/trove-scenario-pxc-single/b3d5d3f/ : FAILURE in 34m 08s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-pxc-multi/bc5c057/ : FAILURE in 31m 16s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/09/665909/1/check/trove-scenario-redis-single/64c60c5/ : FAILURE in 37m 12s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-redis-multi/550a0f3/ : FAILURE in 32m 53s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/09/665909/1/check/trove-tox-fakemodetests/479927b/ : RETRY_LIMIT in 2m 21s\n- trove-tox-bandit-baseline http://logs.openstack.org/09/665909/1/check/trove-tox-bandit-baseline/1fbb322/ : RETRY_LIMIT in 2m 10s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/09/665909/1/check/openstack-tox-pylint/aff2ba9/ : RETRY_LIMIT in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2c8992ca1a633c1fecf05e8d735d8ce11bf2614f","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-26 01:59:53.000000000","message":"Patch Set 1:\n\n\u003e Note: recheck  will not help here, you need a bindep.txt file that\n \u003e installs the missing binaries, see\n \u003e \n \u003e http://logs.openstack.org/09/665909/1/gate/openstack-tox-pylint/fc556d6/job-output.txt.gz#_2019-06-25_11_35_33_163427\n \u003e \n \u003e More details at http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nOK thanks for this, I don\u0027t really know what I\u0027m doing but have copied the bindep.txt from nova see https://review.opendev.org/667488","accounts_in_message":[],"_revision_number":1},{"id":"142388e07ae9af9f43031fe1f95c69343d6f4379","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-26 02:03:02.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e Hi Sam, I\u0027m wondering that are you already using Trove to\n \u003e deploy\n \u003e \u003e \u003e mysql 8?\n \u003e \u003e\n \u003e \u003e Yes I have mysql8 and postgres 11.3 working under a bionic image.\n \u003e \u003e Still a couple more patches required to push upstream\n \u003e \n \u003e Are you using the elements for disk-image-create command to build\n \u003e the mysql 8 and postgres 11.3 guest image? Have you made any\n \u003e change?\n\nNo we have our own image build process, (we build more than just trove for our cloud) You can see it at https://github.com/NeCTAR-RC/nectar-images/tree/master/ansible/roles/trove\n\nHaven\u0027t looked into disk-image-create","accounts_in_message":[],"_revision_number":1},{"id":"92f4616ca55286f482bb93de7cb391b9e53e3ff6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-07-29 01:55:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c89e156c6e30642f0b1da960af01ec073897846a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 02:44:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/09/665909/1/check/openstack-tox-cover/81d05b0/ : RETRY_LIMIT in 2m 48s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/09/665909/1/check/openstack-tox-lower-constraints/8966b89/ : RETRY_LIMIT in 1m 30s\n- openstack-tox-pep8 http://logs.openstack.org/09/665909/1/check/openstack-tox-pep8/acf0af3/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/09/665909/1/check/openstack-tox-py27/67bcdb0/ : RETRY_LIMIT in 1m 45s\n- openstack-tox-py36 http://logs.openstack.org/09/665909/1/check/openstack-tox-py36/09ea234/ : RETRY_LIMIT in 1m 42s\n- openstack-tox-py37 http://logs.openstack.org/09/665909/1/check/openstack-tox-py37/911f316/ : RETRY_LIMIT in 1m 45s\n- openstack-tox-docs http://logs.openstack.org/09/665909/1/check/openstack-tox-docs/0791080/html/ : SUCCESS in 4m 26s\n- trove-functional-mysql http://logs.openstack.org/09/665909/1/check/trove-functional-mysql/46cc753/ : FAILURE in 44m 33s\n- trove-scenario-mariadb-single http://logs.openstack.org/09/665909/1/check/trove-scenario-mariadb-single/149097e/ : FAILURE in 33m 57s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-mariadb-multi/236db6d/ : FAILURE in 39m 04s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/09/665909/1/check/trove-scenario-mysql-single/f603d66/ : FAILURE in 37m 03s\n- trove-scenario-mysql-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-mysql-multi/7a073a7/ : FAILURE in 45m 28s\n- trove-scenario-postgresql-single http://logs.openstack.org/09/665909/1/check/trove-scenario-postgresql-single/21ff6f1/ : FAILURE in 28m 28s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-postgresql-multi/8153a13/ : FAILURE in 37m 22s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/09/665909/1/check/trove-scenario-pxc-single/c1b6452/ : FAILURE in 32m 46s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-pxc-multi/e66e9e2/ : FAILURE in 30m 41s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/09/665909/1/check/trove-scenario-redis-single/4545be4/ : FAILURE in 33m 08s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/09/665909/1/check/trove-scenario-redis-multi/cf53886/ : FAILURE in 33m 55s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/09/665909/1/check/trove-tox-fakemodetests/2776cf0/ : RETRY_LIMIT in 1m 34s\n- trove-tox-bandit-baseline http://logs.openstack.org/09/665909/1/check/trove-tox-bandit-baseline/50fd3d4/ : RETRY_LIMIT in 1m 59s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/09/665909/1/check/openstack-tox-pylint/d531efa/ : RETRY_LIMIT in 1m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d2da9fb2cdf0d5a2aac80637a02fa3a64d90ffa0","author":{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},"date":"2019-07-29 03:04:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"92ee3f178582572204f16f3b1e8f49546f68784c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-29 04:52:13.000000000","message":"Patch Set 1:\n\nLingxian, you need bindep.txt before this can merge.\n\nLet\u0027s merge https://review.opendev.org/#/c/667488/ first...","accounts_in_message":[],"_revision_number":1},{"id":"1176d6d242cef637c52603f8d4c48b3efcf566d0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-07-29 07:32:27.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\n\u003e Lingxian, you need bindep.txt before this can merge.\n \u003e \n \u003e Let\u0027s merge https://review.opendev.org/#/c/667488/ first...\n\nThanks for pointing that out, Andreas, that saves a lot of time","accounts_in_message":[],"_revision_number":1},{"id":"d3e0fb7c83d7009f59122aaa9c9a1def342165cb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-14 22:10:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"236c1e82b1136656ff2d2bcb265ae02a4b803cf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 00:49:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65032a19fe13491ab3b534c13a177d8b : SUCCESS in 20m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fcc336778f84bfeac1697c586e18550 : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdca9793babb462fb1b6e03e22d03917 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed62507fbb6648faa0ebd0224872bfcb : SUCCESS in 10m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c625a72d48024edcae43f0218935e8ed : SUCCESS in 13m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83c72ac76ad444d7b77be26246efbee0 : SUCCESS in 17m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec51568bf0264c24a9bbf5b29f90cf7b : SUCCESS in 5m 24s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/51c4db2a9ed34b4d92f015807652ef75 : FAILURE in 33m 13s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/fd6ba0fb41114f33b368df3ab6d77230 : FAILURE in 41m 27s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/e868cdcfe7e0425396b4dd25430bdeb5 : FAILURE in 45m 32s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/507b5d9cc7cb4b7899859bf13cfa967a : SUCCESS in 1h 54m 30s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a839b323e375443784b2f8757a43915b : SUCCESS in 1h 54m 46s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/f2482512412349b68242d743d90cc3e2 : FAILURE in 39m 38s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ca2ed37fd5bb4dc8887d1da8438aa928 : FAILURE in 38m 03s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/ba42d17f6abd46de8f60fd606aefa92a : FAILURE in 1h 12m 48s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/d3ff53488bd7488ea6b5810de02f9928 : FAILURE in 43m 51s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/425db87cd1a146e38b0eac9624da1e33 : FAILURE in 1h 10m 52s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/92cb000a818f414b941429cb4accab7a : SUCCESS in 1h 21m 27s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/0787a7c277b545d9b2ea0c32b433b1c8 : SUCCESS in 6m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6c0fb36b4cb04ef59159221fccc8aef8 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/09af62b75ceb43dd8db52cda9b9f6ae6 : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":2},{"id":"2c1d7a43184c2780a8470bc413834f1fe6e255c2","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-08-15 06:22:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"85f50cdbab6c997779a800c9c8badefb1caa3b5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 09:10:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cee11f326124e3fb35b6daaae2a5e88 : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a64f842abd27401dae55b4e882f8260a : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4693eb09a86f462a98f924a15f16be16 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5921251ddda84a1e89634ced326311f5 : SUCCESS in 10m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a57330ae24ad49e096d8f16bdc78fb6c : SUCCESS in 12m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6608ca2b48fa48739570f98425607f8a : SUCCESS in 17m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728e7ea6080c444f8de73f41d41f1c19 : SUCCESS in 4m 34s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/51dbe50e2570413cac977786bb845136 : SUCCESS in 2h 07m 16s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/231b6ffcefe74325b6784dcb0a28636e : FAILURE in 49m 10s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/b3b34b789e344ce4a7d2f8b2844c12d9 : FAILURE in 43m 00s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/cf824f8680b24deab015e2cbbfdb6f1d : FAILURE in 1h 58m 54s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1595c68d20ed4569837e5aa9602c4865 : SUCCESS in 1h 41m 47s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/9d99191f08f2406db36b29c9c4234fe0 : FAILURE in 41m 49s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5a3cffd2a51e40acb7fa6db1b6080f52 : FAILURE in 41m 01s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/3d5bcff6078f462899042d4a65886416 : FAILURE in 47m 44s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/4591b19657204551a4abb34ffe1580fc : FAILURE in 41m 44s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/71e35699e01e4e0593f14bd17505db74 : FAILURE in 1h 20m 01s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/76e9c0244d07415a9f0d39cbdfff0302 : SUCCESS in 1h 18m 44s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/948cc09f1a2b4be8aafa0317d0e5e4ed : SUCCESS in 5m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b03d5cd5562047e080a0cb3b0b6c999a : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/759fcd2efc7f41f59944cf4a4b598060 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d52a25914179d30772ec3f3fc60f1d4ceef816eb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-24 10:25:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"475e29eed97eca11c7c43d38d3f1c78e78e5ccae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 12:44:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a5a6db5d042455ca46a37ba1e4f9776 : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9968e5b6ecf7418b8c6550fc1633396b : SUCCESS in 14m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e47b7a8e0e5b44e1b6fb68a509d7005c : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2007c19402294c96ad2c8ae3502bc535 : SUCCESS in 11m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abf5dd16b3ad4a4c82bf991835f319e0 : SUCCESS in 13m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f936df30b264c47bb130779a44e4e41 : SUCCESS in 18m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673254f562984d60943e599d120e186a : SUCCESS in 5m 37s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ee33f411be58491285b1b6ce8e2f4853 : SUCCESS in 2h 13m 19s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/c415138467bc4dc1998020998eef67a8 : FAILURE in 40m 03s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7cc6f166a21647b2bb2a6787b6fe75cd : FAILURE in 40m 21s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/fad7cf39b6a84b3db1751d35bf17f4ad : SUCCESS in 1h 43m 34s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e69d36c7d50c4a14a969505356b2120b : SUCCESS in 1h 36m 16s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/ca480dc077e04d84a2c300a045e37f13 : FAILURE in 41m 46s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/181cf7dd79d14a3cb6c0ec38eee016d8 : FAILURE in 40m 27s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/bf72aeef393f482ebfe6e310049423a0 : FAILURE in 42m 49s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/563f21bd34254bdc967f8cca8378b480 : FAILURE in 41m 47s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/bd037d423f09455ca266f10eaed9f6ec : FAILURE in 1h 12m 09s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/9c58c1dfceab4fada7f78ec855766290 : SUCCESS in 1h 15m 28s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/861e29db65ea4151ac9bee55be81627f : SUCCESS in 6m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d389b83f1a224de1bad05f50ef19c807 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ddfa3dbd2554e308eb21d658f349b57 : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":2},{"id":"743cfc20b28e676e5f285d0b07e30dff7b965264","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-24 22:03:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ffb3b1dff843c600b174227207965427ea6d9513","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 22:03:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f7e39da375cfbdd2be76bc4b9494e6204d495d10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 23:59:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5aa65c887d084442ba563b5d0bc0d536 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260f8a5580404b3b8b8251a59ac4c72d : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fbd4f13397f4c68b8ec9fe3b2f9856d : SUCCESS in 10m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3a1ecc5cc5e4ad5b973ebdae470883b : SUCCESS in 13m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a865fb871029433e86ba4a8c465a9b7e : SUCCESS in 15m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273d52c47bde41aeaa40f6ec7195f87b : SUCCESS in 7m 52s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ea07869e6f0e4b90ae26c2fa1ed5f35e : SUCCESS in 1h 52m 46s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/dbd45d8cc1614738a1ef7ed4b7cb951c : SUCCESS in 1h 48m 06s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/5dce93c6bb7d40c39323ad96a7503bc7 : SUCCESS in 1h 36m 31s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/e4e7bf63b3e7466da5a2db83a46dbfa8 : SUCCESS in 6m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5effbb4d43124118a23133722d2ae409 : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c26fbeade6c91a3d6db3365f4c1ceb338e5eb3dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 23:59:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8edea7d2c52285e12cb1884d3016ff337fe4bf54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 00:01:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28d0d4fa707e40f1b96ff0b556323000 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"73fc73a346027d9341424207bd0062a0cc53531a","revisions":{"998d84e354f2a72482cae4280290f2f17bf383df":{"kind":"REWORK","_number":1,"created":"2019-06-18 08:34:13.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/09/665909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/665909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/665909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/665909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/665909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/665909/1"}}},"commit":{"parents":[{"commit":"49a11dd9c7f8537e0a38f13733962533523050f1","subject":"trovestack guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49a11dd9c7f8537e0a38f13733962533523050f1"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:32:33.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:33:58.000000000","tz":600},"subject":"Support new mysql8.0 utf8mb4 character and collation sets","message":"Support new mysql8.0 utf8mb4 character and collation sets\n\nhttps://dev.mysql.com/doc/refman/8.0/en/charset-unicode-sets.html\nhttps://dev.mysql.com/doc/refman/8.0/en/charset-collation-names.html\n\nChange-Id: I5080de7b4b190053f1bcc1c4c815992c9a5889b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/998d84e354f2a72482cae4280290f2f17bf383df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/998d84e354f2a72482cae4280290f2f17bf383df"}]},"branch":"refs/heads/master"},"73fc73a346027d9341424207bd0062a0cc53531a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-14 22:10:24.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/09/665909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/09/665909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/665909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/665909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/09/665909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/09/665909/2"}}},"commit":{"parents":[{"commit":"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1","subject":"Fix Trove CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:32:33.000000000","tz":600},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-14 22:10:24.000000000","tz":0},"subject":"Support new mysql8.0 utf8mb4 character and collation sets","message":"Support new mysql8.0 utf8mb4 character and collation sets\n\nhttps://dev.mysql.com/doc/refman/8.0/en/charset-unicode-sets.html\nhttps://dev.mysql.com/doc/refman/8.0/en/charset-collation-names.html\n\nChange-Id: I5080de7b4b190053f1bcc1c4c815992c9a5889b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73fc73a346027d9341424207bd0062a0cc53531a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73fc73a346027d9341424207bd0062a0cc53531a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
