)]}'
{"id":"openstack%2Ftrove~665911","triplet_id":"openstack%2Ftrove~master~I0387238ec434073f95dd76f13f1f40c7cec8d1eb","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I0387238ec434073f95dd76f13f1f40c7cec8d1eb","subject":"Use newer style mysql syntax for users/passwords","status":"MERGED","created":"2019-06-18 08:38:26.000000000","updated":"2019-06-24 09:16:11.000000000","submitted":"2019-06-24 09:16:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665911-1561367771429-a6cfe23d","meta_rev_id":"ece8329b15f1f5fdee0a958e65672ee2dfe6359d","_number":665911,"virtual_id_number":665911,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-24 09:16:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-24 06:23:21.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-24 06:23:21.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 06:23:21.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-06-24 09:16:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5894e0b0eb965079db0444f80f1041aeb924fc94","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-18 08:38:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4676996c6f757b2f16c2942708fcd515cfa9ac08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 10:43:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/11/665911/1/check/openstack-tox-cover/9d0688c/cover/ : SUCCESS in 18m 05s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/11/665911/1/check/openstack-tox-lower-constraints/70cf757/ : SUCCESS in 13m 42s\n- openstack-tox-pep8 http://logs.openstack.org/11/665911/1/check/openstack-tox-pep8/52c8c92/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/11/665911/1/check/openstack-tox-py27/7c029ef/ : SUCCESS in 11m 47s\n- openstack-tox-py36 http://logs.openstack.org/11/665911/1/check/openstack-tox-py36/eec78a1/ : SUCCESS in 13m 13s\n- openstack-tox-py37 http://logs.openstack.org/11/665911/1/check/openstack-tox-py37/d0dfb53/ : SUCCESS in 18m 20s\n- openstack-tox-docs http://logs.openstack.org/11/665911/1/check/openstack-tox-docs/116e71f/html/ : SUCCESS in 5m 08s\n- trove-functional-mysql http://logs.openstack.org/11/665911/1/check/trove-functional-mysql/d109dca/ : FAILURE in 1h 32m 20s\n- trove-scenario-mariadb-single http://logs.openstack.org/11/665911/1/check/trove-scenario-mariadb-single/d49767d/ : FAILURE in 41m 40s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/11/665911/1/check/trove-scenario-mariadb-multi/c64c231/ : FAILURE in 41m 42s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/11/665911/1/check/trove-scenario-mysql-single/08b5035/ : FAILURE in 1h 43m 09s\n- trove-scenario-mysql-multi http://logs.openstack.org/11/665911/1/check/trove-scenario-mysql-multi/cee22ee/ : FAILURE in 1h 01m 47s\n- trove-scenario-postgresql-single http://logs.openstack.org/11/665911/1/check/trove-scenario-postgresql-single/b2649fd/ : FAILURE in 39m 24s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/11/665911/1/check/trove-scenario-postgresql-multi/d39fa55/ : FAILURE in 40m 14s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/11/665911/1/check/trove-scenario-pxc-single/a3bd0df/ : FAILURE in 41m 00s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/11/665911/1/check/trove-scenario-pxc-multi/47ea692/ : FAILURE in 40m 08s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/11/665911/1/check/trove-scenario-redis-single/b518eb4/ : FAILURE in 1h 05m 11s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/11/665911/1/check/trove-scenario-redis-multi/12f73bd/ : FAILURE in 54m 14s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/11/665911/1/check/trove-tox-fakemodetests/84b9ea7/ : SUCCESS in 5m 14s\n- trove-tox-bandit-baseline http://logs.openstack.org/11/665911/1/check/trove-tox-bandit-baseline/5e1dda1/ : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/11/665911/1/check/openstack-tox-pylint/e40ae24/ : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":1},{"id":"677645d41e0c75e129778fca123c7d2813b8c332","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-19 05:24:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"038c5c2a62ddbecd83fd97934fe52eb335254676","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-19 06:59:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dcfac429bab9496156cd61e9437566f88a391f54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 09:11:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/11/665911/3/check/openstack-tox-cover/2877687/cover/ : SUCCESS in 22m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/11/665911/3/check/openstack-tox-lower-constraints/0dc7d86/ : SUCCESS in 14m 20s\n- openstack-tox-pep8 http://logs.openstack.org/11/665911/3/check/openstack-tox-pep8/2613a6b/ : SUCCESS in 6m 02s\n- openstack-tox-py27 http://logs.openstack.org/11/665911/3/check/openstack-tox-py27/ff4d004/ : SUCCESS in 10m 58s\n- openstack-tox-py36 http://logs.openstack.org/11/665911/3/check/openstack-tox-py36/c4d1797/ : SUCCESS in 16m 03s\n- openstack-tox-py37 http://logs.openstack.org/11/665911/3/check/openstack-tox-py37/8f3906e/ : SUCCESS in 19m 07s\n- openstack-tox-docs http://logs.openstack.org/11/665911/3/check/openstack-tox-docs/e9f8762/html/ : SUCCESS in 5m 04s\n- trove-functional-mysql http://logs.openstack.org/11/665911/3/check/trove-functional-mysql/14b16d4/ : SUCCESS in 1h 58m 32s\n- trove-scenario-mariadb-single http://logs.openstack.org/11/665911/3/check/trove-scenario-mariadb-single/91ba811/ : FAILURE in 43m 30s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/11/665911/3/check/trove-scenario-mariadb-multi/9383220/ : FAILURE in 40m 55s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/11/665911/3/check/trove-scenario-mysql-single/05cd28f/ : SUCCESS in 1h 44m 20s\n- trove-scenario-mysql-multi http://logs.openstack.org/11/665911/3/check/trove-scenario-mysql-multi/be8cc16/ : SUCCESS in 1h 43m 39s\n- trove-scenario-postgresql-single http://logs.openstack.org/11/665911/3/check/trove-scenario-postgresql-single/abefd53/ : FAILURE in 39m 54s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/11/665911/3/check/trove-scenario-postgresql-multi/cec268c/ : FAILURE in 35m 36s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/11/665911/3/check/trove-scenario-pxc-single/a871035/ : FAILURE in 43m 36s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/11/665911/3/check/trove-scenario-pxc-multi/5550be2/ : FAILURE in 39m 38s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/11/665911/3/check/trove-scenario-redis-single/f27568a/ : FAILURE in 59m 49s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/11/665911/3/check/trove-scenario-redis-multi/6b2c5c6/ : FAILURE in 59m 55s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/11/665911/3/check/trove-tox-fakemodetests/304b9f4/ : SUCCESS in 6m 09s\n- trove-tox-bandit-baseline http://logs.openstack.org/11/665911/3/check/trove-tox-bandit-baseline/aa536a4/ : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/11/665911/3/check/openstack-tox-pylint/fcc7c46/ : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":3},{"id":"7142e15ed9c98a8864e936d85c10db2b38cd1d6a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-24 06:23:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"42de63027a5d4ae24395737c2d417a4570988f5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 06:23:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0cb2b56f1730ccbf9d872f7d0381c756b74fa965","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 09:16:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ece8329b15f1f5fdee0a958e65672ee2dfe6359d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 09:16:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/11/665911/3/gate/openstack-tox-lower-constraints/2b1cacb/ : SUCCESS in 12m 35s\n- openstack-tox-pep8 http://logs.openstack.org/11/665911/3/gate/openstack-tox-pep8/33d7faf/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/11/665911/3/gate/openstack-tox-py27/5a3242b/ : SUCCESS in 11m 05s\n- openstack-tox-py36 http://logs.openstack.org/11/665911/3/gate/openstack-tox-py36/8540949/ : SUCCESS in 12m 48s\n- openstack-tox-py37 http://logs.openstack.org/11/665911/3/gate/openstack-tox-py37/c527dc1/ : SUCCESS in 19m 15s\n- openstack-tox-docs http://logs.openstack.org/11/665911/3/gate/openstack-tox-docs/44d49a4/html/ : SUCCESS in 4m 59s\n- trove-functional-mysql http://logs.openstack.org/11/665911/3/gate/trove-functional-mysql/cd0d208/ : SUCCESS in 2h 46m 49s\n- trove-scenario-mysql-single http://logs.openstack.org/11/665911/3/gate/trove-scenario-mysql-single/66a8c75/ : SUCCESS in 2h 00m 43s\n- trove-scenario-mysql-multi http://logs.openstack.org/11/665911/3/gate/trove-scenario-mysql-multi/f90fd2e/ : SUCCESS in 1h 44m 56s\n- trove-tox-fakemodetests http://logs.openstack.org/11/665911/3/gate/trove-tox-fakemodetests/a3a6744/ : SUCCESS in 5m 37s\n- openstack-tox-pylint http://logs.openstack.org/11/665911/3/gate/openstack-tox-pylint/ed9d9c8/ : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8c951069fa807d6add266b6917aa0437cfa2c13f","revisions":{"56338d0ee75a0a5d4a03b4a629b3d51e621f418d":{"kind":"REWORK","_number":1,"created":"2019-06-18 08:38:26.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/11/665911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/665911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/665911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/665911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/665911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/665911/1"}}},"commit":{"parents":[{"commit":"49a11dd9c7f8537e0a38f13733962533523050f1","subject":"trovestack guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49a11dd9c7f8537e0a38f13733962533523050f1"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:34:32.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:38:15.000000000","tz":600},"subject":"Use newer style mysql syntax for users/passwords","message":"Use newer style mysql syntax for users/passwords\n\nThe sytax used for creating users and setting passwords\nwas deprecated in mysql 5.7 and is not valid in mysql 8.0. [1]\n\nChange to use CREATE USER to create users as opposed\nto creating users with GRANT syntax.\n\nSetting passwords also does not use the PASSWORD keyword\nanymore. [2]\n\n[1] https://dev.mysql.com/doc/refman/5.7/en/grant.html\n[2] https://dev.mysql.com/doc/refman/5.7/en/set-password.html\n\nChange-Id: I0387238ec434073f95dd76f13f1f40c7cec8d1eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56338d0ee75a0a5d4a03b4a629b3d51e621f418d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56338d0ee75a0a5d4a03b4a629b3d51e621f418d"}]},"branch":"refs/heads/master"},"8e7df8b4ecc72abfae833797f0b0582049fac2a1":{"kind":"REWORK","_number":2,"created":"2019-06-19 05:24:30.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/11/665911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/665911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/665911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/665911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/665911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/665911/2"}}},"commit":{"parents":[{"commit":"49a11dd9c7f8537e0a38f13733962533523050f1","subject":"trovestack guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49a11dd9c7f8537e0a38f13733962533523050f1"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:34:32.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-19 05:24:18.000000000","tz":600},"subject":"Use newer style mysql syntax for users/passwords","message":"Use newer style mysql syntax for users/passwords\n\nThe sytax used for creating users and setting passwords\nwas deprecated in mysql 5.7 and is not valid in mysql 8.0. [1]\n\nChange to use CREATE USER to create users as opposed\nto creating users with GRANT syntax.\n\nSetting passwords also does not use the PASSWORD keyword\nanymore. [2]\n\n[1] https://dev.mysql.com/doc/refman/5.7/en/grant.html\n[2] https://dev.mysql.com/doc/refman/5.7/en/set-password.html\n\nChange-Id: I0387238ec434073f95dd76f13f1f40c7cec8d1eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8e7df8b4ecc72abfae833797f0b0582049fac2a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8e7df8b4ecc72abfae833797f0b0582049fac2a1"}]},"branch":"refs/heads/master"},"8c951069fa807d6add266b6917aa0437cfa2c13f":{"kind":"REWORK","_number":3,"created":"2019-06-19 06:59:42.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/11/665911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/665911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/665911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/665911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/665911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/665911/3"}}},"commit":{"parents":[{"commit":"49a11dd9c7f8537e0a38f13733962533523050f1","subject":"trovestack guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49a11dd9c7f8537e0a38f13733962533523050f1"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-18 08:34:32.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-19 06:59:28.000000000","tz":600},"subject":"Use newer style mysql syntax for users/passwords","message":"Use newer style mysql syntax for users/passwords\n\nThe sytax used for creating users and setting passwords\nwas deprecated in mysql 5.7 and is not valid in mysql 8.0. [1]\n\nChange to use CREATE USER to create users as opposed\nto creating users with GRANT syntax.\n\nSetting passwords also does not use the PASSWORD keyword\nanymore. [2]\n\n[1] https://dev.mysql.com/doc/refman/5.7/en/grant.html\n[2] https://dev.mysql.com/doc/refman/5.7/en/set-password.html\n\nChange-Id: I0387238ec434073f95dd76f13f1f40c7cec8d1eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c951069fa807d6add266b6917aa0437cfa2c13f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c951069fa807d6add266b6917aa0437cfa2c13f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
