)]}'
{"id":"openstack%2Ftrove~668828","triplet_id":"openstack%2Ftrove~master~I84aa5a39c9a9d9c410cc780ddc162768d37cab32","project":"openstack/trove","branch":"master","topic":"assertTrue","hashtags":[],"change_id":"I84aa5a39c9a9d9c410cc780ddc162768d37cab32","subject":"Remove invalid assert state","status":"MERGED","created":"2019-07-03 08:44:09.000000000","updated":"2019-08-14 23:14:11.000000000","submitted":"2019-08-14 23:14:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668828-1565824451912-92a0a85e","meta_rev_id":"e3714d4d7d6362f9c63a4980660e711b97f44f35","_number":668828,"virtual_id_number":668828,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-08-14 23:14:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-08-14 22:51:22.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-08-14 22:51:22.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 22:51:22.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-08-14 23:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5599cda5da3310470ecdb288cb2a385224aae7c5","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-07-03 08:44:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10de5a97bb87128e95f286877ea1b620df434a97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 09:14:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/28/668828/1/check/openstack-tox-cover/5216272/ : RETRY_LIMIT in 2m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/28/668828/1/check/openstack-tox-lower-constraints/ebebfe5/ : RETRY_LIMIT in 2m 16s\n- openstack-tox-pep8 http://logs.openstack.org/28/668828/1/check/openstack-tox-pep8/924ca90/ : RETRY_LIMIT in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/28/668828/1/check/openstack-tox-py27/20407e1/ : RETRY_LIMIT in 2m 52s\n- openstack-tox-py36 http://logs.openstack.org/28/668828/1/check/openstack-tox-py36/0ec63e4/ : RETRY_LIMIT in 2m 35s\n- openstack-tox-py37 http://logs.openstack.org/28/668828/1/check/openstack-tox-py37/aa660af/ : RETRY_LIMIT in 2m 26s\n- openstack-tox-docs http://logs.openstack.org/28/668828/1/check/openstack-tox-docs/05e44bb/html/ : SUCCESS in 3m 50s\n- openstack-tox-pylint http://logs.openstack.org/28/668828/1/check/openstack-tox-pylint/7a28dca/ : RETRY_LIMIT in 1m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8f9252168e5cdf8d6d50a37abaeffcf156d4a17f","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-07-04 01:19:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a897894ee0f63bd11d9fb844d97e5823e1b4d990","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 01:49:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/28/668828/1/check/openstack-tox-cover/4c31b2a/ : RETRY_LIMIT in 3m 13s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/28/668828/1/check/openstack-tox-lower-constraints/d0fc87a/ : RETRY_LIMIT in 2m 37s\n- openstack-tox-pep8 http://logs.openstack.org/28/668828/1/check/openstack-tox-pep8/20f484c/ : RETRY_LIMIT in 2m 07s\n- openstack-tox-py27 http://logs.openstack.org/28/668828/1/check/openstack-tox-py27/7a39f55/ : RETRY_LIMIT in 2m 07s\n- openstack-tox-py36 http://logs.openstack.org/28/668828/1/check/openstack-tox-py36/09d7df4/ : RETRY_LIMIT in 1m 46s\n- openstack-tox-py37 http://logs.openstack.org/28/668828/1/check/openstack-tox-py37/e967034/ : RETRY_LIMIT in 2m 00s\n- openstack-tox-docs http://logs.openstack.org/28/668828/1/check/openstack-tox-docs/40cdf17/html/ : SUCCESS in 3m 34s\n- openstack-tox-pylint http://logs.openstack.org/28/668828/1/check/openstack-tox-pylint/e94d315/ : RETRY_LIMIT in 2m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6a7f1286ee7cdeaaf5453b282536035248228039","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-07-31 23:53:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62d31454aabd2a4ce0f713c571738e41b124476a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 00:19:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/28/668828/1/check/openstack-tox-cover/e56062e/ : RETRY_LIMIT in 2m 43s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/28/668828/1/check/openstack-tox-lower-constraints/f0fc561/ : RETRY_LIMIT in 2m 49s\n- openstack-tox-pep8 https://logs.opendev.org/28/668828/1/check/openstack-tox-pep8/8bc309e/ : SUCCESS in 4m 23s\n- openstack-tox-py27 https://logs.opendev.org/28/668828/1/check/openstack-tox-py27/ec2178a/ : RETRY_LIMIT in 2m 24s\n- openstack-tox-py36 https://logs.opendev.org/28/668828/1/check/openstack-tox-py36/776ee8b/ : RETRY_LIMIT in 2m 07s\n- openstack-tox-py37 https://logs.opendev.org/28/668828/1/check/openstack-tox-py37/37073d2/ : RETRY_LIMIT in 2m 23s\n- openstack-tox-docs https://logs.opendev.org/28/668828/1/check/openstack-tox-docs/6346e2b/html/ : SUCCESS in 6m 36s\n- openstack-tox-pylint https://logs.opendev.org/28/668828/1/check/openstack-tox-pylint/d02123d/ : RETRY_LIMIT in 1m 59s","accounts_in_message":[],"_revision_number":1},{"id":"cbc5a2a91697cedea4515ceaa2d77e520e08e407","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-08-07 07:18:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a64f380fbc08268754712ad0114bf637a9c8ad19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 07:46:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/28/668828/1/check/openstack-tox-cover/63ca658/ : RETRY_LIMIT in 2m 41s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/28/668828/1/check/openstack-tox-lower-constraints/a74878f/ : RETRY_LIMIT in 2m 11s\n- openstack-tox-pep8 https://logs.opendev.org/28/668828/1/check/openstack-tox-pep8/ce60d2b/ : SUCCESS in 4m 55s\n- openstack-tox-py27 https://logs.opendev.org/28/668828/1/check/openstack-tox-py27/a2c3843/ : RETRY_LIMIT in 2m 50s\n- openstack-tox-py36 https://logs.opendev.org/28/668828/1/check/openstack-tox-py36/23a65b7/ : RETRY_LIMIT in 2m 29s\n- openstack-tox-py37 https://logs.opendev.org/28/668828/1/check/openstack-tox-py37/105069a/ : RETRY_LIMIT in 2m 15s\n- openstack-tox-docs https://logs.opendev.org/28/668828/1/check/openstack-tox-docs/a2ad962/html/ : SUCCESS in 4m 41s\n- openstack-tox-pylint https://logs.opendev.org/28/668828/1/check/openstack-tox-pylint/19437d7/ : RETRY_LIMIT in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cc1847b74efde36ca55a9de9ead93cd55e6e5e03","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-14 22:09:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1469b088cabbf624cf93df20e662d08645b29c70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 22:38:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea03770b4cc3456bac89a9e0e07564bc : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7809c62f449a4242bae6d714fd685ddc : SUCCESS in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0698092789664ed1a0452b17d07c1bc2 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a74833dc4f404409b994573d459ec1a8 : SUCCESS in 10m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc06d4ea21414ac59f84771015453bb2 : SUCCESS in 12m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcd1df5a61bf4fcbaddb0c68f068bd2e : SUCCESS in 16m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed099044a019419ebc5aae0d0b93aa04 : SUCCESS in 4m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff5d72b898554349ac7bcd77058d3b44 : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":2},{"id":"8094c6b91cec5bd92ff85b3a467aa3be85a86874","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-14 22:51:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"114cfbd90de543b46023e9d1cd99095da74a5edc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 22:51:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"91b1b2cd026f1dc1431c4d169967aa51809249fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 23:14:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be69cec9b1b24967b1791618a278c054 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/911be0879994420ebbcf03943e678fe7 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2b0bcba122148cca4fabc3dab3eb5d5 : SUCCESS in 10m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c97e06e6653a4587926a72a3a1105bba : SUCCESS in 11m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c099489d5f1d4044a00dca2dc33331ab : SUCCESS in 17m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d454565f8df4c1f99e5398ad2b99135 : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98a612a9c75047e3a37d7e56bd3224b1 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":2},{"id":"8e919a13a35cc2a4e74fbd30295dd12d9f758e39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 23:14:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69e6652c92ce95a40d34e7fb181b9ac5a5f55428","revisions":{"547949a0480d0e295b89e677ec59c2df3f301020":{"kind":"REWORK","_number":1,"created":"2019-07-03 08:44:09.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/28/668828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/28/668828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/668828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/668828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/668828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/28/668828/1"}}},"commit":{"parents":[{"commit":"f72015179ea921c8264a77390359d638e0bd8727","subject":"Merge \"Update the outdated content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f72015179ea921c8264a77390359d638e0bd8727"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-07-03 08:38:14.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-07-03 08:38:14.000000000","tz":480},"subject":"Remove invalid assert state","message":"Remove invalid assert state\n\nThis is to remove invalid assert state of self.assertTrue(\u0027cluster\u0027).\n\nChange-Id: I84aa5a39c9a9d9c410cc780ddc162768d37cab32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/547949a0480d0e295b89e677ec59c2df3f301020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/547949a0480d0e295b89e677ec59c2df3f301020"}]},"branch":"refs/heads/master"},"69e6652c92ce95a40d34e7fb181b9ac5a5f55428":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-14 22:09:13.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/28/668828/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/28/668828/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/668828/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/668828/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/668828/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/28/668828/2"}}},"commit":{"parents":[{"commit":"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1","subject":"Fix Trove CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-07-03 08:38:14.000000000","tz":480},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-14 22:09:13.000000000","tz":0},"subject":"Remove invalid assert state","message":"Remove invalid assert state\n\nThis is to remove invalid assert state of self.assertTrue(\u0027cluster\u0027).\n\nChange-Id: I84aa5a39c9a9d9c410cc780ddc162768d37cab32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69e6652c92ce95a40d34e7fb181b9ac5a5f55428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69e6652c92ce95a40d34e7fb181b9ac5a5f55428"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
