)]}'
{"id":"openstack%2Ftrove~669522","triplet_id":"openstack%2Ftrove~master~I1fe23c6115bd6b9aac0bf2bd4fec9c6eaee88c40","project":"openstack/trove","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I1fe23c6115bd6b9aac0bf2bd4fec9c6eaee88c40","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 20:34:55.000000000","updated":"2019-08-15 06:13:20.000000000","submitted":"2019-08-15 06:13:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669522-1565849600854-d4deb22e","meta_rev_id":"eb5ef5f2b5ba280f91cd41cc402265c98841dbbc","_number":669522,"virtual_id_number":669522,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-08-15 06:13:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-08-15 01:21:59.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-08-15 01:21:59.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 01:21:59.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-08-15 06:13:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea6ea37f8626449bfa5ea79b605969e11283d65f","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 20:34:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dac51e3fde7cfff0ea10aea496db2224fabf2f70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:25:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/22/669522/1/check/build-openstack-api-ref/f09f41c/html/ : SUCCESS in 3m 25s\n- openstack-tox-cover http://logs.openstack.org/22/669522/1/check/openstack-tox-cover/633cc92/ : RETRY_LIMIT in 2m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/22/669522/1/check/openstack-tox-lower-constraints/4dfbae0/ : RETRY_LIMIT in 2m 46s\n- openstack-tox-pep8 http://logs.openstack.org/22/669522/1/check/openstack-tox-pep8/7e50368/ : RETRY_LIMIT in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/22/669522/1/check/openstack-tox-py27/54eccc8/ : RETRY_LIMIT in 2m 11s\n- openstack-tox-py36 http://logs.openstack.org/22/669522/1/check/openstack-tox-py36/0c8fd9f/ : RETRY_LIMIT in 2m 05s\n- openstack-tox-py37 http://logs.openstack.org/22/669522/1/check/openstack-tox-py37/211542e/ : RETRY_LIMIT in 1m 47s\n- openstack-tox-docs http://logs.openstack.org/22/669522/1/check/openstack-tox-docs/0559e66/html/ : SUCCESS in 3m 31s\n- build-openstack-releasenotes http://logs.openstack.org/22/669522/1/check/build-openstack-releasenotes/39c348f/html/ : SUCCESS in 5m 26s\n- trove-functional-mysql http://logs.openstack.org/22/669522/1/check/trove-functional-mysql/5775be2/ : FAILURE in 42m 31s\n- trove-scenario-mariadb-single http://logs.openstack.org/22/669522/1/check/trove-scenario-mariadb-single/cded33b/ : FAILURE in 31m 57s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/22/669522/1/check/trove-scenario-mariadb-multi/d72d5ee/ : FAILURE in 41m 46s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/22/669522/1/check/trove-scenario-mysql-single/6f5b716/ : FAILURE in 40m 04s\n- trove-scenario-mysql-multi http://logs.openstack.org/22/669522/1/check/trove-scenario-mysql-multi/c6c2d56/ : FAILURE in 41m 30s\n- trove-scenario-postgresql-single http://logs.openstack.org/22/669522/1/check/trove-scenario-postgresql-single/e010edd/ : FAILURE in 28m 16s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/22/669522/1/check/trove-scenario-postgresql-multi/107a035/ : FAILURE in 37m 31s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/22/669522/1/check/trove-scenario-pxc-single/deb2e18/ : FAILURE in 33m 53s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/22/669522/1/check/trove-scenario-pxc-multi/86eadba/ : FAILURE in 29m 22s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/22/669522/1/check/trove-scenario-redis-single/61aa5c9/ : FAILURE in 29m 49s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/22/669522/1/check/trove-scenario-redis-multi/11d0665/ : FAILURE in 29m 15s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/22/669522/1/check/trove-tox-fakemodetests/182c5f6/ : RETRY_LIMIT in 1m 38s\n- trove-tox-bandit-baseline http://logs.openstack.org/22/669522/1/check/trove-tox-bandit-baseline/9039453/ : RETRY_LIMIT in 1m 57s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/22/669522/1/check/openstack-tox-pylint/c947fde/ : RETRY_LIMIT in 2m 18s","accounts_in_message":[],"_revision_number":1},{"id":"75115c58fc7beec1fef11096a3f49428572d9ea5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-07-31 23:53:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"66db3b84863d1775433b27693c6b5f37b1124c9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 03:08:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/22/669522/2/check/build-openstack-api-ref/cf914e0/html/ : SUCCESS in 4m 36s\n- openstack-tox-cover https://logs.opendev.org/22/669522/2/check/openstack-tox-cover/977e461/cover/ : SUCCESS in 17m 44s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/22/669522/2/check/openstack-tox-lower-constraints/e8f33c1/ : SUCCESS in 15m 20s\n- openstack-tox-pep8 https://logs.opendev.org/22/669522/2/check/openstack-tox-pep8/f4ba614/ : SUCCESS in 5m 29s\n- openstack-tox-py27 https://logs.opendev.org/22/669522/2/check/openstack-tox-py27/e16e4ec/ : SUCCESS in 14m 55s\n- openstack-tox-py36 https://logs.opendev.org/22/669522/2/check/openstack-tox-py36/fe7e8f5/ : SUCCESS in 12m 12s\n- openstack-tox-py37 https://logs.opendev.org/22/669522/2/check/openstack-tox-py37/3e8b5a8/ : SUCCESS in 16m 30s\n- openstack-tox-docs https://logs.opendev.org/22/669522/2/check/openstack-tox-docs/ea60205/html/ : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://logs.opendev.org/22/669522/2/check/build-openstack-releasenotes/9d33a4c/html/ : SUCCESS in 6m 15s\n- trove-functional-mysql https://logs.opendev.org/22/669522/2/check/trove-functional-mysql/76fd587/ : SUCCESS in 1h 48m 03s\n- trove-scenario-mariadb-single https://logs.opendev.org/22/669522/2/check/trove-scenario-mariadb-single/0f4478c/ : FAILURE in 33m 58s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/22/669522/2/check/trove-scenario-mariadb-multi/a9df585/ : FAILURE in 30m 07s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/22/669522/2/check/trove-scenario-mysql-single/b519ec4/ : TIMED_OUT in 3h 03m 34s\n- trove-scenario-mysql-multi https://logs.opendev.org/22/669522/2/check/trove-scenario-mysql-multi/f14c553/ : FAILURE in 52m 07s\n- trove-scenario-postgresql-single https://logs.opendev.org/22/669522/2/check/trove-scenario-postgresql-single/79a66f7/ : FAILURE in 39m 08s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/22/669522/2/check/trove-scenario-postgresql-multi/117f302/ : FAILURE in 32m 42s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/22/669522/2/check/trove-scenario-pxc-single/a1f2843/ : FAILURE in 42m 27s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/22/669522/2/check/trove-scenario-pxc-multi/3333497/ : FAILURE in 34m 16s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/22/669522/2/check/trove-scenario-redis-single/0de97ea/ : FAILURE in 33m 13s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/22/669522/2/check/trove-scenario-redis-multi/4faeb00/ : FAILURE in 35m 27s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/22/669522/2/check/trove-tox-fakemodetests/9af8af7/ : SUCCESS in 6m 01s\n- trove-tox-bandit-baseline https://logs.opendev.org/22/669522/2/check/trove-tox-bandit-baseline/0a434ba/ : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/22/669522/2/check/openstack-tox-pylint/4dda75a/ : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":2},{"id":"edb42594beb4cb347d7cc069906f8aaf5bb23041","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-08-09 14:27:47.000000000","message":"Patch Set 2:\n\n@Lingxian do you know why this is failing?","accounts_in_message":[],"_revision_number":2},{"id":"255464bd15c463a5b0dc51147f8d51abba1ee883","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-10 08:24:17.000000000","message":"Patch Set 2:\n\n\u003e @Lingxian do you know why this is failing?\n\nI\u0027m working on the fix, I will rebase all the ready patch after CI issue is solved.","accounts_in_message":[],"_revision_number":2},{"id":"e429affc53a8d0dcdeeea45d7ff40c776328c286","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-14 21:50:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f0775bb46a6b59ed18d6f1e057df898b2a649692","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 00:07:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1eebb5c50b464e8994e3434d5d34d61a : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92ad75fe8d874752bb8b7f701fdafce7 : SUCCESS in 19m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/094ce7dc8edd49dcb9199c4d912d1264 : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76fe163c2e2f478697c7d1219083dd51 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04cc4a95930442f998a1a0ac7e8344b0 : SUCCESS in 11m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a72e777c18774108a85c556b73d7a134 : SUCCESS in 13m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/011a14fc0d2c478fa8df35e6e4f1f71d : SUCCESS in 19m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5afd733090f4b72acbec6813264a50a : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/876d957fc67045dba610afa00af8379c : SUCCESS in 5m 56s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/dff6f77f177f40b0a97f61b6ab6636d6 : SUCCESS in 2h 07m 18s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/a3536fa314a44ed5804fc68c834725a5 : FAILURE in 42m 23s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/9cd8f22feb6449e48f731534fac49b6d : FAILURE in 40m 34s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6314c89b259143e8afe87566f55c4ab3 : SUCCESS in 1h 56m 48s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d7b262fd8d15464483036b35bc3aeb51 : SUCCESS in 1h 48m 30s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/09c1c78a5d2641d29646047d39241b28 : FAILURE in 39m 05s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/3e734825b4a34dd6a6a41ce1fd56a185 : FAILURE in 43m 07s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/1835bef346b446ddba5014c911595a78 : FAILURE in 1h 04m 35s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/7256d31c5224411389eee6ca21d8774d : FAILURE in 35m 36s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/1f0ae2fff36a4042988ae00647aea2ce : FAILURE in 1h 25m 01s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/b2fa3d0441ec42d0b6a77176a342c4a2 : SUCCESS in 1h 11m 00s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/39e7caeb7fe84487be5d6f35282d6569 : SUCCESS in 6m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d461f91aaaa4b379f33958464fc245c : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a2c951779bdc498081d580dca194d71d : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":3},{"id":"9f3524b4fa865d46c60a4564710bd22473eb36d4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-15 01:21:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9812d46f21d46281633b6b33324f074796d1fbeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 01:22:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"06a29920fb647694ffed4acfbeebcfc4ad38c580","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 06:13:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6d651285b58fc8c8f8082684a33c73ec91ece2a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 06:13:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e6b44fb8a14747839c4a7336f8beb6db : SUCCESS in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae06761deffd46d4a9f166cad6347c93 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9bcebdf0d2f49cd95ccc55f973d8d73 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4842e7a2e1f64575b697ca9324da0ead : SUCCESS in 11m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bed58e833d684a76968c4a379c05f6f8 : SUCCESS in 11m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbccc44c2107410581ce9f64a16c1c72 : SUCCESS in 16m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848e8d2f755b41de975c38c4f5cbad48 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dba57fe7201416f8f24782492c8852a : SUCCESS in 5m 19s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/9819ad98c82141788e033c4c2fe8aca4 : SUCCESS in 2h 08m 54s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/71c97473dc614e79bf47181850a493e3 : SUCCESS in 1h 39m 49s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/b39acb4580e44878aed509f57c0ee12b : SUCCESS in 1h 36m 50s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/fe290c243a5444e98d457825dea1b7e2 : SUCCESS in 7m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/163052c7c92a42a38aeb175a771e69a7 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cc16d3b3bccfb8b5639c54330c846c5e72e3108e","revisions":{"426ec5f433723ef3119df4c0e877d78e151fb211":{"kind":"REWORK","_number":1,"created":"2019-07-05 20:34:55.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/22/669522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/22/669522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/669522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/669522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/669522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/22/669522/1"}}},"commit":{"parents":[{"commit":"f72015179ea921c8264a77390359d638e0bd8727","subject":"Merge \"Update the outdated content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f72015179ea921c8264a77390359d638e0bd8727"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:08:26.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 20:34:54.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I1fe23c6115bd6b9aac0bf2bd4fec9c6eaee88c40\nStory: #2005924\nTask: #34254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/426ec5f433723ef3119df4c0e877d78e151fb211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/426ec5f433723ef3119df4c0e877d78e151fb211"}]},"branch":"refs/heads/master"},"d96dd01ae9c9274315077c115a20e8f3a7996c42":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-31 23:53:05.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/22/669522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/22/669522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/669522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/669522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/669522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/22/669522/2"}}},"commit":{"parents":[{"commit":"6c011a69e0a331ea5c3dff4568e3e6ec9109acc2","subject":"Added a bindep file to install binaries in CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c011a69e0a331ea5c3dff4568e3e6ec9109acc2"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:08:26.000000000","tz":-240},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-07-31 23:53:05.000000000","tz":0},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I1fe23c6115bd6b9aac0bf2bd4fec9c6eaee88c40\nStory: #2005924\nTask: #34254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d96dd01ae9c9274315077c115a20e8f3a7996c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d96dd01ae9c9274315077c115a20e8f3a7996c42"}]},"branch":"refs/heads/master"},"cc16d3b3bccfb8b5639c54330c846c5e72e3108e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-14 21:50:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/22/669522/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/22/669522/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/669522/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/669522/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/669522/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/22/669522/3"}}},"commit":{"parents":[{"commit":"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1","subject":"Fix Trove CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:08:26.000000000","tz":-240},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-14 21:50:49.000000000","tz":720},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I1fe23c6115bd6b9aac0bf2bd4fec9c6eaee88c40\nStory: #2005924\nTask: #34254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc16d3b3bccfb8b5639c54330c846c5e72e3108e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc16d3b3bccfb8b5639c54330c846c5e72e3108e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
