)]}'
{"id":"openstack%2Ftrove~669877","triplet_id":"openstack%2Ftrove~master~I02ecdee3507dfaf949796e79f8919e1233efe6f0","project":"openstack/trove","branch":"master","topic":"do-not-merge","hashtags":[],"change_id":"I02ecdee3507dfaf949796e79f8919e1233efe6f0","subject":"DNM: bindep test","status":"ABANDONED","created":"2019-07-09 13:57:55.000000000","updated":"2019-07-09 15:01:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9d76badf31c74cb028afb70075c482991b72b64c","_number":669877,"virtual_id_number":669877,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-07-09 14:53:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 14:53:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1dfe6bcd091ed53ee720d49718b493250e730aed","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-09 13:57:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8756e3515059e564d2959cdc96146266cfe2caf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 14:53:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/77/669877/1/check/build-openstack-api-ref/628b67f/html/ : SUCCESS in 9m 45s\n- openstack-tox-cover http://logs.openstack.org/77/669877/1/check/openstack-tox-cover/8c8c76d/cover/ : SUCCESS in 15m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/77/669877/1/check/openstack-tox-lower-constraints/a494de3/ : SUCCESS in 13m 24s\n- openstack-tox-pep8 http://logs.openstack.org/77/669877/1/check/openstack-tox-pep8/1788f6f/ : SUCCESS in 7m 16s\n- openstack-tox-py27 http://logs.openstack.org/77/669877/1/check/openstack-tox-py27/1278bf0/ : SUCCESS in 12m 48s\n- openstack-tox-py36 http://logs.openstack.org/77/669877/1/check/openstack-tox-py36/85d52c2/ : SUCCESS in 13m 17s\n- openstack-tox-py37 http://logs.openstack.org/77/669877/1/check/openstack-tox-py37/65492b1/ : SUCCESS in 18m 20s\n- openstack-tox-docs http://logs.openstack.org/77/669877/1/check/openstack-tox-docs/4114c88/html/ : SUCCESS in 5m 39s\n- build-openstack-releasenotes http://logs.openstack.org/77/669877/1/check/build-openstack-releasenotes/d64ac0b/html/ : SUCCESS in 7m 03s\n- trove-functional-mysql http://logs.openstack.org/77/669877/1/check/trove-functional-mysql/32b21f4/ : FAILURE in 35m 06s\n- trove-scenario-mariadb-single http://logs.openstack.org/77/669877/1/check/trove-scenario-mariadb-single/b979a2c/ : FAILURE in 36m 58s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/77/669877/1/check/trove-scenario-mariadb-multi/4a06611/ : FAILURE in 43m 50s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/77/669877/1/check/trove-scenario-mysql-single/05bb046/ : FAILURE in 50m 02s\n- trove-scenario-mysql-multi http://logs.openstack.org/77/669877/1/check/trove-scenario-mysql-multi/3c511f1/ : FAILURE in 48m 13s\n- trove-scenario-postgresql-single http://logs.openstack.org/77/669877/1/check/trove-scenario-postgresql-single/1695074/ : FAILURE in 35m 23s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/77/669877/1/check/trove-scenario-postgresql-multi/fb3eda7/ : FAILURE in 35m 43s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/77/669877/1/check/trove-scenario-pxc-single/110f6fe/ : FAILURE in 37m 31s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/77/669877/1/check/trove-scenario-pxc-multi/f3fd624/ : FAILURE in 38m 36s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/77/669877/1/check/trove-scenario-redis-single/7fa09c8/ : FAILURE in 41m 21s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/77/669877/1/check/trove-scenario-redis-multi/c71bf71/ : FAILURE in 39m 47s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/77/669877/1/check/trove-tox-fakemodetests/73b10e5/ : SUCCESS in 5m 20s\n- openstack-tox-pylint http://logs.openstack.org/77/669877/1/check/openstack-tox-pylint/7368798/ : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d8f75c23ab59e7c9364bd0a7acae8705bad82445","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-09 15:01:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ef1a98c893cfdc1b420990ed777a1f5612cb728","revisions":{"6ef1a98c893cfdc1b420990ed777a1f5612cb728":{"kind":"REWORK","_number":1,"created":"2019-07-09 13:57:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/77/669877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/77/669877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/669877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/669877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/669877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/77/669877/1"}}},"commit":{"parents":[{"commit":"f72015179ea921c8264a77390359d638e0bd8727","subject":"Merge \"Update the outdated content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f72015179ea921c8264a77390359d638e0bd8727"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-09 13:52:40.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-09 13:52:40.000000000","tz":120},"subject":"DNM: bindep test","message":"DNM: bindep test\n\nChange-Id: I02ecdee3507dfaf949796e79f8919e1233efe6f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ef1a98c893cfdc1b420990ed777a1f5612cb728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ef1a98c893cfdc1b420990ed777a1f5612cb728"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
