)]}'
{"id":"openstack%2Ftrove~672796","triplet_id":"openstack%2Ftrove~master~Ie815d3a1f177af35ec1a2f7bb703739a5f0f9f3d","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"Ie815d3a1f177af35ec1a2f7bb703739a5f0f9f3d","subject":"WIP: run tox with stack user","status":"ABANDONED","created":"2019-07-25 17:11:02.000000000","updated":"2019-08-14 11:15:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9122e6d6fc2be3dc634507192bafedb744398a3e","_number":672796,"virtual_id_number":672796,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-08-07 14:04:17.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-1,"date":"2019-07-25 18:03:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 18:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-07 14:04:17.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"bc34159e6733cb5f90ebce6d21e4a906f8304f30","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-25 17:11:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a774a30e68b2e4dedbcf75093635d6c083d4cd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 18:03:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/96/672796/1/check/openstack-tox-cover/bebcfc4/cover/ : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/96/672796/1/check/openstack-tox-lower-constraints/e79e2e8/ : SUCCESS in 14m 15s\n- openstack-tox-pep8 http://logs.openstack.org/96/672796/1/check/openstack-tox-pep8/1e9fb3d/ : SUCCESS in 6m 33s\n- openstack-tox-py27 http://logs.openstack.org/96/672796/1/check/openstack-tox-py27/89a9ff7/ : SUCCESS in 11m 57s\n- openstack-tox-py36 http://logs.openstack.org/96/672796/1/check/openstack-tox-py36/4f0eaa9/ : SUCCESS in 14m 04s\n- openstack-tox-py37 http://logs.openstack.org/96/672796/1/check/openstack-tox-py37/8463b6e/ : SUCCESS in 19m 59s\n- openstack-tox-docs http://logs.openstack.org/96/672796/1/check/openstack-tox-docs/0c39f6a/html/ : SUCCESS in 6m 12s\n- trove-functional-mysql http://logs.openstack.org/96/672796/1/check/trove-functional-mysql/81b0ea3/ : FAILURE in 44m 58s\n- trove-scenario-mariadb-single http://logs.openstack.org/96/672796/1/check/trove-scenario-mariadb-single/e16fbbe/ : FAILURE in 34m 57s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/96/672796/1/check/trove-scenario-mariadb-multi/1946552/ : FAILURE in 36m 32s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/96/672796/1/check/trove-scenario-mysql-single/f1ecc4e/ : FAILURE in 45m 33s\n- trove-scenario-mysql-multi http://logs.openstack.org/96/672796/1/check/trove-scenario-mysql-multi/3329966/ : FAILURE in 41m 41s\n- trove-scenario-postgresql-single http://logs.openstack.org/96/672796/1/check/trove-scenario-postgresql-single/ad7a3b2/ : FAILURE in 36m 15s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/96/672796/1/check/trove-scenario-postgresql-multi/7d57f75/ : FAILURE in 39m 39s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/96/672796/1/check/trove-scenario-pxc-single/7fdb9cc/ : FAILURE in 39m 21s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/96/672796/1/check/trove-scenario-pxc-multi/13b939c/ : FAILURE in 34m 21s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/96/672796/1/check/trove-scenario-redis-single/c3d8212/ : FAILURE in 31m 45s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/96/672796/1/check/trove-scenario-redis-multi/86e62cd/ : FAILURE in 36m 24s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/96/672796/1/check/trove-tox-fakemodetests/e67df9d/ : SUCCESS in 7m 38s\n- trove-tox-bandit-baseline http://logs.openstack.org/96/672796/1/check/trove-tox-bandit-baseline/358a89b/ : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/96/672796/1/check/openstack-tox-pylint/40e35e0/ : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a85ca9395e16740be5a9db6310f280f5307a861b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 14:04:17.000000000","message":"Patch Set 1:\n\nI\u0027m working on a better way to do this, let\u0027s see whether I get it working:\n\nhttps://review.opendev.org/#/c/675098/","accounts_in_message":[],"_revision_number":1},{"id":"9122e6d6fc2be3dc634507192bafedb744398a3e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-14 11:15:30.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/#/c/675098/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63738549bb5ef46fe7e8ebf530addbe448920a97","revisions":{"63738549bb5ef46fe7e8ebf530addbe448920a97":{"kind":"REWORK","_number":1,"created":"2019-07-25 17:11:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/96/672796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/96/672796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/672796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/672796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/672796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/96/672796/1"}}},"commit":{"parents":[{"commit":"b27470bda202b7ebd7d080cc486328016248a631","subject":"Added a bindep file to install binaries in CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b27470bda202b7ebd7d080cc486328016248a631"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-25 17:06:43.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-25 17:09:33.000000000","tz":120},"subject":"WIP: run tox with stack user","message":"WIP: run tox with stack user\n\nChange-Id: Ie815d3a1f177af35ec1a2f7bb703739a5f0f9f3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/63738549bb5ef46fe7e8ebf530addbe448920a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/63738549bb5ef46fe7e8ebf530addbe448920a97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
