)]}'
{"id":"openstack%2Ftrove~673194","triplet_id":"openstack%2Ftrove~master~I8dc53a8c5416b9a4a109e66d5da4f5c470f1d456","project":"openstack/trove","branch":"master","topic":"fix-ci","hashtags":[],"change_id":"I8dc53a8c5416b9a4a109e66d5da4f5c470f1d456","subject":"Remove tools/test-setup.sh","status":"ABANDONED","created":"2019-07-29 02:55:33.000000000","updated":"2019-08-14 21:49:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"81ac5b615116e0585e5655475ed76f798213b5a0","_number":673194,"virtual_id_number":673194,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-08-13 06:43:28.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":1,"date":"2019-08-01 00:24:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-01 01:07:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-08-01 01:07:08.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 00:24:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 06:43:28.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"a70b4f3d711db1e9739ae4d3da77ee1440653526","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-07-29 02:55:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e5c9e830d6b9217377de80054ea8c3fba805878","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 03:10:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/94/673194/1/check/openstack-tox-cover/30c5faa/ : FAILURE in 13m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/94/673194/1/check/openstack-tox-lower-constraints/8f4d9e0/ : FAILURE in 9m 37s\n- openstack-tox-pep8 http://logs.openstack.org/94/673194/1/check/openstack-tox-pep8/be3ffed/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/94/673194/1/check/openstack-tox-py27/b01edaa/ : FAILURE in 8m 07s\n- openstack-tox-py36 http://logs.openstack.org/94/673194/1/check/openstack-tox-py36/9be7ae3/ : FAILURE in 9m 30s\n- openstack-tox-py37 http://logs.openstack.org/94/673194/1/check/openstack-tox-py37/346bace/ : FAILURE in 3m 01s\n- openstack-tox-docs http://logs.openstack.org/94/673194/1/check/openstack-tox-docs/cc80d2a/html/ : SUCCESS in 3m 38s\n- openstack-tox-pylint http://logs.openstack.org/94/673194/1/check/openstack-tox-pylint/8464293/ : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c93e5fbbd416333b2bafd5018b3a08c5dcf781ad","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-07-31 23:54:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d21497ef016f5f101d882d837900822fbcb940da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 00:24:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/94/673194/2/check/openstack-tox-cover/482d91a/cover/ : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/94/673194/2/check/openstack-tox-lower-constraints/80940e4/ : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://logs.opendev.org/94/673194/2/check/openstack-tox-pep8/1249412/ : SUCCESS in 8m 55s\n- openstack-tox-py27 https://logs.opendev.org/94/673194/2/check/openstack-tox-py27/085cf29/ : SUCCESS in 10m 21s\n- openstack-tox-py36 https://logs.opendev.org/94/673194/2/check/openstack-tox-py36/7859c6d/ : SUCCESS in 13m 14s\n- openstack-tox-py37 https://logs.opendev.org/94/673194/2/check/openstack-tox-py37/15b31c9/ : SUCCESS in 17m 36s\n- openstack-tox-docs https://logs.opendev.org/94/673194/2/check/openstack-tox-docs/fd5808e/html/ : SUCCESS in 6m 02s\n- openstack-tox-pylint https://logs.opendev.org/94/673194/2/check/openstack-tox-pylint/17a5ee5/ : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":2},{"id":"071a12834b2bd86a54777059862f00ad4534d409","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-01 01:07:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b4cdf4353e9df3c9014612e3437ad31bef8b11fc","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-13 06:43:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d17b797d482a4e26933966dc240f7c7b35061415","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 07:10:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a20f54ae6a04eb5857b38d4ccf68632 : SUCCESS in 16m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a8f080908bd44a38b3c9aafb092c4f8 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8633716f7dae4e33ba40c737919fd816 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b9bff75852341debd5cb9091285f922 : SUCCESS in 14m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/442e922ea58543cd839f7ac227ed1e5a : SUCCESS in 13m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b07ea81a2f240268514e09a3bb237a1 : SUCCESS in 19m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d234c4e12c8f4af49bad75222fe6e1f2 : SUCCESS in 5m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c6e4c1b883f14615bbe5f40a5ed7fff4 : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":2},{"id":"778a9b9ee9f41c2987b3bab1c30f77ee8ccf3e2e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-14 21:49:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c0ba95563f0908f4cb5cfbef475a6f1d13a2cac","revisions":{"46b43968ca9313dc8bc95192d874d6480bd3bd71":{"kind":"REWORK","_number":1,"created":"2019-07-29 02:55:33.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/94/673194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/94/673194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/673194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/673194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/673194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/94/673194/1"}}},"commit":{"parents":[{"commit":"f72015179ea921c8264a77390359d638e0bd8727","subject":"Merge \"Update the outdated content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f72015179ea921c8264a77390359d638e0bd8727"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-07-29 02:51:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-07-29 02:51:34.000000000","tz":720},"subject":"Remove tools/test-setup.sh","message":"Remove tools/test-setup.sh\n\nThere is a tools/test-setup.sh file for historical reason which fails\nthe Trove Jenkins job with the error:\n\nsudo: mysqladmin: command not found\n\nThis patch deletes that file as most of the OpenStack projects do.\n\nChange-Id: I8dc53a8c5416b9a4a109e66d5da4f5c470f1d456\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/46b43968ca9313dc8bc95192d874d6480bd3bd71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/46b43968ca9313dc8bc95192d874d6480bd3bd71"}]},"branch":"refs/heads/master"},"1c0ba95563f0908f4cb5cfbef475a6f1d13a2cac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-31 23:54:45.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/94/673194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/94/673194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/673194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/673194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/673194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/94/673194/2"}}},"commit":{"parents":[{"commit":"6c011a69e0a331ea5c3dff4568e3e6ec9109acc2","subject":"Added a bindep file to install binaries in CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c011a69e0a331ea5c3dff4568e3e6ec9109acc2"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-07-29 02:51:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-07-31 23:54:45.000000000","tz":0},"subject":"Remove tools/test-setup.sh","message":"Remove tools/test-setup.sh\n\nThere is a tools/test-setup.sh file for historical reason which fails\nthe Trove Jenkins job with the error:\n\nsudo: mysqladmin: command not found\n\nThis patch deletes that file as most of the OpenStack projects do.\n\nChange-Id: I8dc53a8c5416b9a4a109e66d5da4f5c470f1d456\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1c0ba95563f0908f4cb5cfbef475a6f1d13a2cac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1c0ba95563f0908f4cb5cfbef475a6f1d13a2cac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
