)]}'
{"id":"openstack%2Ftrove~673716","triplet_id":"openstack%2Ftrove~master~Ie6022a13c181f5c27e19b54bd49d9a77c90dc1ac","project":"openstack/trove","branch":"master","topic":"postgres-elements-fix-xenial","hashtags":[],"change_id":"Ie6022a13c181f5c27e19b54bd49d9a77c90dc1ac","subject":"Fix 31-fix-init-script for Postgresql","status":"MERGED","created":"2019-07-31 02:19:45.000000000","updated":"2019-08-16 01:45:05.000000000","submitted":"2019-08-16 01:45:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673716-1565919905495-3f7c88c0","meta_rev_id":"945d82f6aabb63e7a66a6bbe65f4d0e5cd14f976","_number":673716,"virtual_id_number":673716,"owner":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-16 01:45:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-15 07:32:50.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-15 07:32:50.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 07:32:50.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-08-16 01:45:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7596a36284603c53016d3a9689c5a13dcb5e8b3","author":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"date":"2019-07-31 02:19:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f8d9489e7c5dc18cbb5483b2e1cc634651f4ffa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 03:06:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/16/673716/1/check/openstack-tox-cover/5ff02e1/ : RETRY_LIMIT in 1m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/673716/1/check/openstack-tox-lower-constraints/3f81220/ : RETRY_LIMIT in 1m 49s\n- openstack-tox-pep8 http://logs.openstack.org/16/673716/1/check/openstack-tox-pep8/310e80a/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/16/673716/1/check/openstack-tox-py27/0f35c0a/ : RETRY_LIMIT in 1m 59s\n- openstack-tox-py36 http://logs.openstack.org/16/673716/1/check/openstack-tox-py36/cc91a3a/ : RETRY_LIMIT in 2m 36s\n- openstack-tox-py37 http://logs.openstack.org/16/673716/1/check/openstack-tox-py37/74a2d0d/ : RETRY_LIMIT in 2m 05s\n- openstack-tox-docs http://logs.openstack.org/16/673716/1/check/openstack-tox-docs/817e7cb/html/ : SUCCESS in 3m 34s\n- trove-functional-mysql http://logs.openstack.org/16/673716/1/check/trove-functional-mysql/6ad96c5/ : FAILURE in 39m 46s\n- trove-scenario-mariadb-single http://logs.openstack.org/16/673716/1/check/trove-scenario-mariadb-single/6c5ba2d/ : FAILURE in 30m 52s (non-voting)\n- trove-scenario-mariadb-multi http://logs.openstack.org/16/673716/1/check/trove-scenario-mariadb-multi/56f7ecc/ : FAILURE in 31m 59s (non-voting)\n- trove-scenario-mysql-single http://logs.openstack.org/16/673716/1/check/trove-scenario-mysql-single/076a6bb/ : FAILURE in 41m 02s\n- trove-scenario-mysql-multi http://logs.openstack.org/16/673716/1/check/trove-scenario-mysql-multi/757b279/ : FAILURE in 41m 05s\n- trove-scenario-postgresql-single http://logs.openstack.org/16/673716/1/check/trove-scenario-postgresql-single/c769363/ : FAILURE in 34m 10s (non-voting)\n- trove-scenario-postgresql-multi http://logs.openstack.org/16/673716/1/check/trove-scenario-postgresql-multi/631be24/ : FAILURE in 32m 39s (non-voting)\n- trove-scenario-pxc-single http://logs.openstack.org/16/673716/1/check/trove-scenario-pxc-single/5531cb5/ : FAILURE in 33m 00s (non-voting)\n- trove-scenario-pxc-multi http://logs.openstack.org/16/673716/1/check/trove-scenario-pxc-multi/66a7083/ : FAILURE in 31m 54s (non-voting)\n- trove-scenario-redis-single http://logs.openstack.org/16/673716/1/check/trove-scenario-redis-single/4c9fc5c/ : FAILURE in 33m 00s (non-voting)\n- trove-scenario-redis-multi http://logs.openstack.org/16/673716/1/check/trove-scenario-redis-multi/e33c9d6/ : FAILURE in 29m 34s (non-voting)\n- trove-tox-fakemodetests http://logs.openstack.org/16/673716/1/check/trove-tox-fakemodetests/47b123b/ : RETRY_LIMIT in 1m 39s\n- trove-tox-bandit-baseline http://logs.openstack.org/16/673716/1/check/trove-tox-bandit-baseline/5df73ce/ : RETRY_LIMIT in 1m 37s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/16/673716/1/check/openstack-tox-pylint/9ceda17/ : RETRY_LIMIT in 1m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e6a60d2a6edd303d42382a2c4bac55b3472f4426","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-14 22:09:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"27050a0c47b79e7f1e38a3a755d005dbf977607c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 01:35:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4382c110018b4cc0b895d718eceaa4cc : SUCCESS in 18m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46bab728b1f74963a9ee1a533a62158e : SUCCESS in 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccc30f5d7d4c4a96893a3e61a70f14d8 : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea069e418e144531a0b8c5b0d228e984 : SUCCESS in 12m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/260d3c24246e4dc5a693e6ae0bbb01b9 : SUCCESS in 12m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/614c66f67353491b8a413577f7196832 : SUCCESS in 17m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a3a9cc52ba04cf9942425e6c38f0bf8 : SUCCESS in 5m 06s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/69a288d2e6f84bca80532a8cdcbdc05f : TIMED_OUT in 3h 03m 34s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/27ed549d550443c197553c262edfdca2 : FAILURE in 1h 12m 38s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/48687b2514d048aabe4d518ad7f6a41c : FAILURE in 46m 27s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/a892b19d64db47fa876c9c03e206e8fa : SUCCESS in 1h 43m 53s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1157d231bca34775b8a141f1c760aa89 : SUCCESS in 2h 05m 22s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/45041e67cd8d4bebb849748fe108e103 : FAILURE in 40m 10s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/d460bdca3a29410bbdd1679ce2881e16 : FAILURE in 43m 04s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/b8c75788ff4447f78d4971b7160e1c23 : FAILURE in 45m 54s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/0fdcaf8311d8433bb93ddbe95c9cfd7e : FAILURE in 49m 29s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/649a9c4323ba43d790fdaf9d326e1c23 : FAILURE in 1h 24m 36s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/dc6d29d8bf304ec1b18745e61a7d0b54 : SUCCESS in 1h 43m 33s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/a0c111344c704e84a22793d85844ce8c : SUCCESS in 6m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b082c74eac3042999e62445142ce3191 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/18ffcccd99494e75baac85756d526d33 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3b0782efc88a8769484db9a9ea66c5172aea4abd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-15 02:02:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e35207cb5584e057d0d42edf622f5ed4014323fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 04:17:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc7e5e07d6fa4e049c8a1dcd16fc4f4f : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddd48507fd974c628d20bdc4811cd0bc : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc9e1869753492c89369ece1e8e1e40 : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c68149d662c244f7a500fae84b2a1f71 : SUCCESS in 11m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/048457e64032408da907d0692af893af : SUCCESS in 15m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b8c5eb0a31f430a97221b9d7059919f : SUCCESS in 17m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7720b613b340c0a4d5dd654a560b6c : SUCCESS in 4m 29s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/a7b69e17bdb34090802a9c2b8797d1c3 : SUCCESS in 2h 13m 16s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/aaf45be7c7884c41863d658075f9526f : FAILURE in 1h 01m 14s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d0b0587e57354a4580ea2c486f930a82 : FAILURE in 44m 55s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/82ddde2cc6f442b5863141569f7d9b9d : SUCCESS in 1h 49m 29s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/24a109e649584d3ca92aa79040593c5a : SUCCESS in 1h 42m 13s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/168ba613e22945a8845beeb7f2792875 : FAILURE in 42m 43s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5df2e533b61a46c4b88b08508dd8a588 : FAILURE in 41m 10s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/a110967489ce4fda8148955e10b64131 : FAILURE in 44m 06s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/4328ec426ba441d7bc19a019c04d503b : FAILURE in 58m 47s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/8deeffc5b8f14ba99941b320fcd1e6f7 : FAILURE in 42m 12s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/c7b764e638a24e1dbe77d2c09afbcd85 : SUCCESS in 1h 23m 47s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/0a25d45c87224b9581eaa910dac0c482 : SUCCESS in 7m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ac20729b9894406b9f6480335d0e87d2 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/507ef272eb1748e0ac47fae6d07f8b6a : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":2},{"id":"ab94ed2d82dcbd0c69383bb5417936e5aadcb714","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-15 07:32:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b3c8687debcc7911b325d849d765e104a93d333","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 07:33:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b3a312a37335933731c22038ea8566a5c555c25a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 10:10:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0f8639cdb3e4b25b03d21aee2505e96 : SUCCESS in 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41cccb17fad440248d87d719f6c6d31f : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bbf489f767b945adb65ca699a9dda86c : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca7d51ae744e4f2f93b6caf8d7e1ddd0 : SUCCESS in 12m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fc190728bb54c7ea9a14f5d7b46aa44 : SUCCESS in 18m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52a20db62834d5882e8e90eae5eda95 : SUCCESS in 4m 54s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/cdb894dc374d41c1b019e8bc764b2177 : SUCCESS in 2h 31m 32s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b309470d061447e4bd31293d00d554c8 : FAILURE in 46m 23s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/cc41ef03f98e460a8f112813d8ea5b1e : SUCCESS in 1h 36m 29s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/e9896b0d43a84036acb2c03a4cf9c256 : SUCCESS in 5m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba13587b7ae84f3d9b1603ee7f2b1490 : SUCCESS in 9m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b8d219ddf335d7a5c79a0a10cfb890fa0b9c203a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-15 10:37:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"abc9ab25d39402a7f037c9f82991a4931c83b080","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 12:58:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb39b3c6c35f4b4d9dc46e95ea184a39 : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3165c56eb3c429a9b2d40e9b637fba2 : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e80a0a4bbc70437395b7274676b3a6b7 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56acfec2312748c7ae9e567608e0bb86 : SUCCESS in 15m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d53914bdcfe48a3af575cdcecec43bd : SUCCESS in 12m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8951e693655e4eaa862c8fb7ef87d45c : SUCCESS in 17m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da7befbec4e4af7a8aa068bc0bee1c4 : SUCCESS in 3m 58s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ed68288e3dfe4e78bb732e291ddc2830 : SUCCESS in 2h 14m 57s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/c85c0e4723ec47c2835ab3a2bffbbfee : FAILURE in 38m 45s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/67f609242f054bd69ad3bbc03da8395c : FAILURE in 39m 05s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/fb24214fec86464fa821a1bf20974658 : SUCCESS in 2h 11m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/705d05ba5fe048449554da608e651ee8 : SUCCESS in 1h 39m 03s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/3dda4530103749698917e153270c0be4 : FAILURE in 39m 53s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/061a980673ad44e78506e49892d3b517 : FAILURE in 45m 20s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/02ef6eee29374fbf8c60fa7a56411472 : FAILURE in 33m 44s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/fc373bfc5b0f46879bef69829e172b4d : FAILURE in 54m 31s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/29a31b8d1f0f41488e942c089e7e9516 : FAILURE in 1h 24m 36s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/76246e2c269f4242a00e9d21b99bd503 : SUCCESS in 1h 27m 09s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/18d3362ce19342e5ac16016cc63c8f38 : SUCCESS in 4m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9a34659b0bd48d7b1b8e0540518265a : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88dca0385eb6493ca2ac98f8c18f6ce8 : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9fd16d4c37a9d9bdadf4ebc809395859e616084e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 12:58:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d00f59244c80449e90df161d319cfdacb9e70eef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 15:33:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0209de379832450d8884732afcc5f124 : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c15389fd10ce432ead1609ad3a4a21b4 : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0df02deb2bcb41a19110fe6b88429fc7 : SUCCESS in 12m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a1d16bc7b2c4277a9358e031780c33f : SUCCESS in 12m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29ce708b6f184501a177e9dc38cd50b3 : SUCCESS in 18m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21e6dbd175aa43479e02534ee2090bab : SUCCESS in 5m 50s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/923e2ede87a34a1a9dfaa63da98e5905 : FAILURE in 2h 32m 47s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b16c9c29ef634330b69bd332cad5f3c4 : SUCCESS in 1h 57m 43s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/72cf9fd0289345588866fa7bf3874266 : SUCCESS in 2h 05m 22s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/f0df5f5bc6c9402fa95a9757f1138360 : SUCCESS in 6m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba0aa00b87a240789e90dbf22c776a6a : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":2},{"id":"e92cfd558dc889d3fda3e39906502a690bf29982","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-15 21:15:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3628fef4278d89f587e67db29b2df9b1cb4f8ddd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 23:35:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf61c9b31a7f4b09a4c57296687341e6 : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b7e8148f15a4823965ca69b914261a1 : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e3a028e3f441a7b607c25d425c32f6 : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93ef9297cdc2497cbd06b867b585f16b : SUCCESS in 11m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09d666ab29444d659edb436934979c3a : SUCCESS in 12m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46b15298bfd6401182d8e7eecebd7215 : SUCCESS in 16m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d782fe4c8a3c4ebcb38427647f1a223b : SUCCESS in 6m 00s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4c532c5b9ee84c5caa4b2269bd72b4a7 : SUCCESS in 2h 13m 02s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/5cfdedcf1b8b4f309363e2c044741f2e : FAILURE in 42m 41s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/850ba0816f4a4583b68d10174c2610cf : FAILURE in 43m 49s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/939d50ee2a7e4bbca6673a7fff7c1e68 : SUCCESS in 1h 44m 00s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/4849d795968b482a99905e88eeb4ada8 : SUCCESS in 2h 18m 32s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/e2234175c122460093bb3c5437271c53 : FAILURE in 38m 36s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/d865323b02f2494c862ea8e13132c23c : FAILURE in 1h 05m 58s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/b6139b81a55c4503ac80a75631fce2e5 : FAILURE in 41m 52s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/d154cefe246049449cb8bf385d8e04d3 : FAILURE in 1h 26m 54s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/4644f8e58c2847deaaa409d29ce46620 : FAILURE in 1h 10m 10s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/3aa20ec29f134bcca1efb035971f4650 : SUCCESS in 1h 19m 59s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/1af001092c3048de95af4cf68f0b55cb : SUCCESS in 6m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b716afe0f734f83b166d89797921325 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f8d5427872f5495090b84bf9d36ea2f4 : SUCCESS in 10m 46s","accounts_in_message":[],"_revision_number":2},{"id":"0853709433a64cfdc9d657d102d1d9862188e23a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 23:36:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d30c2e575aeb9854d49c46fd27f71c69766be64b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 01:45:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42c5a0d34625433392e2bf549b567053 : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcc1df0625894f70ab9f031ee2427779 : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46f03a21db944e4a87e94919709b8765 : SUCCESS in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a250b28fff345e68320e4d55d7b408e : SUCCESS in 13m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43d6b39f94c445c0b7539d308a0df40a : SUCCESS in 19m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75bcc8966718439bb2f9014b323dcde9 : SUCCESS in 5m 05s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3cbf075435054513a4be87202b5f26c0 : SUCCESS in 2h 07m 46s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/4cdd62f1b5454c10b03a965a82a24a6b : SUCCESS in 1h 54m 52s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/5146db68d22c405999daae319334beed : SUCCESS in 1h 50m 10s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/5fee391639f143da8cc432f88cb4a23f : SUCCESS in 7m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e33efe738ca42e090a1ba7f4f30e4ba : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"545722ebde963589ff31cd75522d87d4ca41a6ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 01:45:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d2e986778d71e29dc18be53a89fae7db1c66cf3","revisions":{"02f661dca7a5a38c2407301053d936173ef4eef2":{"kind":"REWORK","_number":1,"created":"2019-07-31 02:19:45.000000000","uploader":{"_account_id":10864,"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","username":"markir"},"ref":"refs/changes/16/673716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/16/673716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/673716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/673716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/673716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/16/673716/1"}}},"commit":{"parents":[{"commit":"f72015179ea921c8264a77390359d638e0bd8727","subject":"Merge \"Update the outdated content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f72015179ea921c8264a77390359d638e0bd8727"}]}],"author":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2019-07-31 00:42:42.000000000","tz":720},"committer":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2019-07-31 00:42:42.000000000","tz":720},"subject":"Fix 31-fix-init-script for Postgresql","message":"Fix 31-fix-init-script for Postgresql\n\nIronically the fix is never applied as the systemd file uses\n/run not /var/run.\n\nChange-Id: Ie6022a13c181f5c27e19b54bd49d9a77c90dc1ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/02f661dca7a5a38c2407301053d936173ef4eef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/02f661dca7a5a38c2407301053d936173ef4eef2"}]},"branch":"refs/heads/master"},"7d2e986778d71e29dc18be53a89fae7db1c66cf3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-14 22:09:56.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/16/673716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/16/673716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/673716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/673716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/673716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/16/673716/2"}}},"commit":{"parents":[{"commit":"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1","subject":"Fix Trove CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}]}],"author":{"name":"Mark Kirkwood","email":"mark.kirkwood@catalyst.net.nz","date":"2019-07-31 00:42:42.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-14 22:09:56.000000000","tz":0},"subject":"Fix 31-fix-init-script for Postgresql","message":"Fix 31-fix-init-script for Postgresql\n\nIronically the fix is never applied as the systemd file uses\n/run not /var/run.\n\nChange-Id: Ie6022a13c181f5c27e19b54bd49d9a77c90dc1ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d2e986778d71e29dc18be53a89fae7db1c66cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d2e986778d71e29dc18be53a89fae7db1c66cf3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
