)]}'
{"id":"openstack%2Ftrove~675068","triplet_id":"openstack%2Ftrove~master~I31d4ee579a554f4c98f9facb9fd4b7779665a3dd","project":"openstack/trove","branch":"master","topic":"devmode-false","hashtags":[],"change_id":"I31d4ee579a554f4c98f9facb9fd4b7779665a3dd","subject":"Improve devmode\u003dflase when building the image","status":"MERGED","created":"2019-08-07 09:35:01.000000000","updated":"2019-08-25 23:36:51.000000000","submitted":"2019-08-25 23:35:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675068-1566776133407-831344a3","meta_rev_id":"bab6ae77228f255e189f401825b51acb820f49d8","_number":675068,"virtual_id_number":675068,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-25 23:35:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-25 21:10:57.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-25 21:10:57.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-25 23:35:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"946c6382b31595ee1726e75ca16399ab0d7fb2ed","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-07 09:35:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd5e98e014efea3634cf61e1310ee042e8d9ad21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 10:28:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/68/675068/1/check/openstack-tox-cover/89f122a/ : RETRY_LIMIT in 2m 34s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/68/675068/1/check/openstack-tox-lower-constraints/92442a9/ : RETRY_LIMIT in 2m 08s\n- openstack-tox-pep8 https://logs.opendev.org/68/675068/1/check/openstack-tox-pep8/1cd8546/ : FAILURE in 6m 58s\n- openstack-tox-py27 https://logs.opendev.org/68/675068/1/check/openstack-tox-py27/917cd63/ : RETRY_LIMIT in 3m 30s\n- openstack-tox-py36 https://logs.opendev.org/68/675068/1/check/openstack-tox-py36/a5bc6c3/ : RETRY_LIMIT in 2m 19s\n- openstack-tox-py37 https://logs.opendev.org/68/675068/1/check/openstack-tox-py37/71492ce/ : RETRY_LIMIT in 2m 39s\n- openstack-tox-docs https://logs.opendev.org/68/675068/1/check/openstack-tox-docs/15f109c/html/ : SUCCESS in 3m 51s\n- trove-functional-mysql https://logs.opendev.org/68/675068/1/check/trove-functional-mysql/ff92c6e/ : FAILURE in 42m 28s\n- trove-scenario-mariadb-single https://logs.opendev.org/68/675068/1/check/trove-scenario-mariadb-single/7f71fd9/ : FAILURE in 35m 17s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/68/675068/1/check/trove-scenario-mariadb-multi/f9e237b/ : FAILURE in 34m 05s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/68/675068/1/check/trove-scenario-mysql-single/5dfa93d/ : FAILURE in 45m 12s\n- trove-scenario-mysql-multi https://logs.opendev.org/68/675068/1/check/trove-scenario-mysql-multi/8b94a68/ : FAILURE in 43m 07s\n- trove-scenario-postgresql-single https://logs.opendev.org/68/675068/1/check/trove-scenario-postgresql-single/22710fb/ : FAILURE in 33m 44s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/68/675068/1/check/trove-scenario-postgresql-multi/cd50bdd/ : FAILURE in 37m 50s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/68/675068/1/check/trove-scenario-pxc-single/8d740ba/ : FAILURE in 34m 00s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/68/675068/1/check/trove-scenario-pxc-multi/354446f/ : FAILURE in 28m 51s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/68/675068/1/check/trove-scenario-redis-single/ccdd740/ : FAILURE in 38m 36s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/68/675068/1/check/trove-scenario-redis-multi/13cd37e/ : FAILURE in 34m 17s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/68/675068/1/check/trove-tox-fakemodetests/6207a14/ : RETRY_LIMIT in 3m 30s\n- trove-tox-bandit-baseline https://logs.opendev.org/68/675068/1/check/trove-tox-bandit-baseline/a7be05e/ : RETRY_LIMIT in 3m 13s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/68/675068/1/check/openstack-tox-pylint/791f908/ : RETRY_LIMIT in 2m 29s","accounts_in_message":[],"_revision_number":1},{"id":"72177cd8136f3b25cdbc5635faea6507de5d697e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-07 11:02:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ccf3e9d79907a1c67fad5e5bb21633edd8f3168","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-07 11:07:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c65732f5a54d4ab4336fa25afa303bf46a6848d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 12:34:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/68/675068/3/check/openstack-tox-cover/cf61337/ : FAILURE in 13m 19s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/68/675068/3/check/openstack-tox-lower-constraints/a9cbfa0/ : FAILURE in 12m 09s\n- openstack-tox-pep8 https://logs.opendev.org/68/675068/3/check/openstack-tox-pep8/5f2459f/ : SUCCESS in 4m 38s\n- openstack-tox-py27 https://logs.opendev.org/68/675068/3/check/openstack-tox-py27/8dacf9f/ : FAILURE in 9m 07s\n- openstack-tox-py36 https://logs.opendev.org/68/675068/3/check/openstack-tox-py36/bbe3085/ : FAILURE in 10m 59s\n- openstack-tox-py37 https://logs.opendev.org/68/675068/3/check/openstack-tox-py37/56e2c05/ : FAILURE in 3m 55s\n- openstack-tox-docs https://logs.opendev.org/68/675068/3/check/openstack-tox-docs/a28bddc/html/ : SUCCESS in 5m 14s\n- trove-functional-mysql https://logs.opendev.org/68/675068/3/check/trove-functional-mysql/84f0e33/ : FAILURE in 1h 10m 05s\n- trove-scenario-mariadb-single https://logs.opendev.org/68/675068/3/check/trove-scenario-mariadb-single/9922d78/ : FAILURE in 40m 37s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/68/675068/3/check/trove-scenario-mariadb-multi/d3c0e04/ : FAILURE in 45m 45s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/68/675068/3/check/trove-scenario-mysql-single/8819e46/ : FAILURE in 49m 26s\n- trove-scenario-mysql-multi https://logs.opendev.org/68/675068/3/check/trove-scenario-mysql-multi/c0745af/ : SUCCESS in 1h 26m 03s\n- trove-scenario-postgresql-single https://logs.opendev.org/68/675068/3/check/trove-scenario-postgresql-single/4a36cb8/ : FAILURE in 45m 08s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/68/675068/3/check/trove-scenario-postgresql-multi/6671cb3/ : FAILURE in 39m 26s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/68/675068/3/check/trove-scenario-pxc-single/b6affdc/ : FAILURE in 40m 19s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/68/675068/3/check/trove-scenario-pxc-multi/825d69a/ : FAILURE in 35m 32s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/68/675068/3/check/trove-scenario-redis-single/f41ff1c/ : FAILURE in 39m 23s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/68/675068/3/check/trove-scenario-redis-multi/49ca565/ : FAILURE in 32m 59s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/68/675068/3/check/trove-tox-fakemodetests/51f46c9/ : FAILURE in 4m 26s\n- trove-tox-bandit-baseline https://logs.opendev.org/68/675068/3/check/trove-tox-bandit-baseline/0be4328/ : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/68/675068/3/check/openstack-tox-pylint/66e2804/ : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":3},{"id":"844f2ee20562700177f64f9e65283021cbcd98c8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-15 01:20:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"74f1895c420cd4bc1565fc152d61492e5cf99abf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 02:10:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c416804ac222424dbdbf0e0ac40604b3 : FAILURE in 15m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62269733a98e47da96aa0a4381427d0b : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99bacb9a9a84a8d85f2d1045b00f53e : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ecc8ea886aec415da39369e4e4a230ba : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/027fbe0be9f54387be5716c882c7e440 : SUCCESS in 14m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4f022520872469d86406e4aae48b961 : SUCCESS in 17m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/592d367b3e474f1f9f8fb759341c5d58 : SUCCESS in 4m 58s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8508014d57d940dc8ce9c6f21b54a01f : FAILURE in 36m 11s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/974385894b39472a898521a3e27c9c16 : FAILURE in 33m 17s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/1919479c242d498a9c3da13b9e9be1ba : FAILURE in 36m 02s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/7bc3c4c43d2347c2902ac744025eaaf3 : FAILURE in 34m 55s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7a6fe5b67c4f42e3a24daefb2369008b : FAILURE in 38m 57s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/594b8acbd53d4a95b2f109e03638347e : FAILURE in 36m 03s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/3af6ce9f7a3a4a22a0b2fe48b265319e : FAILURE in 38m 47s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/10f10f791fd64f0dbdae2e63e90ce73a : FAILURE in 34m 49s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/72baa73575c04d61a44178d1ce43afd6 : FAILURE in 39m 31s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/0a1543a9eb054ee6ae46748322766860 : FAILURE in 34m 17s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/fdd60686a62b4b0eafb23a8a010fd69d : FAILURE in 40m 52s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/93963bed883e496587b5f2f1c04f470d : FAILURE in 4m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/49cddc9cb37c4b3cb96e152ec6108686 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83b7a887bbc445d7a5d4fad41fc66505 : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":4},{"id":"fcd4bf1c1a199b68e48576c517b70556b653d551","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-15 02:47:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"79e054be76ef25ba3e38d1d0dd3ef3d6d5b1f460","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 05:11:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8526de7898004bebaff62619e3fb73e2 : FAILURE in 16m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d091e68fd2f1443f808046447c3090be : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d67caee6544f5da1f549f2827c8c4f : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6776a4971d804670b3107638ba0a9aa2 : SUCCESS in 10m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5d9d830690a480088d5daf4f7c0e971 : SUCCESS in 11m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b8825b2adda487ca00d5f8ebf9ad565 : SUCCESS in 17m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e946ae0c7c324705992fdde241091a2b : SUCCESS in 4m 24s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4d065bba3120465baa6c288517d9b3b0 : SUCCESS in 2h 10m 27s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/5711c9a90a7b49ceb97e51773deab04f : FAILURE in 41m 48s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/46f7f57a137a47888f522db4c6228a73 : FAILURE in 44m 48s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/cd24217233bb4e6e811673b5f88e36e5 : SUCCESS in 1h 58m 19s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/ef9a3226107c489c99c3f8cd51bab5c7 : SUCCESS in 1h 32m 25s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/2ba2ca4863fe46c8bc4994d7449513db : FAILURE in 36m 38s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/f2c709f80b0d4fcba21f7381ec6a11cc : FAILURE in 45m 02s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/5f6685585078498fa503f3b23a82240a : FAILURE in 40m 09s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/d87e08e4afca402085a1f00bdf6abecc : FAILURE in 39m 57s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/d499323f843d4e3fba431c1757a2d7e5 : FAILURE in 1h 12m 38s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/a9c833298d5c47f3a9cd621d64df1fb2 : SUCCESS in 1h 16m 05s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/754512c1404248d2ba4012edabf5b0d2 : FAILURE in 4m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/89a4396560284a2283b8b570f2ac08b1 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f0d579e826294d1d88987308ce5b0183 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":5},{"id":"5dc78393aa7eba13c58b1a561247c701329a9d07","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-16 10:38:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fc2595abf90c7ffddf6ee1968fc377ea74ad8f30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 13:01:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8af90bbe495a43389143516beba66bb5 : SUCCESS in 19m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/327b6d447d034f62af5d94ffb3f36e9f : SUCCESS in 13m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f79ca7358944758f037bdfc63c2717 : FAILURE in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a58fcb3cb8d47998e809fd023a46d2b : SUCCESS in 11m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bec9efdb9c44876b3895c8d78e8f64d : SUCCESS in 13m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22893bf118814af59f8d4db138320a45 : SUCCESS in 18m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddbbe6d397834338af701a126609a579 : SUCCESS in 5m 52s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d946be6b4c6640a99e160d06462aa36f : SUCCESS in 2h 17m 23s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/a7b2edc662494b05b9462e04eff2d367 : FAILURE in 43m 54s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d7cec63a9e6e47d49695c01c179e2393 : FAILURE in 43m 45s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/229792935f2b4d86b32f3392ab75a1a9 : SUCCESS in 1h 45m 19s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/69fa5d973dda4f4a82b7b2fcc039f50f : SUCCESS in 1h 35m 08s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/ad4d7205c4074fa38103622817118e54 : FAILURE in 52m 27s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/968fd654f0c74719b841dec5816d54d7 : FAILURE in 41m 18s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/8971f30a47f14b30b2522833570984c1 : FAILURE in 58m 46s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/89d9745195bf4e75b9f1260c236d7e2c : FAILURE in 42m 31s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/6d6130164e2b4138bf428d692c4eb3ef : FAILURE in 1h 26m 47s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/400ad257f0ff452e980f0b35ad8d0196 : SUCCESS in 1h 20m 14s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/ad358ff50c344a06abf108ebd42c0703 : SUCCESS in 5m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3fd007fae9d486ab57f2f218aa71409 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5315358896324f2894760642384504c4 : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":6},{"id":"686616edbbf00ef0339a23accf851f1f2513942a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-25 11:12:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ce8dfd81871df895d60d0f3513d63eabdcf9fa63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 13:25:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/360ca6386e1f4a9a8d397285c3fece0d : SUCCESS in 17m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a147ca78e0343338b31300aa5eb7417 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7337a7ff573741d9aec1315949272175 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72b0b2657cd24b2bb87e87e2d13fed70 : SUCCESS in 12m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ea9176059944ad3a7dff85633316793 : SUCCESS in 13m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6db2d466f2014bbf9e1d710d25874e1b : SUCCESS in 20m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f7ed358bd5249d3a479ce2866fdb119 : SUCCESS in 11m 26s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/98a938237f78489bba83771855e5153f : SUCCESS in 2h 01m 53s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/211098f5f44b4e29bf004b4314a6be3b : FAILURE in 38m 40s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/e096d8b093d84774a3cf1517e7527474 : FAILURE in 46m 31s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/39e7a766377c42899ef4166002f3df7f : SUCCESS in 1h 30m 08s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/29492a8257ec4736ab92619d20daeaa9 : SUCCESS in 1h 40m 02s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/d0d31b4bf91f4c088a1a96477028a8ee : FAILURE in 43m 12s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/e687404f535b4717aa7e4fc1c9501f0b : FAILURE in 34m 47s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/df6d289d0faf4db4ad9845b7fd6f3d0a : FAILURE in 42m 06s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/01f7477383b4400caa46097f7abc75ad : FAILURE in 37m 36s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/d50a5539f51e4da8a8769c395dcb0313 : FAILURE in 1h 09m 29s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/a6828fb73430463eb6789248ac3bcac9 : SUCCESS in 1h 24m 55s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/1f200928d7d04afdbbc23f0b06908674 : SUCCESS in 10m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c16ce1806c8e4dcb8d66a5f193a9787f : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7879bfd022974547b95a523c3d95b824 : SUCCESS in 12m 25s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6d6d3c9b5174d209722715228a08ca7 : SUCCESS in 40m 27s","accounts_in_message":[],"_revision_number":7},{"id":"f95f1c3e656e068820cd8662e9bca08401710d64","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-25 21:10:57.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"5fa93c95bb3369e45ca0835fe2e7f8a69189edbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 21:11:08.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ace0b08cb46a6e62dba86ee38b9b54cab854c342","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 23:35:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"aaf22b376f2b611e103e6135d544e2a7537da428","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 23:35:33.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d34a238e60d7410dba7921950abb9adc : SUCCESS in 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06722ce1fdf648cf901fc878c8a60d41 : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa7a366e192e4edabea3d871f31ca09d : SUCCESS in 11m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d935d9fe342498fb6eeca87a0ee0cc7 : SUCCESS in 14m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0018b20679d43c5974ea9b275dfca2a : SUCCESS in 17m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c9b5daa8910445fb78f3db4054cefbb : SUCCESS in 10m 56s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/436ee2bdc6594688906f2a90148e9c54 : SUCCESS in 2h 19m 16s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8d30c2c9f4654fd5828f2e5c6b799084 : SUCCESS in 1h 45m 38s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0cff5b445f29409398b336e6a7e5df9e : SUCCESS in 1h 38m 12s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/24678bbda0e6495fa02045459ed97da5 : SUCCESS in 4m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/02fa5fd035024e6280f15fbed7933133 : SUCCESS in 8m 31s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10b4999f865f41f4a09c83ba607ba4d9 : SUCCESS in 46m 43s","accounts_in_message":[],"_revision_number":7},{"id":"6a1d273a1305739f623377775adb7d18b473c648","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 23:36:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a251368fbcd7429cb498c2eda325360b : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"dfa5ce93d5f7caa49950d4565bab12fe2dc56ffb","revisions":{"7dde68c62c27c2a6c3100d9d6620875c5af499e2":{"kind":"REWORK","_number":1,"created":"2019-08-07 09:35:01.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/675068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/675068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/675068/1"}}},"commit":{"parents":[{"commit":"f72015179ea921c8264a77390359d638e0bd8727","subject":"Merge \"Update the outdated content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f72015179ea921c8264a77390359d638e0bd8727"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-12 08:35:29.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-07 09:34:41.000000000","tz":720},"subject":"Improve devmode\u003dflase when building the image","message":"Improve devmode\u003dflase when building the image\n\nDuring debugging, the following changes are also included:\n\n- Support to specify an image ID to run the integration test.\n- Modify the /etc/hosts file in cloud-init userdata.\n- Fix the reboot function bug.\n- Remove flavor support in API. Trove should get flavor information from\n  Nova.\n- Remove the quota related tests which create too many instances.\n- Remove the unsuccessful restart test.\n\nChange-Id: I31d4ee579a554f4c98f9facb9fd4b7779665a3dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7dde68c62c27c2a6c3100d9d6620875c5af499e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7dde68c62c27c2a6c3100d9d6620875c5af499e2"}]},"branch":"refs/heads/master"},"eaac05ed40113a0c771a67449b8029b23af0473e":{"kind":"REWORK","_number":2,"created":"2019-08-07 11:02:50.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/675068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/675068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/675068/2"}}},"commit":{"parents":[{"commit":"f72015179ea921c8264a77390359d638e0bd8727","subject":"Merge \"Update the outdated content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f72015179ea921c8264a77390359d638e0bd8727"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-12 08:35:29.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-07 11:02:37.000000000","tz":720},"subject":"Improve devmode\u003dflase when building the image","message":"Improve devmode\u003dflase when building the image\n\nDuring debugging, the following changes are also included:\n\n- Support to specify an image ID to run the integration test.\n- Modify the /etc/hosts file in cloud-init userdata.\n- Fix the reboot function bug.\n- Remove flavor support in API. Trove should get flavor information from\n  Nova.\n- Remove the quota related tests which create too many instances.\n- Remove the unsuccessful restart test.\n\nChange-Id: I31d4ee579a554f4c98f9facb9fd4b7779665a3dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eaac05ed40113a0c771a67449b8029b23af0473e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/eaac05ed40113a0c771a67449b8029b23af0473e"}]},"branch":"refs/heads/master"},"d39c265fde08c4d163869a48b524c830ec842ca0":{"kind":"REWORK","_number":3,"created":"2019-08-07 11:07:21.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/675068/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/675068/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/675068/3"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-12 08:35:29.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-07 11:06:48.000000000","tz":720},"subject":"Improve devmode\u003dflase when building the image","message":"Improve devmode\u003dflase when building the image\n\nDuring debugging, the following changes are also included:\n\n- Support to specify an image ID to run the integration test.\n- Modify the /etc/hosts file in cloud-init userdata.\n- Fix the reboot function bug.\n- Remove flavor support in API. Trove should get flavor information from\n  Nova.\n- Remove the quota related tests which create too many instances.\n- Remove the unsuccessful restart test.\n- Remove tools/test-setup.sh.\n\nChange-Id: I31d4ee579a554f4c98f9facb9fd4b7779665a3dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d39c265fde08c4d163869a48b524c830ec842ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d39c265fde08c4d163869a48b524c830ec842ca0"}]},"branch":"refs/heads/master"},"69eab0122c41710e35510c1ddbec8e1a3a6ad49f":{"kind":"REWORK","_number":4,"created":"2019-08-15 01:20:50.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/675068/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/675068/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/675068/4"}}},"commit":{"parents":[{"commit":"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1","subject":"Fix Trove CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-12 08:35:29.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-14 23:52:06.000000000","tz":720},"subject":"Improve devmode\u003dflase when building the image","message":"Improve devmode\u003dflase when building the image\n\nDuring debugging, the following changes are also included:\n\n- Support to specify an image ID to run the integration test.\n- Fix the reboot function bug.\n- Remove flavor support in API. Trove should get flavor information from\n  Nova.\n- Remove the quota related tests which create too many instances.\n- Remove the unsuccessful restart test.\n\nHow to run integration test with dev_mode\u003dfalse:\n\n    ADMIN_PASSWORD\u003dpassword \\\n    SERVICE_PASSWORD\u003dpassword \\\n    DEV_MODE\u003dfalse \\\n    /opt/stack/trove/integration/scripts/trovestack gate-tests mysql mysql\n\nChange-Id: I31d4ee579a554f4c98f9facb9fd4b7779665a3dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69eab0122c41710e35510c1ddbec8e1a3a6ad49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69eab0122c41710e35510c1ddbec8e1a3a6ad49f"}]},"branch":"refs/heads/master"},"a4e008b125af61a0e8046ed4dfaa205a81236f3b":{"kind":"REWORK","_number":5,"created":"2019-08-15 02:47:00.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/675068/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/675068/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/675068/5"}}},"commit":{"parents":[{"commit":"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1","subject":"Fix Trove CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-12 08:35:29.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-15 02:46:39.000000000","tz":720},"subject":"Improve devmode\u003dflase when building the image","message":"Improve devmode\u003dflase when building the image\n\nDuring debugging, the following changes are also included:\n\n- Support to specify an image ID to run the integration test.\n- Fix the reboot function bug.\n- Remove flavor support in API. Trove should get flavor information from\n  Nova.\n- Remove the quota related tests which create too many instances.\n- Remove the unsuccessful restart test.\n\nHow to run integration test with dev_mode\u003dfalse:\n\n    ADMIN_PASSWORD\u003dpassword \\\n    SERVICE_PASSWORD\u003dpassword \\\n    DEV_MODE\u003dfalse \\\n    /opt/stack/trove/integration/scripts/trovestack gate-tests mysql mysql\n\nChange-Id: I31d4ee579a554f4c98f9facb9fd4b7779665a3dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4e008b125af61a0e8046ed4dfaa205a81236f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4e008b125af61a0e8046ed4dfaa205a81236f3b"}]},"branch":"refs/heads/master"},"29448f0efcdb7f935ee66a0cacdbd3ab5b1eb3d3":{"kind":"REWORK","_number":6,"created":"2019-08-16 10:38:22.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/675068/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/675068/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/675068/6"}}},"commit":{"parents":[{"commit":"484d1b53b743a7bf637542b62bf20edfeacae4ae","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/484d1b53b743a7bf637542b62bf20edfeacae4ae"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-12 08:35:29.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-16 10:38:06.000000000","tz":720},"subject":"Improve devmode\u003dflase when building the image","message":"Improve devmode\u003dflase when building the image\n\nDuring debugging, the following changes are also included:\n\n- Support to specify an image ID to run the integration test.\n- Fix the reboot function bug.\n- Remove the unsuccessful restart test.\n\nHow to run integration test with dev_mode\u003dfalse:\n\n    ADMIN_PASSWORD\u003dpassword \\\n    SERVICE_PASSWORD\u003dpassword \\\n    DEV_MODE\u003dfalse \\\n    /opt/stack/trove/integration/scripts/trovestack gate-tests mysql mysql\n\nChange-Id: I31d4ee579a554f4c98f9facb9fd4b7779665a3dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/29448f0efcdb7f935ee66a0cacdbd3ab5b1eb3d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/29448f0efcdb7f935ee66a0cacdbd3ab5b1eb3d3"}]},"branch":"refs/heads/master"},"dfa5ce93d5f7caa49950d4565bab12fe2dc56ffb":{"kind":"REWORK","_number":7,"created":"2019-08-25 11:12:24.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/675068/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/675068/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/675068/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/675068/7"}}},"commit":{"parents":[{"commit":"c247e4842eb424214da55ca751535d107d87ceab","subject":"Merge \"Support backup filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c247e4842eb424214da55ca751535d107d87ceab"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-06-12 08:35:29.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-25 11:11:54.000000000","tz":720},"subject":"Improve devmode\u003dflase when building the image","message":"Improve devmode\u003dflase when building the image\n\nDuring debugging, the following changes are also included:\n\n- Support to specify an image ID to run the integration test.\n- Fix the reboot function bug.\n- Remove the unsuccessful restart test.\n\nHow to run integration test with dev_mode\u003dfalse:\n\n    ADMIN_PASSWORD\u003dpassword \\\n    SERVICE_PASSWORD\u003dpassword \\\n    DEV_MODE\u003dfalse \\\n    /opt/stack/trove/integration/scripts/trovestack gate-tests mysql mysql\n\nChange-Id: I31d4ee579a554f4c98f9facb9fd4b7779665a3dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dfa5ce93d5f7caa49950d4565bab12fe2dc56ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dfa5ce93d5f7caa49950d4565bab12fe2dc56ffb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
