)]}'
{"id":"openstack%2Ftrove~675098","triplet_id":"openstack%2Ftrove~master~I35e17afb9e827b1fead9d28dbf32f11ce4034a9b","project":"openstack/trove","branch":"master","topic":"fix-trove-ci","hashtags":[],"change_id":"I35e17afb9e827b1fead9d28dbf32f11ce4034a9b","subject":"Fix Trove CI jobs","status":"MERGED","created":"2019-08-07 11:50:43.000000000","updated":"2019-08-14 21:48:16.000000000","submitted":"2019-08-14 21:48:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675098-1565819296459-e4a4d76b","meta_rev_id":"341ef9e6feb47ee83146564b4cea5cc6cc4a2c37","_number":675098,"virtual_id_number":675098,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-08-14 21:48:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":2,"date":"2019-08-14 19:41:21.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2019-08-14 19:41:21.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 08:38:16.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2019-08-08 12:03:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-14 19:41:21.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-08-14 21:48:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b801a6991a0443a03974f3546d91c5a07077a8af","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 11:50:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8ce714981a4684e2e3f49da41155058e9bf6e84","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 11:51:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"54232fc813f539d9c60debbeed870e6e3d95fcc7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 12:01:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a1cd3ccfbf94d338c6b13aaafcddfad955c18493","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 12:34:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"416b8803874bd9a3a7636c88adb48be49e9c52fb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 13:21:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ecc53ebfab5f3190fc6f14c183730697d94306bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 15:44:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/5/check/build-openstack-api-ref/7b1e55c/html/ : SUCCESS in 4m 45s\n- openstack-tox-cover https://logs.opendev.org/98/675098/5/check/openstack-tox-cover/341f41c/cover/ : SUCCESS in 21m 18s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/5/check/openstack-tox-lower-constraints/b751141/ : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/5/check/openstack-tox-pep8/d872494/ : SUCCESS in 6m 27s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/5/check/openstack-tox-py27/4ce259d/ : SUCCESS in 11m 47s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/5/check/openstack-tox-py36/c1d4315/ : SUCCESS in 11m 43s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/5/check/openstack-tox-py37/5509c7a/ : SUCCESS in 16m 25s\n- openstack-tox-docs https://logs.opendev.org/98/675098/5/check/openstack-tox-docs/5756bf2/html/ : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/5/check/build-openstack-releasenotes/85d86c5/html/ : SUCCESS in 7m 58s\n- trove-functional-mysql https://logs.opendev.org/98/675098/5/check/trove-functional-mysql/3251497/ : FAILURE in 1h 26m 47s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/5/check/trove-scenario-mariadb-single/3c7232f/ : FAILURE in 41m 05s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-mariadb-multi/d6807ec/ : FAILURE in 43m 35s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/5/check/trove-scenario-mysql-single/2ae3833/ : FAILURE in 2h 07m 03s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-mysql-multi/2b82493/ : FAILURE in 1h 08m 45s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/5/check/trove-scenario-postgresql-single/d4b0fdd/ : FAILURE in 42m 39s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-postgresql-multi/4ca7d8a/ : FAILURE in 41m 18s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/5/check/trove-scenario-pxc-single/651ce69/ : FAILURE in 40m 21s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-pxc-multi/b7a777a/ : FAILURE in 59m 34s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/5/check/trove-scenario-redis-single/71e2075/ : FAILURE in 1h 24m 21s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-redis-multi/5f6e676/ : FAILURE in 1h 02m 18s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/5/check/trove-tox-fakemodetests/94fe6a0/ : SUCCESS in 6m 36s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/5/check/trove-tox-bandit-baseline/4db53d9/ : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/5/check/openstack-tox-pylint/5d47aaa/ : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":5},{"id":"4192b01d64f0b7a8137f3bad21fc73384fbe70a3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 15:46:07.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ec2877e04eb6f293d2430cbc9c4d4118ac22d509","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 18:09:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/5/check/build-openstack-api-ref/299e3e0/html/ : SUCCESS in 6m 43s\n- openstack-tox-cover https://logs.opendev.org/98/675098/5/check/openstack-tox-cover/3f9e54c/cover/ : SUCCESS in 17m 57s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/5/check/openstack-tox-lower-constraints/b2b1a0a/ : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/5/check/openstack-tox-pep8/198b1e3/ : SUCCESS in 6m 57s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/5/check/openstack-tox-py27/66e990b/ : SUCCESS in 13m 25s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/5/check/openstack-tox-py36/eb5d6f9/ : SUCCESS in 14m 46s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/5/check/openstack-tox-py37/e7dbd7f/ : SUCCESS in 23m 27s\n- openstack-tox-docs https://logs.opendev.org/98/675098/5/check/openstack-tox-docs/d2f82d0/html/ : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/5/check/build-openstack-releasenotes/bcd02de/html/ : SUCCESS in 7m 43s\n- trove-functional-mysql https://logs.opendev.org/98/675098/5/check/trove-functional-mysql/e447758/ : FAILURE in 1h 33m 22s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/5/check/trove-scenario-mariadb-single/66bc5c8/ : FAILURE in 48m 59s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-mariadb-multi/15c1798/ : FAILURE in 1h 07m 36s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/5/check/trove-scenario-mysql-single/aff9582/ : FAILURE in 2h 17m 08s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-mysql-multi/7ed2b15/ : FAILURE in 1h 01m 27s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/5/check/trove-scenario-postgresql-single/c6cddb7/ : FAILURE in 45m 14s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-postgresql-multi/f77cad2/ : FAILURE in 48m 38s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/5/check/trove-scenario-pxc-single/20dfa69/ : FAILURE in 47m 00s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-pxc-multi/60690f2/ : FAILURE in 39m 25s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/5/check/trove-scenario-redis-single/7e4b33e/ : FAILURE in 1h 08m 09s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-redis-multi/80501ca/ : FAILURE in 1h 04m 51s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/5/check/trove-tox-fakemodetests/01d186a/ : SUCCESS in 5m 30s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/5/check/trove-tox-bandit-baseline/0d88d68/ : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/5/check/openstack-tox-pylint/2ee78c8/ : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":5},{"id":"5c7fd2fb7ba12e7258c37d519be739795d1450cc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 19:27:36.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"048620a9639d1ff855eb6780764d5667ece4177c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 21:51:09.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/5/check/build-openstack-api-ref/ef477ce/html/ : SUCCESS in 5m 09s\n- openstack-tox-cover https://logs.opendev.org/98/675098/5/check/openstack-tox-cover/bab1aa9/cover/ : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/5/check/openstack-tox-lower-constraints/21cc799/ : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/5/check/openstack-tox-pep8/c6ece7e/ : SUCCESS in 6m 24s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/5/check/openstack-tox-py27/0ee5121/ : SUCCESS in 12m 45s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/5/check/openstack-tox-py36/7097a67/ : SUCCESS in 11m 52s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/5/check/openstack-tox-py37/bd035c6/ : SUCCESS in 17m 56s\n- openstack-tox-docs https://logs.opendev.org/98/675098/5/check/openstack-tox-docs/b99e666/html/ : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/5/check/build-openstack-releasenotes/7a871ac/html/ : SUCCESS in 5m 56s\n- trove-functional-mysql https://logs.opendev.org/98/675098/5/check/trove-functional-mysql/3fa8fe0/ : FAILURE in 1h 25m 09s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/5/check/trove-scenario-mariadb-single/7fa0d37/ : FAILURE in 43m 20s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-mariadb-multi/6f74ce9/ : FAILURE in 46m 19s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/5/check/trove-scenario-mysql-single/917c637/ : FAILURE in 2h 10m 07s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-mysql-multi/5e1dc35/ : FAILURE in 1h 09m 37s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/5/check/trove-scenario-postgresql-single/d236e0a/ : FAILURE in 46m 15s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-postgresql-multi/fcf8411/ : FAILURE in 44m 46s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/5/check/trove-scenario-pxc-single/3e33944/ : FAILURE in 46m 01s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-pxc-multi/d20d34f/ : FAILURE in 44m 00s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/5/check/trove-scenario-redis-single/28b9df1/ : FAILURE in 1h 17m 02s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/5/check/trove-scenario-redis-multi/b9980f9/ : FAILURE in 1h 11m 32s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/5/check/trove-tox-fakemodetests/9be4c71/ : SUCCESS in 6m 43s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/5/check/trove-tox-bandit-baseline/b9f4b31/ : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/5/check/openstack-tox-pylint/5024b5d/ : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":5},{"id":"d9e81453093e974e801f68c94dc1d90abdd5912e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-08 01:08:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e8cb8ac075ba30b05369e43c372957d54018f6a2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-08 01:09:11.000000000","message":"Patch Set 6:\n\nThanks Andreas, I will keep working on this if you don\u0027t mind.","accounts_in_message":[],"_revision_number":6},{"id":"c1be8bb5d39ffae5e5995e03ef6d96ef972a4ba9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 02:10:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/6/check/build-openstack-api-ref/6c6d145/html/ : SUCCESS in 5m 57s\n- openstack-tox-cover https://logs.opendev.org/98/675098/6/check/openstack-tox-cover/ce7f3aa/cover/ : SUCCESS in 19m 57s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/6/check/openstack-tox-lower-constraints/20b6154/ : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/6/check/openstack-tox-pep8/d0efc5a/ : SUCCESS in 5m 09s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/6/check/openstack-tox-py27/6063da1/ : SUCCESS in 11m 42s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/6/check/openstack-tox-py36/3fc022b/ : SUCCESS in 16m 13s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/6/check/openstack-tox-py37/11444fc/ : SUCCESS in 18m 12s\n- openstack-tox-docs https://logs.opendev.org/98/675098/6/check/openstack-tox-docs/ffba624/html/ : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/6/check/build-openstack-releasenotes/8afe459/html/ : SUCCESS in 5m 49s\n- trove-functional-mysql https://logs.opendev.org/98/675098/6/check/trove-functional-mysql/275ea42/ : FAILURE in 39m 43s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/6/check/trove-scenario-mariadb-single/b017276/ : FAILURE in 37m 09s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/6/check/trove-scenario-mariadb-multi/3cfe4d4/ : FAILURE in 32m 57s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/6/check/trove-scenario-mysql-single/a8ca830/ : FAILURE in 54m 29s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/6/check/trove-scenario-mysql-multi/12527dd/ : FAILURE in 35m 01s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/6/check/trove-scenario-postgresql-single/668d04e/ : FAILURE in 39m 57s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/6/check/trove-scenario-postgresql-multi/498d0a9/ : FAILURE in 35m 43s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/6/check/trove-scenario-pxc-single/b087b9d/ : FAILURE in 37m 58s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/6/check/trove-scenario-pxc-multi/809fe66/ : FAILURE in 36m 49s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/6/check/trove-scenario-redis-single/796e96b/ : FAILURE in 32m 11s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/6/check/trove-scenario-redis-multi/c3315b2/ : FAILURE in 30m 48s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/6/check/trove-tox-fakemodetests/198d17c/ : SUCCESS in 6m 19s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/6/check/trove-tox-bandit-baseline/060e34a/ : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/6/check/openstack-tox-pylint/291007a/ : SUCCESS in 11m 13s","accounts_in_message":[],"_revision_number":6},{"id":"4d7e83292600935b5af11427eb2807c89d7b8c36","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-08 03:14:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7c39108cb947bd4d74f95a73b024d060fb000ecd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-08 04:41:35.000000000","message":"Patch Set 7:\n\nLingxian, please continue if you like. Happy to hand this over ;)","accounts_in_message":[],"_revision_number":7},{"id":"085a8a2cf0511978726a1856aafb131120d477e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 04:46:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/7/check/build-openstack-api-ref/1ecf708/html/ : SUCCESS in 6m 52s\n- openstack-tox-cover https://logs.opendev.org/98/675098/7/check/openstack-tox-cover/e2664fc/cover/ : SUCCESS in 17m 49s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/7/check/openstack-tox-lower-constraints/f316c22/ : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/7/check/openstack-tox-pep8/9ef12f2/ : SUCCESS in 7m 15s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/7/check/openstack-tox-py27/542eb67/ : SUCCESS in 9m 56s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/7/check/openstack-tox-py36/4dfc745/ : SUCCESS in 13m 25s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/7/check/openstack-tox-py37/f7007d4/ : SUCCESS in 17m 33s\n- openstack-tox-docs https://logs.opendev.org/98/675098/7/check/openstack-tox-docs/5900ed3/html/ : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/7/check/build-openstack-releasenotes/63b0952/html/ : SUCCESS in 7m 11s\n- trove-functional-mysql https://logs.opendev.org/98/675098/7/check/trove-functional-mysql/27d658a/ : FAILURE in 40m 58s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/7/check/trove-scenario-mariadb-single/e0995e1/ : FAILURE in 38m 27s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/7/check/trove-scenario-mariadb-multi/e641498/ : FAILURE in 56m 34s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/7/check/trove-scenario-mysql-single/ed3dfb6/ : FAILURE in 37m 52s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/7/check/trove-scenario-mysql-multi/0ae9dca/ : FAILURE in 55m 07s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/7/check/trove-scenario-postgresql-single/2d1a5d6/ : FAILURE in 38m 25s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/7/check/trove-scenario-postgresql-multi/feb5751/ : FAILURE in 36m 34s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/7/check/trove-scenario-pxc-single/1865872/ : FAILURE in 49m 48s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/7/check/trove-scenario-pxc-multi/af1f760/ : FAILURE in 49m 35s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/7/check/trove-scenario-redis-single/5db0333/ : FAILURE in 1h 30m 41s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/7/check/trove-scenario-redis-multi/0e18a50/ : FAILURE in 38m 05s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/7/check/trove-tox-fakemodetests/d43c547/ : SUCCESS in 5m 26s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/7/check/trove-tox-bandit-baseline/c3052a7/ : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/7/check/openstack-tox-pylint/9af688f/ : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":7},{"id":"7de5d1cf0d2af72c7f56c588762087cb853b3999","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-08 07:22:00.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"316fa56f98aa450ca013e03431224e89b59ef2b7","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-08-08 08:38:16.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\nLooks good, just a minor nit with the comment","accounts_in_message":[],"_revision_number":8},{"id":"d61204e54b8303c84cc5bd1e6d8ad3876a82c2d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:57:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/8/check/build-openstack-api-ref/d4784c3/html/ : SUCCESS in 5m 52s\n- openstack-tox-cover https://logs.opendev.org/98/675098/8/check/openstack-tox-cover/d0829c9/cover/ : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/8/check/openstack-tox-lower-constraints/127bf34/ : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/8/check/openstack-tox-pep8/7d0eb8a/ : SUCCESS in 6m 39s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/8/check/openstack-tox-py27/c5e4285/ : SUCCESS in 10m 45s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/8/check/openstack-tox-py36/68244ea/ : SUCCESS in 12m 43s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/8/check/openstack-tox-py37/f1d2d3f/ : SUCCESS in 21m 59s\n- openstack-tox-docs https://logs.opendev.org/98/675098/8/check/openstack-tox-docs/3e391b1/html/ : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/8/check/build-openstack-releasenotes/c490b21/html/ : SUCCESS in 6m 29s\n- trove-functional-mysql https://logs.opendev.org/98/675098/8/check/trove-functional-mysql/56d1e0b/ : SUCCESS in 2h 24m 20s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/8/check/trove-scenario-mariadb-single/de4949d/ : FAILURE in 35m 25s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/8/check/trove-scenario-mariadb-multi/16d3770/ : FAILURE in 1h 09m 23s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/8/check/trove-scenario-mysql-single/b812240/ : FAILURE in 1h 04m 31s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/8/check/trove-scenario-mysql-multi/cadefb3/ : SUCCESS in 1h 56m 27s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/8/check/trove-scenario-postgresql-single/59fbe10/ : FAILURE in 43m 29s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/8/check/trove-scenario-postgresql-multi/832f679/ : FAILURE in 45m 00s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/8/check/trove-scenario-pxc-single/aa686df/ : FAILURE in 42m 49s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/8/check/trove-scenario-pxc-multi/d3e6716/ : FAILURE in 43m 33s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/8/check/trove-scenario-redis-single/a0c9202/ : FAILURE in 1h 23m 22s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/8/check/trove-scenario-redis-multi/b6a479e/ : SUCCESS in 1h 29m 47s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/8/check/trove-tox-fakemodetests/ae926b6/ : SUCCESS in 5m 47s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/8/check/trove-tox-bandit-baseline/b3e6ef2/ : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/8/check/openstack-tox-pylint/dd746ab/ : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":8},{"id":"0cbf1a73cef20d4aeb29aad5970996b22181d426","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-08 10:34:38.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0d03f7b91f36e06db81f83dec8c6031278e5e6ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 12:34:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/9/check/build-openstack-api-ref/443e095/html/ : SUCCESS in 4m 07s\n- openstack-tox-cover https://logs.opendev.org/98/675098/9/check/openstack-tox-cover/89cbbd9/cover/ : SUCCESS in 20m 32s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/9/check/openstack-tox-lower-constraints/f43bb03/ : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/9/check/openstack-tox-pep8/aa91a7c/ : SUCCESS in 6m 51s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/9/check/openstack-tox-py27/5e15a99/ : SUCCESS in 11m 56s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/9/check/openstack-tox-py36/0aa51df/ : SUCCESS in 13m 40s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/9/check/openstack-tox-py37/2b3d638/ : SUCCESS in 18m 12s\n- openstack-tox-docs https://logs.opendev.org/98/675098/9/check/openstack-tox-docs/6e38d61/html/ : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/9/check/build-openstack-releasenotes/b361c41/html/ : SUCCESS in 6m 32s\n- trove-functional-mysql https://logs.opendev.org/98/675098/9/check/trove-functional-mysql/10a749f/ : SUCCESS in 1h 52m 19s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/9/check/trove-scenario-mariadb-single/4ff45a9/ : FAILURE in 47m 22s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-mariadb-multi/a2d148e/ : FAILURE in 48m 46s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/9/check/trove-scenario-mysql-single/7f03f02/ : FAILURE in 1h 55m 37s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-mysql-multi/a1cd97b/ : SUCCESS in 1h 38m 43s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/9/check/trove-scenario-postgresql-single/4c1cfe7/ : FAILURE in 36m 47s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-postgresql-multi/5a3c119/ : FAILURE in 45m 16s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/9/check/trove-scenario-pxc-single/cadec2e/ : FAILURE in 47m 20s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-pxc-multi/06fe578/ : FAILURE in 41m 19s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/9/check/trove-scenario-redis-single/236e527/ : FAILURE in 1h 49m 02s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-redis-multi/e3e6118/ : SUCCESS in 1h 15m 58s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/9/check/trove-tox-fakemodetests/df687c3/ : SUCCESS in 5m 51s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/9/check/trove-tox-bandit-baseline/fff23b2/ : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/9/check/openstack-tox-pylint/3faa7e2/ : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":9},{"id":"0105f660cae130906a3c05e51c4883a63c345bb2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-08 13:01:11.000000000","message":"Patch Set 9:\n\nrecheck - double checking ;)","accounts_in_message":[],"_revision_number":9},{"id":"bd977acee1fb07485c597211b42e1d5b47ef0904","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:12:15.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/9/check/build-openstack-api-ref/e263dc2/html/ : SUCCESS in 7m 18s\n- openstack-tox-cover https://logs.opendev.org/98/675098/9/check/openstack-tox-cover/4afe78c/cover/ : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/9/check/openstack-tox-lower-constraints/3d770fc/ : SUCCESS in 13m 00s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/9/check/openstack-tox-pep8/fbdd62f/ : SUCCESS in 5m 01s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/9/check/openstack-tox-py27/4587105/ : SUCCESS in 10m 52s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/9/check/openstack-tox-py36/31415c0/ : SUCCESS in 12m 42s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/9/check/openstack-tox-py37/ff43455/ : SUCCESS in 20m 02s\n- openstack-tox-docs https://logs.opendev.org/98/675098/9/check/openstack-tox-docs/b128a50/html/ : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/9/check/build-openstack-releasenotes/a3092bd/html/ : SUCCESS in 7m 43s\n- trove-functional-mysql https://logs.opendev.org/98/675098/9/check/trove-functional-mysql/b122d6b/ : SUCCESS in 2h 06m 33s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/9/check/trove-scenario-mariadb-single/c7829cd/ : FAILURE in 57m 26s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-mariadb-multi/6006477/ : FAILURE in 1h 02m 01s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/9/check/trove-scenario-mysql-single/a1e0973/ : FAILURE in 1h 44m 50s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-mysql-multi/3b6bd22/ : SUCCESS in 1h 34m 11s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/9/check/trove-scenario-postgresql-single/d70497a/ : FAILURE in 42m 03s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-postgresql-multi/fd33f79/ : FAILURE in 40m 26s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/9/check/trove-scenario-pxc-single/a5cfab6/ : FAILURE in 40m 43s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-pxc-multi/9077aef/ : FAILURE in 46m 01s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/9/check/trove-scenario-redis-single/a463a22/ : FAILURE in 1h 22m 50s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-redis-multi/bce43a1/ : SUCCESS in 1h 20m 13s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/9/check/trove-tox-fakemodetests/e0685e5/ : SUCCESS in 6m 15s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/9/check/trove-tox-bandit-baseline/4ee8f0c/ : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/9/check/openstack-tox-pylint/acb74ab/ : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":9},{"id":"cebcb2a0d9ce19ad56b853d3667db55185a13aff","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-10 08:53:18.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"453a9b38d4918712d8a2bb25c5a7b5c1b93ac11b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 10:59:15.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/9/check/build-openstack-api-ref/de4593e/html/ : SUCCESS in 4m 01s\n- openstack-tox-cover https://logs.opendev.org/98/675098/9/check/openstack-tox-cover/d78bcc7/cover/ : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/9/check/openstack-tox-lower-constraints/22067a4/ : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/9/check/openstack-tox-pep8/ae20844/ : SUCCESS in 5m 28s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/9/check/openstack-tox-py27/e70b184/ : SUCCESS in 11m 48s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/9/check/openstack-tox-py36/9bebb6e/ : SUCCESS in 13m 27s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/9/check/openstack-tox-py37/cb6ae31/ : SUCCESS in 19m 27s\n- openstack-tox-docs https://logs.opendev.org/98/675098/9/check/openstack-tox-docs/c96bfe8/html/ : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/9/check/build-openstack-releasenotes/5e0120d/html/ : SUCCESS in 6m 00s\n- trove-functional-mysql https://logs.opendev.org/98/675098/9/check/trove-functional-mysql/c2d4560/ : SUCCESS in 2h 00m 28s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/9/check/trove-scenario-mariadb-single/88bcf22/ : FAILURE in 44m 56s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-mariadb-multi/c2b797d/ : FAILURE in 37m 34s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/9/check/trove-scenario-mysql-single/e65c3b5/ : FAILURE in 1h 36m 22s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-mysql-multi/9859168/ : SUCCESS in 1h 38m 37s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/9/check/trove-scenario-postgresql-single/d697565/ : FAILURE in 35m 03s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-postgresql-multi/ff5726d/ : FAILURE in 51m 23s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/9/check/trove-scenario-pxc-single/26360b2/ : FAILURE in 44m 55s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-pxc-multi/b615187/ : FAILURE in 37m 21s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/9/check/trove-scenario-redis-single/0f0138a/ : FAILURE in 1h 10m 24s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/9/check/trove-scenario-redis-multi/fa8cc4d/ : SUCCESS in 1h 43m 46s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/9/check/trove-tox-fakemodetests/f86d4fc/ : SUCCESS in 5m 02s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/9/check/trove-tox-bandit-baseline/17c45a6/ : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/9/check/openstack-tox-pylint/4b48c5e/ : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":9},{"id":"d9012d73da5850c73f7e53b2582dfc70e2756c4d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-10 11:47:38.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f60ccaede639a032513b8b8ea9bad602c9e59fec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 13:56:26.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/10/check/build-openstack-api-ref/7ba2795/html/ : SUCCESS in 4m 32s\n- openstack-tox-cover https://logs.opendev.org/98/675098/10/check/openstack-tox-cover/6280a5a/cover/ : SUCCESS in 16m 57s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/10/check/openstack-tox-lower-constraints/f0de301/ : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/10/check/openstack-tox-pep8/20842fc/ : SUCCESS in 5m 09s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/10/check/openstack-tox-py27/567a814/ : SUCCESS in 11m 23s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/10/check/openstack-tox-py36/46e3889/ : SUCCESS in 10m 38s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/10/check/openstack-tox-py37/d749904/ : SUCCESS in 16m 23s\n- openstack-tox-docs https://logs.opendev.org/98/675098/10/check/openstack-tox-docs/27fc510/html/ : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/10/check/build-openstack-releasenotes/ea61570/html/ : SUCCESS in 5m 23s\n- trove-functional-mysql https://logs.opendev.org/98/675098/10/check/trove-functional-mysql/26cb747/ : SUCCESS in 2h 07m 39s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/10/check/trove-scenario-mariadb-single/f051eab/ : FAILURE in 41m 45s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/10/check/trove-scenario-mariadb-multi/11a05bb/ : FAILURE in 33m 47s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/10/check/trove-scenario-mysql-single/b5663c6/ : FAILURE in 1h 25m 23s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/10/check/trove-scenario-mysql-multi/7308b2b/ : SUCCESS in 1h 38m 33s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/10/check/trove-scenario-postgresql-single/4b13d88/ : FAILURE in 32m 37s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/10/check/trove-scenario-postgresql-multi/83e46e2/ : FAILURE in 40m 16s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/10/check/trove-scenario-pxc-single/aaefbe3/ : FAILURE in 33m 26s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/10/check/trove-scenario-pxc-multi/aaa0c74/ : FAILURE in 36m 30s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/10/check/trove-scenario-redis-single/ed159f5/ : FAILURE in 1h 02m 47s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/10/check/trove-scenario-redis-multi/52d2502/ : SUCCESS in 1h 04m 00s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/10/check/trove-tox-fakemodetests/0b0596a/ : SUCCESS in 5m 31s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/10/check/trove-tox-bandit-baseline/a56d238/ : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/10/check/openstack-tox-pylint/a5959df/ : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":10},{"id":"63f6bedcc5b747b1540f84bc8ff23175a28bb460","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-11 12:03:44.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f4967c5382c18d1ef1e44c4ccb14bc43679c1d95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-11 14:05:17.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/11/check/build-openstack-api-ref/f51fded/html/ : SUCCESS in 4m 39s\n- openstack-tox-cover https://logs.opendev.org/98/675098/11/check/openstack-tox-cover/63b4481/cover/ : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/11/check/openstack-tox-lower-constraints/9c2db68/ : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/11/check/openstack-tox-pep8/15121e2/ : SUCCESS in 4m 21s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/11/check/openstack-tox-py27/2500b87/ : SUCCESS in 9m 24s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/11/check/openstack-tox-py36/f6ab84b/ : SUCCESS in 9m 14s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/11/check/openstack-tox-py37/1a5f9db/ : SUCCESS in 16m 17s\n- openstack-tox-docs https://logs.opendev.org/98/675098/11/check/openstack-tox-docs/2b5084f/html/ : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/11/check/build-openstack-releasenotes/1b6cc2c/html/ : SUCCESS in 6m 12s\n- trove-functional-mysql https://logs.opendev.org/98/675098/11/check/trove-functional-mysql/60b8ce6/ : SUCCESS in 2h 00m 20s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/11/check/trove-scenario-mariadb-single/be9e979/ : FAILURE in 44m 36s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/11/check/trove-scenario-mariadb-multi/eef4238/ : FAILURE in 35m 08s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/11/check/trove-scenario-mysql-single/5789b51/ : FAILURE in 1h 28m 21s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/11/check/trove-scenario-mysql-multi/3d71263/ : SUCCESS in 1h 19m 44s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/11/check/trove-scenario-postgresql-single/0000017/ : FAILURE in 36m 57s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/11/check/trove-scenario-postgresql-multi/122c256/ : FAILURE in 35m 08s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/11/check/trove-scenario-pxc-single/4ee55aa/ : FAILURE in 38m 25s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/11/check/trove-scenario-pxc-multi/0dcd35c/ : FAILURE in 33m 37s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/11/check/trove-scenario-redis-single/aa5d50d/ : FAILURE in 59m 16s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/11/check/trove-scenario-redis-multi/9937f68/ : SUCCESS in 1h 14m 31s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/11/check/trove-tox-fakemodetests/6dcd3ec/ : SUCCESS in 6m 02s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/11/check/trove-tox-bandit-baseline/7562bc5/ : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/11/check/openstack-tox-pylint/e29c777/ : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":11},{"id":"adaba3a2b3e2f872a7912b83a700b2ade98557a4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-11 23:06:35.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"35098932f6d717879d9167ff8377a53985208ce9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-11 23:29:58.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"1291e3c7a2170e81d3c08b71168443ee15e49103","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 01:05:42.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/13/check/build-openstack-api-ref/448176c/html/ : SUCCESS in 4m 38s\n- openstack-tox-cover https://logs.opendev.org/98/675098/13/check/openstack-tox-cover/79c3f3d/cover/ : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/13/check/openstack-tox-lower-constraints/590f4c8/ : SUCCESS in 11m 52s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/13/check/openstack-tox-pep8/596d800/ : SUCCESS in 4m 38s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/13/check/openstack-tox-py27/1f41c71/ : SUCCESS in 10m 41s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/13/check/openstack-tox-py36/da66209/ : SUCCESS in 10m 56s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/13/check/openstack-tox-py37/eb2cbf0/ : SUCCESS in 23m 36s\n- openstack-tox-docs https://logs.opendev.org/98/675098/13/check/openstack-tox-docs/7bc33b5/html/ : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/13/check/build-openstack-releasenotes/fd729bc/html/ : SUCCESS in 4m 56s\n- trove-functional-mysql https://logs.opendev.org/98/675098/13/check/trove-functional-mysql/c3763f5/ : FAILURE in 33m 56s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/13/check/trove-scenario-mariadb-single/eb9f8ac/ : FAILURE in 44m 31s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-mariadb-multi/2a24d23/ : FAILURE in 38m 43s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/13/check/trove-scenario-mysql-single/02df740/ : FAILURE in 1h 34m 11s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-mysql-multi/a43d1fd/ : SUCCESS in 1h 34m 11s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/13/check/trove-scenario-postgresql-single/6e5166f/ : FAILURE in 31m 27s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-postgresql-multi/54bccfb/ : FAILURE in 39m 21s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/13/check/trove-scenario-pxc-single/aeaf586/ : FAILURE in 42m 01s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-pxc-multi/cdc2478/ : FAILURE in 37m 55s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/13/check/trove-scenario-redis-single/5d18fed/ : FAILURE in 1h 06m 29s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-redis-multi/4fb2535/ : SUCCESS in 1h 04m 21s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/13/check/trove-tox-fakemodetests/d387937/ : SUCCESS in 5m 43s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/13/check/trove-tox-bandit-baseline/e2271f9/ : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/13/check/openstack-tox-pylint/3d8a337/ : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":13},{"id":"455d62172671c18a71f1f7eda74704362f5a2cea","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-12 01:09:55.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"f239d7c9c36636611aab5082d2bab4d5f4c77d97","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-12 02:53:35.000000000","message":"Patch Set 13:\n\nrecheck\n\nLast time failure:\n2019-08-12 01:57:12.101116 | controller | 2019-08-12 01:57:12.095 | fatal: write error: No space left on device","accounts_in_message":[],"_revision_number":13},{"id":"c29128a8bdcfdd8c0532b89107fe9020ee4e6ab4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 03:08:55.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/13/check/build-openstack-api-ref/e38da04/html/ : SUCCESS in 5m 28s\n- openstack-tox-cover https://logs.opendev.org/98/675098/13/check/openstack-tox-cover/9d1f594/cover/ : SUCCESS in 14m 46s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/13/check/openstack-tox-lower-constraints/8cad551/ : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/13/check/openstack-tox-pep8/fc3c3ef/ : SUCCESS in 4m 54s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/13/check/openstack-tox-py27/33bc26d/ : SUCCESS in 12m 06s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/13/check/openstack-tox-py36/f1dd900/ : SUCCESS in 12m 59s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/13/check/openstack-tox-py37/305d602/ : SUCCESS in 17m 24s\n- openstack-tox-docs https://logs.opendev.org/98/675098/13/check/openstack-tox-docs/9515f1e/html/ : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/13/check/build-openstack-releasenotes/8589418/html/ : SUCCESS in 6m 43s\n- trove-functional-mysql https://logs.opendev.org/98/675098/13/check/trove-functional-mysql/989d280/ : SUCCESS in 1h 57m 24s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/13/check/trove-scenario-mariadb-single/bfac211/ : FAILURE in 1h 20m 57s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-mariadb-multi/99f9fa3/ : FAILURE in 36m 41s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/13/check/trove-scenario-mysql-single/554ede8/ : FAILURE in 49m 38s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-mysql-multi/a6aa2ec/ : SUCCESS in 1h 37m 27s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/13/check/trove-scenario-postgresql-single/b3d66fc/ : FAILURE in 38m 46s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-postgresql-multi/b5ccc75/ : FAILURE in 36m 12s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/13/check/trove-scenario-pxc-single/4741ba8/ : FAILURE in 49m 23s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-pxc-multi/de58158/ : FAILURE in 37m 19s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/13/check/trove-scenario-redis-single/9b22aa3/ : FAILURE in 1h 14m 36s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-redis-multi/4f2e52b/ : SUCCESS in 1h 08m 55s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/13/check/trove-tox-fakemodetests/83b31cc/ : SUCCESS in 6m 26s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/13/check/trove-tox-bandit-baseline/df2b387/ : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/13/check/openstack-tox-pylint/2a58bcc/ : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":13},{"id":"47b0a7fde67ed827bf840fa9b44f094c2f0a38b1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-12 03:10:28.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"394ac25565795d1000a6c0453fede9f40b479c8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 06:44:20.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/98/675098/13/check/build-openstack-api-ref/1dba089/html/ : SUCCESS in 4m 07s\n- openstack-tox-cover https://logs.opendev.org/98/675098/13/check/openstack-tox-cover/11cb1cd/cover/ : SUCCESS in 16m 11s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/675098/13/check/openstack-tox-lower-constraints/8928e8d/ : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://logs.opendev.org/98/675098/13/check/openstack-tox-pep8/80c202b/ : SUCCESS in 5m 13s\n- openstack-tox-py27 https://logs.opendev.org/98/675098/13/check/openstack-tox-py27/65a2c92/ : SUCCESS in 10m 34s\n- openstack-tox-py36 https://logs.opendev.org/98/675098/13/check/openstack-tox-py36/32b9aa6/ : SUCCESS in 11m 13s\n- openstack-tox-py37 https://logs.opendev.org/98/675098/13/check/openstack-tox-py37/3b90455/ : SUCCESS in 16m 42s\n- openstack-tox-docs https://logs.opendev.org/98/675098/13/check/openstack-tox-docs/ac506ed/html/ : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://logs.opendev.org/98/675098/13/check/build-openstack-releasenotes/1f4ff47/html/ : SUCCESS in 5m 47s\n- trove-functional-mysql https://logs.opendev.org/98/675098/13/check/trove-functional-mysql/07d1dfd/ : SUCCESS in 1h 55m 14s\n- trove-scenario-mariadb-single https://logs.opendev.org/98/675098/13/check/trove-scenario-mariadb-single/a67ed91/ : FAILURE in 35m 04s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-mariadb-multi/e2b21ba/ : FAILURE in 47m 23s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/98/675098/13/check/trove-scenario-mysql-single/5d36beb/ : FAILURE in 1h 26m 42s\n- trove-scenario-mysql-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-mysql-multi/47f9440/ : TIMED_OUT in 3h 32m 48s\n- trove-scenario-postgresql-single https://logs.opendev.org/98/675098/13/check/trove-scenario-postgresql-single/b995ccf/ : FAILURE in 44m 44s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-postgresql-multi/97fd1f0/ : FAILURE in 32m 03s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/98/675098/13/check/trove-scenario-pxc-single/b29bead/ : FAILURE in 45m 15s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-pxc-multi/c05f55d/ : FAILURE in 33m 55s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/98/675098/13/check/trove-scenario-redis-single/8758d7f/ : FAILURE in 1h 01m 56s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/98/675098/13/check/trove-scenario-redis-multi/991a89b/ : SUCCESS in 1h 06m 42s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/98/675098/13/check/trove-tox-fakemodetests/332d01a/ : SUCCESS in 5m 10s\n- trove-tox-bandit-baseline https://logs.opendev.org/98/675098/13/check/trove-tox-bandit-baseline/dc457c0/ : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/98/675098/13/check/openstack-tox-pylint/aee22b2/ : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":13},{"id":"4edd1df703e647d4be7b866847b8638227504e08","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-13 01:25:46.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"12dfbb461971bc1f5f8200be02bd1e87e19944d2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-13 03:00:39.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"05da0f1168f1db8dde363f70e5a4122b6b3a8688","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 05:15:42.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b7f20a3840cc4cb3a1ef5a384278d091 : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bf0c144010f4656b27236e07fb71228 : SUCCESS in 18m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18b20d5ba84848c6b0806935b954933d : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f592954131a34ae2bf964eeea7a38586 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b71adea920484041a285df1ed410dae2 : SUCCESS in 12m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f180c4608b3e465bbc65548ac6f411fb : SUCCESS in 13m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e574599cfa934991b65a0a90488fffa8 : SUCCESS in 19m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/debc0086c3504f64a46c2dead4b1aa3c : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44b89099f6b64753819137c3959e27fc : SUCCESS in 5m 53s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4ebe9ab52f6c400a968babab433faf80 : SUCCESS in 2h 09m 19s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/0202d4f293594d4782d975163a5b7ce6 : FAILURE in 48m 52s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/5ade7d7843744dd3b5acfebcdb127213 : FAILURE in 47m 04s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/742c7f99f6234ed59f5eb2c96a04633c : FAILURE in 1h 25m 32s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/9adb107267e24a67a72dbf2c13f47b14 : SUCCESS in 1h 40m 18s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/01366a2be4944e438f8e58c46f76fb6d : FAILURE in 42m 25s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/2ef21aab3ea64418af889669ca01ca3d : FAILURE in 40m 54s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/ebb646348a6c4a49894221f356d97558 : FAILURE in 48m 37s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/85674df98343459f95a7a156d8a561af : FAILURE in 41m 12s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/e43c72fc4f28495387d7ecdfc2e227a3 : FAILURE in 1h 25m 44s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/fe3218cdbe2c496c970df47f15affd64 : SUCCESS in 1h 22m 55s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/eb7b649b982148ba9e06760e3c09eee5 : SUCCESS in 5m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/40fb575b600042639ee5538f56b038e0 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/71b0992f25dd4b92b55902ebca2dab35 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":14},{"id":"ec6829005bb4b37b59a89a02065c9527e047cbf2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-13 08:40:07.000000000","message":"Patch Set 14:\n\nrecheck\n\nIt\u0027s annoying, all the jobs are working well in my dev environment :-(","accounts_in_message":[],"_revision_number":14},{"id":"35a63c094e4bfa389759a05356a01f5a3bcdaf5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 10:47:49.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/35fc7dcb56394f9c805b321400d992d1 : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98ad0242df204f6b8e8db0af6b5f0684 : SUCCESS in 18m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e07b6a52e6749a4b2086db63c4137ce : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/607028499d77474d88d3c12c03c8512d : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a56ed0fe3781484eb18d38e96a643356 : SUCCESS in 10m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc210f9db3044700a52c0bd4256e3141 : SUCCESS in 12m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fab04ec226fb4a5ca823657dfd9747af : SUCCESS in 19m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e888c6f37d564870a0c1bfb54dd3651e : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0221ea3b263f4fb79afc1ca7077882b6 : SUCCESS in 6m 31s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c98974aae5694c189311c8712a43b586 : SUCCESS in 2h 00m 04s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/2de3811313834d039b4f8539bae1f04e : FAILURE in 39m 47s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/6daa75b829c14fd381b27964267396a0 : FAILURE in 45m 03s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/856e10a9e285472e9ff351242c95899d : FAILURE in 1h 47m 03s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7b8b588f3f05425b8aef49c21deb0a0e : SUCCESS in 1h 57m 11s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/713428e71d9d4595a2096d2fa0846307 : FAILURE in 45m 58s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/fa7f13f8e3084555beddbb971fe20f85 : FAILURE in 41m 00s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/ba99a26800384b8694603e54dbd2d383 : FAILURE in 52m 09s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/60b3b6d1dfb14b33a879cab17f26af87 : FAILURE in 42m 45s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/0ac1e8186c8d4ed4b32cdf05fccbc613 : FAILURE in 1h 22m 34s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/8168dbc36647472cb5b657a634679f0a : SUCCESS in 1h 16m 25s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/adaf859d01a84a39ac2320af2ce3be98 : SUCCESS in 4m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/edbd8ea7653d4e62bc077aec0f5ce1e3 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/683d91f265444e2ca1bdff1e3b4eef7e : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":14},{"id":"1e858148fbf37ba266821b03e16575697862c8ec","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-13 11:35:04.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"28cbd966b028f0d275e41815b83dc96debe20b7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 14:02:23.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a5306b332afb4dc986c7d129b6517677 : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/133e0ff66f9243f5b662f9c777add8e7 : SUCCESS in 19m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/690a20d1f5c14766a2e158bf6e93e719 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04669fe6628241e2ab7a336acc95e39e : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be9d8e804d584ed1927cd0eb08402d22 : SUCCESS in 12m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6c242b324624377b8fcdee7dcb28be0 : SUCCESS in 13m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c143609803f84818b173726a3d042f58 : SUCCESS in 18m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89bbb1035bc54d49aba2279007f1c9d8 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b69d2c5d4cf846d8bdc0bd997f4fa165 : SUCCESS in 6m 08s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7d61ff52a3854929b598c24c85bbee57 : SUCCESS in 2h 20m 55s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/c228e326f82144b892801c93bc3ebb82 : FAILURE in 42m 44s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/30c8eb34c6e94ce59513566867115551 : FAILURE in 43m 24s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/cfce1249fd8f4f4b8467d1def1b5c376 : FAILURE in 1h 51m 05s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/714008a356d1441d8e93e1b622fed53c : SUCCESS in 1h 36m 42s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/12c44c7d594a4b6db1b7f7fbd76dddc3 : FAILURE in 45m 21s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ece0f46e6c934806a7cb2bfe30d8204c : FAILURE in 41m 22s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/2930a8093f774e7eaa04b70bb4269e4c : FAILURE in 41m 43s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/69161cba261e4cd4a7b3bce5f52fc72c : FAILURE in 40m 52s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/e3ea1b3b50cf4ed0a4592111b23a084f : FAILURE in 1h 11m 20s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/43ac164c858443569754643f65f9147c : SUCCESS in 1h 25m 54s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/6ea3d569009b4a25b91b538f29b8a63b : SUCCESS in 5m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/58d40adb53a04ef0959d75fab0af63aa : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ec8f5349c65a4e0388547de11c15cbcb : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":15},{"id":"93aec9d0c9adff5e39c300a9652fc22fe06cef7e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-14 10:02:46.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"307d53a91a6495b6b44c5a051fcab551ed28b811","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 12:41:27.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e7c694c2c5034ee6b95df479c86a9fee : SUCCESS in 6m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ec13daa6e4347f98b232d7ff32a0c93 : SUCCESS in 19m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cadbd1aba7f44c5b9bed2cf17b0ef73f : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12dee46c9a74507abe47628ef6aeda7 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1efe180fe3714bf4be1cc4040c7dbbe2 : SUCCESS in 11m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9366ea2b39a940608428af413ebc20e8 : SUCCESS in 12m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17f80e8cc5ff45c7adad4d956a810bcb : SUCCESS in 20m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f5e949d9f042aab244be1ba7c611f9 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a20f3a0c1e2948a2b61e39c6e411c045 : SUCCESS in 6m 48s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4d2e04c64351406fb976d5a3374ee73c : SUCCESS in 2h 33m 13s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/7f8107a9a40a4beebf4171b258fbed4e : FAILURE in 45m 51s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/f62734d3c32e42e5a0b014d6a5a78ead : FAILURE in 50m 35s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8f62cc9abd6e499f8a0fdd8e444c1887 : SUCCESS in 1h 46m 25s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/fdda7231d065462e82d80b5027ea8299 : SUCCESS in 1h 41m 41s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/c18608458e474291a41482391322bb4b : FAILURE in 44m 37s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/77ea3f2250fe457588af517d295068fc : FAILURE in 44m 20s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/7afcee96417a42a6904b729fdda8ae39 : FAILURE in 53m 32s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/a52a29ca73b340d2940a2526616a56ea : FAILURE in 43m 56s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/b7244a917c3f4de1818db7a477dda768 : FAILURE in 1h 36m 08s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/4183eb71f8244b3292e305e40c0e1974 : SUCCESS in 1h 25m 34s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/bfddc954e5a6403ca38e8851f1c0c600 : SUCCESS in 4m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/826bed3073654a9a9be704ef8430f7e2 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3462af26304c4cf3832986c726b5b720 : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":16},{"id":"777da243a3e2ebe6f2f5418ffe8c7b42dcf5df3e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-14 19:41:21.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"08f1d1fec12293cef8dec81003d1e2745da3a8de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 19:41:35.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"4d8f9e0f6a262526fe3b0e122fcacea78648d4f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 21:48:15.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9b147b7038b3436b8c0e4f244135c657 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/080e57e955b7417c87033d17df36e070 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df40e7dcc81a415bacaabe86dd0e79ab : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ebfe3dd5143049f2ad008692d392288d : SUCCESS in 11m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dac9580554345dcaba907e0e1f5bd83 : SUCCESS in 12m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e091591d9914435ab8d3bf447bb024b : SUCCESS in 17m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa1d900cd446447e9c823eac591a74f3 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a6e301fdcb347da9d642727ba6194b4 : SUCCESS in 7m 40s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/395dc3e8c283477f8e9d392994359aaa : SUCCESS in 2h 01m 07s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f625c7cf2d744e958716b64616ef6417 : SUCCESS in 1h 56m 08s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/24805387428a434082ce3730a338f8aa : SUCCESS in 1h 42m 28s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/95d7bb224a4c42f7a465340b9cff69af : SUCCESS in 6m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/007b67c39a6946108f99a82f6e41c873 : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":16},{"id":"dd1ab9f4cc9f19ad7f99ccdb14dcc04f72eba756","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 21:48:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1","revisions":{"8f543e74458199e77df2b86256bf2d66c2750ed5":{"kind":"REWORK","_number":1,"created":"2019-08-07 11:50:43.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/98/675098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/1"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"subject":"Make /opt/stack accessible","message":"Make /opt/stack accessible\n\nThe trove-scenario tests write to /opt/stack as user stack, change\nownership so that it can be written.\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8f543e74458199e77df2b86256bf2d66c2750ed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8f543e74458199e77df2b86256bf2d66c2750ed5"}]},"branch":"refs/heads/master"},"d662454cbf2ba9b798a22478711d574803b1361f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-07 11:51:36.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/98/675098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/2"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:50:53.000000000","tz":120},"subject":"Make /opt/stack accessible","message":"Make /opt/stack accessible\n\nAfter [0] devstack no longer changes the ownership of the whole\n/opt/stack tree to the stack user unconditionally.\n\nThe trove-scenario tests write to /opt/stack as user stack, change\nownership so that it can be written.\n\n[0] https://review.opendev.org/203698\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d662454cbf2ba9b798a22478711d574803b1361f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d662454cbf2ba9b798a22478711d574803b1361f"}]},"branch":"refs/heads/master"},"af970b85df113ac5e3f850accebadd5d859a682d":{"kind":"REWORK","_number":3,"created":"2019-08-07 12:01:47.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/98/675098/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/3"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 12:01:36.000000000","tz":120},"subject":"Fix jobs: make /opt/stack accessible, add bindep","message":"Fix jobs: make /opt/stack accessible, add bindep\n\nTwo changes to get jobs working:\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally.\n\n   The trove-scenario tests write to /opt/stack as user stack, change\n   ownership so that it can be written.\n\n2) Add bindep.txt file. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af970b85df113ac5e3f850accebadd5d859a682d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af970b85df113ac5e3f850accebadd5d859a682d"}]},"branch":"refs/heads/master"},"aa8a4b707433f10e2f2039a1124719e788632fdb":{"kind":"REWORK","_number":4,"created":"2019-08-07 12:34:19.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/98/675098/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/4"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 12:32:56.000000000","tz":120},"subject":"Fix jobs: make /opt/stack accessible, add bindep","message":"Fix jobs: make /opt/stack accessible, add bindep\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally.\n\n   The trove-scenario tests write to /opt/stack as user stack, change\n   ownership so that it can be written.\n\n2) Add bindep.txt file. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Set SSH_DIR to ansible_user_dir to use the setup .ssh directory\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa8a4b707433f10e2f2039a1124719e788632fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa8a4b707433f10e2f2039a1124719e788632fdb"}]},"branch":"refs/heads/master"},"99c7e6e601118089b0239c957c6ee8b75b3fc94b":{"kind":"REWORK","_number":5,"created":"2019-08-07 13:21:19.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/98/675098/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/5"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 13:21:08.000000000","tz":120},"subject":"Fix jobs: make /opt/stack accessible, add bindep","message":"Fix jobs: make /opt/stack accessible, add bindep\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally.\n\n   The trove-scenario tests write to /opt/stack as user stack, change\n   ownership so that it can be written.\n\n2) Add bindep.txt file. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Set SSH_DIR to ansible_user_dir to use the setup .ssh directory\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99c7e6e601118089b0239c957c6ee8b75b3fc94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99c7e6e601118089b0239c957c6ee8b75b3fc94b"}]},"branch":"refs/heads/master"},"e73d38b94e1e394cdbfa4485a63a64c3f2962ae6":{"kind":"REWORK","_number":6,"created":"2019-08-08 01:08:16.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/6"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-08 01:07:51.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e73d38b94e1e394cdbfa4485a63a64c3f2962ae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e73d38b94e1e394cdbfa4485a63a64c3f2962ae6"}]},"branch":"refs/heads/master"},"0909b25af4f08e30b7636c357fafd18346842d6e":{"kind":"REWORK","_number":7,"created":"2019-08-08 03:14:25.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/7"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-08 03:14:16.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0909b25af4f08e30b7636c357fafd18346842d6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0909b25af4f08e30b7636c357fafd18346842d6e"}]},"branch":"refs/heads/master"},"bcc6efad8321d2fb56cc3d9d75f23bb331964af7":{"kind":"REWORK","_number":8,"created":"2019-08-08 07:22:00.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/8"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-08 06:32:50.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bcc6efad8321d2fb56cc3d9d75f23bb331964af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bcc6efad8321d2fb56cc3d9d75f23bb331964af7"}]},"branch":"refs/heads/master"},"1f0f3b0f839cbaf5a2087e40a782639e204b3d62":{"kind":"REWORK","_number":9,"created":"2019-08-08 10:34:38.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/9"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-08 10:34:24.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f0f3b0f839cbaf5a2087e40a782639e204b3d62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f0f3b0f839cbaf5a2087e40a782639e204b3d62"}]},"branch":"refs/heads/master"},"439225f4cc513148a2e915c2243f25059af45cca":{"kind":"REWORK","_number":10,"created":"2019-08-10 11:47:38.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/10"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-10 11:47:26.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/439225f4cc513148a2e915c2243f25059af45cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/439225f4cc513148a2e915c2243f25059af45cca"}]},"branch":"refs/heads/master"},"e258c65f7b394618f93d1936eed6139dd65f2efa":{"kind":"REWORK","_number":11,"created":"2019-08-11 12:03:44.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/11"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-11 11:10:06.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e258c65f7b394618f93d1936eed6139dd65f2efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e258c65f7b394618f93d1936eed6139dd65f2efa"}]},"branch":"refs/heads/master"},"b5df0fd0da8d26f2fc3011e70774b8033e4eaa99":{"kind":"REWORK","_number":12,"created":"2019-08-11 23:06:35.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/12"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-11 22:25:58.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b5df0fd0da8d26f2fc3011e70774b8033e4eaa99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b5df0fd0da8d26f2fc3011e70774b8033e4eaa99"}]},"branch":"refs/heads/master"},"00b3247a4e0e162167532cbe8afd1740db647b96":{"kind":"REWORK","_number":13,"created":"2019-08-11 23:29:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/13"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-11 23:29:49.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00b3247a4e0e162167532cbe8afd1740db647b96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00b3247a4e0e162167532cbe8afd1740db647b96"}]},"branch":"refs/heads/master"},"0f88de3bd6e16cefecf06050818436f21907933a":{"kind":"REWORK","_number":14,"created":"2019-08-13 03:00:39.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/14"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-13 03:00:27.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0f88de3bd6e16cefecf06050818436f21907933a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0f88de3bd6e16cefecf06050818436f21907933a"}]},"branch":"refs/heads/master"},"30d9e9c9fa538e613396357aca2e477242c17704":{"kind":"REWORK","_number":15,"created":"2019-08-13 11:35:04.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/15"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-13 11:34:52.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/30d9e9c9fa538e613396357aca2e477242c17704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/30d9e9c9fa538e613396357aca2e477242c17704"}]},"branch":"refs/heads/master"},"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1":{"kind":"REWORK","_number":16,"created":"2019-08-14 10:02:46.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/675098/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/675098/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/675098/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/675098/16"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 11:49:48.000000000","tz":120},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-14 09:20:34.000000000","tz":720},"subject":"Fix Trove CI jobs","message":"Fix Trove CI jobs\n\nChanges to get jobs working\n\n1) After [1] devstack no longer changes the ownership of the whole\n   /opt/stack tree to the stack user unconditionally. Switch to the\n   stack user when running integration test.\n\n2) Add bindep.txt file[2]. The default fallback file is not installed\n   anymore and therefore a bindep.txt file is needed to add install\n   additional packages.\n\n3) Use trovestack script rather than devstack to build image so many\n   global variables could be used for consistency. By default, devstack\n   won\u0027t build image.\n\n4) Remove the tools/test-setup.sh as it is not used any more.\n\n5) Instance upgrade test keeps failing in CI for some reason, although\n   it\u0027s always passed on my local environment. In order not to block\n   other patches, skip the instance upgrade tests temporarily.\n\n[1] https://review.opendev.org/203698\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/006888.html\n\nChange-Id: I35e17afb9e827b1fead9d28dbf32f11ce4034a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
