)]}'
{"id":"openstack%2Ftrove~675803","triplet_id":"openstack%2Ftrove~master~I479628d66a9016750a21d9846358406e097525e5","project":"openstack/trove","branch":"master","topic":"add_check_logs","hashtags":[],"change_id":"I479628d66a9016750a21d9846358406e097525e5","subject":"Add periodic logs during database mechanism","status":"MERGED","created":"2019-08-12 02:57:47.000000000","updated":"2019-09-04 11:54:13.000000000","submitted":"2019-09-04 11:52:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675803-1567597950511-233c4fa7","meta_rev_id":"19434e2bfe9af3f7f8681358f09ab8158650b656","_number":675803,"virtual_id_number":675803,"owner":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-09-04 11:52:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":2,"date":"2019-09-04 09:12:30.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":1,"date":"2019-09-04 09:12:30.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 09:12:30.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-09-04 11:52:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa469ec92eba2e3becf4702641464e07f040d876","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-08-12 02:57:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbe1cb3cb710804b26175d9f0c8c13d950e8baa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 03:49:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/03/675803/1/check/openstack-tox-cover/c7c3087/ : RETRY_LIMIT in 2m 09s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/03/675803/1/check/openstack-tox-lower-constraints/16fe0b6/ : RETRY_LIMIT in 1m 39s\n- openstack-tox-pep8 https://logs.opendev.org/03/675803/1/check/openstack-tox-pep8/ead67b4/ : FAILURE in 3m 57s\n- openstack-tox-py27 https://logs.opendev.org/03/675803/1/check/openstack-tox-py27/7ade283/ : RETRY_LIMIT in 2m 04s\n- openstack-tox-py36 https://logs.opendev.org/03/675803/1/check/openstack-tox-py36/2487e53/ : RETRY_LIMIT in 2m 03s\n- openstack-tox-py37 https://logs.opendev.org/03/675803/1/check/openstack-tox-py37/f2f3fd8/ : RETRY_LIMIT in 1m 52s\n- openstack-tox-docs https://logs.opendev.org/03/675803/1/check/openstack-tox-docs/84d99b0/html/ : SUCCESS in 4m 46s\n- trove-functional-mysql https://logs.opendev.org/03/675803/1/check/trove-functional-mysql/174ff07/ : FAILURE in 35m 27s\n- trove-scenario-mariadb-single https://logs.opendev.org/03/675803/1/check/trove-scenario-mariadb-single/6e044d9/ : FAILURE in 26m 23s (non-voting)\n- trove-scenario-mariadb-multi https://logs.opendev.org/03/675803/1/check/trove-scenario-mariadb-multi/1b82ee1/ : FAILURE in 32m 12s (non-voting)\n- trove-scenario-mysql-single https://logs.opendev.org/03/675803/1/check/trove-scenario-mysql-single/db21c81/ : FAILURE in 34m 00s\n- trove-scenario-mysql-multi https://logs.opendev.org/03/675803/1/check/trove-scenario-mysql-multi/d4c5ea5/ : FAILURE in 48m 49s\n- trove-scenario-postgresql-single https://logs.opendev.org/03/675803/1/check/trove-scenario-postgresql-single/cf780e6/ : FAILURE in 41m 18s (non-voting)\n- trove-scenario-postgresql-multi https://logs.opendev.org/03/675803/1/check/trove-scenario-postgresql-multi/a2b859e/ : FAILURE in 39m 55s (non-voting)\n- trove-scenario-pxc-single https://logs.opendev.org/03/675803/1/check/trove-scenario-pxc-single/8e78833/ : FAILURE in 26m 18s (non-voting)\n- trove-scenario-pxc-multi https://logs.opendev.org/03/675803/1/check/trove-scenario-pxc-multi/7c1f6f4/ : FAILURE in 39m 11s (non-voting)\n- trove-scenario-redis-single https://logs.opendev.org/03/675803/1/check/trove-scenario-redis-single/7108dd6/ : FAILURE in 26m 44s (non-voting)\n- trove-scenario-redis-multi https://logs.opendev.org/03/675803/1/check/trove-scenario-redis-multi/b2b4ebc/ : FAILURE in 40m 40s (non-voting)\n- trove-tox-fakemodetests https://logs.opendev.org/03/675803/1/check/trove-tox-fakemodetests/1131ef6/ : RETRY_LIMIT in 2m 16s\n- trove-tox-bandit-baseline https://logs.opendev.org/03/675803/1/check/trove-tox-bandit-baseline/68ce610/ : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/03/675803/1/check/openstack-tox-pylint/90b687e/ : RETRY_LIMIT in 1m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b236538aec94c5b44da88e3b313a1a636aebf14b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-14 21:49:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b189b268665ade233857eb855ea1de47370d0964","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 00:20:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/586b77a255dc454ea56d121fde5cafee : SUCCESS in 16m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98dbc4a06d09432083e31508d4d93790 : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d1c098b4e164e3798f18f4e00eb2378 : FAILURE in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0486037b2f8a46548aefdb9150c676d3 : SUCCESS in 10m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bdd9c97a04f45909ae3fce5857956cf : SUCCESS in 13m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0b05e211ab44cef949a4033a1b6f243 : SUCCESS in 18m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38cb8f39861d43ac996d7c841a21ec6d : SUCCESS in 4m 39s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/28851a9b4c644e60a5ceb865a1f109da : SUCCESS in 2h 20m 59s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/ee09b68d0a9443c4a164cb94bbc1b3c8 : FAILURE in 50m 21s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d5dc893bc1954be388c0b0e211d78308 : FAILURE in 44m 22s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6cee880f5f3e43acaeab99f2e6657ebf : SUCCESS in 2h 04m 29s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f3fd830e86c84310a6e1d29c1a1276e2 : SUCCESS in 1h 28m 01s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/9545acf63e5942aeb55c52859d46ffd1 : FAILURE in 49m 41s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5ee3fdd64da046de9302ab3334504d6d : FAILURE in 42m 39s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/15501365325646a88693756acacb2a9c : FAILURE in 43m 05s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/076e642378ca41eea46f55fda44b3f43 : FAILURE in 47m 29s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/153c01e3f5214866abdf94603cfd40ec : FAILURE in 1h 12m 44s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/b5239e47fe404733b8cc35a57d9bb9a5 : SUCCESS in 1h 24m 05s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/6ec3c51fcef643158400f5b253ad99bb : SUCCESS in 4m 54s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4da296042ebb4f039918de1143549916 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b318f735718b4c37b1232a1db83d9032 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4ef53c30313ad522aa68d37e57215c778e2c2ad2","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-08-15 00:28:06.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"88885bafd26f4e46ea52dbbd6f0681bec58b247d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 03:07:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b220a8f53ec452a940dc6b06fa0bd9a : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc34f34a76d74fda9ab31aab1584e25f : SUCCESS in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3dfed04bbfe4210bdfd36f8c4a71ebf : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88918750203e40dea288bf15c00fbf1a : SUCCESS in 10m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b9baeb689de45a6bcd06681203d6fa3 : SUCCESS in 13m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df67915e3c434d3dbc8fd0ee12e261d7 : SUCCESS in 15m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6d73248180944d3a384f5cb913b4f14 : SUCCESS in 5m 28s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2a4b82c18dfe41c0ac67ebaab834f522 : SUCCESS in 2h 16m 36s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/7cd3572e8b6449fe9613430089f26632 : FAILURE in 36m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7524187bc0734dc6b17e36f707bf614c : FAILURE in 40m 02s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/81b527ce861243ad90b5ca1e97da0ce3 : FAILURE in 37m 42s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/4aa501efba004b5eba4fc0b681b13902 : SUCCESS in 2h 03m 52s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/2fedd720158c466eabda4074d190a6d0 : FAILURE in 39m 50s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/7534bb2f924b436999583ba270166580 : FAILURE in 43m 34s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/b4e357acf27f4bb0818b42bf0beb1b14 : FAILURE in 37m 46s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/a7fa5352eece45f49e49c8d8198353ce : FAILURE in 41m 22s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/8c2494f5fbf34959ab0206a1e8ed755e : FAILURE in 1h 05m 10s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/9a3033122cf34af393d85f3b6e77f1ea : SUCCESS in 1h 18m 05s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/67720b28f39743dd92a6709854c5cd46 : SUCCESS in 5m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9a61094e02f44a9fa27296812ce3a0b1 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/455b0817c6b343d2a830a71359fd38a7 : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":3},{"id":"1ae3831225996fe0ceab26df6d970f824fb0b64c","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-08-15 03:26:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ad179d96d49733502e36947427dfeacdc6789e62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 05:29:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4e632872a274aa19bbf8d2b70512ef8 : SUCCESS in 18m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78b4a3874309466597314bfc5d8a80fd : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf05c2a932e7459d8531bc4de916d77e : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5ddb1c7160843ed8b227d21fd3a5d60 : SUCCESS in 10m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dfe15c6ae0e4e24a09b4c535beca43d : SUCCESS in 13m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/092fa8848a0e47ed8ac560c5f44b4941 : SUCCESS in 18m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a49595913e4e7d8dbc900c20c243cb : SUCCESS in 4m 54s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/9f1ffc7557fb4fffa8792abdf14e95d8 : SUCCESS in 1h 54m 10s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d0c0c67a8f5b472a85f50f8b773ba2e5 : FAILURE in 42m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/3eef50b5510f461995fc052509487223 : FAILURE in 38m 59s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/498a7bd4c61442d4aed92a2987bf05f5 : SUCCESS in 1h 59m 25s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f546ecc8fd1f488e9029a2789a6fdd25 : SUCCESS in 1h 41m 45s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/fe6f712365274de9bc9df70d1176ba78 : FAILURE in 55m 59s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/9f22086ffb404076ae0f616ca763b974 : FAILURE in 38m 57s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/32fab17e4fbe4dc89d92dfd029ae10e7 : FAILURE in 49m 08s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/5d92eedb20af479a82ecadf659dc0bf5 : FAILURE in 47m 21s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/3952b7a7529d49978abb36594ab97245 : FAILURE in 1h 05m 42s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/50810ef1253f414485d0cfbe2c845f7c : SUCCESS in 1h 24m 45s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/2cb44f4ce00b491b85c22485afa530ab : SUCCESS in 4m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/996a62f7b80445cbaf0ee092eeec1999 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bad8b8c0a38740ff977937a986a3d3a8 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":4},{"id":"d88d60c40f07c6e052716b073726001e41f7608f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-08-24 10:29:40.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7fc0ffe9fd7fa596f605bcbb87358ead98246810","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-08-26 07:59:00.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d5328ef41d667dec3fce529aa1a8fc88223d62cf","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-08-26 08:01:50.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThank you for your review. Trove is working fine with my creted images via DIB and/or Cloud-init. We should add this debug logs to operators can know what exactly kind of thing during wait until active. Thanks","accounts_in_message":[],"_revision_number":5},{"id":"47e04ca2e7a817364086cfeb0d026fd39b9e4862","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 10:26:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02d93adcaefb4c6082b8504baac7ddf8 : SUCCESS in 21m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dbba1274f4847e99885c383792e166f : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da91d94345a54962994cbae610ad3b60 : FAILURE in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e2b261db2ec4cfaab2d80e5e96972b8 : SUCCESS in 12m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/508db85cf3b54160839f95d0afd1b450 : SUCCESS in 12m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8734b49f08d45449443a1ae1bdf98c2 : SUCCESS in 18m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6abb24b526b5497b91b2e930799701d1 : SUCCESS in 5m 52s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/314e164ed51e4358b0884dc2e97718b1 : SUCCESS in 2h 24m 28s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/30fb3f3ff8ca4087b3f2e41e68d9caff : FAILURE in 54m 24s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/80a587b3b8ce4db986670e70d121cdd8 : FAILURE in 37m 53s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c6b519fa14e24fdfbf0f9625eff2b801 : SUCCESS in 1h 42m 35s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/00d1e0d8171a456aa7587d05b5ee4a23 : SUCCESS in 1h 39m 14s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/9faa818cf50a4e35a8825f404bd0a1d1 : FAILURE in 47m 09s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/bd79fa8a9850492d843aa89a4f4ed08a : FAILURE in 42m 58s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/b5f712616be849c69c3eb702e7d7e449 : FAILURE in 47m 14s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/19c2b548e31741b58c908814d89e3b84 : FAILURE in 50m 04s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/e71e1e865ba04adc8188fd862dc4b592 : FAILURE in 1h 29m 46s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/12c49531d9504897a8aad924dc027b48 : SUCCESS in 1h 22m 11s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/f8029f23b60e46698e94855fd53eec7c : SUCCESS in 6m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4f0c27f62e5d4e0ebce8d4991b4cdb7b : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d233e1fa705e417496b4f706dc29d6da : SUCCESS in 8m 43s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82b8e31b6c5248d08bde24192b6878f6 : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":5},{"id":"8f6f56ff92e7e39a25f36fb3e452bd62987a5153","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-08-28 02:26:31.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"90e99087b46d74c3a1a4750a91b3f5a1b57bc5fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 04:41:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82c19510629a47a4a3141ca721fdc26f : SUCCESS in 17m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b903c87eed744da8b2c84f1d46e563f9 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f62a13fb1fb845aeada1b47ea73cd347 : FAILURE in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c2c256ed0544aa88c337400cb92fa2f : SUCCESS in 11m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e506aa70a7d4358b72da1a190de8a63 : SUCCESS in 12m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3fd089129eb546fbaaf97c2eabf61089 : SUCCESS in 17m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6726dfedb3be4bc58d40001c084fc0c2 : SUCCESS in 5m 52s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/fdc189f1cd25434a97dff5d38de01005 : SUCCESS in 2h 11m 03s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/6daf6c401b714b27bbae0b7e4966684c : FAILURE in 41m 03s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/331b6c70252f43c69612d3a4673bf7b6 : FAILURE in 51m 36s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b8dfc41801c545ac9203a4f2250686b6 : SUCCESS in 1h 40m 17s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/749e899ab85f4160ab63a83f7155f442 : SUCCESS in 1h 41m 02s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/2aaa902122d944d1b375724bc8091c88 : FAILURE in 38m 09s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/fe6d11a960d64414868f0625830275a1 : FAILURE in 44m 14s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/b20a16bdb95c4d59bcbbe9d06c53a15e : FAILURE in 1h 02m 39s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/38bcbd9f383a47f481ce6368e011b2cd : FAILURE in 40m 12s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/586fc6ceb6884f3eb5d08e70535db62b : FAILURE in 37m 00s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/a701eec5af6e4c18a22b6140d079218e : FAILURE in 42m 25s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/0df5f2deb8da4276974e044fecc41272 : SUCCESS in 5m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7281ec72fad64514b4f466965d59dd9e : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/957c7b6d65a64c5eae71b781d9badaae : SUCCESS in 8m 47s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b165f00a56b3401a9520ca5d06c8886c : SUCCESS in 37m 13s","accounts_in_message":[],"_revision_number":6},{"id":"7f5cacab01b74236c81a62d8e8a5bbff408dc487","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-09-03 03:51:38.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"4d245c2b5ae6a533005dc1922c697d4ad74f3819","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 06:15:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13483fef210f4b4b87a5e08604357d62 : SUCCESS in 19m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c314edf8096a48f4b679831c473533e5 : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875b0f0d2ff0455f8a283a688c421b57 : FAILURE in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9bdb4bb8cf44d7f808c5fbe17786d30 : SUCCESS in 11m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0b3a8ea6c4141a4a53ff8b0310752c4 : SUCCESS in 13m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/741c4a0fd7dd456d9643a38a1478759d : SUCCESS in 17m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b74c74370b442a84fa529ba750b9f8 : SUCCESS in 5m 24s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d11c3e1388cc46dab8875418d74098a8 : SUCCESS in 2h 22m 31s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/066f389c632443e886e2beb197b520da : FAILURE in 41m 44s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/a14842b01c054b5a944e1b92c3c7b7cc : FAILURE in 45m 09s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/a0853873aa314fa48d3202c6d5ce5360 : SUCCESS in 1h 40m 09s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c795ef36ce364ad4829dba43ba447363 : SUCCESS in 1h 39m 15s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/be01ea28f6bb4c2eb1875450db181368 : FAILURE in 37m 46s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/127a4495149548d694a0e440c7e587cf : FAILURE in 44m 05s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/25b473ff1f5d495391ae1f16b253b419 : FAILURE in 40m 07s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/149ab2ec50584b2e8e1d39c11daa8ae0 : FAILURE in 41m 27s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/8de5e53c403c4b7c9050851283ee61c7 : FAILURE in 40m 41s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/c54d4db2871e4e6c9df083688e0e6bce : FAILURE in 43m 37s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/d50dce17f4154d69ab04f75a486a0e9c : SUCCESS in 5m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d965105227b543489f7c5467916300ab : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b201fd886ed438884064dd8fbde3b29 : SUCCESS in 8m 31s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97d40ba1a49049708cd7e89b97223dea : SUCCESS in 33m 00s","accounts_in_message":[],"_revision_number":7},{"id":"1b8c4c1eca68a0913ec1a54bf5a1cacd306cd442","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-09-03 06:34:51.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"233242fbf5456cba5b4c8c8cacd77615fd6aa627","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 09:07:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfc85f096e354f9ab839bd12ef046e11 : SUCCESS in 21m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b09a00af750e47e6a353f49937a225bc : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebcf13909709486d8a0aa30f67decfa2 : SUCCESS in 7m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55eb26b84bf34807ad1b670ab67cc1e0 : SUCCESS in 13m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b23a85d2e2e43838e04861040efa9e7 : SUCCESS in 14m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0568fc99e65c41628f5d6a6c111784ec : SUCCESS in 15m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46dd48d7efdc42fcaaaed553d7cfb049 : SUCCESS in 7m 37s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/68bd2bf2a7b34a5abf46368221980d58 : SUCCESS in 2h 25m 56s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/3327ab79217d4500ac911c2c3461f5e4 : FAILURE in 45m 00s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/39476a623248493f9e1aba4144318124 : FAILURE in 42m 27s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b1f7f65be929415dbddfe79ef37428d7 : SUCCESS in 1h 48m 03s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1fb041da2ce34f468be4bede4a7f933b : SUCCESS in 1h 46m 32s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/5e8efe0808cd46889fc7ec1fb47e20b6 : FAILURE in 41m 40s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/6be204d99aeb436d977d2b17c86823aa : FAILURE in 52m 06s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/5e05b648d5bb4e2fa5166cfdc1bca027 : FAILURE in 53m 05s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/a1cd3544f2304a5e86bbd509960ec0c0 : FAILURE in 51m 58s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/3bd19132c0ef4a85b7ddc0caf37b3e92 : FAILURE in 51m 47s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/3b81fd4499fd4f279a0a60be04354b8b : FAILURE in 55m 48s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/0d918ad5b1ff4f809b9b5edb3642df88 : SUCCESS in 7m 48s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8dcb4de7f65d485694f25d4cb0c4963e : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38d3fee1227543a08e2d509bf6716aa3 : SUCCESS in 9m 46s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b877e0678164e7fab67de2a870bcdfc : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":8},{"id":"1b9605edcb26116383694c9e04f42fed10b07462","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-03 10:17:47.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"62fe5d671d09375cc5555b563f5fbc8c9a9fb258","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2019-09-03 13:21:18.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"80f9c41645f337b71eaf54275f2fad7cfbb7dbbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 15:57:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3923189afadf47b59f0c80cf38eccd75 : SUCCESS in 19m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abba8567601245f2be9a19b8a82b6a2d : SUCCESS in 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71039fdfd7014de9a9c6dcb206766a41 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08c989b0f9c04064a79e7dffdc07569c : SUCCESS in 10m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c9c93e3bd24077ac70f2673294d78d : SUCCESS in 14m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/627e19d4d60649b6982803a2c0ee85ac : SUCCESS in 18m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a942a806f97049babffa5aacf7fbb331 : SUCCESS in 6m 03s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b6d6632b8ea3418585f42177f06c06e1 : SUCCESS in 2h 34m 28s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/588e8bb49e6b4254bd1732f2fb251ec8 : FAILURE in 48m 10s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d45034dd348945cfb7e3ab3ec1006cb1 : FAILURE in 45m 46s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ee376a398de24b1bb6f29d514ca19e91 : FAILURE in 2h 08m 27s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/618de26f5ee94cf5b2d63ba460bf0a9c : SUCCESS in 2h 16m 19s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/539282c9d4014de582727f52c667b4e2 : FAILURE in 43m 09s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/baafaa6a86cc484ea7c68849f69fb117 : FAILURE in 45m 48s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/1d376c0b48d04177bc55af37bdaa7cf3 : FAILURE in 41m 53s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/e4a35fdbe6e14dc7964dae35becdcb02 : FAILURE in 44m 54s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/33594d99af334ebab1ca294165b2f3a7 : FAILURE in 48m 29s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/79be290345f94cad841797763b224075 : FAILURE in 50m 53s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/c7082d1ae66a409cb2344659ac10924a : SUCCESS in 6m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f2e35fcb6f714176905d27ce86a92c8a : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b893ee9f646404ea6f756a8158ad570 : SUCCESS in 8m 31s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3b44a1c54b34d0c834c625eb1404799 : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":9},{"id":"cf689f63a17cf63957a9754a96875edcd028d550","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-03 21:32:50.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"5ae5318b4ab0951f783ec67381ed63b065d0ffb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 23:42:13.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef6637f1810d40809ece6f8f4aa7b061 : SUCCESS in 17m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad658a758d46413f8efb690841f9397f : SUCCESS in 13m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ba40c1e33c496d91716f0899b69aac : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21f074b83cba480fa954202a6fb58468 : SUCCESS in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d296e1b3a0304691815ba4e949a964d2 : SUCCESS in 13m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c0850348fcf4ea6900c87a7838647ef : SUCCESS in 18m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87e3660936a845d38e383d2a36b5250f : SUCCESS in 6m 47s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/96043604ae3e4934ac5a6ef068547f7f : FAILURE in 51m 16s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/dd4884f12f304f09bbe5d2d3a27e41f5 : FAILURE in 42m 18s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/9c45f909256b4084bde510473d398348 : FAILURE in 41m 47s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5c160fcbc4f64d2fad7bc9fc6a40059f : FAILURE in 2h 06m 48s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/894c72c8647b4cb8abab13a3a02d123a : SUCCESS in 1h 43m 01s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/224a2ad0453141a19573bc01fe6b3aa1 : FAILURE in 41m 38s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/9b4f94eefcf4421684fb864bba5ef82b : FAILURE in 44m 50s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/04833d4bd78e4f9a91e6cc7451006cbe : FAILURE in 41m 58s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/f035bd67f70d4f738ffc1a13e45dfcd5 : FAILURE in 1h 02m 56s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/56085a4ecfbc4b95bb931dc6a913da52 : FAILURE in 41m 33s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/8014c6d247464b0d8e7cf6207db65e9b : FAILURE in 47m 37s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/0fee72fb201b4b32be8a7a8f3f33a17b : SUCCESS in 7m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dbe57e63080847088fb8364f5b09f933 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42d9bb207e1144d88ab35f8303800e52 : SUCCESS in 9m 13s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a42c11fc0f14c839e8b32cc7398e7fa : SUCCESS in 35m 27s","accounts_in_message":[],"_revision_number":10},{"id":"d185f0024187140aeb78b935c5899aeb2ea9e807","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-03 23:44:05.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"8ca3b985702916290056928ae2bec601016373ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 02:12:53.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a48c9cd5af8045e1bc454a6e464ba106 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95aa45401fdf46d9843e98080cf1c491 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a1b6de6d3e431d89e42002648d1e56 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa6f263843de4c2c936979bf6bdf9f7b : SUCCESS in 11m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f17f061dc8e14b82a8fec24a5d4be88b : SUCCESS in 13m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a42ffaf09fb49feb59b04ce0fbd2a09 : SUCCESS in 18m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae99321579e9489c946967542ca34b51 : SUCCESS in 5m 49s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b36930e290534592a1f1f7599a203c87 : SUCCESS in 2h 21m 24s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/e3e0ba07b46f4a33a1944572ec7772b9 : FAILURE in 41m 16s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/14c7b92a2af749749c56e94a5c49e715 : FAILURE in 45m 48s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/dc0c83cf457e4926bbb4bf5e64b25b0a : FAILURE in 53m 48s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/701855bb303b4148b3d59e5a684656ed : SUCCESS in 1h 43m 09s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/66567afebeef49358c15181e8d2aefb3 : FAILURE in 40m 37s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/0b8ed975d4c44c168e7ab70ebd4ed2f4 : FAILURE in 41m 08s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/cd63499b1b4d4eed85deb20819227082 : FAILURE in 43m 06s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/879e1978edff4fc99b1d81eafb44ab01 : FAILURE in 41m 40s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/7841efc1225544f69c69f2274288736e : FAILURE in 41m 59s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/36bcad8358064bdd9774b67f58739d3a : FAILURE in 42m 01s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/e73eb2e1618e492b8b16d598fd4b0fe8 : SUCCESS in 6m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69ea2b135ea94e92a0a083436e362858 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/177a7b7ee4d24390841948cd392d7151 : SUCCESS in 9m 40s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ff8e404857b4cedbed47d475589102f : SUCCESS in 36m 55s","accounts_in_message":[],"_revision_number":10},{"id":"98552c2f45c880e272a9e5fd59814624b7305eab","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-04 02:19:15.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"d45885a4a9bb918586d7971822bc80be0c4f4d89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 04:40:02.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc3a42ccabff4f6293d37efe56c5aac8 : SUCCESS in 18m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/347f1929f115439b823cb05bfe1bfa9d : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7b572fdbb8844718f6fcb605a0a9ae7 : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8dc806cce74b4fcdab25834c3eb71e45 : SUCCESS in 13m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cee0d160705491581318d21c93240b8 : SUCCESS in 14m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/344ba535cd47410da346afe59dc332aa : SUCCESS in 20m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7c7aafee074f358016540fd979edbf : SUCCESS in 4m 55s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/1feb4e591a984226a699270ac0214b4a : SUCCESS in 2h 15m 22s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/740838fe13b445c3a2d6813dc8aaa1c8 : FAILURE in 1h 00m 26s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/6a341b0d0fd94cab87ac69e0f9bde2db : FAILURE in 1h 27m 39s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/3f0b1714fe1d4af9b30144eec82c089c : FAILURE in 2h 02m 15s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/462bb48ba3cf40dfb64e57bdda9a85d7 : FAILURE in 1h 47m 42s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/00ede3fe08444423ad3da89dfebae50a : FAILURE in 53m 13s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/a2e178aa6d8b42cabbc63c3ec78c6a2a : FAILURE in 45m 16s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/a4b86204eecd491cacbb99feb5678987 : FAILURE in 1h 16m 54s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/55a4951e7ff84d92b65324c899260ea3 : FAILURE in 41m 48s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/0b43803a907f4e5791c05176f1a07d07 : FAILURE in 50m 44s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/33db0464adac401c9626fbc5a4af045f : FAILURE in 43m 15s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/9a327b99a5124462b345d4a817f580e5 : SUCCESS in 5m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aeb999b9d48f443881bcd4093c5e9549 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac85ca85f6714466b8af839b916eddba : SUCCESS in 8m 28s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0deeb2c7f404947a89ec577a80f6002 : SUCCESS in 31m 22s","accounts_in_message":[],"_revision_number":11},{"id":"8ca60822ee5acc798e0de3ef7d4d79f67ce3083f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-04 05:48:17.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"cc1f785c37171dcfcfe33d36e1f04e737eba34e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 08:26:05.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/160138238a6643ddbf274e748c1ed96f : SUCCESS in 16m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4284d0a5a85d4117b354fa20d1885aa4 : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8a45ca206c7415bb1ca5fd2853e8fac : SUCCESS in 8m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e6c6b543cb846bd9f449c6b4e2cbb80 : SUCCESS in 13m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/623486d51f7c426bb102878b1e25b3e3 : SUCCESS in 12m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea62b829cd894d0b82ab1826c115fff2 : SUCCESS in 17m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6afdc26e32184f37b5a48fa22637c906 : SUCCESS in 5m 26s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c029b3cb27e447579b87a156b2f60872 : SUCCESS in 2h 20m 38s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d4595b682cb84336ace0685e280b7319 : FAILURE in 2h 11m 52s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/8de56d7687f04874a9b98f06dd782ba1 : FAILURE in 1h 08m 43s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/9a2c00ce3e8847179e558cbd25d0ebf6 : SUCCESS in 2h 24m 08s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2e1a8fc0fa164b8da54e4effda253baa : SUCCESS in 2h 23m 29s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/385e08cbaf664669b4dd40393539cdac : FAILURE in 59m 39s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/188409b704014ddf8afcdee2bbe96f82 : FAILURE in 44m 01s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/c65d5ca2366e474b9ef768c819b7b445 : FAILURE in 1h 44m 32s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/a5d8b9a3e81e44b08ddc5ae397a4e82d : FAILURE in 54m 29s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/ab46dfab9f7740ac8eed63fc1dea16ce : FAILURE in 1h 30m 06s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/6ce41f42f92d422baa0e93f1263cd5a5 : FAILURE in 1h 30m 55s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/c9be919a39274c6784f02fc46b4a6663 : SUCCESS in 5m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/71fb7b14b8db4f20a960578c46e0c4fb : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c51c1df7cd1a4fc3b65fff7cff5f041c : SUCCESS in 8m 27s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90601b60b7f9436e9cd8a7123c202e99 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":12},{"id":"3538d0306a592292cc433b1ee6499674e1140a49","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-04 09:12:30.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"e0b8de5372475eae296e5afb42880eacd567b45f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 09:12:43.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"826c193cc68fcc4c66e21f758aa1d5f3bba31295","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 11:52:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"69ef672692ac45711f414d6c75bcab8d0fc8f3b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 11:52:30.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/253ce75fd2a14a2aa5c5503be8b6d634 : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc43413c354d4f94b6e5e5e85037cb4b : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e4d46afd8254fb580fed709a6288a8e : SUCCESS in 14m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef7ddd8a11634c0db4f70f7bdd4d635a : SUCCESS in 13m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05d70d496b864bcab77c4f2411bfe7b4 : SUCCESS in 20m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/755332b1cf7745698da0d4429bf61100 : SUCCESS in 6m 07s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b198c8c1ca0347aabdf673142aadfcfb : SUCCESS in 2h 22m 20s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/46fbb5e1653842d9b72170b081043698 : SUCCESS in 2h 33m 28s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a55e9391901c433ea3e5e831861a2919 : SUCCESS in 1h 59m 49s\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/9566c4b189bc42a0bfcf86e422aed40e : SUCCESS in 6m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c91cab6265c64533ac92f5c345c00e3b : SUCCESS in 9m 12s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6d404b01a5b4f5cbe9c6db141c8b2ea : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":12},{"id":"24a1236cd44fc0ec8449bb036775944b487a25dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 11:54:13.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc98d2f440149398b7848108d845cfb : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"6c50dec163ce094c4cb27afaef0ada2cb50bbb07","revisions":{"6263a509913275f66ec81803c6dfd81bfd7b57ec":{"kind":"REWORK","_number":1,"created":"2019-08-12 02:57:47.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/03/675803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/1"}}},"commit":{"parents":[{"commit":"dae5bc1bc7922ee35dffe5e283fc8529299b3081","subject":"Re-define the \u0027nics\u0027 parameter for creating database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dae5bc1bc7922ee35dffe5e283fc8529299b3081"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:57:26.000000000","tz":420},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6263a509913275f66ec81803c6dfd81bfd7b57ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6263a509913275f66ec81803c6dfd81bfd7b57ec"}]},"branch":"refs/heads/master"},"5ffef4e93e7f21fd23ef647d09f3d0f249042fdb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-14 21:49:55.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/03/675803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/2"}}},"commit":{"parents":[{"commit":"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1","subject":"Fix Trove CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-08-14 21:49:55.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5ffef4e93e7f21fd23ef647d09f3d0f249042fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5ffef4e93e7f21fd23ef647d09f3d0f249042fdb"}]},"branch":"refs/heads/master"},"07c465d8ce4becf2e2c010bb7414c57daad8b25b":{"kind":"REWORK","_number":3,"created":"2019-08-15 00:28:06.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/03/675803/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/3"}}},"commit":{"parents":[{"commit":"7b3483723a5fdfa56a8cc419ad77c8660ac87ad1","subject":"Fix Trove CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b3483723a5fdfa56a8cc419ad77c8660ac87ad1"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-15 00:28:02.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/07c465d8ce4becf2e2c010bb7414c57daad8b25b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/07c465d8ce4becf2e2c010bb7414c57daad8b25b"}]},"branch":"refs/heads/master"},"580dc24a8dd0d493ccadea7f34400285a649e00e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-15 03:26:26.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/03/675803/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/4"}}},"commit":{"parents":[{"commit":"69e6652c92ce95a40d34e7fb181b9ac5a5f55428","subject":"Remove invalid assert state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69e6652c92ce95a40d34e7fb181b9ac5a5f55428"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-15 03:26:26.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/580dc24a8dd0d493ccadea7f34400285a649e00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/580dc24a8dd0d493ccadea7f34400285a649e00e"}]},"branch":"refs/heads/master"},"c4b5e9712296094cdf368fc5933fe6b8fdf49def":{"kind":"REWORK","_number":5,"created":"2019-08-26 07:59:00.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/03/675803/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/5"}}},"commit":{"parents":[{"commit":"69e6652c92ce95a40d34e7fb181b9ac5a5f55428","subject":"Remove invalid assert state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69e6652c92ce95a40d34e7fb181b9ac5a5f55428"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-26 07:58:09.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4b5e9712296094cdf368fc5933fe6b8fdf49def"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4b5e9712296094cdf368fc5933fe6b8fdf49def"}]},"branch":"refs/heads/master"},"c773d4baa28495b7a5705af15fc2ce2180d4dff1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-28 02:26:31.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/03/675803/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/6"}}},"commit":{"parents":[{"commit":"259e0f6f093ec5397a778bd485379e2532b4cc5f","subject":"Merge \"Support python3 in guest agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/259e0f6f093ec5397a778bd485379e2532b4cc5f"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-28 02:26:31.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c773d4baa28495b7a5705af15fc2ce2180d4dff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c773d4baa28495b7a5705af15fc2ce2180d4dff1"}]},"branch":"refs/heads/master"},"9cf3c3677e03f81d7b41b91d5efe88c24c070eca":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-09-03 03:51:38.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/03/675803/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/7"}}},"commit":{"parents":[{"commit":"86228d9b88082619883f983ef009df7dfeb57138","subject":"Merge \"fix bug report url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/86228d9b88082619883f983ef009df7dfeb57138"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-09-03 03:51:38.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9cf3c3677e03f81d7b41b91d5efe88c24c070eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9cf3c3677e03f81d7b41b91d5efe88c24c070eca"}]},"branch":"refs/heads/master"},"cdfbb642998f902b09385f24ed52b283b4d874d9":{"kind":"REWORK","_number":8,"created":"2019-09-03 06:34:51.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/03/675803/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/8"}}},"commit":{"parents":[{"commit":"86228d9b88082619883f983ef009df7dfeb57138","subject":"Merge \"fix bug report url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/86228d9b88082619883f983ef009df7dfeb57138"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-09-03 06:34:48.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cdfbb642998f902b09385f24ed52b283b4d874d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cdfbb642998f902b09385f24ed52b283b4d874d9"}]},"branch":"refs/heads/master"},"b3e81dc091db5db44c27b2bec7d39c18731f4c78":{"kind":"REWORK","_number":9,"created":"2019-09-03 13:21:18.000000000","uploader":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"ref":"refs/changes/03/675803/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/9"}}},"commit":{"parents":[{"commit":"86228d9b88082619883f983ef009df7dfeb57138","subject":"Merge \"fix bug report url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/86228d9b88082619883f983ef009df7dfeb57138"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-09-03 13:21:14.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3e81dc091db5db44c27b2bec7d39c18731f4c78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3e81dc091db5db44c27b2bec7d39c18731f4c78"}]},"branch":"refs/heads/master"},"9143e92f7ea98400c1d77170a6ac76573d270c4a":{"kind":"REWORK","_number":10,"created":"2019-09-03 21:32:50.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/03/675803/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/10"}}},"commit":{"parents":[{"commit":"86228d9b88082619883f983ef009df7dfeb57138","subject":"Merge \"fix bug report url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/86228d9b88082619883f983ef009df7dfeb57138"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-03 21:32:46.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9143e92f7ea98400c1d77170a6ac76573d270c4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9143e92f7ea98400c1d77170a6ac76573d270c4a"}]},"branch":"refs/heads/master"},"5ab9702c482dff3b8b7697a629c26df29202144f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-09-04 02:19:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/03/675803/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/11"}}},"commit":{"parents":[{"commit":"c23c1b9d0c532b5fb0eff97d99e44fb2fa5156cc","subject":"Improve image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c23c1b9d0c532b5fb0eff97d99e44fb2fa5156cc"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-04 02:19:15.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5ab9702c482dff3b8b7697a629c26df29202144f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5ab9702c482dff3b8b7697a629c26df29202144f"}]},"branch":"refs/heads/master"},"6c50dec163ce094c4cb27afaef0ada2cb50bbb07":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-09-04 05:48:17.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/03/675803/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/675803/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/675803/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/675803/12"}}},"commit":{"parents":[{"commit":"1b159c7d6a40f194e34c2083fb840c620382567e","subject":"Merge \"Improve image build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b159c7d6a40f194e34c2083fb840c620382567e"}]}],"author":{"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","date":"2019-08-12 02:51:29.000000000","tz":420},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-09-04 05:48:17.000000000","tz":0},"subject":"Add periodic logs during database mechanism","message":"Add periodic logs during database mechanism\n\nThis patch will help operator and developer who can\nfollow actions during checking server status and\nservice status of creating/deleting mechanism.\n\nChange-Id: I479628d66a9016750a21d9846358406e097525e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c50dec163ce094c4cb27afaef0ada2cb50bbb07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c50dec163ce094c4cb27afaef0ada2cb50bbb07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
