)]}'
{"id":"openstack%2Ftrove~676671","triplet_id":"openstack%2Ftrove~stable%2Fstein~Ib5445a63825939307d41f34c47756d2f264b604b","project":"openstack/trove","branch":"stable/stein","topic":"stable-ci-fix","hashtags":[],"change_id":"Ib5445a63825939307d41f34c47756d2f264b604b","subject":"Remove unused tools/test-setup.sh","status":"ABANDONED","created":"2019-08-15 10:47:34.000000000","updated":"2019-08-21 15:45:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"808ce16a8a0ff92c9ec8ae66e8d34d054d07e149","_number":676671,"virtual_id_number":676671,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":-1,"date":"2019-08-21 15:37:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 15:37:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2342acffd9a79c79a03c18fa341b069db83a3286","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-15 10:47:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf0445e0d6a30845dbedece06b5967330d990d3d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-15 10:50:47.000000000","message":"Patch Set 1:\n\nLet\u0027s see whether this is enough to fix CI on stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"addaed55c3b519798d3b5956f85d9ea3074c8191","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 11:12:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1987dba61ea44a4ca8b7dcf445d8f59b : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16accad3fc2043c296cd0566ce320772 : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69820db31f7d4d22993034048b6899a8 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7edce7dd07e43d89fb36fa81e5f5232 : SUCCESS in 12m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88173973e4ad41ada411f0572140138a : SUCCESS in 12m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e3fadf9f69b419db30fa16accc67dd3 : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c840e19946664f5ebd69ba5a50a71bf9 : SUCCESS in 5m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88ad8f5337254c8fb90baab98d22a157 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c77e37e83b5e09a078cb2e36f9f281a46eedf909","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-15 11:23:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"efac135e4c75730b3a713392b64d14d6417beb52","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-15 11:36:59.000000000","message":"Patch Set 2:\n\nIt seems that not all the jobs are running for this patch (compared to other patch on stein, e.g. [1]). Maybe those (devstack based) tests need this script.\n\n[1] https://review.opendev.org/#/c/645647/","accounts_in_message":[],"_revision_number":2},{"id":"069f4a20d87577f401cce4843c08cbf11ec0a850","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 11:45:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34c0c2e5478d4b4baa77c4866b859c14 : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6477e9cc1ce243de89da45d71bfb6568 : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af1ef4da07d4ca38da29b8014257868 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/990baa1c9cfa43d4a51558f43ac5ff86 : SUCCESS in 8m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ba6155c38763406095981356aa16f352 : SUCCESS in 15m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4157d8fd51c04829bb9092bc6480e8c7 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc85fb13e61e4703aefa072e8e66f94c : SUCCESS in 3m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/039f9f4da612469ca1085031632a0a5b : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ef2e131bea6d10173ffb0b3c014e042bf327bf7c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-15 12:33:39.000000000","message":"Topic set to stable-ci-fix","accounts_in_message":[],"_revision_number":2},{"id":"496089eb16548b88b009d645dd32aff41a418631","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-21 13:28:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"18bf408cad07b452233f3a0d2cafaff353d680d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 14:12:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6daa5ac4bbb466fa0634f8435e40111 : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc0b4062f7934777ba338fa79e0f0b92 : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de214e1557254d688ee1054539b9be7b : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/266c05df1eb14f56a6ecd9407ff7001b : SUCCESS in 10m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/264a348d3ff24512950f89e3d9a36201 : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1605fde12ba7492fb3309937c742d21f : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08e29fc2521741b09581296bc2535a39 : SUCCESS in 6m 37s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e22fc434075e414da5a87c0765855351 : FAILURE in 34m 08s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f79caa0a8ecc4c5e888516d9cf238a0b : FAILURE in 33m 21s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c54e980c6e464874ad9ad660177a78d3 : FAILURE in 37m 54s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/24bc3a5aa31f4f78bd007a6f8630967a : FAILURE in 37m 24s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e9ae07fbd9b24c4eb805b25f7719b768 : FAILURE in 32m 57s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/71162162129e45798efe7939609f773b : FAILURE in 32m 57s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/8f58cce4f7ee4f41a7c8f8af5850f83d : FAILURE in 30m 40s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/9ed3f679f3f8406fb1d1ca7ef1cd0c0d : FAILURE in 33m 59s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/36bbb87915d3438d87d66065a4a93a00 : FAILURE in 29m 36s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/9078c5f073ee4ca9adb9768ce18473ca : FAILURE in 34m 15s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/da8e145ba755476587e44e255ac6b061 : FAILURE in 34m 27s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/d2584233ca4f450e8d23e4cda9617d46 : SUCCESS in 4m 28s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/4da05819303d4bfdb4ee0ae20e523bf6 : FAILURE in 5m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/382cb2c088674e6ebb565db2f26ad59c : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/864bbb1d691148539d0c11433c4b9563 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":3},{"id":"1b10c2575a519e91daa96387c51bf65510dcb399","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-21 14:46:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cb084a1de43b4f2d3f93044f78b3cd6349120714","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 15:37:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bd9829311bf447b97a5ca58e81996ad : SUCCESS in 17m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3317d0193aab498789d4de40d952f241 : SUCCESS in 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e099e27afb084ba491aa6103ea85c795 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67e6b63a9b9d445a880748709c4bfc2c : SUCCESS in 12m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/070694a684c94562ae28d74e5d32095e : SUCCESS in 14m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4b57762fad043eebacc67bf71a9cd56 : SUCCESS in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b7c5e6993c4a5496147345c728769f : SUCCESS in 6m 32s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d654ef8798e145cb958a14d6c98297eb : FAILURE in 41m 46s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/0da0208ba7a5487290aff6b2928ca71c : FAILURE in 35m 48s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/db89a3c8ad344c39aaae867b4c6c0894 : FAILURE in 36m 02s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/68bcb77945764ea6a3a34b31cde0e5c7 : FAILURE in 34m 47s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/043e9fb2a8ce420181647b5f4528407e : FAILURE in 37m 44s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/15ceb54946034e449212e3139ca2e38b : FAILURE in 37m 08s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/bc027642673f4e2ab9a2740adb326a36 : FAILURE in 39m 29s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/0ea6e3557ea44953af72396ae243d80e : FAILURE in 35m 33s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/3b4d3f9adbde4b728cff4b51f456780a : FAILURE in 33m 39s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/36ad47b407734820be0e376d737cf2db : FAILURE in 38m 07s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/a0c3258cbd694980961cb2490a9c66ef : FAILURE in 41m 55s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/9d95e1d46e094150b2c049876c9b76aa : SUCCESS in 5m 38s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/ca1c416903c54c9687bedd2d71b6854c : FAILURE in 5m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a7aff07f0eb64146bf516baf0592bfef : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1c5dcfbc1faa4da8a7d25e53bf475901 : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":4},{"id":"808ce16a8a0ff92c9ec8ae66e8d34d054d07e149","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-08-21 15:45:01.000000000","message":"Abandoned\n\nas there are more failures, it\u0027s easier to cherry-pick CI fix patch I35e17afb9e827b1fead9d28dbf32f11ce4034a9b to stable/stein","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"438f7e2044784151e5798083bdbf55d34d3b6438","revisions":{"35e65cf22c5b0b1a8a1b0810b02b2d89ccdf1bd4":{"kind":"REWORK","_number":1,"created":"2019-08-15 10:47:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/676671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/676671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/676671/1"}}},"commit":{"parents":[{"commit":"26d1e5ad59eda795e5d1485730bd8a71d620e55c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26d1e5ad59eda795e5d1485730bd8a71d620e55c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-08-15 10:44:23.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-08-15 10:46:43.000000000","tz":120},"subject":"WIP: Remove unused tools/test-setup.sh","message":"WIP: Remove unused tools/test-setup.sh\n\nChange-Id: Ib5445a63825939307d41f34c47756d2f264b604b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35e65cf22c5b0b1a8a1b0810b02b2d89ccdf1bd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35e65cf22c5b0b1a8a1b0810b02b2d89ccdf1bd4"}]},"branch":"refs/heads/stable/stein"},"5f12409fc87f607299e3630dfb6154cd08de4a74":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-15 11:23:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/676671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/676671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/676671/2"}}},"commit":{"parents":[{"commit":"26d1e5ad59eda795e5d1485730bd8a71d620e55c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26d1e5ad59eda795e5d1485730bd8a71d620e55c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-08-15 10:44:23.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-08-15 11:21:24.000000000","tz":120},"subject":"Remove unused tools/test-setup.sh","message":"Remove unused tools/test-setup.sh\n\nCI jobs fail with \u0027sudo: mysqladmin: command not found\u0027 error [1],\nhowever tools/test-setup.sh seems to be unnecessary for the jobs.\n\n[1] https://logs.opendev.org/periodic-stable/opendev.org/openstack/trove/stable/stein/openstack-tox-py27/08d30f2/job-output.txt.gz#_2019-08-15_07_21_37_969034\n\nChange-Id: Ib5445a63825939307d41f34c47756d2f264b604b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5f12409fc87f607299e3630dfb6154cd08de4a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5f12409fc87f607299e3630dfb6154cd08de4a74"}]},"branch":"refs/heads/stable/stein"},"6fe4a2d10466dc4d967dbf56ba7a60b6f0d6bd27":{"kind":"REWORK","_number":3,"created":"2019-08-21 13:28:10.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/676671/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/676671/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/676671/3"}}},"commit":{"parents":[{"commit":"26d1e5ad59eda795e5d1485730bd8a71d620e55c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26d1e5ad59eda795e5d1485730bd8a71d620e55c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-08-15 10:44:23.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-08-21 13:27:36.000000000","tz":120},"subject":"Remove unused tools/test-setup.sh","message":"Remove unused tools/test-setup.sh\n\nCI jobs fail with \u0027sudo: mysqladmin: command not found\u0027 error [1],\nhowever tools/test-setup.sh seems to be unnecessary for the jobs.\n\n[1] https://logs.opendev.org/periodic-stable/opendev.org/openstack/trove/stable/stein/openstack-tox-py27/08d30f2/job-output.txt.gz#_2019-08-15_07_21_37_969034\n\nChange-Id: Ib5445a63825939307d41f34c47756d2f264b604b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6fe4a2d10466dc4d967dbf56ba7a60b6f0d6bd27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6fe4a2d10466dc4d967dbf56ba7a60b6f0d6bd27"}]},"branch":"refs/heads/stable/stein"},"438f7e2044784151e5798083bdbf55d34d3b6438":{"kind":"REWORK","_number":4,"created":"2019-08-21 14:46:19.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/676671/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/676671/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/676671/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/676671/4"}}},"commit":{"parents":[{"commit":"26d1e5ad59eda795e5d1485730bd8a71d620e55c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26d1e5ad59eda795e5d1485730bd8a71d620e55c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-08-15 10:44:23.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-08-21 14:45:25.000000000","tz":120},"subject":"Remove unused tools/test-setup.sh","message":"Remove unused tools/test-setup.sh\n\nCI jobs fail with \u0027sudo: mysqladmin: command not found\u0027 error [1],\nhowever tools/test-setup.sh seems to be unnecessary for the jobs.\n\n[1] https://logs.opendev.org/periodic-stable/opendev.org/openstack/trove/stable/stein/openstack-tox-py27/08d30f2/job-output.txt.gz#_2019-08-15_07_21_37_969034\n\nChange-Id: Ib5445a63825939307d41f34c47756d2f264b604b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/438f7e2044784151e5798083bdbf55d34d3b6438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/438f7e2044784151e5798083bdbf55d34d3b6438"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
